)]}'
{"id":"openstack%2Ftooz~684236","triplet_id":"openstack%2Ftooz~master~I6a18183f092689c8c9d283c19914da9baa59cc32","project":"openstack/tooz","branch":"master","hashtags":[],"change_id":"I6a18183f092689c8c9d283c19914da9baa59cc32","subject":"Drop os-testr test-requirement and pretty_tox.sh","status":"MERGED","created":"2019-09-24 08:27:55.000000000","updated":"2019-10-10 15:39:14.000000000","submitted":"2019-10-10 15:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684236-1570721871994-01385176","meta_rev_id":"fcca7229f6a8e46a9769e50abc6dd58d35255e37","_number":684236,"virtual_id_number":684236,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 15:37:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-24 13:36:44.000000000","_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-09-24 13:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-10-10 13:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"date":"2019-09-26 06:16:08.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2019-10-10 13:59:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 13:56:20.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-26 06:16:08.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-10-10 13:59:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-10-10 15:37:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b0ad9bdb13f728b0bd1946cd33e0d5904b970ce","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 08:27:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b8f02fe240ba4aed244d00ff9aa73c69a734bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 10:20:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1e9d5ce9d6b24b7e8a7487bd9d127d9f : SUCCESS in 4m 21s\n- tempest-full https://zuul.opendev.org/t/openstack/build/378dcf18b3674fb0900e8dc632ceaef9 : SUCCESS in 1h 45m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bd5ff181a9e7494f8121f6b6099df48e : SUCCESS in 1h 25m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c7933847608409b8aeb6e84b0030753 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e533e0c441614952b264127c05b05e9d : SUCCESS in 5m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/043fc4a700f04bbcb2883bcb76232583 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/623bfdc1513849bd91129a5077ea3f8d : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04287c82246f4c8fb95225c811c3edd6 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852dfdabd27c40629d5304747b4874a6 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc696211de214842b795498d43d4f9bd : SUCCESS in 3m 43s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/2d58179608fb4c71a5d82415dbe8e6cd : SUCCESS in 5m 39s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/619ae14b7c1741eb9af1d0ad52753385 : SUCCESS in 4m 24s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/5c7b34d6efb24f22ba1e8a6cafd74e4c : SUCCESS in 7m 25s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/21696704d6b540d3b9cb94f2bce7ebee : TIMED_OUT in 30m 59s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/9eda9a1947f146158f4624bf1dbef8c6 : SUCCESS in 3m 58s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/dd80d83930df4f0eaa0b8778c9167e52 : SUCCESS in 4m 44s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/010a683cebdc48d49083e4e29dcaf5f5 : SUCCESS in 4m 30s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9e230a8429c9462cad41400968d4709a : SUCCESS in 5m 50s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/a9e9ac96c1d84e8dab5443fac5c8430e : SUCCESS in 5m 39s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/ed4f330e2c564699bbc55b411f9c9d6d : SUCCESS in 5m 56s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/0f6a7a34aa714a66a7faf96a866c1dbc : SUCCESS in 5m 34s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2462094b204042b1b1cc795a7390efc8 : SUCCESS in 5m 19s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/2233194cfe0a44f79b4d23483a3abac3 : SUCCESS in 6m 11s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/c3f16570475c4310a454f11b636a9a53 : SUCCESS in 6m 10s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/9735b3f025694426acf8fb059e35467b : SUCCESS in 7m 35s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/3484a2039da9491f9633774449369f6e : SUCCESS in 4m 11s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/371fd0574749468890c7c9d682f54813 : SUCCESS in 4m 03s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/558fe76299dc443e8ac1cd74ccdebd7c : SUCCESS in 4m 09s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/95b8a89fa7924a09b3cf5302c07c2124 : SUCCESS in 4m 08s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/121296ae3f2e478482640327ca2fb41d : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"214e402c2e8a407fa3984c45a6b0476323c4021d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 10:50:17.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65d5c64399c4c02cf4d717cee241caf476660491","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 11:05:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4bb47bb38d54b1d16fb67a411c975cd1c5d96157","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 11:05:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24ebf982e2aaeecf3f2a8653824d513a092fc003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 13:10:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/aeb2f91881ae4e8d88127c3cd9cd94ef : SUCCESS in 3m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/5541152fa54041e792ec372321daddbc : SUCCESS in 1h 52m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2416084050c84653afbef7e6aa889954 : SUCCESS in 1h 39m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f43f63a71fb9488c98ca2ffc8427c3f0 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f3fe60cf884dd7a74c2d43100e692c : SUCCESS in 3m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2936e45fba1d4c1ebc4e1de1ebfca68e : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34d2b54f7494f0e9e4790d33db1bc85 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b160a78d0034876b637bcc5d695c73e : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9c360b8b38f4d048e634c00c15d120e : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61aa9d1e5e374aa9b5795395c2e32f89 : SUCCESS in 5m 54s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/bfd1b60776fd4f6a8ae403d8598ccbb2 : SUCCESS in 6m 01s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/a9c86235c17b417f8095cadf5b37ee21 : SUCCESS in 4m 20s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/61553d9ad9a640c0bf7b58a9c4a6c11f : SUCCESS in 4m 25s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/fa0f1257c42a4a83a0bada3e19a25718 : SUCCESS in 5m 04s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/6f10af33e704404da957bf20c19d9663 : SUCCESS in 4m 31s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/92a9dcc160a8422b9dbdeb4eb28cb8ae : FAILURE in 4m 26s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/d7fd93f2779145ceb840234faf38189a : SUCCESS in 4m 01s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/489c91e7c82a46b298a02470ec3d8ad4 : SUCCESS in 3m 50s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/656fa668c69e479a8eac656989f44423 : SUCCESS in 3m 36s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/b89f1dbabbde42a1b3d45360d0bb532c : SUCCESS in 3m 43s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/cf7a410a3b9c462aa151e64942cdd42b : SUCCESS in 4m 17s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/8a73553db30d452993b37a8dc086f78f : SUCCESS in 4m 20s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/12511f2aed474a458623bdd0d86b5614 : SUCCESS in 5m 54s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/157760f275eb4bd3a10a1bf467e849b8 : SUCCESS in 4m 19s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/757324101d1c464ab61b9e425c8e8dbf : SUCCESS in 4m 52s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/4b7be0312bcb45e79b6a83f33109619a : SUCCESS in 4m 53s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/a505c55ee3f34f2e8e6f0f926d58cdef : SUCCESS in 4m 10s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/b9d44475c25e42c192bd70bccb6612c8 : SUCCESS in 7m 14s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/513d10b54bfe4cc287feecf3234738e4 : SUCCESS in 5m 48s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/1f39b40f24da408da080e3a0b5b09f6a : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":2},{"id":"995794cbcd6bfb49c55769b322c71abc553bad8c","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-24 13:36:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"944748aefb7ce93daa82a1aefc6443462eaa1df8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-24 13:56:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"93224cdf2526815976543783ecef3afa7ccdacff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 15:28:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3cdad94eeafb4dd38dcb12104219d1c2 : SUCCESS in 4m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8256fe324ab8436da5d759422bacc9b1 : SUCCESS in 1h 49m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e5af827da5cc4e62815e99bd4ee94cdb : SUCCESS in 1h 29m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ed3391cea9a47d386a47107665edf8a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c72c5b8f7f7641e092ff1f7f027bf381 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50f672e46a9544a49dd3184760eaec8d : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a209f279b812412ab4cf77acaf9cae79 : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eadf4c421f6c4e7ab41cc6e623827176 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80a07f730824c36bf8c7a138ee211c2 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c6373df0f8a470b82221ed133a9e78e : SUCCESS in 4m 29s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/05051ca460fc437cac85782a809a0fc5 : SUCCESS in 4m 35s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/aaeef157d5894d06aba3082037670c5a : SUCCESS in 6m 21s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/c25ccad1b81f40a290bf4b9469c5ce3e : SUCCESS in 5m 09s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/0eb933d5f64d448a8107579988f02b8c : SUCCESS in 4m 08s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/adb1c05e2ce94637afb82cc61ed09c41 : SUCCESS in 6m 50s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/97a25502665e460599a9c970787f1c1a : SUCCESS in 4m 13s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/9eb368c8611348ceb3292097be778886 : SUCCESS in 4m 39s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d11b5717f4fd4799b41e1a1bd25b0e69 : SUCCESS in 4m 43s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/1605669df3d74a52921a6dc702c7bbba : SUCCESS in 4m 31s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a356f14ee2b747039c74a142e2229316 : SUCCESS in 4m 50s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/165362d652074f0ab8fe22b3e8b34735 : SUCCESS in 6m 21s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/7f49e8dcea7e46ed9c37a321b4093cb4 : SUCCESS in 4m 28s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/00340b7aa142479d9276cd34447a9972 : SUCCESS in 3m 44s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/16dc6e3d4e24487daef8c16cf6e0a105 : SUCCESS in 5m 33s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/4cef60a6320c4a2eb709df2e81a57b14 : SUCCESS in 6m 35s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/da88335b156845f9838f60246f44a5af : SUCCESS in 4m 00s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/a12e58732a8d4d5f8b3b396d01ecfa86 : SUCCESS in 7m 17s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/ea4c8feba8c3479186f0a7efe66bce8a : SUCCESS in 5m 09s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/8c7a428f38ef4e11b547868e76723aff : SUCCESS in 5m 36s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/c2683ccac5cc4d74a5a82063ca517e71 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":2},{"id":"2fec764c2694e523568f44d61e8f4daae4a341ac","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-10-10 13:59:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"921fcb00a1fb63663ef62f0a3ddcac570ea585e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 13:59:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3cc59ed3540731ae09e41d0ef1f524d19fe7af73","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:37:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e69637faf581cb0db164bfead7b5062d90149e6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:37:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a514e67f57b940d08392c96574611315 : SUCCESS in 2m 45s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f50ab4a012f44bacbda5e2095228a2a2 : SUCCESS in 1h 33m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f82204bfe1fb4964bcaadff6f3aec1bb : SUCCESS in 1h 24m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a8712c96f141c683dcc24d0a53a92f : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/51aef8ba20e94e13aede2d38b9cea764 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49f9ba0658644ae2ae91581d94792353 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/18ce938192a04d119a607ba8c8395a42 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aa47028cb9a434798db79399fcbb6d1 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f0cd15702c64ca9971495e31abf2e51 : SUCCESS in 3m 54s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/d877579e0ccf4865a1f529068637f8ca : SUCCESS in 4m 36s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/5c210b27537e4d55b302836805973b31 : SUCCESS in 3m 54s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/7bb66fd337034e5c88b05443b8847275 : SUCCESS in 6m 12s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/ca88e9c7bdbb49ef9307c4c1ebf7c895 : SUCCESS in 3m 58s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/453010ab5af94ad0a9dfc63b4f54a795 : SUCCESS in 4m 03s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/242c5ffae6e2457e9d6991b5176329e8 : SUCCESS in 4m 08s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/2a41909de5574dd881415ad7582388a3 : SUCCESS in 3m 41s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/88e7ce64eec840e9a21057c9fd78df43 : SUCCESS in 4m 02s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/14bed48e803f49569bd8017ce1063005 : SUCCESS in 4m 01s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/6485ea555a2040ecb1b807ef992ec9b6 : SUCCESS in 4m 17s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/e1a1a3bacf5e400f852e05ccaf9f359d : SUCCESS in 4m 03s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/fa0e13efe5f34aadb6d9ee4a5dfd14cf : SUCCESS in 4m 47s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/09417b889559454c8c0fd0b9f279b9d3 : SUCCESS in 6m 12s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/40f6b59afad0480f825951d0f6bb13ae : SUCCESS in 3m 46s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/e4488614922345b7b569bfe70d54e038 : SUCCESS in 4m 03s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/26bb85dff21049e8acb058eacce00133 : SUCCESS in 3m 40s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/ab2bcbdc9ef4494697064846551b8fd5 : SUCCESS in 5m 37s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/c6ccd3d6725349f5aefeb9ee8bf3bd98 : SUCCESS in 4m 09s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/96a2d3efb9ea4aab908c1fa48c818dcb : SUCCESS in 3m 40s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/0442676317504aec9aa7c832fe8e3bb9 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":2},{"id":"fcca7229f6a8e46a9769e50abc6dd58d35255e37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:39:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9144595e3764124a9acccf94f2e9e96 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c27608db635249ac957c656b1a1fc0e1 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4492bbaea810886e0e39109edca198ab10296da2","revisions":{"85c528b1a8ea4719a84628497be478c1f2e1a04a":{"kind":"REWORK","_number":1,"created":"2019-09-24 08:27:55.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/36/684236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/36/684236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/684236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/684236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/684236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/36/684236/1"}}},"commit":{"parents":[{"commit":"5b280a85c548ecc247e4bd491841aef4d8575257","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5b280a85c548ecc247e4bd491841aef4d8575257"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 08:25:47.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 08:25:47.000000000","tz":120},"subject":"Drop os-testr test-requirement and pretty_tox.sh","message":"Drop os-testr test-requirement and pretty_tox.sh\n\nos-testr is only needed to provide the \"subunit-trace\" command which\nis used in the tools/pretty_tox.sh script. And pretty_tox.sh is not\nneeded. \"stestr\" can be used directly in tox.ini. So drop\ntools/pretty_tox.sh .\n\nChange-Id: I6a18183f092689c8c9d283c19914da9baa59cc32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85c528b1a8ea4719a84628497be478c1f2e1a04a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85c528b1a8ea4719a84628497be478c1f2e1a04a"}]},"branch":"refs/heads/master"},"4492bbaea810886e0e39109edca198ab10296da2":{"kind":"REWORK","_number":2,"created":"2019-09-24 11:05:19.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/36/684236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/36/684236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/684236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/684236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/684236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/36/684236/2"}}},"commit":{"parents":[{"commit":"5b280a85c548ecc247e4bd491841aef4d8575257","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5b280a85c548ecc247e4bd491841aef4d8575257"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 08:25:47.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-24 11:05:10.000000000","tz":120},"subject":"Drop os-testr test-requirement and pretty_tox.sh","message":"Drop os-testr test-requirement and pretty_tox.sh\n\nos-testr is only needed to provide the \"subunit-trace\" command which\nis used in the tools/pretty_tox.sh script. And pretty_tox.sh is not\nneeded. \"stestr\" can be used directly in tox.ini. So drop\ntools/pretty_tox.sh .\n\nChange-Id: I6a18183f092689c8c9d283c19914da9baa59cc32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4492bbaea810886e0e39109edca198ab10296da2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4492bbaea810886e0e39109edca198ab10296da2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
