)]}'
{"id":"openstack%2Ftooz~707681","triplet_id":"openstack%2Ftooz~master~Ie22e944e270dfd37988724f76823f6fd38d19b21","project":"openstack/tooz","branch":"master","topic":"bug/1752225","hashtags":[],"change_id":"Ie22e944e270dfd37988724f76823f6fd38d19b21","subject":"Adds heartbeating to the consul driver.","status":"MERGED","created":"2020-02-13 16:12:46.000000000","updated":"2020-03-04 03:03:52.000000000","submitted":"2020-03-04 03:02:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707681-1583290936130-dc82b96e","meta_rev_id":"210d64a8977cf58b05ff230dd470786714573a8b","_number":707681,"virtual_id_number":707681,"owner":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-03 22:17:47.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-03-03 21:04:31.000000000","_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-04 03:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-03 23:58:22.000000000","_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":2,"date":"2020-03-03 20:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-03-03 20:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":1,"date":"2020-03-03 20:43:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 15:30:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-02 10:14:13.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-03 22:17:47.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-03 23:58:22.000000000","updated_by":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"reviewer":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"state":"REVIEWER"},{"updated":"2020-03-04 03:02:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2d2ba5d8410d656b68bb98172b1bcd369b74202","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-13 16:12:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dbab1f3456d8fa60575cef2552084f9b3a93ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:51:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6faa69d5885e45faa4b5f73cec891946 : SUCCESS in 1h 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/248877a999df425f9f59f5c12e7f9e88 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e6ddfb564b2481aa27eac907ff5f5bc : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e92096873684a98ac852a81991573e2 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cefa19ecf34dd7876e075e2ab063a0 : SUCCESS in 7m 35s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/e4e55539cc4945ed8d9ab7b3800fbc95 : SUCCESS in 8m 49s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/93521d49ca3943768e3000856e2b5000 : SUCCESS in 7m 20s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/8263a8fdb14c471fac716bac30f219e0 : SUCCESS in 7m 03s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/3652268049dc4800ab70f511f8325a52 : SUCCESS in 6m 13s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/9bc8be57ec18476e9d6b67f9a0e0d2b2 : SUCCESS in 8m 06s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7440a3edd7f44ee1ad31b89a6cf9d055 : SUCCESS in 7m 55s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/17d45ee1b57748119e0a1ad731e4c02a : SUCCESS in 7m 04s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/e3c3cd0266b84dd396e09af09656a119 : SUCCESS in 8m 11s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/1de6add24fa54d10b2a012876f462b10 : SUCCESS in 8m 51s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/7b5875e9d13e43649b6a5e48cfa673ce : SUCCESS in 6m 52s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/b4c2e58241004b3687f1f01ab3f840b0 : SUCCESS in 7m 18s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b3869b9fdd3446b198c8e6cb35ffca42 : SUCCESS in 10m 22s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/b6cb8c7d5ab64525a2ace1c6da59d497 : SUCCESS in 9m 54s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/733b42a2abca42d9ba7bab1ad0fee54f : SUCCESS in 9m 57s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/f79bd6b616df4a66b71e28e9c58907d9 : SUCCESS in 9m 23s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/79843e7585234679bbedffdfc70b5f16 : SUCCESS in 9m 18s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/108c489a7e7f4d53970b17ebb79685d5 : SUCCESS in 9m 41s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/e62a1632f54b4fb8ac48e40ded4869d7 : SUCCESS in 8m 39s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/e040859851014f9bb157c412a96d8c3f : SUCCESS in 8m 22s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/c0590f9c24d34b0ebe3f4ec2e529f1a9 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6bcddd5a7b2ffb3b2b5360f93b9133f4529eade4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-18 08:59:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nHello,\n\nThanks for these changes, they looks fine, I just have some suggestions and changes requested.\n\nFirst, in your previous abandoned patch [1] you spoke about a tenacity log issue, I suppose it correspond to this code [2], what\u0027s the related issue you found? is it still an issue?\n\nAFAIK these changes seems to make sense [3].\n\nConcerning the maximum returned `sleep_sec` it seems okay to me.\n\nConsul driver don\u0027t seem to have related tests so is not easy to introduce related tests to your changes.\n\nPlease add a related release note `fix` [4] to inform users concerning these changes.\n\n[1] https://review.opendev.org/#/c/648557/\n[2] https://github.com/openstack/tooz/blob/a4f1288ee4234865a4db8b9c191ba275e2deef3d/tooz/coordination.py#L184,L186\n[3] https://python-consul.readthedocs.io/en/latest/#consul.base.Consul.Session.renew\n[4] https://docs.openstack.org/reno/latest/","accounts_in_message":[],"_revision_number":1},{"id":"0d1cd570fdce7b5c407aa50362dfaa6371579489","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-18 13:06:35.000000000","message":"Patch Set 1:\n\nHerve,\n\nThanks for the review! I\u0027ll try to answer your questions in order:\n\n- Tenacity log issue: This seems to have been resolved by someone else. The changes i made  are present in the current master branch, so no changes needed on my end after i rebased.\n\n- I will work on adding release notes.\n\nThanks,\nNick","accounts_in_message":[],"_revision_number":1},{"id":"23af8d948436c90f32a3f42a9a012a5c2b7bd932","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-18 13:18:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1fc6e2ce77084cd8fe1705a41d082fdf3226efff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 14:48:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6e102c6a91e4e78bda7e760b69e05e0 : SUCCESS in 1h 23m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/698d3cd7000b45bc9ca304610ece6bf0 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3668eaae9bd24ffb9ce22ee67a010dcb : SUCCESS in 6m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09f2270f59154ea08cde25cbe8ff44c7 : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98b1fc283e24015857bfdc7e36f3cdf : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3319642a396f434cb97bcf67b9352f48 : SUCCESS in 5m 38s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/5d3fe6f3eb7e4a0c9cd6a7052cf376da : SUCCESS in 6m 58s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/204337aab44d461886cab37b7c1c56bc : SUCCESS in 6m 44s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/77a79468a2e74b26b99532a3abe5ff08 : SUCCESS in 5m 54s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/0d2cb281c5a149cc9d6d2e5dda2a6dcf : TIMED_OUT in 43m 20s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/9582c187c3ea4a478cb0cc8f470b7107 : SUCCESS in 6m 02s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/0ad3cbeee60c4d99bd01c6a4ccb9a5dc : SUCCESS in 5m 50s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/91dbc095f45f4b0887dcf3fb4d924f6e : SUCCESS in 6m 29s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/359fd538e8ca4a8592e11659edd88b6f : SUCCESS in 6m 36s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/e9eea19cc024438a83edfa9661ddb016 : SUCCESS in 7m 10s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a4e145baa86c47fe927e89c59fbf840a : SUCCESS in 5m 58s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/91648f0c9e954ad1a81a8209239e460d : SUCCESS in 9m 31s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/53d3dc31cd344375ac952831f0b3d425 : SUCCESS in 7m 11s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/df6d603bc5ae4f0194464599d93708a1 : SUCCESS in 6m 49s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b921eef3cf0e43fdad893c7cbf881728 : SUCCESS in 6m 48s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/78fc77048c034279ac452d90626ce24d : SUCCESS in 9m 57s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/a87144e402614f128ccc889ec9b0b834 : SUCCESS in 5m 40s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/20dbdd18a84b4f50b9874db169600c6c : SUCCESS in 7m 12s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/ae4d59b84d1a4c4b9793ae6d18a0c24b : SUCCESS in 6m 05s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/c80f991d728b42419df0e285135aab40 : SUCCESS in 6m 05s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/fc9414dbe2df471d8a900c2b8711d1d4 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1c9f31ee6f62983f3130c395cb027c339b311497","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-18 14:55:39.000000000","message":"Patch Set 2:\n\nIs there a way to restart these tests? Looks like one timed out and i\u0027m not sure how to bump it.","accounts_in_message":[],"_revision_number":2},{"id":"a1fa3cc501f4e056a6cc969728d980d2a091ed85","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-18 14:55:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c1858acbed00cc9ec117d0c2ba386dd8fee36d68","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-18 15:57:59.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4cc4586ff0e8ca6228c531074431a5bf53ceef35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:38:16.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7823b44241e74c6fb43d37f672ffe982 : SUCCESS in 1h 33m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2138e6426e54490aa612779ea2575f1 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a76cdad5ec774f879129cdad3ad41c8b : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccfb548713114de3a3e039fbf86e30e6 : SUCCESS in 7m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e324d6d77f1d4133a9034f830fbb8a7e : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64d75279115743819d07c9d71de8cc49 : SUCCESS in 5m 03s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/7a283b55f80744b3a14d2367703a4499 : SUCCESS in 7m 17s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/3d79187af528460ea6f334ed0cc1cec8 : SUCCESS in 6m 55s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/28259779a0c749a1bab008004adfd0ea : SUCCESS in 6m 31s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/e77dd11d0019407daec01709e9de3036 : TIMED_OUT in 41m 28s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/594432b57a4a4b6b8b01c9ca76013b11 : SUCCESS in 7m 29s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/75361672d4da4d6d88602295ff5a6a0b : SUCCESS in 7m 15s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/928fe91162bb4700850a33c0088f56a1 : SUCCESS in 9m 00s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/cf66c42a101b4fae9301c593af79834a : SUCCESS in 7m 16s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/d6328777d27748c2a7e9582f2a364ffb : SUCCESS in 7m 36s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/94dbe66968e64279aa545055cee69389 : SUCCESS in 7m 02s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/5cdaa45035844e119c8d2071ed9d1d1e : SUCCESS in 6m 43s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/a85cb6a4d9d9420191e73dbd04058754 : SUCCESS in 5m 58s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/3bd4156f4d904e3eaa82edba23d3f3a7 : SUCCESS in 7m 40s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/4d8968778dd24ac2802a6e5b572f8908 : SUCCESS in 5m 25s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/46aaca1a138e4c56ad9cff9d9b033f9d : SUCCESS in 9m 45s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/abc8948ed4624accb9d3dfddc5b822e4 : SUCCESS in 5m 18s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/77668fb24d6049198c547c47fd2d7e6e : SUCCESS in 5m 36s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/184de54e8d5347cd8ca02b79d14d3571 : SUCCESS in 6m 05s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/21b0d2aa824742fa81974a35f9fbdc5e : SUCCESS in 5m 43s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/a50d32b8b75042a2b89ce093f3b25c87 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"162d42f9b60e4947e3f731086e87ef9496094c9a","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-18 17:42:54.000000000","message":"Patch Set 3:\n\nHervé,\n\nAny advice on the etd3 test timing out?\n\nI tried doing a force push to see if i could get the build to pass, but no luck.\n\nAny thoughts here?\n\nThanks,\nNick","accounts_in_message":[],"_revision_number":3},{"id":"32ed85fdd1acf50b9e5d55bef3fcb1920470bf6f","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-18 20:57:40.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"6ffc74d60e3bad3edfc29c35030c7dac50a007b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 22:28:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/596fa771914e42b1b78248f57454e93f : SUCCESS in 1h 25m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/575588090cca40d5925a1ff9622ebe77 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdfc960b76844da3818a71fcbd4b1c09 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d98b2ac92a224de4b0e99fd7929d9394 : SUCCESS in 9m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45dfce38b5684e89a5c85508fb7e7650 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c20f05c2294419381b7476133491477 : SUCCESS in 8m 10s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/b271a9f947384f22ab9469d362f5f128 : SUCCESS in 8m 09s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/b1d4c456ff164da39a3ce9c2a7c2e5e8 : SUCCESS in 7m 26s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/2a838f3075d04fdcb609a9eed5ba6bc6 : SUCCESS in 6m 58s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/f26e42b0471c4ed5b401bf8319b7982c : SUCCESS in 8m 34s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/c4a2a8d223d1433c91df1a32725c006a : SUCCESS in 7m 57s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/f5a0a399ee0b464daec0f6905057fe39 : SUCCESS in 5m 47s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/95cb5b4ea4954cb1b68f7a7f3ca6b3d5 : SUCCESS in 10m 49s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/8721d982215b47eb8f76a0108009e89e : SUCCESS in 7m 09s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/af31718b51204a7c9b753838efb624a3 : SUCCESS in 8m 24s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/15c0d0c404dc4558858f1ae602010e9b : SUCCESS in 9m 09s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/5c4fc86365fb4124a58ca8a4e7f82723 : SUCCESS in 9m 08s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/0e17fa6364584ce093c3548008872efd : SUCCESS in 8m 15s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/6151fac9bcf94361afcf0f0975d498e5 : SUCCESS in 6m 32s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/2701ae0da7754e36affe7e9ed9802983 : SUCCESS in 6m 51s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/78fad9f9c5534158bb7d28c445265017 : SUCCESS in 8m 51s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/2b27b56ecc8c4374a3020c715e7b7f91 : SUCCESS in 8m 20s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/6736f9b3cea94510965b277939cb7752 : SUCCESS in 8m 49s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/33a898a66fcf4963a98bda967ce019aa : SUCCESS in 7m 18s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/0f83eef164f149dba5ee315179f5505f : SUCCESS in 5m 53s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/f13ef5053d294789b43b80a5561298dc : SUCCESS in 9m 30s","accounts_in_message":[],"_revision_number":4},{"id":"64cf2e19a57def53206a711ca44284e0a0a85314","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 06:54:24.000000000","message":"Patch Set 4:\n\n\u003e Hervé,\n \u003e \n \u003e Any advice on the etd3 test timing out?\n \u003e \n \u003e I tried doing a force push to see if i could get the build to pass,\n \u003e but no luck.\n \u003e \n \u003e Any thoughts here?\n \u003e \n \u003e Thanks,\n \u003e Nick\n\nSorry for my late response.\n\nNo sorry I do not have good advice to give to you... you can try to `recheck`.\n\nAnyway tests are now successful!","accounts_in_message":[],"_revision_number":4},{"id":"ab1fb038a3a29444aa95cd88eace5a9da1c1dd95","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-02-21 04:36:18.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2e9a0a1b6e700750fd94564f8da0ecd01675ca4c","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-02-21 04:45:16.000000000","message":"Patch Set 5: Code-Review+1\n\n\u003e Patch Set 5: Published edit on patch set 4. Fix Typo","accounts_in_message":[],"_revision_number":5},{"id":"a7b3dd6f9ec4dc21329a4671fe6eb7903aeca2f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 05:58:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d9e92e9e7c042668042ed578dc63204 : SUCCESS in 1h 21m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a7e2c25ea2c4dab8eaccaa08b55d5be : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d932c3c5f2b437dbeead59e193a5d3d : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/680476242c00488f922d3111966b42ae : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b52d8cd5f2448dcb3ddc9bd5e369097 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7e682171907494ca839aaecebd277cf : SUCCESS in 5m 21s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/dde4edfaed0045849c852f789a39d0de : SUCCESS in 5m 44s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/ac5f96c7ab4242bd8ca452e873ff896b : SUCCESS in 5m 30s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/d1f6adcbb68743c183ae4b34a6c46ab9 : SUCCESS in 5m 48s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/17cf79c6ae82419b9dd26df8633882b4 : SUCCESS in 5m 28s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/d55b572726f741c395e945f22faa6bb6 : SUCCESS in 6m 28s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/e5728999fd4c4690a03aaff5e22e2f5c : SUCCESS in 5m 21s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/ee2c5783cb154df78056769ab18b76db : SUCCESS in 5m 52s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/e28c99c3142b47758de6c7c94fc7cc1a : SUCCESS in 5m 29s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/42bd244760ba40edb9313279ddcb3e4a : SUCCESS in 5m 47s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/e12a700aea5f483ca752f2517da2207d : SUCCESS in 5m 12s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/c447b58561c94fb4bb152bc152217eae : SUCCESS in 5m 28s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/015631bf3e9341bb8965cfada2657506 : SUCCESS in 6m 04s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/37a196b7e4054af7948ada65cdf0c188 : SUCCESS in 5m 33s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/7ec22dc97aff43c2902a9abbf7c7f29d : SUCCESS in 5m 24s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/49c0a0c85c7349428657c3b122fdbfa1 : SUCCESS in 5m 43s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/a162836083e34eebbf963c9963544068 : SUCCESS in 5m 17s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/85685d9df51949e9b3aed366ca150fc3 : SUCCESS in 5m 20s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/66d520fa74ca4c34b77aa8a51ec25b11 : SUCCESS in 5m 08s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/f7603f645e1545e78163f79f8779b8c1 : SUCCESS in 5m 12s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/a33198b4d26c460d9595ed9cea59fc7e : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":5},{"id":"cc41aa2e75e80799c0ea6889c392816cfb8adf13","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-02-22 15:34:16.000000000","message":"Patch Set 5:\n\nHerve - anything else needed to get this merged? Or just wait for you guys to merge?","accounts_in_message":[],"_revision_number":5},{"id":"57069fa9e9a177fb7e2b2a8055ee03ab20809298","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-22 19:19:43.000000000","message":"Patch Set 5: Code-Review+2\n\n\u003e Herve - anything else needed to get this merged? Or just wait for\n \u003e you guys to merge?\n\nThanks guys for your changes :)\n\n@punkrokk: We need another review from another core developer before these changes to double check that everything is fine, also thanks for nit fix.","accounts_in_message":[],"_revision_number":5},{"id":"48f9f8e29f76f3cb8fbf5182307c1b0a151b6a7a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-22 19:21:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a73ce3d4923a6b84e50390796eaa7fef7a3a4c8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-24 14:28:29.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)\n\nLooks good overall, but we should get the release note cleaned up before merging.","accounts_in_message":[],"_revision_number":5},{"id":"64bd826d5d4b95dc056786a75de44641f209232b","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-24 14:34:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5a0b345a427f311e63dad26750651c06595af600","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-24 14:35:56.000000000","message":"Patch Set 6:\n\nSean,\n\nI\u0027ve removed the \u0027prelude\u0027 section.","accounts_in_message":[],"_revision_number":6},{"id":"e5d5ff9f4c0fcff5ffb6ca5993a79304c55b06b3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-24 14:37:02.000000000","message":"Patch Set 6: Code-Review+2\n\nThis looks good to me. Assuming zuul will be happy since it\u0027s only a release note change since the last rev.","accounts_in_message":[],"_revision_number":6},{"id":"28bfd3b00430e00e0c0f3e64f7ab894d60f18681","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-24 15:18:22.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0938d3af0f44e6836fca52800fa50a3f6f3e926b","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-24 15:28:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fea9b1129d60e31509b2149664cc70f12736441d","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-02-24 15:29:17.000000000","message":"Patch Set 7:\n\nHerve, I fixed the spelling mistake.","accounts_in_message":[],"_revision_number":7},{"id":"48168a47d7b2cdb2bd1de1035edca1d0f7df480b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-24 15:30:46.000000000","message":"Patch Set 7: Code-Review+2\n\n\u003e Herve, I fixed the spelling mistake.\nThanks","accounts_in_message":[],"_revision_number":7},{"id":"5ac8a4ba4a4ccf60732ff70414ffaba77e2048b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 16:59:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9dc428ecb86f48609cc0691a06474e66 : SUCCESS in 1h 24m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aff3016c89e046c580fcf0e00241118e : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4095b55efabf44c18352375b73118e22 : SUCCESS in 10m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23e98761e1a44b7d819abae6224617f5 : SUCCESS in 10m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0182538c320d4dfcb18939ccadf8c5a9 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b1cfa9f164e4a3ab822faeba67ec8bd : SUCCESS in 9m 42s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/c1885ef26e874d30b4f6b12eec6c2948 : SUCCESS in 8m 35s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/508ae95ab4dc4a3994e09806bd2844ac : SUCCESS in 6m 39s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/15f3f308ba964a2a83a4ee8275d8cf09 : SUCCESS in 8m 23s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b964308e440a4b5abafc1e644997ad43 : SUCCESS in 7m 51s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/f00178687ffa4acbbd2452eaad254cf0 : SUCCESS in 7m 14s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/bd8d1419e2ad41ff8bc89c6440951398 : SUCCESS in 6m 05s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/55013d14aa0444f3b1a6a8546c82d916 : SUCCESS in 7m 29s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/7ab7f978f2c34b55a8c7ed3a6da8f940 : SUCCESS in 7m 10s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/d66babd8d2224006bac8bedfc1b9ee89 : SUCCESS in 10m 33s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/0dc09bbcdf2e4d578816f48ea13d6a0e : SUCCESS in 6m 51s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/609a02b68302462ca2da49528c0896c9 : SUCCESS in 8m 59s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/33059614505e48deb839f8b298b0e2e9 : SUCCESS in 10m 53s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/6eae0c7c4b4743a9af14e0129c5ec4a1 : SUCCESS in 11m 07s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/7e430b808e294b6daddcac1d23f6cbc9 : SUCCESS in 7m 21s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/c46121a1aeb349b99aacc28943388c8b : SUCCESS in 7m 38s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/58d5dac575954102a2dc33c1dfd5d7ee : SUCCESS in 10m 13s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/578e544d112742359b3066f1f394edac : SUCCESS in 6m 18s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/03f4a701c55e458891aa870160a024a5 : SUCCESS in 7m 08s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/5cd789bbacb648a8b7d28516a9957f44 : SUCCESS in 9m 15s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/0b75c0fa378c4e99a7ac6b9efad323db : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":7},{"id":"56dea86bff76f1899b7deb13f34f552d5d3c66bf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-02 10:14:13.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"bb30bd10969f59bdad24e1384fee090434599e19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 10:14:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"302a7e393ead079e2bd89b93ec290b8ad59a8224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 11:59:13.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6dafcceb9ff84f348493954403fd6971 : SUCCESS in 1h 37m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/975ae0b7a84f48dcacc97128e99cd484 : SUCCESS in 8m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9a13c73f023404c81317d704498fe81 : SUCCESS in 6m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4c1b85bfe3647a1962113b1d87923b1 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dbb67cb51054455977dc476065e1d29 : SUCCESS in 5m 57s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/4d8991a4c117492eaf0aa71aa59f4ce6 : SUCCESS in 6m 41s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/39559ba4a5a5471285a78904a22cb289 : SUCCESS in 10m 12s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/c90ac5f27df44abd9533d1d001d4fd7f : SUCCESS in 7m 20s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/1c1537804a0a45fdac760056963d77c3 : TIMED_OUT in 41m 56s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/a1273a3adc9c441692f621be2e67468a : SUCCESS in 11m 28s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/e56cebd34616478fb69a10a2734e9272 : SUCCESS in 5m 04s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/9c57cbfc4f5f4e3d9bc7f9836d880e29 : SUCCESS in 8m 08s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/19467ec44d1f4c10861997fbfb937874 : SUCCESS in 7m 52s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/c5c5cdb69e974438afa445aaa06d2418 : SUCCESS in 8m 08s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/bc10ab4db9954b71be73501dbac5685a : SUCCESS in 6m 58s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/4da0139fc7ee412b9a1f047a89910fce : SUCCESS in 6m 45s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/c03e1d0a66ba4560a78ac880d897b03b : SUCCESS in 6m 12s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/59a87021fa0349909dc452410c6625b5 : SUCCESS in 8m 13s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/6417822b0ddd4489aa63a09ec13a9123 : SUCCESS in 7m 46s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/6e0b46b578444451b2f1f61164c84967 : SUCCESS in 6m 29s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/c4eebfe074274b119956c45e36e0605a : SUCCESS in 5m 48s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/25dc7c7f78e1446ea6040afe985fbdc1 : SUCCESS in 8m 43s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/c7289a257ebc42c788d0d4da5cecb5e0 : SUCCESS in 5m 59s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/1199c4d6c2fb4e1ca52eead332e9de08 : SUCCESS in 8m 10s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/130f8d7bb57444ddb8d5370ce73ffdde : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":7},{"id":"32413a4f1d1e4712b1a690357380731cf7d57e8b","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-03-03 20:43:13.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"a733bc9177ee4c14bbada791feb1ec9e3c78d7c8","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-03-03 21:04:31.000000000","message":"Patch Set 8:\n\nLooks like Zuul had a random timeout failure. I force pushed again to kick off CI. Any chance we can get this merged?","accounts_in_message":[],"_revision_number":8},{"id":"794381e7efa043f2786d7a41fd2b3a567d1f18a7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-03 21:09:51.000000000","message":"Patch Set 8:\n\n\u003e Looks like Zuul had a random timeout failure. I force pushed again\n \u003e to kick off CI. Any chance we can get this merged?\n\nIf that happens again, just leave a comment on the patch of \"recheck\" and it will kick off the jobs again.","accounts_in_message":[],"_revision_number":8},{"id":"9b7708b619b2c9629eaf65a8315c9a3cc8ba96df","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-03 21:30:58.000000000","message":"Patch Set 8:\n\nLooks like an intermittent error in the gate again. Once zuul reports back, just comment \"recheck\" to get it to run again.","accounts_in_message":[],"_revision_number":8},{"id":"437d3b48d8c017b4975d66edb115083d97ac7415","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 22:16:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bde8b682cccb46378a1bbefd1b485a48 : SUCCESS in 1h 27m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24d10c95e3ed4161b5d9f1ec665550e8 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c95dc84f4df3453887518337a95af2f4 : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/50b8252c0a6d456288000dfc5403924a : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1f64d9419240eab42f5d67c6e634bb : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52eb0c048c384c9a806893e14afe2ef1 : SUCCESS in 5m 00s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/ed257d1e14b04fb4ab4d92d6d8ed6fba : SUCCESS in 6m 14s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/dfd212fc05424df5ab47d53598a6e41a : SUCCESS in 5m 53s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/845daadf6da94826855a55cb6519d1e9 : SUCCESS in 5m 47s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/ac993b8bb96749348ff00396e3bdfc31 : TIMED_OUT in 41m 11s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/c99f425290d8471380208d3fcdd7239b : SUCCESS in 6m 04s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/d9d850c09ef244cda8f56bced960afa0 : SUCCESS in 5m 03s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/18955d7fb32e43b892c6f4c2af4a3f86 : SUCCESS in 5m 26s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/a4b6af15f39c42f39a459de185ac6ee4 : SUCCESS in 5m 23s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/b13fdce60aa142898e26f4e3c9a3a443 : SUCCESS in 6m 04s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/4f40366009fd42c68aea4bdc8bc12ec1 : SUCCESS in 6m 07s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/8cf0fbb81f164992afc7f066d7a016b7 : SUCCESS in 6m 08s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/878d9904d9254002ae2da1f88df39f42 : SUCCESS in 6m 10s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/da760098e6a34d7a8ca46d3f2c7777c9 : SUCCESS in 6m 19s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/1068b080c03d4a1cb95ab3db9e3b2716 : SUCCESS in 6m 03s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/f96f38396925431ebc73c6e41e7136d0 : SUCCESS in 8m 45s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/0be48394bf7a43d5aaac41009b953f46 : SUCCESS in 5m 14s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/b131b51d5e3c42e9863a2fee6bd18033 : SUCCESS in 6m 51s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/2dbfc59588b94e48ba1150b1e8474d7e : SUCCESS in 6m 52s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/b2040acad56140eea0e21bb78f7855aa : SUCCESS in 7m 03s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/0112b9c53a5d40fbb0031fc0914a8899 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":8},{"id":"4fb9ac525eb6a0de244bb6ba49c86750de42e369","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-03 22:17:47.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f2ef8a0505d0e313e67a20b774b8b8718dc13bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-03 23:43:56.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5d1cf68ccf5458599f59d2c79c70cef : SUCCESS in 1h 21m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a541c8d8d40a47548de22b308f8f85ef : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/223e358a2d9b4cc1ab9013b61f258526 : SUCCESS in 6m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b28d3593db80458c8308b3068065e128 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91283e8079ae47afa1ae82516fb32747 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd031034b1644722b14c2b6f1f7c9f1e : SUCCESS in 4m 31s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/1e2e3367e8d149bfb0af56e7ac105809 : SUCCESS in 5m 31s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/bd3a3289c2eb4061acc5acb74ce864fc : SUCCESS in 5m 15s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/d18aead417e84d3da4a566ce1b87ccc0 : SUCCESS in 7m 16s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/645e2698f66d429c886d1a829e04fb79 : TIMED_OUT in 41m 42s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/06c4ea4efe42412f87628f2f10790c6b : SUCCESS in 5m 42s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7dd2fb50278843d69f4b22841c685f9b : SUCCESS in 7m 18s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/d4a13e0d793b49439e3a062b376dc207 : SUCCESS in 7m 23s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/8a20e39335f24dd69522c9e3d15a0966 : SUCCESS in 6m 34s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/4ed2a52d512a4672814c4684ead55461 : SUCCESS in 6m 34s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/77616cda1cdb4612b8a44315131d5820 : SUCCESS in 4m 49s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/7ac4c3542e1e4e80a0198c3e3725f6bc : SUCCESS in 5m 16s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/d065ffcba5004f83a2c56af323fbb0e7 : SUCCESS in 5m 04s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/eb2bef8630724d3a88ab3bbeac96037f : SUCCESS in 4m 59s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/29b261a7a1954352a68c6f98662b1e29 : SUCCESS in 6m 12s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/49a9f9e3c4ee4b5481f82f08fe9e4e4e : SUCCESS in 7m 04s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/df24d5a792fc46cbae78a4d55df1d6b5 : SUCCESS in 5m 25s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/16b8a53ace124aaa839a33c52ed0597c : SUCCESS in 6m 55s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a710374ad3494f56b5b86ae2d3810e43 : SUCCESS in 5m 20s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/d88d5174693a4a8d926fe608b0eb0a0e : SUCCESS in 7m 44s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/a2306ec059b6432ab4f11999d0facdd9 : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":8},{"id":"d89afe8008882dd2dc797f0b75249accc18d24b8","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-03-03 23:53:31.000000000","message":"Patch Set 8:\n\nRecheck","accounts_in_message":[],"_revision_number":8},{"id":"bf480b19c679259d796008c1076ef9b19d60e219","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-03-03 23:58:22.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"ba06070655b595ad65fb1e2807d9eb0c336e869d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 01:33:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c323652cb6d34409a83ca4be8b25b860 : SUCCESS in 1h 34m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/126de0f93c89440eb1fd2b42e003e527 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c72df3f7bc14d06adff0a9422786951 : SUCCESS in 8m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d15aaf843d1e49ee807377f87f8c4dc5 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fbf108d71bc4ada88107ed3c33f81b4 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc064a5b8c6e4301b53e457f74b29598 : SUCCESS in 5m 15s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/d6aa71607acf4f0084f07defd06576c9 : SUCCESS in 7m 39s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/f6514a23cb514812b23851c517ead8cb : SUCCESS in 6m 14s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/556efa229aef4b6fbb20bbf4cc0a6ead : SUCCESS in 7m 26s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b70d3d7454e44cc0a5904b15a2be986d : SUCCESS in 4m 49s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/20fbcf3fadb4435cb4799f0aa32e03d5 : SUCCESS in 6m 29s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/0dcdf7cd463a44f29fa975c124f4596a : SUCCESS in 5m 18s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/e316ce62337e4492bbf0bbb5f0e258c0 : SUCCESS in 5m 52s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/798cd2ee69a443508464e3e3a1e06478 : SUCCESS in 6m 11s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/024aed2094454d56b3bb937662d4bdda : SUCCESS in 5m 09s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/dffa8bfc25e64339a92f53b6ee39ea08 : SUCCESS in 5m 09s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/ad99a4542f8b455ab0156e42f0105c34 : SUCCESS in 7m 25s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/6bf42a5274b94b58a2ca27f667bbb1d2 : SUCCESS in 8m 33s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/be481d3283354534ba0860e42128a097 : SUCCESS in 6m 02s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/4f94d67abab54179b8d69359ec86da65 : SUCCESS in 5m 05s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/aa90606fcde64eb28d85035b4ffeaa0b : SUCCESS in 6m 43s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/d10cfde98fff4c498482b072e11edaa4 : SUCCESS in 6m 10s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/fac5e5b790d34fce83a1a6583dc916ad : SUCCESS in 7m 00s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/03294a4c95f54d96b8781c3fb88ca82a : SUCCESS in 6m 27s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/cf00f88ce73b4092a6ad729a8817c9af : SUCCESS in 7m 29s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/42e8c0b2412541cdb8bbf03ecdfc13e7 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":8},{"id":"b4c7cd1a61362b86fa94d5d89a09a4f2102226c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 01:33:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"21a8f58739c369754f397a14bf7450d37d87dfc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 03:02:15.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/685e8714394046e3a38eeb76c0a07153 : SUCCESS in 1h 23m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e12cf8f8e24663813e99793d80cc02 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dfcffa797854a2b9709ff93ea548bff : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10adf74d726547a889f64c254e3a3cbd : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18d1f94a97e04521bef5c6bfee7b45af : SUCCESS in 4m 09s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/12f6cc2836934ee4ba3a5939a7d8ad60 : SUCCESS in 5m 10s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/303cd381afd54e7680b5ea1012f5283c : SUCCESS in 4m 18s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/d9b2183934d84d90a4852818f7bdceb8 : SUCCESS in 5m 15s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/37a8fcc2f26745a59b24c0b9e0dabc32 : SUCCESS in 4m 45s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/034e672a654b481a80091b439fbebda2 : SUCCESS in 4m 58s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/87ca79018ae4478bb4339515e668e04d : SUCCESS in 4m 18s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/1d7aa38538cf494486e03a0cb9e2df3a : SUCCESS in 5m 03s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/5a558954de42429b9c9b6d5aba3c29a0 : SUCCESS in 4m 46s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/926db898a726400a9a7cc455903103a3 : SUCCESS in 5m 10s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/245f7f2ec12643f0add0e675f894b429 : SUCCESS in 4m 51s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/7b7b0d7b0cc24b6c8bdb5b01da2456a2 : SUCCESS in 4m 27s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/00019daa97f94214bed02ef18258ff67 : SUCCESS in 4m 41s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/c1fdf87459c24160803e0bea6df3aa2c : SUCCESS in 5m 03s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/7bb82a1122b04918984473d9a59a0e69 : SUCCESS in 4m 34s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/9080111a9f0d42dd97133b67ed86b263 : SUCCESS in 4m 33s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/bb99842c73984380966691091037066e : SUCCESS in 4m 17s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/574a265ecf094d90874eb4e890c18bd5 : SUCCESS in 5m 17s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/de8028006462498fb7e612ee1094ea41 : SUCCESS in 4m 23s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/03cfe94e74e54fd0b2083b1d0625911e : SUCCESS in 4m 58s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/d4fe572334784fa2a125c1886a616aad : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":8},{"id":"1d20f4e79d17913dacd3517dcc80411a9176e85a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 03:02:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"4a468c190a27624227ba315244a088efe349e8a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-04 03:03:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936afda4c26d419697e9e445eb3be2c7 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/777b2f77cf0a4f8fbc82e6fa7907fecf : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6e750b6921c42f3c2ffed2e135be4d9c65e96f36","revisions":{"1a22fe77a173b8f3e531bc8281013a7a29837940":{"kind":"REWORK","_number":1,"created":"2020-02-13 16:12:46.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/81/707681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/1"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:12:29.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a22fe77a173b8f3e531bc8281013a7a29837940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a22fe77a173b8f3e531bc8281013a7a29837940"}]},"branch":"refs/heads/master"},"812387acad0e61ce533cd771294f8d2a1a7aa3ee":{"kind":"REWORK","_number":2,"created":"2020-02-18 13:18:12.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/81/707681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/2"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-18 13:16:38.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/812387acad0e61ce533cd771294f8d2a1a7aa3ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/812387acad0e61ce533cd771294f8d2a1a7aa3ee"}]},"branch":"refs/heads/master"},"1c24283d515da7047ad64d43078052c3ad668806":{"kind":"NO_CHANGE","_number":3,"created":"2020-02-18 15:57:59.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/81/707681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/3"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-18 15:57:41.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1c24283d515da7047ad64d43078052c3ad668806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1c24283d515da7047ad64d43078052c3ad668806"}]},"branch":"refs/heads/master"},"e373bdb392e4eec74589bcdca68fceeae34436a3":{"kind":"NO_CHANGE","_number":4,"created":"2020-02-18 20:57:40.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/81/707681/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/4"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-18 20:57:30.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e373bdb392e4eec74589bcdca68fceeae34436a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e373bdb392e4eec74589bcdca68fceeae34436a3"}]},"branch":"refs/heads/master"},"1b67b76ebbad4f7793a9558e78cf302caa875e5a":{"kind":"REWORK","_number":5,"created":"2020-02-21 04:36:18.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/81/707681/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/5"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"punkrokk","email":"jp.bourget@gmail.com","date":"2020-02-21 04:34:25.000000000","tz":0},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1b67b76ebbad4f7793a9558e78cf302caa875e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1b67b76ebbad4f7793a9558e78cf302caa875e5a"}]},"branch":"refs/heads/master"},"74fd02abcb88b6e16261a700917ff8ddf8ed4afe":{"kind":"REWORK","_number":6,"created":"2020-02-24 14:34:57.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/81/707681/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/6"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-24 14:34:46.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/74fd02abcb88b6e16261a700917ff8ddf8ed4afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/74fd02abcb88b6e16261a700917ff8ddf8ed4afe"}]},"branch":"refs/heads/master"},"987fc610586c54ec7e85410236f798b6d98a3bf5":{"kind":"REWORK","_number":7,"created":"2020-02-24 15:28:59.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/81/707681/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/7"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-24 15:28:51.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/987fc610586c54ec7e85410236f798b6d98a3bf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/987fc610586c54ec7e85410236f798b6d98a3bf5"}]},"branch":"refs/heads/master"},"6e750b6921c42f3c2ffed2e135be4d9c65e96f36":{"kind":"NO_CHANGE","_number":8,"created":"2020-03-03 20:43:13.000000000","uploader":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"ref":"refs/changes/81/707681/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/81/707681/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/81/707681/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/81/707681/8"}}},"commit":{"parents":[{"commit":"ba27954b0645398dc7ccd297d47343464f4b1268","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba27954b0645398dc7ccd297d47343464f4b1268"}]}],"author":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-02-13 16:11:35.000000000","tz":-300},"committer":{"name":"Nick Maludy","email":"nmaludy@gmail.com","date":"2020-03-03 20:43:00.000000000","tz":-300},"subject":"Adds heartbeating to the consul driver.","message":"Adds heartbeating to the consul driver.\n\nChange-Id: Ie22e944e270dfd37988724f76823f6fd38d19b21\nCloses-Bug: #1752225\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6e750b6921c42f3c2ffed2e135be4d9c65e96f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6e750b6921c42f3c2ffed2e135be4d9c65e96f36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
