)]}'
{"id":"openstack%2Ftooz~714929","triplet_id":"openstack%2Ftooz~master~I4593085e60930f039d13db2a4c7799b16a1f055a","project":"openstack/tooz","branch":"master","hashtags":[],"change_id":"I4593085e60930f039d13db2a4c7799b16a1f055a","subject":"ignore reno generated artifacts","status":"MERGED","created":"2020-03-25 12:18:10.000000000","updated":"2020-04-29 11:26:41.000000000","submitted":"2020-04-29 11:20:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714929-1588159239210-5004bf0a","meta_rev_id":"185f249de224618a86e9a9ac47d86c649a99c65c","_number":714929,"virtual_id_number":714929,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 11:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-04-29 09:49:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-29 03:55:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-04-29 09:49:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 03:55:43.000000000","updated_by":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"reviewer":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"state":"REVIEWER"},{"updated":"2020-04-29 09:49:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-29 11:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12f0be0a95f202308d0e348fed0391cd8eaabf57","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-25 12:18:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb1a9c4ac582f1ba36b1109bfb851259ea05bd8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 13:49:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/12f686c554cc490b9703efd52bd74564 : SUCCESS in 1h 30m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddb6723bd4974d7087775173fae83d7c : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0af1705d71c1443f83cd33bac4fa5201 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ef2a2294cbca4db386d95f4116a65b64 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e7721e742da4408a1db25321efc9d93 : SUCCESS in 5m 15s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/c76d999cd66d4a6ab73598b8092a9248 : SUCCESS in 6m 15s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/a84fe935a3f748e8921e9c6bb74ee943 : SUCCESS in 5m 26s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/14479dca6eaa425e8fac427a3dec5d9f : SUCCESS in 5m 12s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/8d0580ff86a04db78bfa12e60ed7539d : SUCCESS in 5m 42s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/3ce2a4a2848642cbab8062172328095c : SUCCESS in 5m 59s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/34e0dfddda0c45f297bbf5cc52748527 : SUCCESS in 4m 21s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/b46fc11aca2f436196f895de5fa5d793 : SUCCESS in 5m 14s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/217aea85cd564b5fb8a76813044551b8 : SUCCESS in 4m 45s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/0df47ec3162e403db308533091a53a17 : SUCCESS in 5m 36s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/b036a2013ab7421e8fc2463525844c61 : SUCCESS in 5m 41s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/5069fb0c006d4e13ab49ae569ecc6282 : SUCCESS in 4m 36s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/7ec9016d827849d5aeb0e6afe1066c89 : SUCCESS in 4m 57s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/2ef18aaa8f8441eeb6002014051b3656 : SUCCESS in 5m 32s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/3e3fb572b062408c84950236936249a9 : SUCCESS in 4m 38s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/5a6ae299f78746d49b57bffa545decd2 : SUCCESS in 5m 21s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/50961578254b4d0194f0ce79eef9e8c7 : SUCCESS in 4m 30s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/806da1d55ffc45abb70d67214f42f7fc : SUCCESS in 4m 53s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/ce18bbda84aa4f2493b3244f5456ef6a : SUCCESS in 4m 54s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/c209ad6ebd4f45fdb24ff0ffc5a444a9 : SUCCESS in 5m 54s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/347aa74c8b144d5db83e9d332926dc80 : SUCCESS in 12m 21s","accounts_in_message":[],"_revision_number":1},{"id":"db95d8da6e363e620d639a1e78f667a4608647bd","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-29 03:55:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"32a6be92ec2b9e221d1326463febe604605ba0bb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-29 09:49:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf2cb5ef7caea188ea74fb23da42926375dc0ccd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 09:51:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"946bc118c06abc6e9db9e595a094cffa5d717f4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:20:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82ad5de3032a4be48b80fe4d3cba7ee5 : SUCCESS in 1h 27m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c79398d95e0a4ec682d30ca742dc90b0 : SUCCESS in 8m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0181334206db43c7b60b94c53f827ab1 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8aa0ed265b4bdca055aaff45c65357 : SUCCESS in 5m 28s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/fdc9f4b5a23a4b5dabf7aaca95658f7d : SUCCESS in 5m 43s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2b0c7b14dba34295bd7a7e1557413276 : SUCCESS in 5m 56s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/781541eaaecf4b47814f2d2214344e0f : SUCCESS in 6m 31s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/f19dff3deafc4e2f9e6359d7ed6d9ebf : SUCCESS in 5m 14s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/d75e66b7085d4391951a2df6aa87e14c : SUCCESS in 6m 11s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/449449ba621545ae9aafb293a2803eb9 : SUCCESS in 5m 18s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/45278013c7b7499ba9b5083f93e97ca6 : SUCCESS in 6m 09s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d37a42a627d34e4a83db544b8bb239fa : SUCCESS in 9m 08s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/a54367e702024ac4b908c75b24ebc718 : SUCCESS in 9m 46s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/14817949da204810b2c11d2d4f1826ab : SUCCESS in 8m 59s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/c0dd1f4e25274e18919b5bfb51017fda : SUCCESS in 8m 00s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/c861ab58d6464cbf815495145eb64cf5 : SUCCESS in 7m 48s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/fe710a362b294a00bac2fac38f88401c : SUCCESS in 6m 34s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/cde23983c0f54bc8a081377ceb124355 : SUCCESS in 5m 53s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/8006fce2a44d4889b762a5fc6ae8996b : SUCCESS in 6m 54s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/f0c442045c5a4b68898f47e636705b52 : SUCCESS in 5m 40s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/464ee6592a094166a1a91b0957b17af1 : SUCCESS in 6m 43s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a1d66e538d2443768f52a2c9fdf15693 : SUCCESS in 8m 09s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/596372f0c6434e279502bf428964f965 : SUCCESS in 6m 27s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/408e9e1522ed46b7bd606851c562e8ae : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":1},{"id":"f18454ec52a0ac67d287f70c89a495c28c52efc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:20:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"185f249de224618a86e9a9ac47d86c649a99c65c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:26:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80df467cf14a413a8fcdc4f096742f0f : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4be7780adbd9235233143880044a56e8586b5f68","revisions":{"4be7780adbd9235233143880044a56e8586b5f68":{"kind":"REWORK","_number":1,"created":"2020-03-25 12:18:10.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/29/714929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/29/714929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/714929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/714929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/714929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/29/714929/1"}}},"commit":{"parents":[{"commit":"6e7a8508fbcbe9ceacb04490dc8330501bfa2477","subject":"Merge \"Adds heartbeating to the consul driver.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6e7a8508fbcbe9ceacb04490dc8330501bfa2477"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-25 12:17:59.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-25 12:17:59.000000000","tz":60},"subject":"ignore reno generated artifacts","message":"ignore reno generated artifacts\n\nChange-Id: I4593085e60930f039d13db2a4c7799b16a1f055a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4be7780adbd9235233143880044a56e8586b5f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4be7780adbd9235233143880044a56e8586b5f68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
