)]}'
{"id":"openstack%2Ftooz~715922","triplet_id":"openstack%2Ftooz~stable%2Ftrain~Ifc4b8c291e757d143806cdf16823ff8f3dbe71b2","project":"openstack/tooz","branch":"stable/train","topic":"do-not-merge","hashtags":[],"change_id":"Ifc4b8c291e757d143806cdf16823ff8f3dbe71b2","subject":"DNM: CI health test","status":"ABANDONED","created":"2020-03-30 11:37:07.000000000","updated":"2020-04-16 10:11:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16082137f7c40f1eb42975257b1115ed431b36d9","_number":715922,"virtual_id_number":715922,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-30 13:37:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-30 13:37:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7422290787d32b9ad38166e4ff2a9137ee53aa2d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-30 11:37:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4d8b0730e1788d2033a8f0b5a098a8540579dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 13:37:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/56f170388f7844c8a9543270ce922680 : SUCCESS in 1h 54m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ba5166a15e5417d89a53c3a25ff2db5 : SUCCESS in 1h 29m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b33b27c77c34143808048fa8abf500d : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0604f17728b3482e9603d56570081450 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d12591adf3f4bb7959a54d10a74e293 : FAILURE in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd19a5c5e07e47a3af6c13c0bd355bdf : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b7baca1930ac4968a74285493d0dd09c : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30a1a3bbfc654d668092809df6ff4b02 : SUCCESS in 4m 43s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/27a3195fcacf4e5b9f147bb0e10dd6d9 : FAILURE in 4m 54s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/1f8aabd485624d53a85fe247c110219b : SUCCESS in 4m 36s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/55a33012fccf41548c3c4497729ec477 : FAILURE in 6m 05s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/610800bd70a14e7bb1be3cd13d045608 : SUCCESS in 4m 41s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/5d51feaa804d4942ab2cd92401a668da : FAILURE in 4m 15s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/757ba744025b49d98578892c1b712d30 : SUCCESS in 6m 23s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/31f085fc406643fc860f4f5b1b197cdd : FAILURE in 4m 22s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/f93fc08b3819485193b726eb9693e408 : SUCCESS in 5m 27s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/77bcefff3f4d42ad81d5c9dcacb1114b : FAILURE in 5m 58s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/46d09260bab3425facec34a29f62d703 : SUCCESS in 4m 34s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/6c5562a287214f3597c604ae065042be : FAILURE in 6m 37s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/77722ff09cf84ed2979cf1abda1733b7 : SUCCESS in 5m 03s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/6f0e3c3590a848c2975abbdb94b0ba9d : FAILURE in 8m 05s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/98f84698157a498c9e432bfce8507664 : SUCCESS in 5m 42s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/1fc4676db8be44119175c9c56ce8edc6 : FAILURE in 4m 47s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/905d9b54713a43ebb59911ff3571aa78 : SUCCESS in 5m 13s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/aa0262e943c54b9baf9bbebfeb4b3b51 : FAILURE in 8m 47s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/5ce37796d65949f482c5799793884de6 : SUCCESS in 4m 45s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/918165bf64324471aca4133ffbf2e52c : SUCCESS in 4m 36s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/c05d718e7b8043eba85e00503507ccb4 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9232667464ef3e1d73c5e3d471c7f8dd4018a30f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-16 10:11:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0469be14416f722bf22d2254105db5941a3b8ac3","revisions":{"0469be14416f722bf22d2254105db5941a3b8ac3":{"kind":"REWORK","_number":1,"created":"2020-03-30 11:37:07.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/22/715922/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/22/715922/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/22/715922/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/22/715922/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/22/715922/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/22/715922/1"}}},"commit":{"parents":[{"commit":"966f25804fd2b32bc49993e8faaed25339bccb35","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/966f25804fd2b32bc49993e8faaed25339bccb35"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 11:35:23.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 11:35:23.000000000","tz":120},"subject":"DNM: CI health test","message":"DNM: CI health test\n\nChange-Id: Ifc4b8c291e757d143806cdf16823ff8f3dbe71b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0469be14416f722bf22d2254105db5941a3b8ac3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0469be14416f722bf22d2254105db5941a3b8ac3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
