)]}'
{"id":"openstack%2Ftooz~715940","triplet_id":"openstack%2Ftooz~stable%2Ftrain~I7a75be587efc2098e919f1ea34f9ffc357203dfb","project":"openstack/tooz","branch":"stable/train","hashtags":[],"change_id":"I7a75be587efc2098e919f1ea34f9ffc357203dfb","subject":"[stable-only] Cap stestr for python 2","status":"MERGED","created":"2020-03-30 13:13:50.000000000","updated":"2020-03-31 18:08:16.000000000","submitted":"2020-03-31 18:02:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"715940-1585677745019-93ed6149","meta_rev_id":"db98a90540cd8e2afd508d515b7c445f459f29e7","_number":715940,"virtual_id_number":715940,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"date":"2020-03-31 15:27:16.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 18:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-31 14:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-03-31 14:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-03-31 14:46:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-03-31 14:49:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 11:49:00.000000000","updated_by":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"reviewer":{"_account_id":9796,"name":"ChangBo Guo","email":"glongwave@gmail.com","username":"gcb"},"state":"REVIEWER"},{"updated":"2020-03-31 14:44:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-31 14:45:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-31 14:49:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-31 18:02:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35a697fb6a6ff910fab9412cc64f49b9753e2214","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-30 13:13:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b0b892178f75d3acf7086c60a41ac56be753ea1","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-30 13:14:14.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fb4249dbbe98d05eb5148a445490dda764e370cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 15:18:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b147fd96c8cd4a7c9d8977ba7d032f75 : FAILURE in 4m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/19d9029b048a430ab60741339a1bde68 : SUCCESS in 1h 56m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d856ee52b024681b69d44f70d7e6770 : SUCCESS in 1h 44m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ae19ed66aa24b2ba9c75695077cb5de : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c31e736e3d459098e8bccbdf7aacb1 : SUCCESS in 5m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc6fe73b7f344c0d8deaa06c10961817 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c781be8e65e2428db9800b8145f0e683 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c58cc38a13d14dc782a7780797fd1fd9 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c60655bb2b4cebad49d09e2b50f1db : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a066847f91d148e6a49048498a0cda40 : SUCCESS in 6m 08s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/ebd6aab4f7f94e6fbbb41b36a69d1274 : SUCCESS in 5m 59s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/8cd9cc49e1674cc2928bdb51e2235f81 : SUCCESS in 6m 28s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/9e60cef9719f49eba5df36ae1892a448 : SUCCESS in 5m 58s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/edd1b57920d14b35ae05038d76b8e783 : TIMED_OUT in 31m 30s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/a23db85d99e34dfa945b991c95b90fd0 : SUCCESS in 5m 30s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/bb25c5d2f35348deb7f487820348de68 : SUCCESS in 6m 32s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/ed0eb34f11354054ac1073c5877f3dc6 : SUCCESS in 5m 00s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/b5e8f68214f541ee8dcfc3db38133a15 : SUCCESS in 5m 07s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/2ab70d17c3ec4a5588cdfd22df7549e3 : SUCCESS in 6m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/0c588172c6204c0cbad3f1dc4285ec00 : SUCCESS in 6m 33s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/235bb2e3b7164ea1b4c1303ad14d2cc9 : SUCCESS in 6m 19s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/6b17c833603b4bf8974640d71ad3e894 : SUCCESS in 7m 25s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/f0617b1e4c314d21aa778fd721abcea3 : SUCCESS in 5m 53s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/08a21d4f08f142b1834d80c28b4d38cf : SUCCESS in 4m 44s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/fcdf0f07b23240a0adca57d2333ce383 : SUCCESS in 4m 39s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/d4635fd648744b28ba02a1d68518d07c : SUCCESS in 4m 16s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/a7fe972cd17d4437a6e7b4493e106752 : SUCCESS in 5m 11s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/86ef841a78424bc9991e8e79d6e0e751 : SUCCESS in 5m 32s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/f0e2adead91448618d2d02364229c708 : SUCCESS in 5m 47s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/8a88f97088a94806b707f6d84c8c60e6 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":2},{"id":"20e2ea6fdedb4dbf67f8ef631b596e4b836bef8f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-31 11:49:44.000000000","message":"Patch Set 2:\n\nStable periodic jobs are failing for tooz in Train and Stein, so do the gate jobs [1] with the new stestr release. I am a bit confused what to do here with the requirements-check job as\n\n* either requirements should be aligned with openstack/requirements\n* or in case tooz do not want to do that, then I think requirments-check job should be deleted from .zuul.yaml\n\nAny suggestion what to do?\n\n[1] Ifc4b8c291e757d143806cdf16823ff8f3dbe71b2","accounts_in_message":[],"_revision_number":2},{"id":"2dc1ed637b006954245376f7181693d9a1cfbfb1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-31 14:28:24.000000000","message":"Patch Set 2:\n\nI agree that the requirements-check job should be dropped if we\u0027re not consuming upper-constraints here","accounts_in_message":[],"_revision_number":2},{"id":"8eae296f6a43b93eb9a52f061d45a9eaac284dc2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-31 14:31:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5aa1624bc2142b79c290aaf9d12d2976a8afd8d8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-31 14:32:26.000000000","message":"Patch Set 3:\n\n\u003e I agree that the requirements-check job should be dropped if we\u0027re\n \u003e not consuming upper-constraints here\n\nThanks Stephen, then I\u0027ve removed that check job.","accounts_in_message":[],"_revision_number":3},{"id":"863d5f760d1a360c552e8571539573121679dd3b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-31 14:44:49.000000000","message":"Patch Set 3: Code-Review+2\n\nThis seems like a sensible thing to do, IMO","accounts_in_message":[],"_revision_number":3},{"id":"5a5f6e7597ee2db94b8c08cc86f867fc201e2714","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-31 14:45:00.000000000","message":"Patch Set 3: Code-Review+2\n\nWe should make sure to drop the job from master as well.\n\nGood that our docs state:\n\n\"It should be noted that even though it is designed with OpenStack integration in mind, and that is where most of its current integration is it aims to be generally usable and useful in any project.\"","accounts_in_message":[],"_revision_number":3},{"id":"6f48d42d3eab3378007f5cd2f509dde807789b4b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-31 14:46:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c23a0afef425a9488a68e3fc2a2f21f0bd60fb5f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-31 14:49:42.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1331dc830af42857dee1374999aa3f7577f67906","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-31 15:27:16.000000000","message":"Patch Set 3:\n\n\u003e We should make sure to drop the job from master as well.\n \nGood point, I\u0027ve created the patch: I372cf0c9c9b1ff3d4b83cd84997f780e5de9709a","accounts_in_message":[],"_revision_number":3},{"id":"35fe48929b234a0780abda4c55e2546ac8f2e7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:14:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/52b9726572b34a488d718716781a4727 : SUCCESS in 1h 32m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c75e9d78129b4c2e8be084111df5478d : SUCCESS in 1h 19m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b98ed53780e4e14aeddaf357a7a8287 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e271f57b9834265bb480c5edad8502b : SUCCESS in 5m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab7c12ffb9804467bdf8368e0d0688a5 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92552f4d075748f58e11a190454bac99 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/10b4dcd45f7c4ce89982fec9a779e11c : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10cbc3fa2b14514bbe24f478f0f7494 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/550324a2c30f40b2b5c57441d5ccaffb : SUCCESS in 4m 59s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/4c01fb56d9d54f599c27ebe3b148127e : SUCCESS in 4m 46s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/7f714ef103e04006b2c7eaa6f29c45f5 : SUCCESS in 7m 36s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/5f54730f800c46b4a06841b373753152 : SUCCESS in 6m 12s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/7125cff5c305416e865253acd8d76a52 : SUCCESS in 6m 25s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/fd8d9836a99648a1a46958e389a4fcd2 : SUCCESS in 5m 52s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/e7a4b8cd92af49599b047bbebba448b5 : SUCCESS in 6m 48s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/f7656bad03ef4439a9e87c541460407f : SUCCESS in 6m 16s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/075f554eb8f644569bc13fd92c97fefc : SUCCESS in 5m 56s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/58a86b1b100b4e44838310e804cfcf64 : SUCCESS in 6m 08s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/32270ede37af43a1b82abd2d2148abe6 : SUCCESS in 6m 21s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/406851f809044f1b965b15a9c2947ec8 : SUCCESS in 6m 03s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/f40e7285436f44e78e795a3d2e305267 : SUCCESS in 5m 36s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/89f715690f9349318c39e683abca3522 : SUCCESS in 5m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/692690581573439cb46744771c61c2b6 : SUCCESS in 5m 42s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/4eb606cb37bc4eca982e54e33e1e4566 : SUCCESS in 5m 08s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/50fdcf6b44604f83aa793a53a4d9e535 : SUCCESS in 5m 11s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/a021240d0c984c909d6d651a0bf88fb6 : SUCCESS in 6m 13s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/2f44e2b4244b46d9b0b3f7a77eaba11e : SUCCESS in 6m 02s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/8c9826b029c74b9e896762263193bc3b : SUCCESS in 5m 53s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/1d942fef3bd6437e8532d75d6f939da4 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c9bfcd5867bddd2f810c1380eb2fcd0ffedae196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 16:14:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"57ca16b6ae0cca0eb8ba51e4b0eb89bee591517b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 18:02:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/482f69c2a0104de6ababe00c791982c2 : SUCCESS in 1h 42m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/381b6949c00f48aea27cc7cdceac00f9 : SUCCESS in 1h 29m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68a96f4926c647f1a7cc744f68b856f7 : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/775e26ace3544a84983767f3f397c09a : SUCCESS in 7m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2bb98339cf149c28fcd873d270a3f5c : SUCCESS in 7m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d927fcec4b24f8d925454a476c11098 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c596507c2f64446aa0bf967767211bc2 : SUCCESS in 8m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a0c0699aa3d4d9a855a19f019068ecc : SUCCESS in 6m 49s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/50a93953db7d4321a9b89cecaa79c45c : SUCCESS in 8m 49s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2909fad3868849b6b652e5650f4a17b7 : SUCCESS in 5m 55s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/c6f514b6d64d4cee8eaeeb1acb3104f9 : SUCCESS in 7m 01s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/1e4ce141cfb44977b4229ebd5389531f : SUCCESS in 6m 06s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/27a7133edb3641f59dcaec1dfe6737e3 : SUCCESS in 6m 31s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/2a6c849b2feb42e0b9ea1b224188a831 : SUCCESS in 6m 22s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/2fa4850007a343d68f1ae9c4b207e956 : SUCCESS in 6m 05s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/2fe89605f450489e909996135a23d4cf : SUCCESS in 5m 56s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/a6a9c7fc10314598a5e73fe181f18301 : SUCCESS in 6m 05s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/71e08091e22d4fb5bbcfb940e3e02721 : SUCCESS in 6m 48s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/394f92e0f29947eb95b1abd8d6f5c6ae : SUCCESS in 8m 34s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2223901f54a24494a8bde25c481fdeb8 : SUCCESS in 7m 16s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/aeb1bb8b709e4b5bb22a4b16b7a307ac : SUCCESS in 5m 34s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/a98379c65b074242806235bedc7d09c2 : SUCCESS in 7m 58s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/9a025cbf49b346da95e2335d7a90bcd9 : SUCCESS in 8m 32s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/58771ebe7de34943867636a131b4ccbe : SUCCESS in 6m 18s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/5c490db8267244b29d42a6e464dec154 : SUCCESS in 7m 29s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a008fc0176e7443b84fa7489a3b0f242 : SUCCESS in 5m 43s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/15e561824d864b649697c9c5493c8cea : SUCCESS in 8m 23s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/62a131aebba84e279437fa2a16e5cccb : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":3},{"id":"81c06c42cda9a49b603c5a47d71b4683dd65d166","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 18:02:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"db98a90540cd8e2afd508d515b7c445f459f29e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 18:08:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4c2382cca774a9a8ec9f0040a6341e8 : SUCCESS in 2m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b1653cc0a2d422497ad16d264bd992a : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"13a6dffa32789ac91b724977ed1d73375baf0824","revisions":{"a0ec0a624c33df0e74128fd54ed5c2773da26e8a":{"kind":"REWORK","_number":1,"created":"2020-03-30 13:13:50.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/40/715940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/715940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/715940/1"}}},"commit":{"parents":[{"commit":"966f25804fd2b32bc49993e8faaed25339bccb35","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/966f25804fd2b32bc49993e8faaed25339bccb35"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 13:10:09.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 13:10:09.000000000","tz":120},"subject":"[stable-only] Cap stestr","message":"[stable-only] Cap stestr\n\nstestr dropped python 2 support with version 3.0.0 and tooz does not\nuse constraints from openstack/requirements, so test-requirements.txt\nis updated to use stestr\u003c3.0.0 in case of python 2.\n\nChange-Id: I7a75be587efc2098e919f1ea34f9ffc357203dfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a0ec0a624c33df0e74128fd54ed5c2773da26e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a0ec0a624c33df0e74128fd54ed5c2773da26e8a"}]},"branch":"refs/heads/stable/train"},"a64c289ba60d8ddd5db2c1d6178eb042ee007a74":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-30 13:14:14.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/40/715940/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/715940/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/715940/2"}}},"commit":{"parents":[{"commit":"966f25804fd2b32bc49993e8faaed25339bccb35","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/966f25804fd2b32bc49993e8faaed25339bccb35"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 13:10:09.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 13:13:50.000000000","tz":120},"subject":"[stable-only] Cap stestr for python 2","message":"[stable-only] Cap stestr for python 2\n\nstestr dropped python 2 support with version 3.0.0 and tooz does not\nuse constraints from openstack/requirements, so test-requirements.txt\nis updated to use stestr\u003c3.0.0 in case of python 2.\n\nChange-Id: I7a75be587efc2098e919f1ea34f9ffc357203dfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a64c289ba60d8ddd5db2c1d6178eb042ee007a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a64c289ba60d8ddd5db2c1d6178eb042ee007a74"}]},"branch":"refs/heads/stable/train"},"13a6dffa32789ac91b724977ed1d73375baf0824":{"kind":"REWORK","_number":3,"created":"2020-03-31 14:31:22.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/40/715940/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/715940/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/715940/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/715940/3"}}},"commit":{"parents":[{"commit":"966f25804fd2b32bc49993e8faaed25339bccb35","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/966f25804fd2b32bc49993e8faaed25339bccb35"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 13:10:09.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-31 14:29:31.000000000","tz":120},"subject":"[stable-only] Cap stestr for python 2","message":"[stable-only] Cap stestr for python 2\n\nstestr dropped python 2 support with version 3.0.0 and tooz does not\nuse constraints from openstack/requirements, so test-requirements.txt\nis updated to use stestr\u003c3.0.0 in case of python 2.\n\nSince tooz do not consume upper-constraints there is no need for\nrequirement check job either so it is removed, too.\n\nChange-Id: I7a75be587efc2098e919f1ea34f9ffc357203dfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13a6dffa32789ac91b724977ed1d73375baf0824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13a6dffa32789ac91b724977ed1d73375baf0824"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
