)]}'
{"id":"openstack%2Ftooz~716610","triplet_id":"openstack%2Ftooz~stable%2Fstein~I7a75be587efc2098e919f1ea34f9ffc357203dfb","project":"openstack/tooz","branch":"stable/stein","hashtags":[],"change_id":"I7a75be587efc2098e919f1ea34f9ffc357203dfb","subject":"[stable-only] Cap stestr for python 2 and unblock gate","status":"MERGED","created":"2020-04-01 14:40:16.000000000","updated":"2020-04-16 18:22:05.000000000","submitted":"2020-04-16 18:16:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716610-1587060982767-dd145dbb","meta_rev_id":"3f2ff53175487552dedaae8bb19ccad50e5f2d7c","_number":716610,"virtual_id_number":716610,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 18:16:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-04-16 16:23:46.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-04-16 13:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-04-16 16:23:46.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-16 13:07:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-16 16:23:46.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-04-16 18:16:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e19d11c41e396448ad32e48e6999a05526b70532","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-01 14:40:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ce3c6001a0d2a9bb47c6b7b67182b6537b63a31","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-01 16:03:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"16883bb4c0590b45820404ff51e58c535835cb2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 16:22:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0ab0e8a9c1ee4863b1a754b2e46bde8a : SUCCESS in 5m 49s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4430d2dd99e34cc0a956ed8b5d04e356 : SUCCESS in 1h 34m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d123c71f605438aa210a9b6d7451246 : SUCCESS in 1h 24m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f91f9d8e77f64dbb9d9862a5b61d00dc : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de314171d9e4307930ae3378fe2eac1 : SUCCESS in 8m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e45095abc794314b9b5991d123e1339 : SUCCESS in 9m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/43564041079446ff96bde9b67b23f521 : SUCCESS in 9m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eacd992e425348f88e6f8ceb577ce275 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1573ba53b842fd870251d8f0c2274e : SUCCESS in 5m 57s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/7bc2d4963ccc4276b6e391e2c14c6545 : SUCCESS in 7m 30s\n- tooz-tox-py35-etcd https://zuul.opendev.org/t/openstack/build/b179c844f0a74e0199ced163536aded1 : SUCCESS in 9m 09s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/2e1e0f6fe4ee498694b91a11e2c1aa89 : SUCCESS in 9m 18s\n- tooz-tox-py35-etcd3 https://zuul.opendev.org/t/openstack/build/06aa4aaf9c124c8296c1a7286e8b4b11 : SUCCESS in 7m 14s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/0043ab90b87148c7ba79180f8d1ccba9 : SUCCESS in 6m 57s\n- tooz-tox-py35-etcd3gw https://zuul.opendev.org/t/openstack/build/6029c5755389469c8b094534413109fd : SUCCESS in 7m 20s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/acc7c2b755a14e418711b40e7d31a8b4 : SUCCESS in 6m 54s\n- tooz-tox-py35-zookeeper https://zuul.opendev.org/t/openstack/build/7bbf49336b7f4cce84459a878046c7f7 : SUCCESS in 7m 34s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/890e44f466c24aed9c281e60d9c0ad17 : FAILURE in 6m 39s\n- tooz-tox-py35-redis https://zuul.opendev.org/t/openstack/build/a9f001e58bcc468086a6362cc1784b44 : SUCCESS in 7m 43s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/32c9baa221bc403e90bdfd5bb650225e : FAILURE in 7m 20s\n- tooz-tox-py35-sentinel https://zuul.opendev.org/t/openstack/build/eae23052c5ab40e2be07a7d38ef43157 : SUCCESS in 7m 55s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/aa741a0671d24a24b26c7efa85dd3123 : SUCCESS in 8m 45s\n- tooz-tox-py35-memcached https://zuul.opendev.org/t/openstack/build/a6527b03b39a481aa77fa8af31cfe529 : SUCCESS in 9m 12s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/c1f9185f550d47e99e4716512c5c9192 : SUCCESS in 9m 13s\n- tooz-tox-py35-postgresql https://zuul.opendev.org/t/openstack/build/20cf39ec34a447e6b4434e7150d02458 : SUCCESS in 7m 32s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/76323e18a68d46c6bd64d25217ee550c : SUCCESS in 8m 17s\n- tooz-tox-py35-mysql https://zuul.opendev.org/t/openstack/build/4e40c429a78e40ddbe3070993a664897 : SUCCESS in 8m 28s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/14f78de741cd4fc1988e55667d56396a : SUCCESS in 7m 28s\n- tooz-tox-py35-consul https://zuul.opendev.org/t/openstack/build/aec1dd90776648cf9ce35a0f51d01349 : SUCCESS in 9m 10s","accounts_in_message":[],"_revision_number":1},{"id":"766b88d9a9251a254963ec69855bded8f2c94f97","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-01 17:32:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9eb164c5c651a448206f560b0fae045fd7996b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-01 19:13:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c627ca1cc624d55920d92126c6a9071 : SUCCESS in 5m 50s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b11a6858984348b79c6461584d2ce8c3 : SUCCESS in 1h 31m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cb508514bb74b839d94a54ceddd12f8 : SUCCESS in 1h 35m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dc61aabf2944f2fbb5d608ac9249bca : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86bddeb3aa3f4c7e822c4b892aacd07e : SUCCESS in 8m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6b20b7fb6394d07a0ccd8ea2b325233 : SUCCESS in 8m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/def6e671652d4b179f3628ec1cae1e71 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1dd24f221b941b4a4cbbd8a630968a4 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ed368c59e24415aa908c066347cd555 : SUCCESS in 5m 41s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/0f3b62d853384ab5a9f66e398b8e481e : SUCCESS in 5m 23s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/b8ec7b9659b24c1f89e8bd73e1f3cf73 : SUCCESS in 8m 51s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/57bc014b72384a95b122dd4ad5157c77 : SUCCESS in 6m 57s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/9b51905ff24d4c6e9d0a92a642b856bc : SUCCESS in 9m 37s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/6ac3edce146d43e0b4756875efc2fcf3 : SUCCESS in 5m 08s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/72a9f144ce754af2a8f11b3d8f7faeef : SUCCESS in 7m 10s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/ab7f7dcec8994794926de7b7eac57c71 : SUCCESS in 7m 12s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d312eb31e7a34339981cb17c34a39581 : SUCCESS in 7m 23s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/2a92ffbb5c6d44c3a292765f395f86d6 : SUCCESS in 7m 31s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/9a5910c1f40b4c1581409da00cc505cc : SUCCESS in 5m 49s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/a1623ea56b6846deb0144dea80ee5b37 : SUCCESS in 9m 34s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2325357255ac4f659a28e5b3b5ebae84 : SUCCESS in 5m 28s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/60fcfe650c65451596c7640978cbb789 : SUCCESS in 6m 46s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/6acd1caa0a8b4963aa76f9b2fa3d1c5a : SUCCESS in 8m 01s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/c85b6431fe4c49eaad36535657bccc9f : SUCCESS in 4m 58s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/6e9c36a0066044f3b3035d57546fae5a : SUCCESS in 8m 23s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/628cbee35aa248cba633ba3e6e4590f4 : SUCCESS in 6m 03s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/8f24c4b7b7f14cbf96196bd68ce4a5d0 : SUCCESS in 5m 02s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/796af888e3ba4e83ba8cada36461c490 : SUCCESS in 8m 40s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/e4d85e4d839943bbacc5ae546c0fdce0 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":2},{"id":"bec5e3980cc4f52f9f4333da36eda759ed19987a","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-04-01 20:34:41.000000000","message":"Patch Set 2: Code-Review+2\n\nHeh, it\u0027s squash-ception. This all seems sane though and is passing ci.","accounts_in_message":[],"_revision_number":2},{"id":"c52808e38674411f1036ad3c3b9423b5ab817dd5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-02 14:45:33.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)\n\n\u003e Heh, it\u0027s squash-ception. This all seems sane though and is passing\n \u003e ci.\n\nAh right patch set 2 is weird...\n\nMy main concern is the possible confusing part of these changes but if CI is happy, split not possible, and if Ben think it\u0027s ok to continue with that, then I\u0027ll approve too.\n\nPlease, just can you confirm my inline question.","accounts_in_message":[],"_revision_number":2},{"id":"d3222e3831bcfb2e8057aa5c141e54f7af850cc1","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-04-02 15:57:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a5e1f961a89b829a4dfa226ad187fb6c39c8cb1c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-02 16:59:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f41b0e843f521595a75261483f39af79b41bdba3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-02 17:01:57.000000000","message":"Patch Set 2:\n\nbut of course I can update the commit message if you want!","accounts_in_message":[],"_revision_number":2},{"id":"6544a8f370398c3a0c9738d8c1e8337c3da94add","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-16 10:05:23.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"fb787769dae385eb80de36d29d35e390ccfcde42","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-04-16 10:16:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61993828921aa410daa8a10da37c061622f1b5e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 11:50:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f66396546cbf4cb88540f9cb9ea7b240 : SUCCESS in 3m 13s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0563179514464af8b288ac0cddd82846 : SUCCESS in 1h 38m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c684bb66ba794965a6ec7a7dd45ca225 : SUCCESS in 1h 18m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d69062682af94a3fa0f0d5ae147ba9ff : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d0871428ef4c07900f34c282df5576 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e6cf93e61c845b585c2c0d99720e853 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ee34b171c134d03b26e3d25584125b6 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d65755cc7874344a6800961663bf2cc : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89c25a65c2134b41841f6e45589f1d1d : SUCCESS in 4m 11s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/ae7c73a1f46e4ed78b823bac487e3f31 : SUCCESS in 4m 31s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/a881088763bb4bcd92ee014305232dc7 : SUCCESS in 5m 40s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/dd2ddbf5c8c64b02966562604afce624 : SUCCESS in 5m 09s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/939589d6000844fd902a7e072bb742a1 : SUCCESS in 5m 11s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/c16717164a3a497d97b8e9739ff4b55a : SUCCESS in 5m 22s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/ac7aff78a9a44dacae40afe993ae82d6 : SUCCESS in 4m 59s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/1ea1960c93a0456cb79ebad4e71e4f2c : SUCCESS in 5m 46s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/b239a664b06b4d4cb2713babad9e8fe0 : SUCCESS in 5m 10s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/e456612520a4449a92ede78d0e905d7c : SUCCESS in 5m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/f846ab12d2b341038d68810ee9c68643 : SUCCESS in 6m 26s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/8cd304b1596a415e9e86a01c48a962bc : SUCCESS in 8m 27s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/f5d512b265a24e52820773435df5a3fc : SUCCESS in 6m 20s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/4a0041cfcdb94eeabe084f95a5ac1334 : SUCCESS in 4m 51s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/3ab1a647283b4812a9d0fd6d7b44b8fe : SUCCESS in 4m 47s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/0b99e59ec5474ff3944312c3eb07f31d : SUCCESS in 4m 43s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/51ae92cbe45b487baf88680d23171aca : SUCCESS in 5m 17s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/d1e79ad6941c45b2b3c63c46a9949323 : SUCCESS in 4m 55s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/5c4c5f6b863f4387b25cc392ec3f4078 : SUCCESS in 4m 32s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/dab9db604a804000b079135f984b752f : SUCCESS in 6m 57s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/94ef2e9cd4ba44b9b52cecd8274214c1 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"edf70f4f73263733db64999deea45bf6b33780c7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-16 13:07:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"46a0dfd696e5fc5e8c8553961339ed054dcf3c99","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-04-16 16:23:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"941eb814be1f2c152f1a423e3074c23d263e87c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 16:24:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"22a13ca45fad7e36d0c533e2c43426a4b759d597","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 18:16:22.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/afa5b51d78304e6a836339eb910b0767 : SUCCESS in 3m 16s\n- tempest-full https://zuul.opendev.org/t/openstack/build/b7fad434afa84173a319bf0a85bb2313 : SUCCESS in 1h 47m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a4b6e76207443c4ad77b1e8864a245f : SUCCESS in 1h 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9227e8c972214bd3a4c9ad674c561ffa : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba25ace88b1c4c138cdf1582a0171290 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9b60ef96f95432e8381a405d5653bc8 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bb88f4c7a348ecbcd9bf7480a13648 : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ff76a028d474a2a8b454d576f8e76ee : SUCCESS in 5m 04s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/5256fdb982094ae7a5269933da64126c : SUCCESS in 6m 12s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2924d866c22d45d5bef439deabcafaa7 : SUCCESS in 4m 56s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/84ace3e596724cb6a7eee1193d36a34e : SUCCESS in 5m 08s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/927da1b1d9514134861df3b64e722c9f : SUCCESS in 7m 37s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/46b5827036c044f187a0da659bdd6eb4 : SUCCESS in 5m 41s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7b4f1453e5834fb2b9279b4baa8c9d92 : SUCCESS in 5m 33s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/db552544df2f40bfb609a7774fe0ecf0 : SUCCESS in 4m 44s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/3ccbcde1cb7a4419bc44709b1832bea5 : SUCCESS in 5m 26s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/2166628e459f444c8bb3e4a96dd75802 : SUCCESS in 5m 03s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/438d2f6662254df998623a9c4850dcc1 : SUCCESS in 4m 34s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/7ef4d72c34ce46e4b3f1c17b028a1e25 : SUCCESS in 5m 23s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/a11ca3cb694245d6a4fc3c5269a39f7c : SUCCESS in 5m 31s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/ce7a98047671458c8b4eb96eee813081 : SUCCESS in 5m 17s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/a394a83b5aa349d5b89c49173874835b : SUCCESS in 5m 19s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/a8e16e0d5530404c997c0a1440b691e5 : SUCCESS in 5m 19s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/8da76aee2c7b450c92a45ec87ebb4527 : SUCCESS in 6m 01s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/d7591c700c7847f28542dc0ffaa27043 : SUCCESS in 5m 19s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/0ddcc96538204c319072f957d5e9c0a7 : SUCCESS in 5m 01s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/0934848aacc7437da03c61d2c4d6acc0 : SUCCESS in 5m 04s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/e4f06f38ea20493b9a5ea40ee3e513fb : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8efac6caa2afc0d2b7fcb0321b89bc5ae85b2271","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 18:16:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3f2ff53175487552dedaae8bb19ccad50e5f2d7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 18:22:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77c4383ae45d4618b1ff22baa6d8f41d : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e237154fea0e4040adec5dd8d8bec45d : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4d24f41e47c6a16fbb85ffce974e90e70d07ff76","revisions":{"331b848bf349bbd049ec20fdc7c4a37044e38710":{"kind":"REWORK","_number":1,"created":"2020-04-01 14:40:16.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/716610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/716610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/716610/1"}}},"commit":{"parents":[{"commit":"6204beeb035b6107545fbac1677c4b29ced8156c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6204beeb035b6107545fbac1677c4b29ced8156c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-03-30 13:10:09.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-01 14:35:20.000000000","tz":120},"subject":"[stable-only] Cap stestr for python 2","message":"[stable-only] Cap stestr for python 2\n\nstestr dropped python 2 support with version 3.0.0 and tooz does not\nuse constraints from openstack/requirements, so test-requirements.txt\nis updated to use stestr\u003c3.0.0 in case of python 2.\n\nSince tooz do not consume upper-constraints there is no need for\nrequirement check job either so it is removed, too.\n\nConflicts:\n    test-requirements.txt\n\nNOTE(elod.illes): test-requirments.txt is not present in Stein,\nsetup.cfg is used instead.\nAlso: py35 based jobs need xenial nodeset so replace openstack-tox with\nopenstack-tox-py35 to force the use of xenial nodeset in Stein.\n\nChange-Id: I7a75be587efc2098e919f1ea34f9ffc357203dfb\n(cherry picked from commit 13a6dffa32789ac91b724977ed1d73375baf0824)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/331b848bf349bbd049ec20fdc7c4a37044e38710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/331b848bf349bbd049ec20fdc7c4a37044e38710"}]},"branch":"refs/heads/stable/stein"},"042fb470d4aa3c40c0066c081e10aa4e9175335b":{"kind":"REWORK","_number":2,"created":"2020-04-01 17:32:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/716610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/716610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/716610/2"}}},"commit":{"parents":[{"commit":"6204beeb035b6107545fbac1677c4b29ced8156c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6204beeb035b6107545fbac1677c4b29ced8156c"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-09 20:20:13.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-01 17:29:17.000000000","tz":120},"subject":"[stable-only] Cap stestr for python 2 and unblock gate","message":"[stable-only] Cap stestr for python 2 and unblock gate\n\nThis is a combination of 2 commits that are needed to unblock gate:\n\n1. Unblock tooz gate\n\nThis is a squash of two commits needed to get gate tests passing\nagain on tooz. See below for details:\n\nMove testing to py36\n\nThe base infra images no longer have python 3.5 available and are\nall failing. We should be testing on py36 now anyway since that is\nthe minimum supported version for Train.\n\nStop redis-server before running tests\n\nJust installing redis-server on Ubuntu Bionic starts the service,\nwhich means when we try to start one via pifpaf it fails due to the\nport already being in use.\n\nThis change adds a pre-run playbook that stops the redis-server\nservice so ours can run successfully.\n\nConflicts:\n    tox.ini\n\nNOTE(elod.illes): conflict is due to py37 is not yet added as tox\ntarget in Stein.\n\nChaddnge-Id: I3f7f7ea9069d8c890a82f31ad14c9663e98c09dc\nCloses-Bug: 1828610\n(cherry picked from commit 8207427fffc72547387369ce7ba2696a86bf7626)\n\n2. [stable-only] Cap stestr for python 2\n\nstestr dropped python 2 support with version 3.0.0 and tooz does not\nuse constraints from openstack/requirements, so test-requirements.txt\nis updated to use stestr\u003c3.0.0 in case of python 2.\n\nSince tooz do not consume upper-constraints there is no need for\nrequirement check job either so it is removed, too.\n\nConflicts:\n    test-requirements.txt\n\nNOTE(elod.illes): test-requirments.txt is not present in Stein,\nsetup.cfg is used instead.\n\nChange-Id: I7a75be587efc2098e919f1ea34f9ffc357203dfb\n(cherry picked from commit 13a6dffa32789ac91b724977ed1d73375baf0824)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/042fb470d4aa3c40c0066c081e10aa4e9175335b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/042fb470d4aa3c40c0066c081e10aa4e9175335b"}]},"branch":"refs/heads/stable/stein"},"4d24f41e47c6a16fbb85ffce974e90e70d07ff76":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-16 10:05:23.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/716610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/10/716610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/10/716610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/10/716610/3"}}},"commit":{"parents":[{"commit":"6204beeb035b6107545fbac1677c4b29ced8156c","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6204beeb035b6107545fbac1677c4b29ced8156c"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2019-05-09 20:20:13.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-04-16 09:56:13.000000000","tz":120},"subject":"[stable-only] Cap stestr for python 2 and unblock gate","message":"[stable-only] Cap stestr for python 2 and unblock gate\n\nThis is a combination of 3 commits that are all needed to unblock gate:\n\n1. Move testing to py36\n\nThe base infra images no longer have python 3.5 available and are\nall failing. We should be testing on py36 now anyway since that is\nthe minimum supported version for Train.\n\n2. Stop redis-server before running tests\n\nJust installing redis-server on Ubuntu Bionic starts the service,\nwhich means when we try to start one via pifpaf it fails due to the\nport already being in use.\n\nThis change adds a pre-run playbook that stops the redis-server\nservice so ours can run successfully.\n\n3. [stable-only] Cap stestr for python 2\n\nstestr dropped python 2 support with version 3.0.0 and tooz does not\nuse constraints from openstack/requirements, so test-requirements.txt\nis updated to use stestr\u003c3.0.0 in case of python 2.\n\nSince tooz do not consume upper-constraints there is no need for\nrequirement check job either so it is removed, too.\n\nConflicts:\n    tox.ini\n    test-requirements.txt\n\nNOTE(elod.illes):\n  * tox.ini conflict is due to py37 is not yet added as tox target in\n    Stein.\n  * test-requirments.txt is not present in Stein, setup.cfg is used\n    instead.\n\nFirst two commit was squashed in train (original patch\u0027s change id is\nI3f7f7ea9069d8c890a82f31ad14c9663e98c09dc and cherry picked from commit\n8207427fffc72547387369ce7ba2696a86bf7626).\n\nCloses-Bug: 1828610\nChange-Id: I7a75be587efc2098e919f1ea34f9ffc357203dfb\n(cherry picked from commit 13a6dffa32789ac91b724977ed1d73375baf0824)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4d24f41e47c6a16fbb85ffce974e90e70d07ff76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4d24f41e47c6a16fbb85ffce974e90e70d07ff76"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
