)]}'
{"id":"openstack%2Ftooz~718488","triplet_id":"openstack%2Ftooz~master~I98fc96468b21368ce66365e3fc38c495b1f2918a","project":"openstack/tooz","branch":"master","topic":"consul_acl_token","hashtags":[],"change_id":"I98fc96468b21368ce66365e3fc38c495b1f2918a","subject":"Add support for Consul ACL token parameter.","status":"MERGED","created":"2020-04-08 15:19:43.000000000","updated":"2020-05-15 10:03:46.000000000","submitted":"2020-05-15 10:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718488-1589536907366-d6aca499","meta_rev_id":"4694c8820bfa665012f7878684191eef87c0173e","_number":718488,"virtual_id_number":718488,"owner":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-15 10:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-15 01:49:18.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-15 01:49:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":2,"date":"2020-05-15 01:49:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-05-15 08:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-05-15 08:26:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 20:33:47.000000000","updated_by":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"reviewer":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"state":"REVIEWER"},{"updated":"2020-04-28 22:18:53.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-07 09:37:44.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-05-15 08:26:06.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-05-15 10:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31abcd798cc3482bb9cad49e2d1f49afc01d1f33","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-08 15:19:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cff21353cf1951f2d42f853fcdf9b7dc12a616e","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-08 15:27:42.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f5146a54f30574027baa56beabe578294c322c2e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 15:28:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8507f4be6bddd9ad09885da5b258102b9b689a73","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 15:29:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1354867194bfb366ef5637ab4b1ac7951fa5ad26","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 15:29:44.000000000","message":"Patch Set 2:\n\nAny unit tests that can be added here?","accounts_in_message":[],"_revision_number":2},{"id":"72d42649cac3126961105397744b8db8e85de433","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-08 15:44:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60aa75f4511682be3957aa6075d09d1bf66485be","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-08 15:46:45.000000000","message":"Patch Set 3:\n\nQuestions:\n1.) There are no consul unit tests at all, are you asking me to add them? \n2.) Is this error related to #1 above? \nERROR: InvocationError for command /Users/punkrokk/repos/st2-infra/tooz/run-tests.sh stestr run \u0027\u0027 (exited with code 1)\n_______________________________________________________________________________________________________________ summary _______________________________________________________________________________________________________________\n  pep8: commands succeeded\nERROR:   py37: commands failed\n(tooz) punkrokk17:tooz punkrokk$ \n\n3.) I had to fix a pip8 error in the mysql.py driver. Not sure if it\u0027s ok to commit here also.","accounts_in_message":[],"_revision_number":3},{"id":"c82ad4f095512fc35c29bfd81a00e45985ad7561","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-08 15:48:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7bc24272723eeeb20bb6689ac18b9f80a3c23aca","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-08 15:49:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4a553589c251a17df5d2d448b2aed1a4cf97b1d1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 16:00:38.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nJust some minor comments.","accounts_in_message":[],"_revision_number":3},{"id":"551aafcc854ebec4be88c743886fe75ce690f603","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-08 16:01:17.000000000","message":"Patch Set 3:\n\n\u003e Questions:\n \u003e 1.) There are no consul unit tests at all, are you asking me to add\n \u003e them?\n \u003e 2.) Is this error related to #1 above?\n \u003e ERROR: InvocationError for command /Users/punkrokk/repos/st2-infra/tooz/run-tests.sh\n \u003e stestr run \u0027\u0027 (exited with code 1)\n \u003e _______________________________________________________________________________________________________________\n \u003e summary _______________________________________________________________________________________________________________\n \u003e pep8: commands succeeded\n \u003e ERROR:   py37: commands failed\n \u003e (tooz) punkrokk17:tooz punkrokk$\n \u003e \n \u003e 3.) I had to fix a pip8 error in the mysql.py driver. Not sure if\n \u003e it\u0027s ok to commit here also.\n\nOK, I won\u0027t push for tests then if there\u0027s nothing existing. Unless of course you feel like adding them. ;)","accounts_in_message":[],"_revision_number":3},{"id":"7c1b599044f20cc038146b7d4541924c5d04edb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 17:36:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10271109771044b5888e3f3d5a47a1e9 : SUCCESS in 1h 37m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f797d0e617ca4d2a8f31ffbb5d26e5b7 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1321fffde6cb4defb2396e9bd25fc013 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f8d00f9e8b9469ebbb5388815280540 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ec6e263a4014f43a85be97a18e3cba5 : SUCCESS in 7m 21s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/32f8c3a1c14d4d6da7e36c59e5a5d92e : SUCCESS in 5m 51s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2456e0f9207d47b89c7204849810bd8a : SUCCESS in 5m 37s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/3cf4a7e8b07749cbbd4f37c9a425f265 : SUCCESS in 5m 35s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/f7cae64e03bd46d69397f0d5dd41ea27 : SUCCESS in 6m 52s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/a837296aad4645ccb0f09dcb7d9d1d92 : SUCCESS in 7m 13s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/648bc094485c4fecb15226f13ce01983 : SUCCESS in 6m 50s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/fa6c1b54cb20496f946b4c6935509ad6 : SUCCESS in 7m 40s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/b608b43c53074b8abc2672f47ca7a301 : SUCCESS in 6m 46s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/317062f1a33f41f7884e7394904b7e0c : SUCCESS in 7m 20s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/beb12d5cbc814f5cbbb06ad2a88e8e7c : SUCCESS in 6m 38s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/f5fb898d06da4c96a1bd5653ed257906 : SUCCESS in 8m 26s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/237a7d1418334ef9aed4d0b03ec8298c : SUCCESS in 5m 07s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/2e38ebab118349a8b785042293e68c20 : SUCCESS in 6m 58s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/103074e95c5643ff8df60e176bfa12dc : SUCCESS in 5m 05s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/88bb40fb52e24dd387e52d43ced09834 : SUCCESS in 5m 26s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/469c3d11913d46089c60839d69b8fcf4 : SUCCESS in 7m 07s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/3882e698d8a249c38dd45a4e3dc23154 : SUCCESS in 7m 05s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/9ecfb2161e5f419eb404ef5c85bebb53 : SUCCESS in 6m 37s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/db53a84c88284e4bae7d2b7ec13df0a7 : SUCCESS in 5m 23s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/48dbf81ae42f4beb9e51c4bd5cd54c63 : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":3},{"id":"08383dbb2ff8ccaa5c9fb630adb03ad61ca583a1","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-27 02:59:19.000000000","message":"Patch Set 3:\n\nI updated based on the comments, and backed out the mysql pep8 error for a different PR. But it ended up over here: https://review.opendev.org/#/c/723300/1","accounts_in_message":[],"_revision_number":3},{"id":"a07fd36fd2fa2c1124b0068d13dd47b14cdd581b","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-28 20:08:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"41ec3898407500078aa1da21f9711147f7353ad9","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-28 20:28:09.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"c25222141fee366ed8e526b3a60361238b087615","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-28 20:29:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bb63f4f8472d29fcb8f7563033f2e1dee7cb7155","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-28 20:31:32.000000000","message":"Patch Set 6:\n\nI moved my changes from https://review.opendev.org/723300 and abandoned that PR. \n\nRECHECK","accounts_in_message":[],"_revision_number":6},{"id":"8aaf941a779801c2526f63d256c6fc0e28116cb5","author":{"_account_id":26867,"name":"Nick Maludy","email":"nmaludy@gmail.com","username":"nmaludy"},"date":"2020-04-28 20:33:47.000000000","message":"Patch Set 6: Code-Review+1\n\nGreat job. I would love to see this to support secure installations of Consul!","accounts_in_message":[],"_revision_number":6},{"id":"33d758a24348b78d97419ec30dc5f62b004145f7","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-04-28 20:47:33.000000000","message":"Patch Set 6:\n\nRECHECK","accounts_in_message":[],"_revision_number":6},{"id":"d4ec41b82358681e07b5ee9e26e589e078057257","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-28 20:49:09.000000000","message":"Patch Set 6:\n\n\u003e RECHECK\n\nZuul hasn\u0027t reported yet, so no need to recheck. If zuul fails and it does need a recheck, then it is just lower case. Not sure if the regex that reads the comment stream recognizes upper case versions.","accounts_in_message":[],"_revision_number":6},{"id":"3e74741726311dee6c21a0de7cf836a990b80d6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 22:04:03.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/eeb46e8923444a23a8c19efe0d8c8423 : SUCCESS in 4m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41047ff99b1f4e41b618414934768d9b : SUCCESS in 1h 31m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d653488f2de847dc85319ecdf53fd7e1 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4560d431904749b383376deef3157a : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ceb479f31d444e19aad428420b55ae96 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f97f8054d69444d7bca3aa98b9b9334b : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1b49c0628824395bcc3ac2567bcfdcd : SUCCESS in 7m 07s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/161a844f67814114a0100ae5dedd29ce : SUCCESS in 8m 20s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/879b3a35b1934d6897ab35d64daf5c62 : SUCCESS in 10m 10s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/c5d83ed9895648a0bcd8c1a1a1d40788 : SUCCESS in 8m 33s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/c70cebe63a7947478fc3f8891861f4fe : SUCCESS in 7m 29s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/207ab93ca0e0438381823dddaa73bfe6 : SUCCESS in 7m 32s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/3ff097c3a3004da38fc18fa73356f695 : SUCCESS in 8m 41s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/07615bb93000491f9ad1371afa5a0f76 : SUCCESS in 9m 07s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/1bef3bc3074940c0a0c927ab0faa6253 : SUCCESS in 7m 18s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/93b12f3aa9e94763a9747332c314f102 : SUCCESS in 9m 16s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/cf11949817dd49d9a9c9d3fbe272d69b : SUCCESS in 7m 10s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/3ab906e0e1e241bb8edc72f209b7c547 : SUCCESS in 7m 26s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/7c8be3a150ba43ed9f2a9cc1b1dbbcd7 : SUCCESS in 6m 31s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/514d67fdeba941b886c501d9e7551ceb : SUCCESS in 6m 31s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/6bcc6c2c9958450dafa089c39d1a44ad : SUCCESS in 6m 54s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/962ba60072884a17a9e358b349d1f1c8 : SUCCESS in 7m 06s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/832df068566c449e874e09cf67c69a73 : SUCCESS in 5m 20s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/a6ca044e41784c59a0ba43ef5a122b52 : SUCCESS in 9m 58s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/1aa4013cd89147809656c451d054196e : SUCCESS in 8m 26s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/30c00d9c81044b87b50c3c8d54f9a29f : SUCCESS in 8m 18s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/55ec736a7a2b448ab31665dd6d374379 : SUCCESS in 7m 29s","accounts_in_message":[],"_revision_number":6},{"id":"528f1c94c2bb23691c1f0c0c665bf10f636826b4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-28 22:18:53.000000000","message":"Patch Set 6: Code-Review+2\n\nThis looks good to me. I don\u0027t have a setup to test this locally, but CI jobs are happy with it and nothing stands out as an issue to me.","accounts_in_message":[],"_revision_number":6},{"id":"68f61c2d53d8927436a11c6788b18bcbe87e1af4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-07 09:37:44.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0b4a8765cf63304e5518ec892d309d707d756a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 09:37:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3e0c60f1d1998cfa629ac498f7b64d83f8afad3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-07 11:12:09.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d8ef638e25574065ba4f3735facc51ca : SUCCESS in 4m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5afdc652bb264383a61f0fdcc8cfcf64 : SUCCESS in 1h 29m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e4d4bd9e7234d739de07a9e23e2c368 : FAILURE in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f429682e52564cada854969b35d8ec98 : FAILURE in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730a27e7e4d04a35a7e8f7306d2a92c2 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d88d0b5318b436286f34cbddf54b50d : SUCCESS in 5m 16s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/189fd09a0f5e42e0817da887218c0b47 : FAILURE in 4m 49s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/1a0d9e88d2bb4b00b27c28f5a55b91f0 : FAILURE in 4m 36s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/84c9adacb3c24645bbb884a23dc19d89 : FAILURE in 6m 17s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/f6a7802a6cd0463bb076f8083670e281 : FAILURE in 7m 24s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/5146e9f3ea554ca399b49bed0e48bb31 : FAILURE in 5m 08s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/54e5500704874c29a16e20111b6a9659 : FAILURE in 5m 08s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/7c9d83c76d364ffc8267f213c8c0c208 : FAILURE in 7m 37s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/980a16cf90a140659ba5d0477fcbaf82 : FAILURE in 4m 04s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/b97a041d14ae4ce29fb9bb7791b70034 : FAILURE in 5m 59s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/bee3a3788be843adb35b8164cb1122db : FAILURE in 5m 52s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/6241dff6fa374a2e897d546acb94d94a : FAILURE in 7m 02s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/d943975c42854c6f9a62ef8af882164b : FAILURE in 4m 56s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/52a1b8b0ef5c4751b04cec9917162e0c : FAILURE in 6m 20s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/47d768058e2b45828e515e436ec4cf90 : FAILURE in 5m 00s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/94cea7aceee245b89c0a603f888c9d6d : FAILURE in 7m 39s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/96c8d0680e4c4ba89259c5fdd64427e3 : FAILURE in 4m 42s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/a091b64ae1b749ba9b4784263c783cb3 : FAILURE in 7m 23s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/87b92ffd00d448459774a927f1dc6a2d : FAILURE in 4m 44s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/c01e948b585f47c8a980f2772912a9dc : FAILURE in 4m 36s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/6a92ee872a6e49db9a853515f5e378e9 : FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":6},{"id":"3beb8323d49a8bec8c3f921cfa3f28dd3af277cc","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-05-10 19:03:25.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"f6c1d9cd1051a635867285368947a96b1617ca3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 20:38:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/813e261af22e49c0997a2faa2e4942a1 : SUCCESS in 2m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/31d30fec74d8408483f9e123270a5dab : SUCCESS in 1h 30m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f7012379de04924b335c0ee7a1047e7 : FAILURE in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f66965500db24a8495598a65104455dd : FAILURE in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b6824f7109bd47e3b0cb7db4048339ae : FAILURE in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52d3f20736c3440eb0d8cc189d19a290 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb16aeb5922c4e6ba6ebe30c1b27d59e : SUCCESS in 4m 35s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/55e36e61a8c843cb87dea5ca8f648efa : FAILURE in 5m 10s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2cf7aeeac8f545d6a8af1d11753485e2 : FAILURE in 4m 09s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/09cd2080b8534c45a01b869c79d4d5fe : FAILURE in 4m 54s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/8562e85e30a34974a29bb8ba3d544805 : FAILURE in 4m 06s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/e4cf1125f70b4a2a82ebb96028be2c3c : FAILURE in 5m 03s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7775bbc802054a3cbe300a29d02d4afb : FAILURE in 4m 05s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/924854bc5e124306a9fbbca8a7bdb865 : FAILURE in 4m 34s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/2aa1519b5c734c37aa1b5f93797d1805 : FAILURE in 4m 46s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/d4e64c99f81e419788ecc3295edd3bc3 : FAILURE in 4m 35s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3e14993c7a7f4788b4f3cf2c24da6a9e : FAILURE in 4m 17s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/c67b2cc5f1a248819e0e38f5b2589abc : FAILURE in 4m 44s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/5b2774aea1434a7ca096f93f6cafd627 : FAILURE in 4m 35s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/b0123afac509453ab086cb5d3be95c2e : FAILURE in 4m 16s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/aa38eedd407c4e578aebe5689c38757f : FAILURE in 4m 04s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/a3d84d29c4ad48cd8bb3c482b5af837e : FAILURE in 5m 21s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/5c0dd944050c4e538a4c67edd1c3cc76 : FAILURE in 3m 56s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/f30ee26974c447c185485945ee9cc8a4 : FAILURE in 4m 30s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/34a80839864844f59b57db58a503e429 : FAILURE in 4m 41s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/32333b75476347a08a0193ef797c5884 : FAILURE in 5m 05s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/e5d8faee8cdb4597a5807c6c8b21eab1 : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":7},{"id":"0a4cf176aa04d3fbbd77c70574669b869c98f3c6","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-05-11 03:45:47.000000000","message":"Patch Set 7:\n\nCan anyone help me understand what changed that this started failing?","accounts_in_message":[],"_revision_number":7},{"id":"4b600fa1de741a00648a5b32331b712f5c9aed7c","author":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"date":"2020-05-15 01:49:18.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"71217ba84f89efae898e9a600a6649a7c12e86b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 03:16:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a1d011c838294a1bacf3b9b76be764f0 : SUCCESS in 3m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6b297586b74c44a0b76ab8983e1044e8 : SUCCESS in 1h 22m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3471d25a77d84aad94eb32ab690dd8cc : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49fca64160eb41c7b8960cb0da217b74 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ae633e7fab845ecb48411a6ea1ce194 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b033f54066cd42d0b0d3c7573ee9d9cd : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fa8a0f106264fec891c10f93dcad169 : SUCCESS in 5m 23s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/13b11cba551d4bb99701fa5464950ac3 : SUCCESS in 5m 37s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2b7cd3cfbd1e41d6b03da59e325b527e : SUCCESS in 4m 38s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/e0e5f2124f3d48a9b2d9fc264c86c949 : SUCCESS in 5m 10s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/5a666649054545149c14e5df2bb00cc4 : SUCCESS in 5m 00s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/9b961d89f63f49d7aee7ee862ab47c69 : SUCCESS in 4m 58s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/23447e2632344427837d93cb7386d6e4 : SUCCESS in 3m 55s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/932978fb7b3e4f02a7da288e5c35149a : SUCCESS in 7m 21s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/1eb030b59bf14794adaad7775efbea0e : SUCCESS in 4m 48s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/b4b2efa218e646bba1604e8154e69b38 : SUCCESS in 6m 08s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/b339dafb89f645dd9eef84ac20568f32 : SUCCESS in 4m 50s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/90160604295f4ba7bb7f791d93bba360 : SUCCESS in 5m 03s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/336bccd7254649568746e8ea319c4c97 : SUCCESS in 6m 55s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/f703616218914bb49d15afeba7fdc7e3 : SUCCESS in 6m 09s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/503c24f6a9e84c71b1d04fd366527533 : SUCCESS in 4m 46s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/ca0a2d61318f4a54bd04bd4afa9a729d : SUCCESS in 5m 15s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/75a8ab1a44624083bc3e5c6f6fc14845 : SUCCESS in 4m 12s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/fc844ade29004588913eb2856d137f66 : SUCCESS in 5m 08s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/6d0ba7ee103c4e3093f89d3448682bcb : SUCCESS in 6m 58s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/2f7b3455841c4ab8af94026e63607092 : SUCCESS in 5m 53s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/220cd187e21f4f72ab905a4b3546677b : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":8},{"id":"4b3bdaa15100eb6a5a597900956a051898080526","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-15 08:26:06.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1d7cff9ebb37fa37ddd67e50751012ec06a6aec0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 08:26:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"fcaed242f92925cbc42ae5aab0ad8d037df70348","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:01:47.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c063db7a59243f2ba8742d5ac872d29 : SUCCESS in 5m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25755e42dd714dd0a0f5b93a31d169fa : SUCCESS in 1h 29m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f8ac860f43c44228f3480991b081ca6 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5fb4fa4f3e3f4a14b72700913b29f342 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a224af20948e4556b784310be1018bf9 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/089cf4331a4343a0bd66a568b01628a0 : SUCCESS in 8m 28s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/454a6f3b57da4410b8db2829d27566ad : SUCCESS in 5m 46s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/de41dca92bd34fbaac179c556da3ac99 : SUCCESS in 6m 42s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/6cd7763886284622984d645825873edf : SUCCESS in 5m 55s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/a2ab80c98b114ba9bb80d99863cb5b3a : SUCCESS in 6m 56s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/d04258cb4f664cafb357226514e3d6aa : SUCCESS in 7m 27s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/9870aa17b471450781e07972a7f6ab31 : SUCCESS in 5m 44s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/86f8bec9105b4aaba5305068a47332b6 : SUCCESS in 5m 43s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/a0c8aa9798aa49bc90f7fcd79b83eaef : SUCCESS in 5m 02s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/bdc86f7a785a4f7eb669b15118f4402a : SUCCESS in 6m 23s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/00a4fc2e76474daaaf3f18b60ff5676f : SUCCESS in 5m 31s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/2c369f5ab4b54f679fcc2ea987f64973 : SUCCESS in 5m 55s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/d18e3f129eab42d7abf7d17f1b146b30 : SUCCESS in 6m 36s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/e41f4507592f4950b8c8b61bd1c27e2c : SUCCESS in 5m 58s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/12bd4d99b5cd40aa9e62fdce6343cc97 : SUCCESS in 5m 34s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/ccb4ca68feb642148e81d43f6ab73a5a : SUCCESS in 5m 14s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/c2611e15004041b99c702ad00ced0aa3 : SUCCESS in 8m 03s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/fbea9517efce4ecc9db689f295511004 : SUCCESS in 7m 03s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/d24344dba61a40a9bb81d9d17c54e21d : SUCCESS in 4m 11s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/8bf826a6117a46379238a123273cc74e : SUCCESS in 5m 03s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/49a3f712198540298c77416d69ba76c9 : SUCCESS in 7m 39s","accounts_in_message":[],"_revision_number":8},{"id":"af084410485d7826348ff30923a2028736b7871d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:01:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"c831f92bf6f7e50a1ff6d6b42b3ad23c3a86759d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 10:03:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30afb83d78ad4fce8e5202881df217c7 : SUCCESS in 1m 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2650813936e4b82a74d4746ab627144 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"08b2d93226071c74ce8bf85265429511e417af73","revisions":{"901f140149286cdaec51ac6621b56e06bb52ba56":{"kind":"REWORK","_number":1,"created":"2020-04-08 15:19:43.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/1"}}},"commit":{"parents":[{"commit":"1880be2c000b9a831f58c9f29e13d700f415db62","subject":"Merge \"Add TLS support in etcd3 and etcd3gw drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1880be2c000b9a831f58c9f29e13d700f415db62"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"subject":"add Consul ACL token paramater","message":"add Consul ACL token paramater\n\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/901f140149286cdaec51ac6621b56e06bb52ba56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/901f140149286cdaec51ac6621b56e06bb52ba56"}]},"branch":"refs/heads/master"},"f824502e3bee245da8d582f5fe2a1a4a49f7d9aa":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-08 15:27:42.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/2"}}},"commit":{"parents":[{"commit":"1880be2c000b9a831f58c9f29e13d700f415db62","subject":"Merge \"Add TLS support in etcd3 and etcd3gw drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1880be2c000b9a831f58c9f29e13d700f415db62"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:27:34.000000000","tz":-240},"subject":"Closes-Bug: #1752205 add support for Consul ACL token parameter","message":"Closes-Bug: #1752205 add support for Consul ACL token parameter\n\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f824502e3bee245da8d582f5fe2a1a4a49f7d9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f824502e3bee245da8d582f5fe2a1a4a49f7d9aa"}]},"branch":"refs/heads/master"},"b72b54d0b1d2b0c46f0ea0579056b02bca58dbc8":{"kind":"REWORK","_number":3,"created":"2020-04-08 15:44:58.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/3"}}},"commit":{"parents":[{"commit":"1880be2c000b9a831f58c9f29e13d700f415db62","subject":"Merge \"Add TLS support in etcd3 and etcd3gw drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1880be2c000b9a831f58c9f29e13d700f415db62"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:44:02.000000000","tz":-240},"subject":"Add support for Consul ACL token parameter.","message":"Add support for Consul ACL token parameter.\n\nWhen running Consul with ACLs enabled, the best practice is to default deny all. In order to support applying a policy to the tooz coordination KV store path, passing through an ACL is needed.\nFixes pep8 error in mysql.py.\n\nCloses-Bug: #1752205\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b72b54d0b1d2b0c46f0ea0579056b02bca58dbc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b72b54d0b1d2b0c46f0ea0579056b02bca58dbc8"}]},"branch":"refs/heads/master"},"8800ffe08eb5334ac0892215fce7a2ef82149b2e":{"kind":"REWORK","_number":4,"created":"2020-04-28 20:08:19.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/4"}}},"commit":{"parents":[{"commit":"1880be2c000b9a831f58c9f29e13d700f415db62","subject":"Merge \"Add TLS support in etcd3 and etcd3gw drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1880be2c000b9a831f58c9f29e13d700f415db62"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-28 20:07:59.000000000","tz":-240},"subject":"Add support for Consul ACL token parameter.","message":"Add support for Consul ACL token parameter.\n\nWhen running Consul with ACLs enabled, the best practice is to\ndefault deny all. In order to support applying a policy to the\ntooz coordination KV store path, passing through an ACL is needed.\n\nCloses-Bug: #1752205\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8800ffe08eb5334ac0892215fce7a2ef82149b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8800ffe08eb5334ac0892215fce7a2ef82149b2e"}]},"branch":"refs/heads/master"},"0c317795b20a59348564bf3105bd4d4b2f702f9c":{"kind":"NO_CHANGE","_number":5,"created":"2020-04-28 20:28:09.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/5"}}},"commit":{"parents":[{"commit":"1880be2c000b9a831f58c9f29e13d700f415db62","subject":"Merge \"Add TLS support in etcd3 and etcd3gw drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1880be2c000b9a831f58c9f29e13d700f415db62"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-28 20:27:53.000000000","tz":-240},"subject":"Add support for Consul ACL token parameter.","message":"Add support for Consul ACL token parameter.\n\nWhen running Consul with ACLs enabled, the best practice is to\ndefault deny all. In order to support applying a policy to the\ntooz coordination KV store path, passing through an ACL is needed.\n\nCloses-Bug: #1752205\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c317795b20a59348564bf3105bd4d4b2f702f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c317795b20a59348564bf3105bd4d4b2f702f9c"}]},"branch":"refs/heads/master"},"0a4f1909cd7eb3436e930d89464765c34f3efd74":{"kind":"REWORK","_number":6,"created":"2020-04-28 20:29:27.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/6"}}},"commit":{"parents":[{"commit":"1880be2c000b9a831f58c9f29e13d700f415db62","subject":"Merge \"Add TLS support in etcd3 and etcd3gw drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1880be2c000b9a831f58c9f29e13d700f415db62"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-28 20:29:20.000000000","tz":-240},"subject":"Add support for Consul ACL token parameter.","message":"Add support for Consul ACL token parameter.\n\nWhen running Consul with ACLs enabled, the best practice is to\ndefault deny all. In order to support applying a policy to the\ntooz coordination KV store path, passing through an ACL is needed.\n\nCloses-Bug: #1752205\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a4f1909cd7eb3436e930d89464765c34f3efd74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a4f1909cd7eb3436e930d89464765c34f3efd74"}]},"branch":"refs/heads/master"},"15315ad38d7bd1fb8047f012e8379ea16d2171b2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-05-10 19:03:25.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/7"}}},"commit":{"parents":[{"commit":"c47e8db010bef0f1fb85aed6fbd3dcedacb0e400","subject":"Merge \"ignore reno generated artifacts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c47e8db010bef0f1fb85aed6fbd3dcedacb0e400"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-05-10 19:03:20.000000000","tz":-240},"subject":"Add support for Consul ACL token parameter.","message":"Add support for Consul ACL token parameter.\n\nWhen running Consul with ACLs enabled, the best practice is to\ndefault deny all. In order to support applying a policy to the\ntooz coordination KV store path, passing through an ACL is needed.\n\nCloses-Bug: #1752205\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/15315ad38d7bd1fb8047f012e8379ea16d2171b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/15315ad38d7bd1fb8047f012e8379ea16d2171b2"}]},"branch":"refs/heads/master"},"08b2d93226071c74ce8bf85265429511e417af73":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2020-05-15 01:49:18.000000000","uploader":{"_account_id":31615,"name":"punkrokk","email":"jp.bourget@gmail.com","username":"punkrokk"},"ref":"refs/changes/88/718488/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/88/718488/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/88/718488/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/88/718488/8"}}},"commit":{"parents":[{"commit":"977c7b9ff9c9f9acba588df947d14d92a12b1aea","subject":"Add nose as test-requirement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/977c7b9ff9c9f9acba588df947d14d92a12b1aea"}]}],"author":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-04-08 15:19:34.000000000","tz":-240},"committer":{"name":"JP Bourget","email":"jp.bourget@gmail.com","date":"2020-05-15 01:48:49.000000000","tz":-240},"subject":"Add support for Consul ACL token parameter.","message":"Add support for Consul ACL token parameter.\n\nWhen running Consul with ACLs enabled, the best practice is to\ndefault deny all. In order to support applying a policy to the\ntooz coordination KV store path, passing through an ACL is needed.\n\nCloses-Bug: #1752205\nChange-Id: I98fc96468b21368ce66365e3fc38c495b1f2918a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/08b2d93226071c74ce8bf85265429511e417af73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/08b2d93226071c74ce8bf85265429511e417af73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
