)]}'
{"id":"openstack%2Ftooz~719902","triplet_id":"openstack%2Ftooz~stable%2Fussuri~Iaae95bf720ad2c80bf112c969091fdd303c9ac23","project":"openstack/tooz","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"Iaae95bf720ad2c80bf112c969091fdd303c9ac23","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-14 10:20:27.000000000","updated":"2020-04-14 11:47:53.000000000","submitted":"2020-04-14 11:37:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719902-1586864244483-354c625b","meta_rev_id":"48c82e092439a331eb9688951a43d164cca49ff5","_number":719902,"virtual_id_number":719902,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-14 11:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-04-14 10:45:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-04-14 10:47:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-04-14 10:45:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 10:45:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-14 10:47:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-14 11:37:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42cd0fda9fa1683cb28f2ba7b845c7fe6b105a3e","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-14 10:20:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd341d4d1f9789e88025db96f2cb764c8bc21ca4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-14 10:45:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f92fe24dc4d131e7e5672a91317ef71355ea2e0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-04-14 10:47:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"146116309b8dd675ba5bfd076e94a91f74ed322a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:24:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b90c634451334f719ff1ad84eca0bf9b : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b4d7bfc08534357b362a4537e8481ee : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a16ffe87ea140218f8c869c34716909 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e430f6003d48a4bbad60adbdb64668 : SUCCESS in 5m 12s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/ed3c2e6920ea4c5b9f5398b1818aaa0b : SUCCESS in 7m 11s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/5bee160a255c496cb3a1b0fbcb28e22c : SUCCESS in 6m 50s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/7c1aa3a0bf0c47679089af9a274eb210 : SUCCESS in 5m 43s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/74881f280cce477babcb1e595e2b1e55 : SUCCESS in 6m 48s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/cb30f63ede264278aa581f67a623f38a : SUCCESS in 8m 49s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/2b794348315746feaa66d120ced1445a : SUCCESS in 7m 32s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/3eee08367327499ea8ab24b6984c35e5 : SUCCESS in 7m 02s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9d2aa57dac3a4e47ad365d7658382827 : SUCCESS in 8m 00s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/cca64beeba054e5fbc10fef08b282ada : SUCCESS in 6m 21s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/c24ea10050554c33a2222ad5d5611c14 : SUCCESS in 8m 04s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/72119b058c1746cb83d2a2230c1f2f43 : SUCCESS in 9m 15s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/57aefbc3e57447fb81a32eefddb96dd0 : SUCCESS in 8m 22s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/7dd3df069c6a4f6e90d398166c804de6 : SUCCESS in 7m 15s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/e616d131431047e48cb7d9b25eb79bb3 : SUCCESS in 8m 05s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/af508523bd324bacbcf61dbedf2edfd9 : SUCCESS in 7m 01s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/412da527881c4d78bc0e92479dc18ccf : SUCCESS in 8m 20s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/20e85d3cf8294e1d99123ef664050fa8 : SUCCESS in 7m 56s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/d0b5ca3ca15545e79445a79d3d9928f0 : SUCCESS in 8m 30s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/ed851ac99b9e430281272d6e6136aba5 : SUCCESS in 9m 19s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/cc9929422b6b458faefab54d82234242 : SUCCESS in 8m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b190e33aa5991bdcb65edf8d36bb204e448892d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:24:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"939758c75c6a749dfafa3ab1608182e9c811362e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:37:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d6b69f8f4514bddbac8ab2172f79a57 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/daa89f0e7f58419daf68f62edf65a3ec : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632c620b031e4a9da5cfc4aaa1a8a70b : SUCCESS in 8m 01s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/09050c12e4fd4803b4c10699023e9a4b : SUCCESS in 8m 23s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/a9004a09a80c4e41b92b1b746e89187a : SUCCESS in 7m 49s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/833f744980cb4b3fb219b139784654d6 : SUCCESS in 8m 01s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/6d1d833637de40ca90c4ff35bcbc75eb : SUCCESS in 5m 59s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/4fd2cf56d3604c10bbfe405cd3ed91dc : SUCCESS in 5m 28s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/92c8dc89905e4f8caba0eb29eea2354c : SUCCESS in 5m 13s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/143937117b75438fb93111aa3f9c732e : SUCCESS in 8m 03s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/960a977c743f4d57a02c12a042bf6b6f : SUCCESS in 5m 47s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/fc00fe4a412348e38421943c6bd370b5 : SUCCESS in 5m 06s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/bde6f5eeeb1a42949994d3f6c91fb488 : SUCCESS in 7m 02s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/63a1f21a8df94f63bac10025a3c4f2f0 : SUCCESS in 6m 07s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/6f710c5cd734490fac18ca74d27a9d94 : SUCCESS in 5m 25s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/a0357204ce4c4347ad03776c8459f514 : SUCCESS in 4m 59s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/722a07259da7449790efeef089d0fba7 : SUCCESS in 5m 03s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/736371ec578640c1ad46254696ab9e45 : SUCCESS in 5m 08s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/7f339f7a0f5b43b8b94bd6185648aea1 : SUCCESS in 7m 22s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/69cb7a9e2efa4e39a4d25a3c2f49718f : SUCCESS in 7m 43s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/070b0b9cdd3c477485e3ef53fccd151e : SUCCESS in 4m 49s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/93377f80967345c99782e525d2917b5f : SUCCESS in 4m 46s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/dec4474b2a7744419c778e1aef5fb5ea : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d82df8c71f738a7ce0fa4e1677a463c8a3ec7b8f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:37:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0c6c3ed7cd1b493af4a753d501939a632ec0dbfb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 11:47:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7af1460133946d18e745f068adf6ffa : SUCCESS in 2m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b47e68772756e5d538d2f09f35f6fef05a860a8b","revisions":{"b47e68772756e5d538d2f09f35f6fef05a860a8b":{"kind":"REWORK","_number":1,"created":"2020-04-14 10:20:27.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/02/719902/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/02/719902/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/02/719902/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/02/719902/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/02/719902/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/02/719902/1"}}},"commit":{"parents":[{"commit":"1880be2c000b9a831f58c9f29e13d700f415db62","subject":"Merge \"Add TLS support in etcd3 and etcd3gw drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1880be2c000b9a831f58c9f29e13d700f415db62"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-14 10:20:25.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-14 10:20:25.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: Iaae95bf720ad2c80bf112c969091fdd303c9ac23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b47e68772756e5d538d2f09f35f6fef05a860a8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b47e68772756e5d538d2f09f35f6fef05a860a8b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
