)]}'
{"id":"openstack%2Ftooz~728032","triplet_id":"openstack%2Ftooz~master~I99c029d7a153548b9cfc94f5d80d331ca87d2b70","project":"openstack/tooz","branch":"master","topic":"victoria","hashtags":[],"change_id":"I99c029d7a153548b9cfc94f5d80d331ca87d2b70","subject":"Add nose as test-requirement","status":"MERGED","created":"2020-05-14 11:14:09.000000000","updated":"2020-05-14 18:10:49.000000000","submitted":"2020-05-14 18:09:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728032-1589479756583-299f2a5a","meta_rev_id":"f6497dcf5999297b1c240c40e6169a0ddea18dd7","_number":728032,"virtual_id_number":728032,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 18:09:16.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-05-14 11:27:30.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-14 16:32:18.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-05-14 11:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-05-14 16:32:18.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 11:22:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-05-14 11:27:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-05-14 16:32:18.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-05-14 18:09:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d7dedcaf3b549f6299cd2222d2bce2d1f8cc60c9","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-14 11:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31a6f786637e8671ad2e3fa809a5734edfdde98b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-05-14 11:22:56.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good to me, especially because this patch needs to be backported to stable/ussuri, as it started to fail [1] there, too, with:\n\n  File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py36/lib/python3.6/site-packages/ddt.py\", line 15, in \u003cmodule\u003e\n    from nose.tools import nottest\n  ModuleNotFoundError: No module named \u0027nose\u0027\n\n[1] https://zuul.opendev.org/t/openstack/build/5ea42ad0e1a5443b9e79a365fa73c470","accounts_in_message":[],"_revision_number":1},{"id":"852962a5c584dd7224e757d46fd5ce3de27dc8bd","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-05-14 11:27:30.000000000","message":"Patch Set 1: Code-Review+2\n\nIt may be worth waiting for a new ddt release since they have since fixed this issue:\n\nhttps://github.com/datadriventests/ddt/pull/86\n\nBut we can get things working for now, then clean some of these up later.","accounts_in_message":[],"_revision_number":1},{"id":"50ba134f1fc9bb3fd38c6872883d8c4d4ea5c163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 13:07:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/656fdec2ba864686b3dca23105981a2c : SUCCESS in 1h 52m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbb805fc0844417698fddf86dd5f3c8e : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e618c73939574dd18374b5bd07624a8b : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ead1bfe0cb4d438e8bd64cc5de33e71f : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdff936dc86f42669c92c8650a4ee4df : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d84b3b0d1014ab3bf7458351b5d2f2a : SUCCESS in 4m 29s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/6de5bc18ddf9437d8c8b57ddbcee4e4d : SUCCESS in 6m 04s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/136ed982523d46dc9f9e5a17414acf72 : SUCCESS in 5m 42s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/fe23c344efbf4d989662d438fbc0d43b : SUCCESS in 6m 28s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b92ef2871c024fe88dfb08561f0ca24b : SUCCESS in 5m 18s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/cf42d7e013ea4262822d3b75cba4e9f2 : SUCCESS in 5m 36s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/0547d1059fbb4d5bb73b97dacd368b8e : SUCCESS in 8m 57s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/b4a15a082fe64c3a8b23b057cc9c19b6 : SUCCESS in 5m 56s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/e3e1b10f2671417986ef10dfff7d55b4 : SUCCESS in 4m 59s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/de8d0b5c79e74c01b29c67a3add85310 : SUCCESS in 5m 04s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/ede39fa6ab8844a3bf7944471a7b8438 : SUCCESS in 5m 32s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/0490ec16ddb147538fa51b9d875ee4da : SUCCESS in 9m 48s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/15ad871237ce4971926025298fd0befc : SUCCESS in 5m 48s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/3293d282e8cd4b4b85f3e54bad3da024 : SUCCESS in 6m 03s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b70b3ade23044700a3a2ab38278754ff : SUCCESS in 9m 33s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/864e809cb4d6486ebb261c4ef466e596 : SUCCESS in 9m 07s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/c7c20bca6f444dbfaa5c16147dac4ef2 : SUCCESS in 8m 33s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/413ba8ccf7524324b3f27d9447544556 : SUCCESS in 9m 04s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/c620273f60a24afc9a8f9bd1435bc809 : SUCCESS in 8m 53s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/dfd73f49806d450796f845472591594f : SUCCESS in 8m 40s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/be2b1e495aff40abb50b664c6c1c913a : SUCCESS in 9m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a3b26d9ba0b95bee28aa0046412ef1d8c73cc21c","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-05-14 16:32:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bff69efffda63368ffaa954f18e180d2a2190301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 16:32:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48e96eb0604f8614fb74d542493f0a8be591f4b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:09:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7116c603804f406d956f4e9ebe3f1149 : SUCCESS in 1h 31m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7ff8a3fa59b44fcb35fd7f20306ea85 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85b8d8b5df644e3eb1d8e5d346d8eeb0 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5800623d2f6346a3a368b910ded7f0c0 : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35df1ba0eac74a37a1dd2a0386e9401b : SUCCESS in 6m 38s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/e7449efe1e31476cae88715d631d3e87 : SUCCESS in 5m 16s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/7a2915b31c50485f8373c7c56bc87605 : SUCCESS in 6m 03s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/6e3f8329bcb54e0e882a965ecd2cae14 : SUCCESS in 7m 05s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/09c920d4a7a5419baf12c4ed1ac1936a : SUCCESS in 5m 03s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/3673793d8f574baf9ddb79910ea1749e : SUCCESS in 6m 13s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/99f9f09b91cc463fb236b1a3511e8f14 : SUCCESS in 4m 47s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/f043c737f4e4411faf9b8c0637432e1b : SUCCESS in 6m 15s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/173eed16391a4eb69e44ac605e0d4086 : SUCCESS in 7m 15s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/f6acc21634444f3ebaa446dfdde13215 : SUCCESS in 8m 45s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/86e2c2595fbc498a965f7cc51811719c : SUCCESS in 5m 15s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/e116ff8a1aed463697d595afbf3eeed5 : SUCCESS in 11m 07s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b3ede7f7b62344b5b7cadeadf80112ee : SUCCESS in 6m 26s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/407523a374b44a2c810b4422d9aeed9c : SUCCESS in 7m 04s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/d3364a2ec87b4f7ebd85443c1454a244 : SUCCESS in 6m 08s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/665d8722fc6e4bddb72ff8f1d3039ffc : SUCCESS in 8m 05s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/a80ecf165be64694815b453d0da07045 : SUCCESS in 5m 53s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/03a5dafe97ef40b5accf3f0457ad0215 : SUCCESS in 5m 55s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/8de15c0eb0e743c2827a0971bb190069 : SUCCESS in 5m 45s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/9a5b5970bd4c4d4da99bcdda1d76f6d5 : SUCCESS in 9m 59s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/417ac500add845a9961cd8c1e843598b : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c92923979148d14dd8f78642426301ffd36862a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:09:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"493a52b29fd5cc132b5587a0a527387bedbca207","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 18:10:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8002b67e21425895f224a5893e8c06 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20fff9d6492f4b61816c037b87aed8d1 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"977c7b9ff9c9f9acba588df947d14d92a12b1aea","revisions":{"977c7b9ff9c9f9acba588df947d14d92a12b1aea":{"kind":"REWORK","_number":1,"created":"2020-05-14 11:14:09.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/32/728032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/728032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/728032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/728032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/728032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/728032/1"}}},"commit":{"parents":[{"commit":"c47e8db010bef0f1fb85aed6fbd3dcedacb0e400","subject":"Merge \"ignore reno generated artifacts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c47e8db010bef0f1fb85aed6fbd3dcedacb0e400"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-14 11:11:37.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-14 11:11:37.000000000","tz":120},"subject":"Add nose as test-requirement","message":"Add nose as test-requirement\n\nAdd nose requirement, this is now needed by ddt for testing here,\nbut not required from ddt.\n\nChange-Id: I99c029d7a153548b9cfc94f5d80d331ca87d2b70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/977c7b9ff9c9f9acba588df947d14d92a12b1aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/977c7b9ff9c9f9acba588df947d14d92a12b1aea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
