)]}'
{"id":"openstack%2Ftooz~730767","triplet_id":"openstack%2Ftooz~master~Ife3021fe203fdb6a9a6f2f51d45a64773f5fe89c","project":"openstack/tooz","branch":"master","topic":"py38","hashtags":[],"change_id":"Ife3021fe203fdb6a9a6f2f51d45a64773f5fe89c","subject":"Add python 3.8 to classifiers","status":"MERGED","created":"2020-05-26 12:07:05.000000000","updated":"2020-06-08 23:37:31.000000000","submitted":"2020-06-08 23:36:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730767-1591659363128-f90b450b","meta_rev_id":"dabaf90caa44675d39f43568ce6beb6da870d509","_number":730767,"virtual_id_number":730767,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-05-26 14:17:54.000000000","_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 23:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-06-08 15:46:24.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-05-28 09:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-06-08 15:46:24.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 09:40:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-08 15:46:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-06-08 23:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a793e8170909b1431d6efdc23c83a63e6f9e0346","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 12:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14ceeea14d717f1320abb1193bd4de4802b023e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 13:43:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbb97df7e4b94fb79526970c7880477d : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eee9633fd8e8403b97c01bb8ec294a50 : SUCCESS in 1h 34m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57dd6553124048ab9816d7deaf8a3838 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/868792f4dcc5416488274647192fea49 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a4ab07fb6a649d5948f13b7f0098d53 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d03f5ff5be3b49b3bfe965f2c1fa3bb0 : FAILURE in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8db820f846ce4b5b9f9fa828d0465e9d : SUCCESS in 7m 53s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/da37ec6959334351bcea9b4e2207810c : SUCCESS in 6m 28s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/1472abba928b41d3aef1ef7c97a45b5e : SUCCESS in 5m 07s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/33be7a5d823847e69caaacf672ee55d2 : SUCCESS in 5m 52s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/a5e433dd71ca487fad7cf3562ee2b19a : SUCCESS in 6m 29s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/851846023adc423694da42e9ecfaef28 : SUCCESS in 7m 12s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/0beb6afe4d36444fa468214046d341e1 : SUCCESS in 7m 34s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/40b4075c865a4581b2e59c2d3a948cd3 : SUCCESS in 6m 40s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/e552209c5fea499db1c8e12d24c572a6 : SUCCESS in 7m 39s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/2068c69f313545f5a586dca7dd0acefc : SUCCESS in 7m 04s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/31c800c9a7284c068b5cca6cc0b34a15 : SUCCESS in 7m 34s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/75047a072ccf4aec9d67497ce5a403c7 : SUCCESS in 7m 13s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/c1e18cb305af43308d36b0fb8071ef5a : SUCCESS in 9m 15s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/fc48b96061c442d4b624258c128b8192 : SUCCESS in 9m 53s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b9afcc291c544859b02291353e0a191c : SUCCESS in 6m 19s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/fe9227769a854f0c9409815b81e2b115 : SUCCESS in 7m 23s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/ad25363d4c584371b16970541da43e9c : SUCCESS in 7m 09s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/6b18b7256be44c959f06838a68928d9a : SUCCESS in 8m 18s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/8233f674cec84975bb714c55b4e6ded4 : SUCCESS in 9m 08s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/e868a558765a47d483c94ab48f9486af : SUCCESS in 5m 39s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/29db517bbd8a470880c2949a8588386f : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1086fe3f18d9d6893a985c7305fb30a7256df216","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-05-26 14:17:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"def2dd10bb58244f32b20152d314c2a700baf082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 15:43:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7228ab7e0f9e49de90a20fc7269bf723 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13a5655075fc4ee2969d39c5a9db8edf : SUCCESS in 1h 20m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ed8733ed72b43a29c05fb752d50a287 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff077e1c74db42f5bbf2c962efaf0a7f : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88498e40db4d4bbeb69ff0fb7d9c4468 : SUCCESS in 8m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05570180e0934590b68d11d293a8b2e1 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6861d6cd097a4913b65caef3a39848cd : SUCCESS in 6m 44s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/7188e467d48f411dbd0fbc0fd95d866b : SUCCESS in 7m 18s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/3848ec3efeb04fad8992c9a6199835e7 : SUCCESS in 7m 50s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/0d02d8b200324f3d82a944646a14b63d : SUCCESS in 7m 52s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/6fbed633a1874d92bacb4880d8e9da8f : SUCCESS in 5m 57s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/8cf06efdc5474c978c63fbd3602f8d10 : SUCCESS in 7m 02s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/66e6f3c75cd44bb5a16e8c83b3544375 : SUCCESS in 8m 12s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/189bcba4572940878e97f2dd308e4d31 : SUCCESS in 8m 51s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d82d809493274d0f8baff1959755bdc6 : SUCCESS in 5m 15s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/14f0455e21d74cb8a3ff7d6963e8ce13 : SUCCESS in 8m 59s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/084ae2f047534a9ba2e4e27d638da125 : SUCCESS in 6m 41s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/cc43b7e490774a929561ccf9837fff2c : SUCCESS in 10m 46s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2caebee9a30142b6b8785cd29197d44e : SUCCESS in 7m 18s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e667d222661c44ecbcce4d11af88cf17 : SUCCESS in 7m 15s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/f90baf1679b543afb0e84baecddfe196 : SUCCESS in 7m 48s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/dfac9d039f214e23a90912ea2f49bb8b : SUCCESS in 8m 21s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/d08c0efd645f4a208970248b19f2d56f : SUCCESS in 7m 19s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/8a2215a3cbb542c28b474df1f5ac3d78 : SUCCESS in 6m 24s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/845846cbbacb4758b741f46787cf0750 : SUCCESS in 6m 26s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8f2535e8a54849b58832eeb51faad025 : SUCCESS in 8m 26s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/89995d23c7444b048c1e3db4577b8665 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7105bb22f55103486b3559bc032cf70216c46f80","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-05-28 09:40:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"84291cd2ec8cec690fc3a71bf31a649ebcc90901","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 15:46:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nVictoria template is being used, so this should be good to go.","accounts_in_message":[],"_revision_number":1},{"id":"fe82af3ab3b742b1ec7aff0e39f2070d487a2db4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 15:46:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd9f9a70e260db25bbd91f108b185807ec98956e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 16:51:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ad2cdcb684c47a0993c3e6ad273c4aa : SUCCESS in 7m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b49f78500c8e47f6a25a45a55f4d5d44 : FAILURE in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e93505af73944f9d83d9972525434ec3 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2694da4ea94a4e2686aac792ac7c48a4 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97e635eb51c44d5f9f323002586886a8 : SUCCESS in 9m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fcf9bcff1c544d28a8a9d08c0aac6a0 : SUCCESS in 7m 33s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/08d19503ce20411e8bf6ca085d3b04e5 : SUCCESS in 8m 14s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/c57fe873b24f467ca7d2a83952c0aa94 : SUCCESS in 9m 14s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/5e8b3cb98b3e43ddba965f66e3e7597f : SUCCESS in 6m 19s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/5887ede5144949afae9cd63c2043b711 : SUCCESS in 6m 23s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/98c993529d414d60b9a0cb208eb514cf : SUCCESS in 8m 00s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/601e27a3f5614f17a8c236c23a53007d : SUCCESS in 5m 47s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/d78d001d682d42c1abd1f81f63462365 : SUCCESS in 6m 46s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d93fc55264c841029ff8695499adf947 : SUCCESS in 8m 59s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/7b3382d2f41046f080ab2f774bb7a045 : SUCCESS in 7m 37s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a678f787d19646f9800dfaafe0941263 : SUCCESS in 9m 10s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/1500674a6b5e407caaa03e60a6ea7ff8 : SUCCESS in 7m 10s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/c5079552052c458ca25976d9690eab81 : SUCCESS in 5m 47s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/708425c4594e4b86b8229d1731332265 : SUCCESS in 7m 29s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/c0346b5e2350447aa4f82a1797e69c2b : SUCCESS in 8m 29s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/97c49452382d4ef888f0ad0d5c63133a : SUCCESS in 6m 50s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/7cdbac27ac6143e6adbfdbc1e6df041d : SUCCESS in 6m 56s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/8f64f81c98b44ed1b1253dd155296ef4 : SUCCESS in 8m 56s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/0e6d1dadbb384b9eb1c213ff77825b47 : SUCCESS in 8m 44s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/371b7fa352b54e01b144eb1b7ef9d4f2 : SUCCESS in 6m 54s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/ec90d7d578394f7689e1007089d3adf6 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f9e84482c050dd901f9ae1d89028b9b4a5ab0974","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 16:55:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3a210ec1dee7ae2e68f730f5d2d168cbe2c045ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 17:55:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd59f5f5e7bd45f59df4682b3813e25a : SUCCESS in 3m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93c513e9bbdc4bd2be7512e079cd5192 : SUCCESS in 59m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fef854e078a45688cae4dcf61bc18f4 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849f5dc0e993452eb2d4a9d440697696 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7704d7028027444f99556fb208449dc6 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a747e3dd4d14d588a2d64aff35db0db : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5b653b08aa47afa539f9e9069dfd4b : SUCCESS in 5m 39s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/19e9a6dcb2af463caad7342f9989dbe9 : SUCCESS in 6m 08s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/efa21d2fd35b4088b0f1f2b5e9bbfcc8 : SUCCESS in 5m 34s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/cbb69d9bdf1e4723a5219113738f76c9 : SUCCESS in 5m 32s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/97e67453bf704039a47a0b037f0a0efb : SUCCESS in 5m 30s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2413830f1d894402a38973ee2e4cb6b0 : SUCCESS in 5m 18s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/dd127c8a243a42599d9eb36d8e780c4b : SUCCESS in 5m 36s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/2ade74e986274d658c7c1932f6a09fd4 : SUCCESS in 5m 38s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/89491ea450534993b64ba9783d989f86 : SUCCESS in 6m 07s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/7ef37f78153249d4a56299912d0aa4f5 : SUCCESS in 6m 09s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/8144b6de506d4fdb8422967bbdb2dce6 : SUCCESS in 5m 27s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/905ee6dbcca04d8eb2d3706cf183eb96 : SUCCESS in 8m 03s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/96b522068fdf4207946418f9d10adaca : SUCCESS in 5m 13s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ff25a795d74f4c71a2e2626a020e4d66 : SUCCESS in 8m 29s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/fefcf6b18f644f498dfcd394c03de833 : SUCCESS in 4m 32s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/839a6772263949aba2b041d005333408 : SUCCESS in 5m 54s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/e7e591c05dee4f5b9f0602b69c491845 : SUCCESS in 6m 17s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/d03c6f46c08f4d94b0b144beb3056296 : SUCCESS in 6m 16s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/0e257ff525f542c6952a7e1fe298e0ef : SUCCESS in 6m 12s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/719262c78b214de68e3011d479926037 : SUCCESS in 4m 58s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/e159d32a46da4b6d97f752bf2f7f3c70 : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5417720bb1abe4650a7b44f5e40363dd2ce96148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 17:55:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"248f112f1e95ec8b87431b8c0caf20eb83196787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 19:10:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7007afa2d0af4f4ea7babefaa455dbec : SUCCESS in 3m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dfa57f2d923471cada8bf6b6c31de89 : SUCCESS in 53m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf26dd96d1a44a79c899a8e618ea8f5 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d61845c36f34df1881b5be6d60dd756 : FAILURE in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bb245546fa44ecdb905ba0b08b8f2e2 : FAILURE in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48aaedd0f0ec42a5880d62b58fad24ad : SUCCESS in 6m 10s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/1e2f10e2c8564f69a29aae6da4afe954 : SUCCESS in 6m 23s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/57ecb9ce9ada4899bdfc31db4ec2f1b4 : SUCCESS in 7m 25s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/2386e3c3e17d4dffb577f0a440cf4ff2 : SUCCESS in 4m 50s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/e93eeb4d380e49aca746118f3c606f3d : SUCCESS in 4m 17s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/43834bba47fe4ccb83070e01a31a8ea3 : SUCCESS in 5m 40s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/62b47ec3cea14573b1a7ae29d765e9d1 : SUCCESS in 6m 19s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/0051ec756b62415a96913355edf0b828 : SUCCESS in 7m 45s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/c0fe23dcfacf4e25b5abaa3f7c7f8276 : SUCCESS in 8m 01s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d7e002207ced4a6896e322ff1c81fa96 : SUCCESS in 6m 58s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/d1dd69f2def74463a4872f9b47412396 : SUCCESS in 7m 00s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/e271944c4eaf49a6b9b407af9f15d85d : SUCCESS in 8m 27s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/499d928d4bf54208a96d388a923bac88 : SUCCESS in 7m 08s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/57e27e08b11646d88d9fc6825419f0db : SUCCESS in 8m 29s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/7008e9e5b920497bb4ab211b8a997008 : SUCCESS in 6m 23s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/2ab83bde92e443a19d0bbebc91336ba8 : SUCCESS in 6m 36s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/d9292a712e3544abb4529df3c40a8a02 : SUCCESS in 7m 48s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/888c565648b649d69fdaf9bf2eee5527 : SUCCESS in 7m 03s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/5c125fb51a244db4be122cc1c9704c34 : SUCCESS in 6m 05s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/acc75dbc30594826ac28a6afb4ce5601 : SUCCESS in 7m 10s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/45da6da2dca347859fb345b1b2ce0a41 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"dca557e3cb891362635b58165e827c4fbbbe1565","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-06-08 20:10:06.000000000","message":"Patch Set 1:\n\nrecheck\n\nSame error in both test runs, but let\u0027s see if this is a fluke. We may have had a dependency update that is now causing a failure. I wasn\u0027t able to reproduce this failure locally with a fresh venv, so fingers crossed.","accounts_in_message":[],"_revision_number":1},{"id":"4f2f3debeaddf38f88aa8b9079497f1d31d96fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 22:06:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d97f3ac36bbd49d09cb20c9d4a6e9f0b : SUCCESS in 6m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e6423933c3649faad98ccdb53622c90 : SUCCESS in 1h 31m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f6e92a2f5d145558c9c9a1624a20183 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/308058879b87443eb05595fb4b0fab25 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d63ff616a06d4aaaa54193d7f1ccb06c : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b93dd7c43e3c4628bb2369fd045eccc3 : SUCCESS in 9m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b745eb2f849425fb577cac1e8204c08 : SUCCESS in 9m 11s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/180ed69201354f7abe015c156be903ed : SUCCESS in 8m 37s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/0ec18d219d4641acb8a0969ecb731154 : SUCCESS in 8m 34s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/49386fe828d3401684d66662a89614ec : SUCCESS in 7m 00s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/a6d66ccbbc464e8a89ff891d302afb9d : SUCCESS in 6m 55s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/5ac90aa1ad4d4506a71aad04a47a4623 : SUCCESS in 8m 47s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/1a08f086082c478cbc0400df47fe408d : SUCCESS in 8m 29s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/06b73f0c7b7043c99330065139d75802 : SUCCESS in 9m 58s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/ff952e5e9a3547469c329d28a5ed0be5 : SUCCESS in 9m 30s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ea74f507be2d49e5875e3c50a89f2018 : SUCCESS in 7m 57s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/60f99f31520c4a4aa00ae35fdc105aa9 : SUCCESS in 9m 12s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/05c047545a6548549c8c0614393a3ec1 : SUCCESS in 9m 09s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/4a76dc08df8242ada6923e4499462107 : SUCCESS in 6m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/d68d4e7021974a1b81734d593266c96a : SUCCESS in 7m 23s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/3caf75ec4dc94e1ca3feb496fb75cd1a : SUCCESS in 6m 21s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/89876e86f7d449f6adc442bf9a56aa75 : SUCCESS in 9m 56s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/9a839253f12549cd9e56be651177e37e : SUCCESS in 10m 03s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/cfd5bbd1a8074969a336bf51fff8893a : SUCCESS in 7m 17s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/472a6d198ea041d5826b01b77cdd26dd : SUCCESS in 8m 50s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/79d0df899de346f28ae69c0a42ff7f76 : SUCCESS in 9m 03s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/acce0d34e0364975bc71266f3b9c50fd : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"d976aa9c5937f249a15f690600b82d6a4a62c408","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 22:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ee59ca478e685fb7bcd26d91eb31362b3e6d7fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:36:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8494a3553f5445a580e13ef48b6a427f : SUCCESS in 4m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a007a4d5ea444829a6172c1d0b0dbdb7 : SUCCESS in 1h 23m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d5bb3ceef8845d3aee8c57c726cde36 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7dcaa82a604b4ddaac2545bd59ec34b7 : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc04ef024e4a42a5b9ab32bf330c70a2 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1e5980dbe44bf98b6fa2970815cae5 : SUCCESS in 7m 06s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/5c731df74e9f4133a8d06ced6176bf49 : SUCCESS in 8m 54s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/8c26658f88d44f8792d3bdddfcf3bb7e : SUCCESS in 7m 38s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/5d4016f6998a476087420e075bd95fde : SUCCESS in 9m 43s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/21e00ffb705f4a8e83096c7370b21a49 : SUCCESS in 7m 16s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/eaa9443c862742e58ad0ebee3fbbec55 : SUCCESS in 9m 31s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/4ace7e19b237489595502f14b3f1cbc3 : SUCCESS in 6m 52s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b3c6625a9ed840a7b021d5741e1370f9 : SUCCESS in 7m 54s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/70fb1638a8f44396bfc26aa87dd3fe6d : SUCCESS in 8m 03s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/0d85b986e9f94c0d8a5ec61b4348d3ee : SUCCESS in 9m 49s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/df0c781aebd84347b01cc3cc5f2cc218 : SUCCESS in 8m 53s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/3901b4dc8bc640b28b2c0147530f03bd : SUCCESS in 9m 26s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/a071567c5cdc458ba68d65e064ca52f4 : SUCCESS in 8m 12s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/5c843c1bc06d4f3f85a559cf33639382 : SUCCESS in 8m 05s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/4b44b44470a34dd8bd8688c83b02cd41 : SUCCESS in 9m 56s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/782291e5057a48c8898360d965fa33b8 : SUCCESS in 8m 11s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/94973196ad2748328008964e72a376fa : SUCCESS in 7m 16s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4bdf8d4971054a5ea58061ac94a63731 : SUCCESS in 7m 51s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/57f9405e5a3e470fb43460932a447998 : SUCCESS in 9m 19s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/b2c7702d03ad49fb83ca32fdae9ff138 : SUCCESS in 8m 14s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/cf923ef1549243339cb2ee6ae561f247 : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":1},{"id":"2fec2016ad35d9460650c0013630ba6a1bca411a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:36:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"055af27f435d6eceb52889295f2104e08d03290f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 23:37:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b0604f7d3fc4e549a0f45cc61e5f064 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d60388fd204f11ef9100e88e036e7711e43a681","revisions":{"5d60388fd204f11ef9100e88e036e7711e43a681":{"kind":"REWORK","_number":1,"created":"2020-05-26 12:07:05.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/67/730767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/67/730767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/730767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/730767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/67/730767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/67/730767/1"}}},"commit":{"parents":[{"commit":"94bf257392a759cf9dd00ff4a0971e7aef663b94","subject":"Merge \"Update master for stable/ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/94bf257392a759cf9dd00ff4a0971e7aef663b94"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 12:06:14.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-05-26 12:06:45.000000000","tz":120},"subject":"Add python 3.8 to classifiers","message":"Add python 3.8 to classifiers\n\nChange-Id: Ife3021fe203fdb6a9a6f2f51d45a64773f5fe89c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5d60388fd204f11ef9100e88e036e7711e43a681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5d60388fd204f11ef9100e88e036e7711e43a681"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
