)]}'
{"id":"openstack%2Ftooz~732961","triplet_id":"openstack%2Ftooz~master~I95809cf4035a6bc487b24f9e91aaa715921ad842","project":"openstack/tooz","branch":"master","topic":"drop_future_imports","hashtags":[],"change_id":"I95809cf4035a6bc487b24f9e91aaa715921ad842","subject":"Stop to use the __future__ module.","status":"MERGED","created":"2020-06-02 19:02:25.000000000","updated":"2020-06-03 10:20:07.000000000","submitted":"2020-06-03 10:17:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732961-1591179474435-c5fe9358","meta_rev_id":"ed7ac010887fe04f61987813b5f24d08842d4c44","_number":732961,"virtual_id_number":732961,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-03 10:17:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-06-03 08:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-06-03 08:59:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 08:59:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-03 10:17:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f25763420e6168ee58296af23197c5d61a3754b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-06-02 19:02:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b4ee8945afd02a977e9cbc03a6f8adb0a61851c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 20:40:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf8dfdc0d3cc4f80813240b7bc3a369d : SUCCESS in 1h 31m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f86c4ccdebf9447d9b89f83b4f5c3a8b : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dfa8e3b8e9641798de2f638b15c7b59 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b31de5bf37724ab4bbb40e1f3eaf2cef : SUCCESS in 8m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/867c4d181daf453c8f1e9f9fb34eff09 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d48c2ac03294abebb726795748a9547 : SUCCESS in 5m 27s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/71792904bf2a4466bf6c3426b3657892 : SUCCESS in 8m 39s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/05763627aaa8488f8056d4028248f336 : SUCCESS in 6m 27s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/efbf2c97d19c4c0d828b977986b7e997 : SUCCESS in 9m 07s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/5aee1821763747d68d6cd8be4a37ccd8 : SUCCESS in 8m 55s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/644d941a4ce549fbb18644c8ff540e2e : SUCCESS in 9m 08s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/a034e054bd81460dae59a31cf8089cfe : SUCCESS in 7m 28s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/141e682163944c3ba11a8a94c97d3ecb : SUCCESS in 8m 26s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/032d668b76df41f2a1a5872b0bb5eefa : SUCCESS in 8m 04s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/afb3c62d72014939a674d1abb4341188 : SUCCESS in 9m 05s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/ef9105274aaa4c05be653757e4d82c11 : SUCCESS in 7m 24s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/0f548932cb404237b7ac970349509fab : SUCCESS in 7m 52s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/1c086d9885c94d33a63e00510bef5e15 : SUCCESS in 8m 46s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/bb2d1409c62145bfae83a2d691545c75 : SUCCESS in 8m 11s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/6227f946105840208aa765c107c81110 : SUCCESS in 8m 03s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/efba54ae01cf464889bda8803ae1db62 : SUCCESS in 8m 37s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/10d62c08ac1a461a8871f18167eb0a8b : SUCCESS in 6m 13s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/e689f9bf177c49c799647c0b67af3327 : SUCCESS in 8m 52s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/b998d344021d406e9a233ec753fce311 : SUCCESS in 7m 30s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/ec06e420726549988951e4aa1e243708 : SUCCESS in 6m 09s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/507f3017adc143ec881caaf574544d96 : SUCCESS in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6f8095b5f36145116ce22cb8188a5235b1031e4e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-03 08:59:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"582ef72a68bffc198a014cefa8fe2e0a4b5b9e24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 09:00:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df2f9af95f0065c034c5dc5fb80ed60fff9e5ca8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 10:17:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ede114a0280447d9aaf662644bbb9069 : SUCCESS in 1h 17m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2894a9f836e4c19a41e2140561bd85c : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc05b1438c7f459388ab16c7b5b5c99b : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f618fc2771e413e95b24608c015aad1 : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473fc7fe7baf4c18aa2cc1b3c3dffafa : SUCCESS in 5m 48s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/4723a4b3110f4c5689f45f7f460a109b : SUCCESS in 7m 12s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/fda15f3139a34cc891260d4e23bd172f : SUCCESS in 5m 18s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/902ad02c595147608ebf8ad13b8768b0 : SUCCESS in 7m 21s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/4a1c47a3b85b4ddd93904dd29590fc47 : SUCCESS in 5m 15s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7812f52c53b249d1851cac87801cffa6 : SUCCESS in 7m 27s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/147b909cfa2748d7af2cc91373907546 : SUCCESS in 5m 09s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/16e3a5f7f71a4f0cafebb2163b3631ea : SUCCESS in 5m 52s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/0e16d23ceaeb431d8c56128555dddf0a : SUCCESS in 6m 42s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/5e8402b7eb7f4d239707fda2f827bf5b : SUCCESS in 6m 14s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3ebf554ef72345b0a46339e299b83744 : SUCCESS in 6m 47s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/054ab1b9b619473d9ac63321f139315a : SUCCESS in 5m 37s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/1c7233ee7ccf4a8ba109930730f3965e : SUCCESS in 7m 54s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/5ff7675bcf314db3b222a728ef78436e : SUCCESS in 5m 20s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/ecd59f4f4a8c4e818bf96c2600b92007 : SUCCESS in 9m 01s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/9d714c6a68f841bcbbba17ac4891d88e : SUCCESS in 5m 39s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/1fbb47b271224f5c89e289bb2bbae8d0 : SUCCESS in 6m 55s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/1a6a5ae0cc7a4787a4e7c5b0166df0b4 : SUCCESS in 8m 03s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/4771027a15a74f0c8ff4a8c0017e5e65 : SUCCESS in 6m 23s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/28b3740f18f04e7b9ffec63de2bca8a8 : SUCCESS in 6m 02s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/66b2e981fc4a48d2943d49f55affd22a : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"314bdaa38ebc08b14107e6c0f4e01d5a32643b40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 10:17:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d3c9d6cb5f359229bd053e8117bc83ea4836b24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-03 10:20:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e108249c456d4ab09c5f4e33e7323c38 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c4f23a01923563d43b2ddf5048340cc386e2c33","revisions":{"9c4f23a01923563d43b2ddf5048340cc386e2c33":{"kind":"REWORK","_number":1,"created":"2020-06-02 19:02:25.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/61/732961/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/732961/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/732961/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/732961/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/732961/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/732961/1"}}},"commit":{"parents":[{"commit":"94bf257392a759cf9dd00ff4a0971e7aef663b94","subject":"Merge \"Update master for stable/ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/94bf257392a759cf9dd00ff4a0971e7aef663b94"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 19:02:18.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-06-02 19:02:18.000000000","tz":120},"subject":"Stop to use the __future__ module.","message":"Stop to use the __future__ module.\n\nThe __future__ module [1] was used in this context to ensure compatibility\nbetween python 2 and python 3.\n\nWe previously dropped the support of python 2.7 [2] and now we only support\npython 3 so we don\u0027t need to continue to use this module and the imports\nlisted below.\n\nImports commonly used and their related PEPs:\n- `division` is related to PEP 238 [3]\n- `print_function` is related to PEP 3105 [4]\n- `unicode_literals` is related to PEP 3112 [5]\n- `with_statement` is related to PEP 343 [6]\n- `absolute_import` is related to PEP 328 [7]\n\n[1] https://docs.python.org/3/library/__future__.html\n[2] https://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n[3] https://www.python.org/dev/peps/pep-0238\n[4] https://www.python.org/dev/peps/pep-3105\n[5] https://www.python.org/dev/peps/pep-3112\n[6] https://www.python.org/dev/peps/pep-0343\n[7] https://www.python.org/dev/peps/pep-0328\n\nChange-Id: I95809cf4035a6bc487b24f9e91aaa715921ad842\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9c4f23a01923563d43b2ddf5048340cc386e2c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9c4f23a01923563d43b2ddf5048340cc386e2c33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
