)]}'
{"id":"openstack%2Ftooz~739725","triplet_id":"openstack%2Ftooz~master~Ia3d9035abb645e82083fe9a796d7d7df7cf4d7a7","project":"openstack/tooz","branch":"master","topic":"fips","hashtags":[],"change_id":"Ia3d9035abb645e82083fe9a796d7d7df7cf4d7a7","subject":"hashring: allow choosing hash function","status":"MERGED","created":"2020-07-07 12:32:52.000000000","updated":"2020-07-21 12:14:43.000000000","submitted":"2020-07-21 12:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739725-1595333531214-26181de4","meta_rev_id":"924fa9f95cecfc1b4ecbd879023a4c5762da8674","_number":739725,"virtual_id_number":739725,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-21 12:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"date":"2020-07-21 09:21:40.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-07-20 10:51:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2020-07-15 10:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-07-20 10:51:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"date":"2020-07-20 10:14:17.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-07-20 10:14:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-08 16:06:18.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2020-07-15 10:08:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-20 10:14:17.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-20 10:14:23.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2020-07-20 10:51:37.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-07-21 12:12:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a134f70e39f608670d8de49ebe8185995aa45704","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-07 12:32:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a9d1cd9a4aa4c61f1589694ad5bd33472048284","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 14:17:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aebe160aad6f40bbac02f6d6958ea133 : SUCCESS in 1h 38m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98fb3d5d090f49a89cdbf243cde96ff2 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5025e071e65486599c32a510aebdd89 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edc144a78b094cf8be783c59d82577ee : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39293c72aa884c41ac4efe4733795e81 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb49833de469489cb2be3ed431a7f3af : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87adcc2b413e480abcd1603856b4ebdb : SUCCESS in 7m 23s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/a7d2df1b03b54b4eac8f941990eea665 : SUCCESS in 10m 02s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/160f1108f17043508e417a6a058981d9 : SUCCESS in 9m 21s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/6b4ff91b287f46008907d5d40c8966bb : SUCCESS in 6m 11s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/55787277adec4fb89d5aa2e3f5969bd2 : SUCCESS in 7m 08s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4891b5ca10744ffeb3606c02f395393a : SUCCESS in 6m 32s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/4f0c1bc6389846b4b0bf2ea589a8be4d : SUCCESS in 7m 15s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/61bd705aacb641c3aba2a749c26f31ea : SUCCESS in 6m 37s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/394b68724e8f4691a9ce7b7c20877221 : SUCCESS in 8m 32s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/7d9e0677e6b941b8922b292750986ce4 : SUCCESS in 5m 40s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/8569e442feda40abb33d170274a4d82a : SUCCESS in 5m 45s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/fd855e9131ee4045a93b1e377f385acb : SUCCESS in 6m 02s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/6e7a2bd16cc1443593079f7ce0230450 : SUCCESS in 5m 46s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/cd786a86618c4f039d236d99cacb9396 : SUCCESS in 5m 57s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5c56f9b69f47447eb9692ad1afec07a9 : SUCCESS in 7m 27s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/30e444906af6451b981b07e2d837bee2 : SUCCESS in 5m 43s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/ff98d99340094b409a09fc0ac12a9c78 : SUCCESS in 8m 12s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/ad27b2a4c8504cb6a93dccc1b997aaab : SUCCESS in 9m 05s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/1c7e39e67a4c4cf1a6c532834fb01089 : SUCCESS in 5m 11s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/9d44ffb53818471b8c90c261b412ce30 : SUCCESS in 7m 21s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/50f4d592a9584d25ba9464c32b39279f : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"142baeba1057c41aa585300762f86e149a0e905c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 09:46:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"57286b1e26d153b392cac928da25ddce1497950f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-08 09:47:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbd05b824712e57739e811934008076fc737c523","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 09:48:01.000000000","message":"Patch Set 1: Code-Review+1\n\n+1 for now and for the rationale of these changes.","accounts_in_message":[],"_revision_number":1},{"id":"5a7450f10a6422495a89331483110d55617fa12a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-08 09:51:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b866e3fb7858fc9833bbd1a7d95c06ffdb5b3697","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-08 10:04:05.000000000","message":"Patch Set 2: Code-Review+2\n\nAlright! Thanks Dimitry, nice changes!","accounts_in_message":[],"_revision_number":2},{"id":"89dbec4dffebb9639c5d65a40f9b18c1d97ec068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 11:20:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72d417c0db0b4e7c8a54dcb14f3b430b : SUCCESS in 1h 21m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/200891d89fb2488d8ba5e4dd236ef60b : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab769d77e74c43eabe9dff57eeff8f62 : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5164a5e99984d548aef82d99e44beee : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec005a8d9105437db2d7489088763218 : SUCCESS in 7m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd736a239bb4438e9b1451cab671240e : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14c212c1cdf24101a4a72fedf2674bb6 : SUCCESS in 9m 24s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/dc9558165a884d3bb5e754a8386964c0 : SUCCESS in 6m 20s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/93cb01d403074fca8a04995d19d23f6d : SUCCESS in 7m 53s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/59b2656b11b443869e3ef27e1f1c6588 : SUCCESS in 7m 33s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/bed0a38e32c547d48ce9dc5f025c0b4e : SUCCESS in 8m 59s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d97591cf37ac42419ac471cbd1ecdaf6 : SUCCESS in 8m 17s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/96db11e0de82483889062e8717af3800 : SUCCESS in 9m 03s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/85d9001793aa4775a47edec194e21614 : SUCCESS in 8m 20s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/6f6efed690c24d89899679e11faa2be3 : SUCCESS in 7m 09s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/c2a3d930e6b244d8856c6d853ee89080 : SUCCESS in 7m 56s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a063c9fd5e384035bf0fd4a4addf0171 : SUCCESS in 8m 50s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b9a986dc4b154412a0e44a44d955f44c : SUCCESS in 9m 37s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/7aa15273776b455b9700c81c91c1fca6 : SUCCESS in 8m 35s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/5bfc246416ac4cacaf3e02c3aa144044 : SUCCESS in 7m 23s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/957550b4f8bd473fb0dc973c885fb70a : SUCCESS in 7m 08s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/ef46b9630c7f43f48ae8dc3542596fa1 : SUCCESS in 10m 15s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/0ea8b3bc6d6f49279981d5d1fbd6d1f4 : SUCCESS in 8m 29s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c144978aee744bc2ad442cc847ff25dc : SUCCESS in 6m 08s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/189b8c55603f4454af52b6ee46975540 : SUCCESS in 9m 34s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/e55b7a081da144c3a3c242f73457005e : SUCCESS in 6m 53s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/9d97b12d89f44628a0fbfb165ecafae6 : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":2},{"id":"0a3b3eb2f08e6880a3beb4fa5c9d033356701879","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2020-07-08 16:06:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cbb56b8341e417912e528e10f7e63944b4076e58","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-12 12:56:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4273eb70fbf19272e27bdfc0eb621430e1b5cc40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-12 14:14:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0ca8ad2defc42c784ddd302e050725a : SUCCESS in 1h 12m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9266ce2aa9347679d100f3fe5945be2 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f98d7973b6d24dbea8594e9d5f4610f8 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0779ffac08ce404b9b4db8c8f10ae292 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb6f55f3429b447798f370ae50e83544 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0df1cf370646dab265185ace2b732e : SUCCESS in 4m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ccd4913010349e3a43090fa9f02d51a : SUCCESS in 7m 35s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/1884494a014246a8bded64a45687ff04 : SUCCESS in 5m 16s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/8d1e95cd594f405ba89dc8944aae141b : SUCCESS in 5m 16s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/b6a97c2cf6de41eabe5da0d26445c774 : SUCCESS in 7m 21s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/db81162194a547fbadb9a29198a01e1f : SUCCESS in 4m 50s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/54331bd29fda4e339280327ec9604944 : SUCCESS in 4m 55s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/67c944f2d4074d82b5712410f9309870 : SUCCESS in 8m 01s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b1da8ac77bef407595bfb0a0a4de6906 : SUCCESS in 4m 43s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/98a83bf6e58043f08155d5b542168b8d : SUCCESS in 8m 09s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/467c2df680234745bab7fbbb4fbc0e4b : SUCCESS in 5m 43s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/122438c8d2f34fb881ad8665830695cd : SUCCESS in 8m 06s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/c04316f68e8f4d9a8c7ca4bc16e2814e : SUCCESS in 5m 06s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/1f1d2624d7d344f2bbf461239001f506 : SUCCESS in 7m 06s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/f6e92f3d34e848b684087c5bbfcdbae8 : SUCCESS in 4m 49s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/c05ea7982cc046a78c759f39a6903726 : SUCCESS in 4m 42s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a5ecd616db56457eb63eab7f1327dbae : SUCCESS in 7m 26s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/901ece1c3ddb4db0827b3653d6fe787b : SUCCESS in 4m 50s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c2692af54ec44f60b536562c4dd2fed2 : SUCCESS in 7m 26s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/1c8e2463e89a4811902e2b51ff67cd29 : SUCCESS in 6m 52s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/749fe7bde2604042aa522fce4d0f3772 : SUCCESS in 5m 23s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/6c1642b401e74ea2ba6635974532b940 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":3},{"id":"217f2e8a503856cb8d7a669b041ca13460e7403e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-15 10:08:50.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6aa1821c1fbd327d0145bd245660f7481fefcc95","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-20 10:15:02.000000000","message":"Patch Set 3:\n\nSorry for mass-adding reviewers, this issue makes ironic completely broken in FIPS mode.","accounts_in_message":[],"_revision_number":3},{"id":"b9aa3622f0ea6d8e064f27f0a7f636f1c713e24f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-20 10:51:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"62ba34d73cc3c7a70f366caaecc61da9b1611278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 10:51:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e4f0415ddc368facc3a21ef03e0d79663aeeddfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 11:50:56.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fe5a709c58f549c192b34f3ddff0d803 : SUCCESS in 57m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ff1b7c88d84b1ea408b2f430dd1968 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bc4adbcd4e7485e80b6fd8fb5dbb1be : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/552f39b466ac4fc3be8ee6b1de7cc883 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5d2b7bcdcc4fc6a57aa76214cbf877 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c4d7cc484014758aa459e120872daa2 : SUCCESS in 4m 01s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/7e26878d55254de79d76a97987f8b7c0 : SUCCESS in 5m 01s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/7bc205af8ffe47e088ac9d12a63fa311 : SUCCESS in 7m 17s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/5ff16ed846664f9cb295470c2083ac1c : SUCCESS in 5m 29s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/79fef2c43c8d4defa9004c5c123edcc6 : SUCCESS in 5m 03s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/0f62559b44c2484aa5483f22183cf863 : SUCCESS in 6m 17s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/793184cd9da84c9daa9571521242767a : SUCCESS in 5m 15s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/0e930a62ea93452ab8040f7c2b96db47 : SUCCESS in 5m 09s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9f904d85540e4801b7a8df5a18eb5a30 : SUCCESS in 4m 52s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d3fb9b2293d949739965486d255ca681 : SUCCESS in 9m 02s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/2c4ce8342ed0455a89fd70196a0fb613 : SUCCESS in 8m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/43c4d48e652d4ac1b98e1b956560eeeb : SUCCESS in 5m 01s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b14ebfdc904049b79671ef63991ab7f8 : SUCCESS in 4m 14s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e9f7481a1c7e4922ab545d6f383beaab : SUCCESS in 8m 02s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/61e97eb541a84a69a96fca60ed38f1ca : SUCCESS in 4m 46s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/2490487ef7f241a39881a81079bb48f3 : SUCCESS in 6m 55s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/10c6d9c032de49a095c2cad9236fe9b9 : SUCCESS in 5m 49s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/faab86f568f04ba181cd1aea13b70727 : FAILURE in 5m 12s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/4752f93ba1d44b97b8fbdb5b0830e104 : FAILURE in 6m 07s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/ee42c2b3836a4d979387b02f3f2449fd : SUCCESS in 5m 41s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/ac4336b43fd543098d4118c133408a03 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"7e2a0b1995d545afa70defea7dcb68e71cd94aff","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-20 11:52:47.000000000","message":"Patch Set 3:\n\nrecheck\n\nmight be a breakage in the mysql library","accounts_in_message":[],"_revision_number":3},{"id":"4197ce437c138ce3e65c64b59a12821bab01a36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:28:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8db60873a7ce4318b13838707ee243f0 : SUCCESS in 1h 27m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ded1cbc4c744855a8033d9e4a271eb1 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d538a53117f24b6b91a3f024fbbf5378 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36509d8e299b4fb18915fea670a05fa6 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cdc9d5773a6473392a43d8ca9fa9907 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ec1d459ad34d20803ef8541c83510d : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27607113d6114e3ebd34af9e43e4b058 : SUCCESS in 6m 10s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/8f5e40883d99425499e1cec8dee43f26 : SUCCESS in 5m 49s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/8a69a6681fe847e9ac0463cde3dd5430 : SUCCESS in 6m 22s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/b9bc69c2550b4cc58fdeb32f47678d33 : SUCCESS in 5m 18s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/ae9170ddc08a42d7b4d03c624310e251 : SUCCESS in 5m 44s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7b0dc6afea074a63b27156f824932723 : SUCCESS in 5m 04s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/acab8ab694b64d729ae61c92f32c8332 : SUCCESS in 6m 17s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/07ec0118aa7b4ee5ae7efa072e1c3a04 : SUCCESS in 5m 47s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/eb57df0ac35b4261b23510369f53919a : SUCCESS in 5m 42s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/05fc1335aa814f868fb4630cbe6f2738 : SUCCESS in 5m 44s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/86eb49080f8a4041840c62b2660fc23f : SUCCESS in 6m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/54aa531f32974f2f808ac32dc9824192 : SUCCESS in 6m 22s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/9cbc8781691c4a37a2877b20a2cb9b90 : SUCCESS in 5m 54s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/9afd1eaf6ede42d3af550199143f15c8 : SUCCESS in 6m 24s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/c60a8f3dd734475ea2017a9af20b62e5 : SUCCESS in 6m 22s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c855e7d0164043f3861b8a1ecf351ce5 : SUCCESS in 7m 32s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/6f5d0909cc094e04beb061f946e17a98 : SUCCESS in 5m 23s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0f310fd42bd6475ab1328ca056c4ada1 : FAILURE in 5m 19s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/de86bcd738ed4d4b8bcfdb7a6a22de9a : FAILURE in 5m 58s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8e60d04c804a448b9f666fce6703ce9b : SUCCESS in 6m 07s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/ac9eeba81f3543eba3506543b20c3fe2 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"b0875d4772e66c603efdfbe80828f98b805e1215","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-21 09:21:40.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"838586d80279faa1f32fe3eea240822f491b5bc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 10:50:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8f3071c170044efa5729fdda7c63d5a : SUCCESS in 1h 19m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bedd61bccc9d4ce0b4fbac1aa3a75bd7 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fa3470acb19463a8d0a3074c1f9c56a : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a563cfba86af4388ba0d70824b8be2d8 : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0e08e69a6f14c50ab7c2d1ba3f22e93 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/659d8e7be4684be4bfedc5978e7c4fb4 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30de41b82495455bbc0b61c227b3cb17 : SUCCESS in 5m 24s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/cd8f2f36ab8e4caeb428a2f2dfeca479 : SUCCESS in 7m 20s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/71661b0a0177479bb675e82454fdef59 : SUCCESS in 5m 37s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/0c590f34e61f4d5290ef4a1802b1438a : SUCCESS in 7m 28s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/95f1c3703c7f4f3f88e13e37341b0e1c : SUCCESS in 4m 34s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/6ddfdb590c7a4121929c326287f7c565 : SUCCESS in 6m 45s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/ddb756171a2f48809ab1aa918cc9a399 : SUCCESS in 5m 41s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/3dfec60e79474cb9823b8114de0548f8 : SUCCESS in 6m 01s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/48fc4d05ea87465cb6839b5d4e4bfea0 : SUCCESS in 6m 46s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/9c5cad2ff3b644d3aacc262bf0e81a6a : SUCCESS in 5m 46s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/587c170a42d64b6bb3699c920128752f : SUCCESS in 5m 19s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/be44d71c63924fd780dbfb6464f6324b : SUCCESS in 6m 18s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/19fefb7b2d3243a588a25df15fe9d642 : SUCCESS in 6m 31s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/6b16fee98e884c6180a9c653838c9764 : SUCCESS in 5m 18s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5f409ce3124748d5ad4d91ee5f8070fc : SUCCESS in 6m 42s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/eb0b9c8cca534e15a3f860f741ca49e7 : SUCCESS in 6m 19s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/865ab6f7f0fa4e3985cda54c76d9709e : SUCCESS in 6m 14s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/650aaa5363784131a5019fb38bb59e5c : SUCCESS in 5m 56s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a142d60c4f2b4b808b4848be554b811f : SUCCESS in 6m 28s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/9a94b583e0314e5797510b219d03808f : SUCCESS in 4m 52s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/2222e682caa842ccb0bb6dc4f56cc049 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"d6bc78d338d48a575a4554009efdd5bdd3352f01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 10:50:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"26ab68621eb636a9b553ab08a99bb45b936e52d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 12:12:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7dcec1dc80b4cd4bc23995d3c5b7e8d : SUCCESS in 1h 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa801c49ed9849aba7d4e807d12fb7f9 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca976e063a2c4c0eb8f8b2f7832324af : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca5c8808d75494fb356dc95bfb480fd : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d6cfc236c11437b85e87a5b41d8125b : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42bf191417294b7493694ac6ce239d09 : SUCCESS in 5m 37s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/ccaed294b3e2494da443613db10a974b : SUCCESS in 5m 50s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/c00dcd5e38294e6babb129c3a041d407 : SUCCESS in 5m 03s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/968716d55d9943bd83e4f8d164d238ba : SUCCESS in 6m 12s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/343a29cf8edd47df9f405b10d7d84a9d : SUCCESS in 5m 44s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/59dfc228e6dc44fc84d24aa09df40703 : SUCCESS in 6m 08s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/ee160f560d06410f8c3eb02e38c500f3 : SUCCESS in 5m 49s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b86a1893a6374623b904b413c4a4eaf4 : SUCCESS in 6m 57s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/476684fbff624ee6b922c7498269534d : SUCCESS in 5m 00s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/100c7ad229b948309e13e13afa4cb1c6 : SUCCESS in 5m 52s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/6df1ddcd9719460a8255b6c4c15755fa : SUCCESS in 5m 58s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/204309c795444c0c975e93b7fa5ca0fc : SUCCESS in 6m 35s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/40f5e7f8c58343b1ac79ab3170a9f1fd : SUCCESS in 5m 59s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/53bbd52e39a347fd9b77ebeae779398d : SUCCESS in 8m 15s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/30ada001187a4fc49622a1529b142e18 : SUCCESS in 5m 24s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/562f37e3a000493ba08de85ebec97c1f : SUCCESS in 4m 53s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/568a514e4c584ffaa81102c902d73ed9 : SUCCESS in 6m 02s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/db7c36b259514df98ed770ebfc707285 : SUCCESS in 5m 33s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/ca87180c6a644516b24c87216f18534d : SUCCESS in 5m 36s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/5eca396a7b714d5e825695cfc489f6d5 : SUCCESS in 6m 54s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/17f1bca5c8f24dccb952b24975cf6a94 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"6d82d89a585bf2f89accc766ca474218db8b6614","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 12:12:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2a3df73879880ebe29c8884fe8964de731c21807","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 12:14:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630c1999cc164004bd5c293fa104c26e : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa97a3d45bc04155b75d2cb663c0a337 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f5d32480d9cd76e33f3a88e00441bc39b2a55ff1","revisions":{"2e061c96c94f78b901d04e097893f727426c93ac":{"kind":"REWORK","_number":1,"created":"2020-07-07 12:32:52.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/25/739725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/25/739725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/25/739725/1"}}},"commit":{"parents":[{"commit":"80f379bfa2af9c9a80ce369beb275361ed78963c","subject":"Implements Group API for the Consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/80f379bfa2af9c9a80ce369beb275361ed78963c"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-07 12:21:24.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-07 12:31:18.000000000","tz":120},"subject":"hashring: allow choosing hash function","message":"hashring: allow choosing hash function\n\nIn FIPS mode it\u0027s not possible to use md5 for any purposes.\n\nChange-Id: Ia3d9035abb645e82083fe9a796d7d7df7cf4d7a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e061c96c94f78b901d04e097893f727426c93ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2e061c96c94f78b901d04e097893f727426c93ac"}]},"branch":"refs/heads/master"},"613a80c9d729cab812c0b347f98d0940a947446c":{"kind":"REWORK","_number":2,"created":"2020-07-08 09:51:55.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/25/739725/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/25/739725/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/25/739725/2"}}},"commit":{"parents":[{"commit":"80f379bfa2af9c9a80ce369beb275361ed78963c","subject":"Implements Group API for the Consul driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/80f379bfa2af9c9a80ce369beb275361ed78963c"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-07 12:21:24.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-08 09:51:07.000000000","tz":120},"subject":"hashring: allow choosing hash function","message":"hashring: allow choosing hash function\n\nIn FIPS mode it\u0027s not possible to use md5 for any purposes.\n\nChange-Id: Ia3d9035abb645e82083fe9a796d7d7df7cf4d7a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/613a80c9d729cab812c0b347f98d0940a947446c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/613a80c9d729cab812c0b347f98d0940a947446c"}]},"branch":"refs/heads/master"},"f5d32480d9cd76e33f3a88e00441bc39b2a55ff1":{"kind":"REWORK","_number":3,"created":"2020-07-12 12:56:59.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/25/739725/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/25/739725/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/25/739725/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/25/739725/3"}}},"commit":{"parents":[{"commit":"5ac72fcb3d2071d8fd4c382a48efe5b9dbd4dec9","subject":"Remove six library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5ac72fcb3d2071d8fd4c382a48efe5b9dbd4dec9"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-07 12:21:24.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-12 12:56:17.000000000","tz":120},"subject":"hashring: allow choosing hash function","message":"hashring: allow choosing hash function\n\nIn FIPS mode it\u0027s not possible to use md5 for any purposes.\n\nChange-Id: Ia3d9035abb645e82083fe9a796d7d7df7cf4d7a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f5d32480d9cd76e33f3a88e00441bc39b2a55ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f5d32480d9cd76e33f3a88e00441bc39b2a55ff1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
