)]}'
{"id":"openstack%2Ftooz~743770","triplet_id":"openstack%2Ftooz~master~I22ce4212381511e2b2631a97057d6da938024b99","project":"openstack/tooz","branch":"master","hashtags":[],"change_id":"I22ce4212381511e2b2631a97057d6da938024b99","subject":"Use safe_decode for decoding in zookeeper driver","status":"MERGED","created":"2020-07-29 14:10:39.000000000","updated":"2020-07-29 18:41:39.000000000","submitted":"2020-07-29 18:40:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743770-1596048010614-4e4c0980","meta_rev_id":"683ca3c7986203ad2197c1332703ae7b46659c1e","_number":743770,"virtual_id_number":743770,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-29 18:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2020-07-29 16:06:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-07-29 15:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-07-29 16:06:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-29 15:49:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-07-29 16:06:23.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-07-29 18:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0073caad136dc6b742d2df2a1cbe8c07ec7b2de","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-29 14:10:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e5e8e9c46e3bfdf6fdc3c78f657b8bd92392670","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 14:49:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8de8eade76dd954e886f1265d35a6c8363e5ff3d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 14:50:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7bde63694ba3a1582185d8d3d2e6948a24dca5d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-29 15:09:26.000000000","message":"Patch Set 1:\n\nHi, \n\nYes, I face them on gnocchi:\nhttp://paste.openstack.org/show/796436/\n\nIn the meanwhile this works for the ceilometer, so I think some are binary objects, and some are not - all depends on the project I guess...\n\nI\u0027m not sure about removing decode at all, since it decodes to ascii, while on py3 it\u0027s supposed to be utf-8 by default. I\u0027m really not sure, but that might have some influence.\nAlso I\u0027ve found, that for redis there\u0027s still decode used.","accounts_in_message":[],"_revision_number":1},{"id":"b8d1c70f2011e40332f6201c4e2675240acaa0e3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-29 15:34:33.000000000","message":"Patch Set 1:\n\nEventually, in [1] we don\u0027t fail, but return text at once if it\u0027s text. If it\u0027s not we go further and do decode, so we handle both scenarios which is safer.","accounts_in_message":[],"_revision_number":1},{"id":"ccd839295f09379e340c840ddfbe866f4efecbcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 15:39:58.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2b0267e316542daa1df2035e6e468b0 : SUCCESS in 1h 20m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/206b2a12f200428e998293247cf1eca2 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351c511c230a426f80a1e6c042dc2240 : FAILURE in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0a11bc06a494e1dadf9b8f1f39cdb83 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a625e3c4c9f4abfae9595fb2f59cacd : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e37988ff8041878ed3f22a3baba358 : SUCCESS in 5m 20s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/f8a990c6cfa74e2f81f129d098fe103e : SUCCESS in 5m 30s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/4f9b947adf4a4616824b82db793ddb7a : SUCCESS in 6m 57s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/c8f349f646dd46c6ad877ccbc76024c8 : SUCCESS in 6m 18s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/92818051e5544264ae92ec2144eac488 : SUCCESS in 4m 40s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/3f25cfec2c964237a237d2c5d427ca85 : SUCCESS in 6m 56s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/2de318646af44934ae3fda35055588bf : SUCCESS in 4m 15s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9046a16457f7427cb121b451cd610fc8 : SUCCESS in 5m 19s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/adda3084598a46cbbbac4ac0d3abe31a : SUCCESS in 5m 40s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/fd1eaf4eb4884630b80e9b16818712f0 : SUCCESS in 4m 53s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/7d790cfffc9a442caa76eff4951ed0ae : SUCCESS in 5m 56s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/755a3c6648224f58b353e7a2ce530ecf : SUCCESS in 5m 17s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/a0206a6454c749d3a0993e6995cd5aaa : SUCCESS in 4m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/b169c1a3ba2448da8a12a83b8a4d6729 : SUCCESS in 5m 43s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5b4159354e9a4042b0ad7478980db198 : SUCCESS in 6m 15s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/dca4fe6b8be64b76a12bf4df9081b025 : SUCCESS in 5m 53s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/dbbb5e01419043e381a7a9481577e4f9 : SUCCESS in 4m 07s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4385bd20a7284c04b3f145c66889e48e : SUCCESS in 5m 27s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a44cf292ece34f5796572dd812b0e176 : SUCCESS in 4m 27s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/4e7fd26bd469473b9ab437c54d97f821 : SUCCESS in 5m 56s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/89070d3c937f42bcbe24fe0fc1206587 : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6eda9e25d40af42831054a48abc075db1cdbde93","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-29 15:43:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3fa2e53ca4806caf461b05f58b0b523786c57b24","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-07-29 15:49:28.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e Eventually, in [1] we don\u0027t fail, but return text at once if it\u0027s\n \u003e text. If it\u0027s not we go further and do decode, so we handle both\n \u003e scenarios which is safer.\n\nYep, I don\u0027t think it can hurt, so let\u0027s continue with these changes and oslo.utils\u0027 encodutils.","accounts_in_message":[],"_revision_number":2},{"id":"d396bcc6069afb03dd041ec041a1446973f9b3a8","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-07-29 16:06:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3865cf3e9d881ea196e94c045fff36423d317040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:02:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4007b728aed3440a8b658401c01176be : SUCCESS in 1h 15m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5eb04f82d7fc42ca9a1a06e693907b33 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8971958a7924b099b8105b27b3cc71c : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14a2fdfa457b4655bc489732d64b24c4 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c75f54bc5749ec8e0bc4d4cf1ef2d6 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017d89f99add457e8f56043f5705d5d0 : SUCCESS in 5m 20s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/99b7166bbd4e45b590b7aa051992c64e : SUCCESS in 6m 13s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/8620f24384de4870ad5cb152f1375bd7 : SUCCESS in 5m 05s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/d9aa10c6ee884a21b089042cdf4c990d : SUCCESS in 5m 21s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/aa862a0b9ea74e45b4854cf8259e2f31 : SUCCESS in 6m 43s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/3f73061a306a4777a91563c6f70ee364 : SUCCESS in 5m 26s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/6c2be9cdf745477e83690a1b8923c6bf : SUCCESS in 4m 26s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/aceb4aa3950145ed8b0733807892023a : SUCCESS in 7m 33s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/0cccfd55e2b84398943bcd89ce802bd1 : SUCCESS in 5m 30s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/0b551fd7fc81423c881d9a45d86f5955 : SUCCESS in 7m 04s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/9f838fcf14f14fc0a73c3c2416957a3c : SUCCESS in 6m 14s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/a498d145432142fabb119d4be1cdd141 : SUCCESS in 5m 15s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/75179b1dfcce410d8ad0800d14a9b7fa : SUCCESS in 4m 30s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/c8a925f16f5641f9bb938806460e05f7 : SUCCESS in 5m 50s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/c0970eca539142229ec4d42fd1379b51 : SUCCESS in 5m 03s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/9da4cf5da2b4467c8171c94d88b6d80f : SUCCESS in 5m 13s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/c3a2365144c840f6beec2dbf99b43c3c : SUCCESS in 6m 34s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bcdc0cdd789a49c1ad59cefc6bb24f58 : SUCCESS in 5m 37s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/3ad87de7d11e40fd88e68742c39eb6f9 : SUCCESS in 6m 39s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/11066cacc0db4258a1e0ef2d05e0c270 : SUCCESS in 5m 07s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/b04af68ec719407faa7385a747c207f6 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ea85e2c36d4170eb5ccd18da44201e391fcc8f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 17:03:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aabecbec98f6ccd6985b2ca27f2a1901a662aa04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 18:40:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae7a98f29a964d9cb4f3210098c774de : SUCCESS in 1h 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f56c35988a4f1988a4dd0849d3d0d0 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f4a1a52e4045f8b3deb258ce86db49 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1705089a39d24cbe9b301f240fcec6f4 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb9729385b544bdbb7e5fa41adce719b : SUCCESS in 5m 18s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/00c209099e7443299a51efe792863a4f : SUCCESS in 5m 41s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/6089e52dedd543b8b83c34202d822cc5 : SUCCESS in 5m 25s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/d33cbc20bfff42f58ebe70626fc3e178 : SUCCESS in 7m 00s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/7d72738aba2e40389af8faad9090bfa1 : SUCCESS in 5m 30s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/70db05bf911644a8bdf65b9da9be7cc5 : SUCCESS in 5m 38s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8222e100080e4ee086a76c84c932efa8 : SUCCESS in 4m 50s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/e4d7d1b9a8114fed94a90c93e708f891 : SUCCESS in 5m 20s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/3abe0a5e291f45c6a2ac76e555efd393 : SUCCESS in 4m 57s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/196b283058b04de6a7a0c1bdf8dd40ac : SUCCESS in 6m 16s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/495a226cf14f4ef0a065186ed40c7d03 : SUCCESS in 5m 30s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/df69aeac5c7b42e8ae361bf9988fb85c : SUCCESS in 5m 36s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/1d3ac4e77cbf40cfb8a6b10283ce713b : SUCCESS in 5m 37s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/8410e0ccdfdb4659afe75731b38548d0 : SUCCESS in 6m 32s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/3b4d6c2e7ecd44ffa02a2de46173a07d : SUCCESS in 4m 44s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/0a8abcb67398453abbca9cc06fb094f6 : SUCCESS in 5m 50s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/80ed502abd454a3580620740363adc87 : SUCCESS in 4m 23s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/244b886ed4ee42b6b4a10f1291f2db5d : SUCCESS in 5m 47s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/9f59d2a32e354ece98510399b9a12ac7 : SUCCESS in 4m 36s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/09fafba1893c45c08df95bd27132d851 : SUCCESS in 5m 39s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/4bbec13bc3f84be8807db668b9568230 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2b39e389a5ed851f4a39f2b79eb83aa1e95fdbb3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 18:40:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"683ca3c7986203ad2197c1332703ae7b46659c1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-29 18:41:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f51c8146294f5aa1abe4ddffcc2f18 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9aa22c34ca29d0adcc9ad0154a6476ef799f058a","revisions":{"924168f93ecf1dddc07f0ee9c86dea8b117c63f6":{"kind":"REWORK","_number":1,"created":"2020-07-29 14:10:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/743770/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/70/743770/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/70/743770/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/70/743770/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/70/743770/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/70/743770/1"}}},"commit":{"parents":[{"commit":"fca43df8973f20d075cddecc6b7f65381e0e7a13","subject":"Merge \"hashring: allow choosing hash function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fca43df8973f20d075cddecc6b7f65381e0e7a13"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-29 14:10:19.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-29 14:10:37.000000000","tz":180},"subject":"Use safe_decode for decoding in zookeeper driver","message":"Use safe_decode for decoding in zookeeper driver\n\nCurrent behaviour leds to AttributeError because py3 don\u0027t have decode\nattribut for the string objects. So it\u0027s better to use safe_decode from\noslo_utils.encodeutils for that purpose\n\nChange-Id: I22ce4212381511e2b2631a97057d6da938024b99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/924168f93ecf1dddc07f0ee9c86dea8b117c63f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/924168f93ecf1dddc07f0ee9c86dea8b117c63f6"}]},"branch":"refs/heads/master"},"9aa22c34ca29d0adcc9ad0154a6476ef799f058a":{"kind":"REWORK","_number":2,"created":"2020-07-29 15:43:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/70/743770/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/70/743770/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/70/743770/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/70/743770/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/70/743770/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/70/743770/2"}}},"commit":{"parents":[{"commit":"fca43df8973f20d075cddecc6b7f65381e0e7a13","subject":"Merge \"hashring: allow choosing hash function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fca43df8973f20d075cddecc6b7f65381e0e7a13"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2020-07-29 14:10:19.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-29 15:43:06.000000000","tz":0},"subject":"Use safe_decode for decoding in zookeeper driver","message":"Use safe_decode for decoding in zookeeper driver\n\nCurrent behaviour leds to AttributeError because py3 don\u0027t have decode\nattribut for the string objects. So it\u0027s better to use safe_decode from\noslo_utils.encodeutils for that purpose\n\nChange-Id: I22ce4212381511e2b2631a97057d6da938024b99\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9aa22c34ca29d0adcc9ad0154a6476ef799f058a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9aa22c34ca29d0adcc9ad0154a6476ef799f058a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
