)]}'
{"id":"openstack%2Ftooz~744337","triplet_id":"openstack%2Ftooz~master~I242655a557e72b8325825c8f2a8363b479a662a6","project":"openstack/tooz","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I242655a557e72b8325825c8f2a8363b479a662a6","subject":"DNM: testing gate on ubuntu focal","status":"ABANDONED","created":"2020-08-01 23:07:43.000000000","updated":"2020-09-07 13:29:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb201e686c8d1e7818f390f64724a2ce19cc0292","_number":744337,"virtual_id_number":744337,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-05 20:33:26.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-02 21:45:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-02 21:45:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c691c79d48b3ae250316cf78cba39bc675e49868","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-01 23:07:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5c0dc8f31f2347e4601662fee5844682d0c414fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 00:05:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da2ec3e1d5a64bcbbb9d3fa74bc916ed : SUCCESS in 53m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26de2d61aebd413e91dd31c44f705499 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06c8e0335d2344cd91a38b5aa51f0041 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/240e4dbe7bbe40f79c4d5b5ec6f59202 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/554f31d03d89423ba14000bfd8c65fa5 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09e3625b720495bbffaf118ffb099a0 : SUCCESS in 4m 05s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/1b8ae6329ab747c18c69227258264e5f : SUCCESS in 4m 06s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/131bfb615533437fb0cc631e5ec74b81 : SUCCESS in 4m 52s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/6db6c6403ed84dff8d0ed524b9ab3cf0 : SUCCESS in 4m 22s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/767fb076fd92493f8b30907fdc82a205 : SUCCESS in 6m 00s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/f2bfadefe099469ca7ce4568afb5f2d6 : FAILURE in 5m 25s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/1a0c8246c8bc42bf85b8af59bc484fcc : FAILURE in 3m 47s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c7567183ca384d4bab05eb9003671d36 : SUCCESS in 7m 24s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/a443f89d89624be08d21d6ff1434ddf8 : SUCCESS in 3m 52s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/88e1461f7a844dca877b7278c8a7d903 : SUCCESS in 4m 38s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/33bb8ae2a2024f939ee1cb1f23559131 : SUCCESS in 5m 09s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/a65b3498d53848b0aa736becd9ca0f11 : SUCCESS in 4m 38s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/e9887aab68cd401da696ec1273f829a0 : SUCCESS in 5m 02s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/8cc8b2029dc8434f9ba2c466257af028 : SUCCESS in 7m 13s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/d83939c2f8e1402bba97f913d5c3cfbe : SUCCESS in 4m 01s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/4d4fd735a5f54b7cb2ebdeb1a00123f1 : SUCCESS in 5m 03s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/9afe43b8d6244a198f8e970c358f7e1d : SUCCESS in 4m 37s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/61957ee351014341b08b012890f06150 : SUCCESS in 5m 25s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/b1307ccaf5ed4c66a970e26260277b2e : SUCCESS in 4m 10s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/436b1d031db14c6cb28ba5c559f048c8 : SUCCESS in 5m 29s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/b88ef770ffdf490588d70dbd79c278cc : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"02dfeaa02f06acf6f862d580394b28eb91e44286","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-02 19:54:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0a0a404afdb973de6b005406f78392373adebb62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-02 21:45:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c117ae9485d4e86adb10a9fd10b98bf : SUCCESS in 1h 14m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59bbb5fbf825430687d204f60d181150 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a587e4a0b814a9a95cf77bbd266841c : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10b7b1b06ddb49c18486ebcec61a9d9c : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c797a7a028a475f9c5167174cf1a442 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b3486e35110409d979f728d41da0b6c : SUCCESS in 6m 31s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/e1f3610175f04475881a4dcdeed83b4f : SUCCESS in 6m 12s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/66ee8e3f18664c9e9a4efa74ae0f8cec : SUCCESS in 5m 25s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/efae58c136ea4a119577915b2632ef3e : SUCCESS in 5m 08s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/592c835a5e5a41728653297a9f61d83d : SUCCESS in 4m 57s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/b94ee10af3fe4d46a4bbbac0b9651022 : FAILURE in 4m 48s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/1ed5150b24764dcd92a7fa0a13034a5d : FAILURE in 6m 38s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/6ee0f3bfaadf4476bd9aba6f2ae45c16 : SUCCESS in 5m 54s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/60a2c00a095e4fa982b581bc300c61d6 : SUCCESS in 4m 32s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/bcf4158d5b7b47ba8e62a8d76ff937e5 : SUCCESS in 5m 50s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/2280cfc45c2a43de9d4246c39fed147b : SUCCESS in 4m 37s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/2f3dda444ecd4e01aeab651da56c8fd8 : SUCCESS in 5m 01s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/a5038a7eca5e45a2b2a949ba25c03920 : SUCCESS in 6m 09s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/6965022f49f94adf8e6a25828e1a605a : SUCCESS in 6m 03s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/0a2457d93ee842dd97d18dbec0961619 : SUCCESS in 5m 37s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/2a8ae404f3a847b7977d341b3a21bacd : SUCCESS in 5m 45s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/4743365c49e74954a22a31852caecea2 : SUCCESS in 4m 31s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bf779daae5bb4b91a56d2e561b8a9560 : SUCCESS in 4m 32s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/1253e3d7bf2341d59ab050b9e4352617 : SUCCESS in 4m 16s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/552dcf9e30d2452c8ce162b5be595b7a : SUCCESS in 6m 48s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/06883ef403f44a65959b338cab702853 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":2},{"id":"00f91fb74f1457df12eb914d7cd72ba5726aed77","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-08-04 00:26:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"880d716212239575ea72e21b86cd2ea9e5338368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 02:28:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52373a91ea4f4e24b9f192159656b760 : SUCCESS in 1h 57m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbcd9ed84ce34661a04afabc81f38151 : SUCCESS in 8m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/632e958338ab4b1681e1a87dda51cabb : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ef6b84aa7a643a5b3d794b90b78be49 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad7e363483fa4851a5d4e9c4a3a80480 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9a498b208664d04b8432e1adfc6127a : SUCCESS in 7m 43s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/ddab2a0435984f99bea8aa67dc63735f : SUCCESS in 4m 37s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/671f901ff68145e49f416b3d865b0520 : SUCCESS in 6m 14s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/5cf8db24b19140e9928613737a03d162 : FAILURE in 4m 55s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/ae7d2a2ed9f2419a9111cedad891300d : SUCCESS in 4m 20s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/f3c96b8c44684401a6510909f746b36e : FAILURE in 4m 20s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/04561c2c0ca54a48b7ff75b0adf86150 : FAILURE in 5m 36s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/577effd458c04ef0bfa2ce6f55eb6b65 : SUCCESS in 6m 12s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/eb1beadad392424ab64a52e264b5a726 : SUCCESS in 6m 09s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/379272b67d26444586b91f39f122525f : SUCCESS in 6m 22s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/66d0095bc529455c87b08213eadb8ad3 : SUCCESS in 5m 53s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/667fbafa38b54415845de1e662e65146 : SUCCESS in 6m 12s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/03107f72c9c0493d906c548f6a54735d : SUCCESS in 6m 05s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/b065fd38040145e0b4a4505f78e43352 : SUCCESS in 5m 54s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/ff230ac3f4be494bb630110b86071fd5 : SUCCESS in 5m 56s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/14a6948c275e48d1b9aa3a069c5184c8 : SUCCESS in 6m 25s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/65ebf89cc054433d8c68e9e92a025a98 : SUCCESS in 5m 54s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/03fcc63374cb485788336edd3fa39227 : SUCCESS in 6m 19s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/d1f4e599615441c98edc0b2e3951ec28 : SUCCESS in 5m 43s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/34d41be46b6e47c8ab69972f77e43dd8 : SUCCESS in 6m 21s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/e982d122fb6741f1bea8b5f72856ce0f : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":2},{"id":"dcb7108def81b5a357472c105d7f02ccf97c7cd2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-05 20:33:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3ceb1b80ce1f73b78a1ea7799d260cf943a6d971","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-05 21:42:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c92e832d78545f4aabd949f5d840dee : SUCCESS in 54m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad2062fe0b2149079b65022374a2ccf1 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fce72dbdffa04cb793db2d44647b4ec5 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/216750db39e7415db2983339e2067b65 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff656ede8be64d37acd7b56349f4eb36 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e7cf839f7a4cc29f0e2b6a856a875a : SUCCESS in 4m 07s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/2187db8ca4b64e9aab73e48dea6ec531 : SUCCESS in 4m 30s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/6e7f0143389a497a851aa4e3f3f19feb : SUCCESS in 4m 36s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/1d3337b18bb342e2a93ac419483327db : SUCCESS in 4m 41s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/58f95114d3fb446e98e217c8277be1d6 : SUCCESS in 4m 07s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/9e6759aa99764f2b83df859b44fbfff7 : FAILURE in 4m 07s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/c7fbb3ae865042089f6437953a11a540 : FAILURE in 4m 05s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/d4c769a377da46b0b4fd235157610bf9 : SUCCESS in 4m 27s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/938c47fb47604cc69ad279ee9e259449 : SUCCESS in 4m 10s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/2c66a0700b304fddbea74347500328fe : SUCCESS in 4m 24s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a41b46ae31d342218041abe2c7700262 : SUCCESS in 4m 07s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6f89220d53fa410fb1593bb0e6f0fead : SUCCESS in 4m 24s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/459e2bd782774f87866eca972019ad44 : SUCCESS in 4m 30s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ea73d6d4b6574a6cb3ae3880bd397190 : SUCCESS in 4m 24s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/d4bb1aa40f6e4add8eba95b390113b59 : SUCCESS in 4m 40s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/154cb707c4ca4ae89809ed37f5fbe1b2 : SUCCESS in 5m 15s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/c9613faabc8047109304cd57988695d4 : SUCCESS in 5m 54s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/eb1a45f0ef214f3f88c1e347bdb7c752 : SUCCESS in 4m 35s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/54162c4c06f04f2c95fd1209d794ee71 : SUCCESS in 6m 19s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/db0dca44a96145db8469e6a3a109f982 : SUCCESS in 6m 10s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/c08f6d130d1744238749ac47f16dfb66 : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2},{"id":"7975fff7ac4e4da638550c09578e12edcfa4ada8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-07 13:29:17.000000000","message":"Abandoned\n\netcd3gw jobs are broken already. there is no other work needed here for Focal.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a31dcf552d9e8b0c99b61140da67769ad5d336df","revisions":{"2641fd1cc956d4b271163055537ffb4718dbaece":{"kind":"REWORK","_number":1,"created":"2020-08-01 23:07:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/744337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/37/744337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/37/744337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/37/744337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/37/744337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/37/744337/1"}}},"commit":{"parents":[{"commit":"9aa22c34ca29d0adcc9ad0154a6476ef799f058a","subject":"Use safe_decode for decoding in zookeeper driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9aa22c34ca29d0adcc9ad0154a6476ef799f058a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:06:29.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:07:42.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nChange-Id: I242655a557e72b8325825c8f2a8363b479a662a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2641fd1cc956d4b271163055537ffb4718dbaece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2641fd1cc956d4b271163055537ffb4718dbaece"}]},"branch":"refs/heads/master"},"a31dcf552d9e8b0c99b61140da67769ad5d336df":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-08-02 19:54:37.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/37/744337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/37/744337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/37/744337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/37/744337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/37/744337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/37/744337/2"}}},"commit":{"parents":[{"commit":"9aa22c34ca29d0adcc9ad0154a6476ef799f058a","subject":"Use safe_decode for decoding in zookeeper driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9aa22c34ca29d0adcc9ad0154a6476ef799f058a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-01 23:06:29.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-08-02 19:54:35.000000000","tz":0},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- unit tests jobs\n- functional integration tests jobs\n- docs jobs\nStory: #2007865\nTask: #40207\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I242655a557e72b8325825c8f2a8363b479a662a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a31dcf552d9e8b0c99b61140da67769ad5d336df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a31dcf552d9e8b0c99b61140da67769ad5d336df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
