)]}'
{"id":"openstack%2Ftooz~750112","triplet_id":"openstack%2Ftooz~master~I0505b4ee3b3d277337f9b64a304b7d2420907824","project":"openstack/tooz","branch":"master","hashtags":[],"change_id":"I0505b4ee3b3d277337f9b64a304b7d2420907824","subject":"DNM: testing gate","status":"ABANDONED","created":"2020-09-06 22:29:34.000000000","updated":"2020-09-07 13:27:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c7cf07b9225b3fa8dc6740e7974a57a7709cbe48","_number":750112,"virtual_id_number":750112,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-06 23:49:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-06 23:49:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"81a1939d42774a769387891ac9151d2823d4a160","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-06 22:29:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d0aaea4c20420f8fb559764d5def51644c7098d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-06 23:49:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/135585320e6d447b9c8254360c77b7d4 : SUCCESS in 1h 18m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcdf8fa04c014fd9a4b9d68db7bf7575 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b903e0b0294948e8a3b6fce03cc0f2af : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faeb57317e9f4ba4ac8c113d7c0f791c : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8193657ad6fe4c0f987c1f9fbf408898 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008a98842df049b6820bae46a4e7b7ed : SUCCESS in 4m 12s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/5c8c1e484a03491088e53c9aae2e989d : SUCCESS in 4m 46s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/242072920be24973897cd02c734bd927 : SUCCESS in 4m 10s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/14ab2f5056c44c80a362b3cccfb39658 : SUCCESS in 4m 33s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/aacd7a5f68e343d5ad27d88a9f3a292c : SUCCESS in 4m 11s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/079d60ce25c74dab8383370026aac768 : FAILURE in 4m 25s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8f6c8381fbfc42a7a44c4ca157c74d97 : FAILURE in 3m 44s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/ce7e53e5683d4ec09c59001db86d751f : SUCCESS in 4m 13s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/e2f504cf8dc34903b9d74407ed24d323 : SUCCESS in 4m 20s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a9140479da6e483c84680c225a73ef59 : SUCCESS in 4m 56s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/cf1a402f0e394962803d653b9865966d : SUCCESS in 4m 16s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/a073dfdad4c446cba86c576096148c73 : SUCCESS in 4m 44s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/77af37f116864d09ba03668087ab6ae2 : SUCCESS in 4m 25s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/df897bf7012246b1bd8655193e73f3d1 : SUCCESS in 4m 34s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/574d86bef3fd4703829d3c23167f80d6 : SUCCESS in 4m 10s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7a28291efff34b53863077b9d51e1376 : SUCCESS in 4m 14s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/32a385059ff14a17b53b236cb2db07cf : SUCCESS in 3m 59s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/ca8e7efa9cc34b6b8d7e4b0a00a0b825 : SUCCESS in 4m 15s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/07e50e39784f4c44b64409d2d2e5f51a : SUCCESS in 3m 52s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/640ed9b5268947fa8f4b6ba9be1dbf46 : SUCCESS in 4m 40s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/6372fb8d723140aeb489db609518ec49 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c7cf07b9225b3fa8dc6740e7974a57a7709cbe48","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-09-07 13:27:47.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f44d5b4bafe266544b647264049d19d9f21aecd","revisions":{"7f44d5b4bafe266544b647264049d19d9f21aecd":{"kind":"REWORK","_number":1,"created":"2020-09-06 22:29:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/12/750112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/12/750112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/750112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/750112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/750112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/12/750112/1"}}},"commit":{"parents":[{"commit":"9aa22c34ca29d0adcc9ad0154a6476ef799f058a","subject":"Use safe_decode for decoding in zookeeper driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9aa22c34ca29d0adcc9ad0154a6476ef799f058a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-06 22:29:14.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-09-06 22:29:33.000000000","tz":-300},"subject":"DNM: testing gate","message":"DNM: testing gate\n\nChange-Id: I0505b4ee3b3d277337f9b64a304b7d2420907824\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7f44d5b4bafe266544b647264049d19d9f21aecd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7f44d5b4bafe266544b647264049d19d9f21aecd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
