)]}'
{"id":"openstack%2Ftooz~751540","triplet_id":"openstack%2Ftooz~stable%2Fvictoria~I463460a800ba29fc125a6dc9032c565ad54a6c8d","project":"openstack/tooz","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"I463460a800ba29fc125a6dc9032c565ad54a6c8d","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-09-11 21:04:20.000000000","updated":"2020-10-22 11:37:19.000000000","submitted":"2020-10-22 11:35:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751540-1603366559692-298625ce","meta_rev_id":"24a27bb85192b870ab4c0e9a1297e33da34064df","_number":751540,"virtual_id_number":751540,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-10-22 08:20:09.000000000","_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-22 11:35:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-09-11 21:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-09-14 16:12:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-09-14 09:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-09-14 16:12:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 21:11:19.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-09-14 09:33:05.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-09-14 16:12:01.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-10-22 08:20:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-10-22 11:35:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9431299aed18f99fdb675cb887b62b224b8e7936","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-09-11 21:04:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc9627650f6426951619b6073f1c631fd326e754","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-09-11 21:11:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e97cb1299c7b4232bab8cd94d26eedcdf8c5c82f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 21:44:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13edf71f7cf84f3e87e838287a09097c : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9d2869002e4ea0812408c96548f06d : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d53d89b94d74958bb2c208c25ca5c36 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12a5f1423d2a44c3a7f5afa2d7b60ac3 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149034900c414470bb1ce31f18f18552 : SUCCESS in 5m 41s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/4a9461a3daea4b109d7c01ea772ba9e9 : SUCCESS in 4m 55s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/cf8a3ae247bf4690b25ee806c876f743 : SUCCESS in 4m 52s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/49dab9c127e64547bab378fa17b7c99d : SUCCESS in 4m 55s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/91061aadfd3747d9af10d919dc9eb725 : SUCCESS in 4m 48s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4eabce88d6b14e838104d73df4f59dbf : FAILURE in 4m 55s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/a94310e3a1294751aedefe1a8bd61a11 : FAILURE in 4m 44s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/6cfb7246ba944fc2896a1d7f86b76275 : SUCCESS in 5m 28s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/5caf9e563d6a42f5895cac2c2a24cd33 : SUCCESS in 5m 14s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ce5698e7da344063a52bf4ff9a7c9cd9 : SUCCESS in 5m 26s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/c8c38d5331de43589e2eceb18e42b870 : SUCCESS in 5m 59s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b2272fbd0de5473ebc873ed34cfdf6e7 : SUCCESS in 5m 07s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/f07d3a7b6c5f479aa9b5329232402430 : SUCCESS in 5m 32s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/1a993a60917a463faf627f201c9d762f : SUCCESS in 5m 22s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/62abc78115564104a6030ac18066f710 : SUCCESS in 5m 25s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/d766fe636b354a1bb2fef1d655dd6132 : SUCCESS in 5m 17s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/e092301bba304c72aca40485805bea4e : SUCCESS in 5m 35s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/30b46eb89f5b4ab881702f3313f6f641 : SUCCESS in 5m 14s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/833ce9a53007485fb834b715eb8b0011 : SUCCESS in 6m 10s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/65db8c0a362f42c395ec365fa6d2178b : SUCCESS in 6m 14s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/482a07eee7d94ce485ad5a147247e95d : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f6afcbf521228d938d57fcf3a673bf5ab6938466","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-09-14 09:33:05.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"76d13b62b4b9df0149ee38592557b209f4258637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 09:47:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59e4ed4e009c461f802fd1d48520f628 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d18874f8a94d77abf75d61b6141b9a : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e047c7ef3bec4d6ab4ebdf99eed010ca : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a51311148b8477b8dc81085dc03388d : FAILURE in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b091b9c9e60146dc9af9bacba5726d31 : SUCCESS in 7m 58s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/379fd83599084c6da5d4d0eb989339a3 : SUCCESS in 6m 46s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/416798c1f94c43d69d412b8e1fb146b7 : SUCCESS in 5m 58s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/a83f21cf57614927b2bb6a2009351315 : SUCCESS in 6m 52s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/569d39167f4a4500a47fe82e1fdcf7eb : SUCCESS in 6m 04s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/cbe162841ed74cc5990dce7f759f4638 : FAILURE in 6m 22s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/81ea5e2f4acd40fb9382b9ba4e8c2fcf : FAILURE in 7m 28s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/bc0aa8fc9fb34e559898e24d6b82c951 : SUCCESS in 7m 12s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/914448ab05514c69954922d0a3aeb974 : SUCCESS in 6m 58s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/37003d19fc8c47d983ebf67671ad0e30 : SUCCESS in 6m 49s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/71341051ae9f4dcc8d3c462ead6b097d : SUCCESS in 5m 12s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/3cf9de9843ec4b719ade3f124c38b978 : SUCCESS in 5m 22s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/304b33a8e94847d485496abaf1e2c42c : SUCCESS in 6m 35s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/087894825fd54b699e63931e336a9fab : SUCCESS in 6m 54s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/dd30a82cb8db4fd6b845a6c60b79811a : SUCCESS in 5m 39s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/453b40144388414fb480381767be2d6d : SUCCESS in 7m 12s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/4e417f6fbea644538fec076000e55f0d : SUCCESS in 5m 02s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/3d50c51a52764cacb5ea73708fc7995a : SUCCESS in 5m 18s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/55e2f6b4ad9d46af8b5cbd1479ef3b35 : SUCCESS in 4m 56s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/643d7cd91d69422a9df9808b7fc114ed : SUCCESS in 5m 25s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/18798ee129df4a979a59b5cf6f5a6f43 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"06adc27ac7f469c24722d7839f5eea18c2d24dbe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-14 16:12:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClearly we need to do some work on our tests here too :-\\","accounts_in_message":[],"_revision_number":1},{"id":"c18196ab8a602e7912e773a8312691d866a71527","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 16:43:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a80ea8fb6d5944da8a581f591d49e052 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43300aea22bf46f99ae48c5ca84af3f4 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2351f8077a784263be72fc6cad1e3e2a : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/873db954dc894d3d879c2d108c52057a : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8198d23d208473b9ed149bf3de588f3 : SUCCESS in 5m 31s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/569863375fb8447191a670d7e1f203c1 : SUCCESS in 6m 15s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/3d397323c5d94aee8e60f6eccac9ee07 : SUCCESS in 5m 25s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/1798dcd505f44c448c9852367241921d : SUCCESS in 5m 00s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/2242294a3f7e4d9ab7b3cb5e6bfed4de : SUCCESS in 5m 56s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/e600796ecc924eb9adffa9cf3faad47f : FAILURE in 6m 09s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7f68f8270feb4d06bee0d3870d5a52d6 : FAILURE in 4m 49s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/4264301428d94e87bfcd257695a4a187 : SUCCESS in 9m 12s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/691774da3d71448ca2e97f1098655c26 : SUCCESS in 6m 14s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/87e80a9a8eb64d5fbde0e6933e4cf5bf : SUCCESS in 6m 12s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/558427c6c2dd49f18a80930b4d6c022f : SUCCESS in 6m 00s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/abee35b799f944f2b4f6ba3e094ed7ea : SUCCESS in 5m 59s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/e1e24fffa50f4f039709ee3e6b6d23e5 : SUCCESS in 6m 07s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/38c477e165784c61ab29a45c0b3a1a90 : SUCCESS in 5m 30s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/2bd5f82c6f1340b182f51b02f6d064cc : SUCCESS in 5m 45s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/2b27c89924204677b49f9f179f949080 : SUCCESS in 4m 53s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/6db1e3cf907d45e7b8c2ddbac343ad8a : SUCCESS in 5m 24s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bc847a1e349849988275e3c5a41f1853 : SUCCESS in 6m 06s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/44d82b7331f84538b2a18b08576f7852 : SUCCESS in 4m 37s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/65e01d61646343999faa8d5e3785bfaa : SUCCESS in 7m 32s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/88472e4417504459887b93ce2a311b58 : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6e7527d356bf9a3154787bd256243bd1ca6e1f65","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-10-02 15:10:16.000000000","message":"Patch Set 1:\n\nNeeds etcd3gw job fixing patch to be (merged and then) backported first:\n\nI22b955419014dd34c63e406c488e0636ffe9013b","accounts_in_message":[],"_revision_number":1},{"id":"ea11c4e76020d61e9c36ebb6064d3a9260c0bed9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-10-22 08:20:09.000000000","message":"Patch Set 1:\n\nrecheck - the etcd3gw restricted patch has merged, zuul should be happy now :)","accounts_in_message":[],"_revision_number":1},{"id":"2d905b3360d445aec02b36fced7d7ee20bcd4244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 10:03:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8264c874d89e404999b7f47640b1b532 : SUCCESS in 1h 31m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa0781688db1461f86b68d34fd0c402c : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc22b6e2a5334f67b4e30c578cf923a2 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d172269d7f3b4f08a8e21caee78aa711 : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/688b575a83774ff6b5875e4995874ea6 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec2b666f368440c0a360e2ccbafccc5e : SUCCESS in 8m 35s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/083e74e203a14435b6d21f969a53d04f : SUCCESS in 9m 01s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/c44a276bdc2742cfbc8d52a68d5e46ca : SUCCESS in 7m 32s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/7e5e35f3356f432da63f360fe886d1ed : SUCCESS in 9m 38s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/e03e797d7f0e4dbc9b9bf17e1b846956 : SUCCESS in 8m 19s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/588a83c3419247268b2745a84c1a792f : SUCCESS in 9m 06s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8253eb8a1da94e35afa9dec6a789b3aa : SUCCESS in 6m 04s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/bff04c47910444239a7af2764fb63e0d : SUCCESS in 6m 57s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/a91bfcf11fda4dd191233f6ed266ec0e : SUCCESS in 6m 54s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/b998e6df9225433cb612b5617109ccf0 : SUCCESS in 8m 40s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/d6b62ff835bc4f25a27db27f34de5683 : SUCCESS in 6m 45s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/d09d32c24ecb41e9b2450b82a4d41885 : SUCCESS in 6m 09s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/4d60ac1619244327991ee0be8ccaba7c : SUCCESS in 6m 24s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/3bd8f4b8280f4d64944e5b9ee3f30caf : SUCCESS in 5m 53s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/e725a27824df4b998920e65c40851a8f : SUCCESS in 6m 49s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/31fb93fdb9ba4147b0c36fabb8f34acd : SUCCESS in 6m 44s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/4bb12d2e4df14475924b6bc2b594f92c : SUCCESS in 6m 08s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/d632492c9ada4fbbb789bfb54f44815e : SUCCESS in 6m 11s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/10c5848136594afabdf97e5d864ad807 : SUCCESS in 6m 41s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/941a87c9b07844b7980fe51ddb257179 : SUCCESS in 6m 23s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/725ccf316c564bc498c6e7e9879b5c9a : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":1},{"id":"5971ec01b335d61e1288af463322fc84627f12a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 10:04:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2efa3895913172fc42b7d5869ac3d24552965f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:35:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d97c96bf45bf5fdcc4ee2c45274494a4aa9c786b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:35:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41c67b091e4b4c87984c083e52f8171e : SUCCESS in 1h 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6c2aab5e6b4472abaf987bcb02fb4f : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/037a9a21729b497993c8dc805d5a29b9 : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a69a3250fbc420d98203e097afca27d : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aecf487565a4fc3940a6b05ef693468 : SUCCESS in 9m 50s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/283e8de3fad74c0786e19f00fe3abd03 : SUCCESS in 9m 00s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/858ac602c1044463888de3066196987a : SUCCESS in 8m 16s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/d52628229fdf4b8e93c05d1cb4aefd45 : SUCCESS in 8m 03s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/c6913073b41947a1b8d48d188f0b3b30 : SUCCESS in 8m 28s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/b5519d455e054feca30eb781b8a1d744 : SUCCESS in 8m 09s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/681ebdaa1ffb49b1a91a15599152f4ca : SUCCESS in 7m 33s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/6afc814bc2fc4c07a9b7fed4446f2446 : SUCCESS in 7m 37s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/245f3c919b0a46d0afd7ad9844866834 : SUCCESS in 7m 32s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/85ff7f31bdae495baf757c77feea5a60 : SUCCESS in 7m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/18cd4362e15f4c5abe2f40892e5fba04 : SUCCESS in 7m 36s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/366c8e44e64a43a4acf0be80fe7ead5d : SUCCESS in 9m 02s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/81fa4b165dc94c2889537f134b710392 : SUCCESS in 7m 32s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/763446e5efd045fdbb23e031428be07b : SUCCESS in 9m 27s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/d512f8e19542424489d93d78b4ed26a6 : SUCCESS in 7m 20s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7bf7fc7cf42343c6a10135372dc21a55 : SUCCESS in 8m 09s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/32ebbdaff00c414f9bb8325b3a01901d : SUCCESS in 8m 53s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/08897263a5d14f23a7dfba06b3e9861b : SUCCESS in 8m 15s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a2393e1785774eecb4bbcdbbe8242e3a : SUCCESS in 8m 36s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/c7bfa4b54f31498a864bec0da0554f5f : SUCCESS in 9m 44s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/f7f5ec58ceef46d4a7f76be5c4c08f8e : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7f3323b3244ffed57ac4fcd80cd596dce40712a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-22 11:37:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5519e6bc1964ab68856a287e5372c40 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"affcbdaad3b58f92afb927a88e0108a8982d470b","revisions":{"affcbdaad3b58f92afb927a88e0108a8982d470b":{"kind":"REWORK","_number":1,"created":"2020-09-11 21:04:20.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/40/751540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/751540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/751540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/751540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/751540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/751540/1"}}},"commit":{"parents":[{"commit":"9aa22c34ca29d0adcc9ad0154a6476ef799f058a","subject":"Use safe_decode for decoding in zookeeper driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9aa22c34ca29d0adcc9ad0154a6476ef799f058a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-11 21:04:17.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-09-11 21:04:17.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: I463460a800ba29fc125a6dc9032c565ad54a6c8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/affcbdaad3b58f92afb927a88e0108a8982d470b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/affcbdaad3b58f92afb927a88e0108a8982d470b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
