)]}'
{"id":"openstack%2Ftooz~756432","triplet_id":"openstack%2Ftooz~master~Idbef0f0896753765372c8dfac8ab15e6be49922f","project":"openstack/tooz","branch":"master","topic":"replace_md5_for_fips","hashtags":[],"change_id":"Idbef0f0896753765372c8dfac8ab15e6be49922f","subject":"Replace md5 with oslo version","status":"MERGED","created":"2020-10-07 05:43:21.000000000","updated":"2020-11-02 13:34:47.000000000","submitted":"2020-11-02 13:28:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756432-1604323726801-94ba6c35","meta_rev_id":"2d54f0ecf29955974fac947893e4fdfdaa33cff2","_number":756432,"virtual_id_number":756432,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-02 13:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2020-11-02 12:03:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-10-30 09:30:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2020-11-02 12:03:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":-1,"date":"2020-11-02 12:03:58.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-30 09:30:34.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-11-02 12:03:58.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-11-02 13:28:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d5ec22ea20f84a22566a29b3e08b4bef3ab6841","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-10-07 05:43:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3bc377d6edb3c7af6f921335b8e79a7786b6f0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-07 06:16:21.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90aecd3779eb4af8a311593fcfbf6059 : FAILURE in 27m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2d2e379b8a2448ab1ea7b052eda08dc : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd9f8f9e9d7480fa9fe2757d2792562 : FAILURE in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dab68b5fcd394c5584103ab43a38fe7c : FAILURE in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf27210ac94e4cc3a9f9c5299c95dee0 : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc27a7a25983491e9d74c87278666d1b : FAILURE in 5m 15s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/9bdb0be452784701ba78b56284bbde80 : FAILURE in 4m 12s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/11c42c1bb2814051ab5d9b5f1f0b6f1b : FAILURE in 4m 33s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/fa7d2d17263642cabb057e1aefe28c99 : FAILURE in 4m 25s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/aa52c2582b1f4f09aebef9f9ef801914 : FAILURE in 5m 18s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/ea91eddb8f9a4481af536a216bdf0f31 : FAILURE in 4m 57s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/6fd32b12c38542f9bc90b8dc7191dff8 : FAILURE in 3m 50s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/67eae54c3e7c4114b6803b01c5a1c4f7 : FAILURE in 4m 04s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/f4bd61c73e914a0eb7b4bf94d6fb0353 : FAILURE in 4m 49s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f8d0cf224a6e4849a5fe1a8e883efda7 : FAILURE in 4m 17s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/c5d67539af9a434ab43e4427ecddf833 : FAILURE in 3m 36s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9600827c5b8b4c9dae5002832ae2f009 : FAILURE in 4m 54s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/49cdb59a34c0436988aa71dd2e73db9b : FAILURE in 3m 48s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e45c215922ad458f8daacb95929ec5de : FAILURE in 4m 05s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/91bd23fa15c94af48498476db6dc7836 : FAILURE in 3m 29s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/4ddcdad718d247579c9d0308054e0928 : FAILURE in 4m 15s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/865762e3efb24a3c8d369972273be14d : FAILURE in 5m 09s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4c19c8fba9184ad0b6144b8c8bcfc19d : FAILURE in 4m 10s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/25cb840b2ae24e79b4dbd89e7a1a2e3e : FAILURE in 5m 23s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/91ceace4ae6441b28e5768ef24bcc783 : FAILURE in 4m 19s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/f2798cb265b44b579278afb394720813 : FAILURE in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"da74ba2350f2d2ce358504b094e7267f38c163d0","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2020-10-29 19:45:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d88a52deaf42762be34b90e89f927ec3e260b7db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-29 21:36:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa9eedb3125e4bdf8e09dbceb5563bac : SUCCESS in 1h 32m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/034783b25dde4fc395e212af7b9822a6 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cbb6ed747d04b068d863f10e5084dc6 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4d1c0cbadae40769f9466c197718ad1 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98f73e2b8fb648759ad27a92157f2236 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f10a6972bd49cb98dbd33d31b337b2 : SUCCESS in 5m 41s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/da971c8e8e624bd3923ae6fc9f7839ad : SUCCESS in 9m 00s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/832f7a74b58b4a51b1145f3c7536d93d : SUCCESS in 6m 37s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/74eda7e2c91d4591a0084a9646fdacd8 : SUCCESS in 4m 51s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/046f9668f5e04c9b81fc365533cbb0c1 : SUCCESS in 5m 14s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d3fa9d526db44ffdbdf5efaa77771092 : SUCCESS in 4m 51s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/79f2c7b2f4f4461c8ec225143e059af4 : SUCCESS in 11m 59s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/15a62ae284604d299731ad7a3c72606f : SUCCESS in 5m 11s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/33a1ffe60c6642a6b2fabed19b6859a9 : SUCCESS in 5m 46s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ac632d521e7e4748a61a23726ce9880b : SUCCESS in 6m 18s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/2eae92a368e34f6da51ff0978e415cc5 : SUCCESS in 5m 13s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/e0ec15a0daba463380f8302327b9eee7 : SUCCESS in 4m 48s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ab7a91de5b7c47c7837bc8bc8da279c4 : SUCCESS in 4m 45s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e1086d3a1d0042838e5e3f0aa5ad9727 : SUCCESS in 5m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/0ca4fbf6a6784bb099d8d6e7ff612d70 : SUCCESS in 4m 38s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/16fec29ea4cd445d9ce9d83d854888a3 : SUCCESS in 5m 07s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/1113ee1fb3b54d63926fffc75e15ce60 : SUCCESS in 7m 43s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/b3a7456b375345c5b5ec3154a19f6473 : SUCCESS in 15m 40s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/44281f24dd2946f58c139e603e41f789 : SUCCESS in 4m 46s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/3b745b15fda34facb61a5a07713035e3 : SUCCESS in 4m 26s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/271197ef2f40473d912c1e6e417e9e6a : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2d640c68e6238ef5ee8260b303babbd4a4ebb4b0","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-10-30 09:30:34.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0db0d6fd16b4aacf9ff2c53b1ef0d736962217a2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-11-02 12:03:58.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d170faddf4ef3b98d181e5fb748811023a7267f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 12:04:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ede1548e7197522a081cb8a83b048dea4d28d0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 13:28:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09f8090214cd43fbacc6bff5324b087e : SUCCESS in 1h 22m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/962183e30b584325a6e732bbfa817836 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/782e98d01536488ca75fcb1b5e21b6a8 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a3fdf75a3c24b429f923883676d2f60 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c28e87d00ef84ddf8c9811e890f823af : SUCCESS in 5m 48s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/27992534334f4428b4841b9bf221f03d : SUCCESS in 4m 22s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/34017c05699b4091afe204d517387ae1 : SUCCESS in 4m 35s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/ee59d7cf12d14b1388c825036f427176 : SUCCESS in 5m 16s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b9c70eae407348f1b512788c0f2cff0f : SUCCESS in 5m 44s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/b7282d9c97914375b75a419590458cf7 : SUCCESS in 4m 51s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/0d27b2bbd48a4c01b31ed5f5881c590b : SUCCESS in 5m 31s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/0b911c4a004f432094bf51630e6aae29 : SUCCESS in 6m 03s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9e2316842093469686ecb621c18e7a98 : SUCCESS in 7m 31s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/425e0148c3c24e8f8cfd0c242e3a6d00 : SUCCESS in 5m 29s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/28eda1c38119443f88711b714c168884 : SUCCESS in 6m 15s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/de2e8ee0017e4ca88c01440773f4438f : SUCCESS in 5m 01s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/e11ff1661d134a919f50d024383eb10b : SUCCESS in 4m 30s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/5444951a4b494fe9a5543204e7ff41ad : SUCCESS in 6m 24s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/21c65ceac0994fbc8900335fd4facd0d : SUCCESS in 5m 44s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/aedaf2306baf45708150a57d554dd165 : SUCCESS in 4m 48s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/6ec65b58749a40ecb26d6a82947a107e : SUCCESS in 4m 31s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/ceeef1ca659b43888d8fd529495e8d28 : SUCCESS in 5m 21s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/05beb7b9b1614011a8c7d3fc518d2589 : SUCCESS in 4m 39s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/06e3d355f1ce4debbd39d93ba8a3f5e6 : SUCCESS in 5m 51s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/b1d1fa33f4da454ab5865349d6bdb844 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"42c4adfee306c9795f3fd6022fde7f84931f1476","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 13:28:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"8e36e73ef8c07d27e5b5556c5351ca7f71a3079c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-02 13:34:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d96733d3a3493a968401157669ce3e : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"54448e9d8b968c6d537a4a3b648ed280174b4f2c","revisions":{"0e98c2d5e0cbcbd4870ea1a981c2be9f0a3c838e":{"kind":"REWORK","_number":1,"created":"2020-10-07 05:43:21.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/32/756432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/756432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/756432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/756432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/756432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/756432/1"}}},"commit":{"parents":[{"commit":"b7fe722e9727455881c1a37e540252efa3a0e41b","subject":"Update master for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b7fe722e9727455881c1a37e540252efa3a0e41b"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2020-10-06 18:35:00.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2020-10-07 05:43:20.000000000","tz":-240},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nReviewers need to pay particular attention as to whether the keyword\nparameter (usedforsecurity) is set correctly.\n\nChange-Id: Idbef0f0896753765372c8dfac8ab15e6be49922f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0e98c2d5e0cbcbd4870ea1a981c2be9f0a3c838e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0e98c2d5e0cbcbd4870ea1a981c2be9f0a3c838e"}]},"branch":"refs/heads/master"},"54448e9d8b968c6d537a4a3b648ed280174b4f2c":{"kind":"REWORK","_number":2,"created":"2020-10-29 19:45:21.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/32/756432/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/756432/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/756432/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/756432/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/756432/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/756432/2"}}},"commit":{"parents":[{"commit":"c3d8c2ed9629522291da69ab61f38a8009466a04","subject":"Adding pre-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c3d8c2ed9629522291da69ab61f38a8009466a04"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2020-10-06 18:35:00.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2020-10-29 19:44:36.000000000","tz":-240},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nReviewers need to pay particular attention as to whether the keyword\nparameter (usedforsecurity) is set correctly.\n\nChange-Id: Idbef0f0896753765372c8dfac8ab15e6be49922f\nDepends-On: https://review.opendev.org/#/c/760160\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/54448e9d8b968c6d537a4a3b648ed280174b4f2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/54448e9d8b968c6d537a4a3b648ed280174b4f2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
