)]}'
{"id":"openstack%2Ftooz~762797","triplet_id":"openstack%2Ftooz~stable%2Fussuri~I417dbc5bd3f591905d9b1bf9a6ecaa978812ad32","project":"openstack/tooz","branch":"stable/ussuri","topic":"755753","hashtags":[],"change_id":"I417dbc5bd3f591905d9b1bf9a6ecaa978812ad32","subject":"Revert \"Set some jobs jobs to not vote\"","status":"ABANDONED","created":"2020-11-16 08:24:33.000000000","updated":"2020-11-16 18:47:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"95f874d7921f2d5323abda3b924728281ab3aa1d","_number":762797,"virtual_id_number":762797,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-16 10:04:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-16 10:04:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67583612c4b2c5eba5b00ef218c08c6ab679ef41","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-11-16 08:24:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74dd18f8f3b890bc482d21f5c0dc645789653733","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 10:04:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b1aba0d067d4600b7a41074b699e759 : SUCCESS in 1h 21m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a444261ebbc4c419fb378dedbcf944e : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8cad499b7714b61a7222ea9c187612a : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b93ecaa3b8eb442184dae23c7b3993ec : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153769faca4b4f4b87b8f4da0e53534a : SUCCESS in 4m 42s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/988bc8c906d848029888592e8ae2b22a : SUCCESS in 5m 26s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/b6e907f182df45a9a53c1d1ba77ccec6 : SUCCESS in 4m 54s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/c19f29404dcb4cd7bee8ba6d9ecfe400 : SUCCESS in 5m 03s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/2f3fd5dddc5a4052b2d9dd41c7003d74 : SUCCESS in 4m 53s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/d038cd8a791d49a985b5a043624746b6 : SUCCESS in 5m 10s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/a36e444891a3444f9e03ebaa8df23fab : SUCCESS in 5m 01s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/ea8ad1f56b7744f79ecfc26e1a627846 : SUCCESS in 5m 07s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/b440a1ed79fc49a18917eb8c08d4a100 : SUCCESS in 10m 12s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/6cb0a372bbd6444192b21509f5fcfede : SUCCESS in 6m 16s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/074b3d234bc54e159a6d368999925a7f : SUCCESS in 5m 47s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/8d12ff062ac449fc8310174ceaee8926 : SUCCESS in 5m 02s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/525697da22f644799db9686cdf7d3253 : SUCCESS in 4m 10s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/74c53916a7c1430a9abd27cf280e9f9e : SUCCESS in 4m 52s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/fa1f15aee22a4938a70f348f31d87a4f : SUCCESS in 4m 11s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/5cccff89ad51440ab6010b429bf6e700 : SUCCESS in 5m 08s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/7889c480c7d3489d864cd97111d8961e : SUCCESS in 5m 33s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/93b995cd86454069b024d3e375c74820 : SUCCESS in 5m 24s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/550465d8c4cd411b8c5c250157e8e215 : SUCCESS in 5m 37s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/1e6d733da6d0467b839d82c05adaed8a : SUCCESS in 5m 48s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/0023cfac59c24e1e801dd23583ae96df : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a42ec47e8786bd34777e232a10e7e0703ad625ef","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-11-16 18:47:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48a26fd421883978c8e1d50973e0776fd58d692e","revisions":{"48a26fd421883978c8e1d50973e0776fd58d692e":{"kind":"REWORK","_number":1,"created":"2020-11-16 08:24:33.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/97/762797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/97/762797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/97/762797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/97/762797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/97/762797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/97/762797/1"}}},"commit":{"parents":[{"commit":"de0094785503ef76f16b32964c79db64a2bbd7ec","subject":"Fix breakage with PyMySQL 0.10.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/de0094785503ef76f16b32964c79db64a2bbd7ec"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-11-16 08:20:26.000000000","tz":60},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-11-16 08:24:22.000000000","tz":60},"subject":"Revert \"Set some jobs jobs to not vote\"","message":"Revert \"Set some jobs jobs to not vote\"\n\nThis reverts commit a3b31631bfea05666d76d0fdf909ca821d9de9cd.\n\nChange-Id: I417dbc5bd3f591905d9b1bf9a6ecaa978812ad32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/48a26fd421883978c8e1d50973e0776fd58d692e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/48a26fd421883978c8e1d50973e0776fd58d692e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
