)]}'
{"id":"openstack%2Ftooz~779161","triplet_id":"openstack%2Ftooz~master~I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f","project":"openstack/tooz","branch":"master","topic":"tooz-gate","hashtags":[],"change_id":"I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f","subject":"Cap tenacity to unblock the gate","status":"MERGED","created":"2021-03-08 07:24:33.000000000","updated":"2021-03-29 13:09:34.000000000","submitted":"2021-03-29 13:08:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779161-tooz-gate","meta_rev_id":"df6fe5f76fcb27887344465f94a243cddc7b9dba","_number":779161,"virtual_id_number":779161,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 13:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-16 17:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-29 11:15:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-03-10 08:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-03-29 11:15:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-03-16 17:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 09:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 13:46:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-03-10 08:42:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-03-16 17:00:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-03-16 17:00:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"a48123663d33d3f3631fb906f61ae2a0d61b017e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-08 07:24:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0707836d4b2e364b3c1f4be6169bec7f21ee1ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/189b1fe1b4914a41914d75eeccef1041 : SUCCESS in 1h 36m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5d79c3a19294efab3e7718b0fa985b3 : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0a270537b04edbba6efa17f894bb51 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0530cbdd67be40dc8daaf3ae34c07c29 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52ceff8bffe34ee3b052d8f1aae8d9be : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c73abc1960454f19aa4cb5e70b74cb75 : SUCCESS in 6m 03s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/4b4c0509f9e64f7692b3ca6ffe3b80df : SUCCESS in 7m 09s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/f575d68ffad64c0b88d8f915f343a4d1 : SUCCESS in 6m 23s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/0aa2e881e3b14eaca76cd727c0a6778b : SUCCESS in 8m 31s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/ba545a47ce6e452cb8cf2fd920f94ae8 : SUCCESS in 7m 16s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/a3d51cde29404fa79d06015f7e9b939c : SUCCESS in 5m 47s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/f1e43146b3eb4d44812070764b0dd98a : SUCCESS in 7m 01s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/fc548d21dcce4c45add503600646722c : SUCCESS in 5m 55s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/4e1c364eab0e42e588093df6813a60bd : SUCCESS in 6m 11s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/89fabaa3f2ae4dcaaf44ab4f1579cb07 : SUCCESS in 7m 14s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/cb62fd72011d4eb488fb68934b26f81c : SUCCESS in 6m 02s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/0f0d5e1a61a2415d85032b9ed35645bd : SUCCESS in 6m 08s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/427975adabfe4f4baa1e19688b69e95b : SUCCESS in 6m 12s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/b53834a7fc2d4a28861e85fa29444e00 : SUCCESS in 6m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/2911542376a74ba29573e595d094f329 : SUCCESS in 5m 56s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/0b91b7c95fe04ecb88429e3c17142096 : SUCCESS in 5m 12s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/e6abfa66f4404e528d7218629a586718 : SUCCESS in 5m 52s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/448c4b73cef54072b99ee352aa44e6bf : SUCCESS in 6m 14s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/5cdba44f3f89426e9300965432f0856a : SUCCESS in 5m 25s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/34057fe9d86c45a28d4698e21a485f04 : SUCCESS in 6m 09s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/0e1368075541422db7f94ac5feaa8584 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b32a685d09dbdd192307169ac257b0fbaa0bc072","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-09 13:46:04.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello","accounts_in_message":[],"_revision_number":1},{"id":"03672ef106bed2bcfbe76debb78723cf14f91fc6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-09 16:11:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7ac7d12696c58fee74ea26befdeef6d807a92501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-09 16:14:42.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1e583fbcfbe5887732e4bd954335d1644f1ab64f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:02:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92f732af57944bc2a7882078fae7a308 : SUCCESS in 1h 39m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a786e346cf0b4d519bb0760a8aa17ebe : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf0b94688a84443b322de00c5afc14a : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff98e1c6b78244b0b83b4f0564c6f6cf : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5261c11125a473b9d2b92fc26f8f7ea : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3944941b94c4597b3fbcd75f03bf55d : SUCCESS in 5m 47s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/3a26f95175a94184a7cda944f9018d10 : SUCCESS in 5m 37s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/5df37290941844ef90c81a6927c93d06 : SUCCESS in 5m 40s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/9c71a1df4cca4b7091fe636e0be024d7 : SUCCESS in 6m 16s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b121c300ac0d4f80a58bfb9865c05f23 : SUCCESS in 5m 49s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/243d95fea83f40d09fd7dd7fa3105bae : SUCCESS in 5m 49s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/0d96b409b26c4df8b5adf73ff9b4c73e : SUCCESS in 5m 30s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9f6e2bf1edbc49808c09c606a4df8cba : SUCCESS in 6m 23s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9b5eea71bd0342c6921581ac2efeeee2 : SUCCESS in 6m 28s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/9bf28a1245fe456cb8575e0b353c1226 : SUCCESS in 7m 15s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/44189fa4e1dc4c938071000aff17dc5a : SUCCESS in 6m 26s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/75d70af75d944c9ca03bb5995db65c18 : SUCCESS in 5m 51s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b3dd9bd21b664a178cdb2f78af0b882f : SUCCESS in 9m 42s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/74925932233b47f5be9ed1cf7986b6ee : SUCCESS in 6m 34s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/514d7f2ee7c14d3ba53f7706f3312ca6 : SUCCESS in 6m 24s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7790df8479fd419f8897cf9e67170d02 : SUCCESS in 7m 16s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/d4d3ff331b4d4d14868addf2078a1b1f : SUCCESS in 7m 07s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/d25ac35f75a040f59c89c14bb6bce760 : SUCCESS in 6m 15s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/1f19452a31d6406c9a2f4b2dbc50616f : SUCCESS in 7m 59s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/9d30eab2f4df41c2a16069316ec6e797 : SUCCESS in 5m 19s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/ab6a62015bfe43729643b6dead8fdabc : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"171ffd26dd57cbe626111b395729a2208c000f73","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-10 08:42:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9b8aa250dd4b04150ce51baaa715aaae503cd702","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-03-10 08:43:16.000000000","message":"Patch Set 2:\n\nThanks for details this is more crystal clear for me now","accounts_in_message":[],"_revision_number":2},{"id":"e08c295d5e66ee584f9a9dcbfd163cb6e130f772","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-10 11:20:05.000000000","message":"Patch Set 2:\n\nas a data point, this is the commit breaking tooz tests with later tenacity: https://github.com/jd/tenacity/commit/250b27e47712e42e096a68a3c3518da8de0e90ba","accounts_in_message":[],"_revision_number":2},{"id":"e249388cb817b0952ce67ac9d9d062becf3294c2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-16 17:00:09.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\nAs I see currently \u0027tenacity\u0027 is upper constrained in openstack(/requirements) as well [1]. So this patch is OK for now.\n\nThis issue also affects stable/victoria [2], stable/ussuri [3], stable/train [4] and maybe older branches, so needs to be backported.\n\n[1] https://opendev.org/openstack/requirements/src/commit/c4279a3522ef251b0426dd0a0a5dbc70ec009081/upper-constraints.txt#L480\n[2] http://lists.openstack.org/pipermail/openstack-stable-maint/2021-March/thread.html#33729\n[3] http://lists.openstack.org/pipermail/openstack-stable-maint/2021-March/thread.html#33650\n[4] http://lists.openstack.org/pipermail/openstack-stable-maint/2021-March/thread.html#33493","accounts_in_message":[],"_revision_number":2},{"id":"0a4107f18308c186997029beba1b5066556b2b3e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-29 05:56:48.000000000","message":"Patch Set 2:\n\nthank you for your reviews. This becomes a bit more pressing, it\u0027s blocking the gate, not just for tooz.","accounts_in_message":[],"_revision_number":2},{"id":"d41555cdcbcb571f342612c95cc92114070b7193","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-29 11:15:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"35a9421cd991b8cc07866f0845f32328e7b874be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 11:15:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14af71229abcdcaf195ab99655e1f210abedde98","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-03-29 11:16:13.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e thank you for your reviews. This becomes a bit more pressing, it\u0027s blocking the gate, not just for tooz.\n\nI\u0027ve pushed this through for now, but this shouldn\u0027t be blocking the gate if you\u0027re using upper-constraints, right? That will cap the dependency at whatever\u0027s stored in there","accounts_in_message":[],"_revision_number":2},{"id":"868bb543b17e5bb72973d4529a93be25d5508122","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-29 11:52:36.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e thank you for your reviews. This becomes a bit more pressing, it\u0027s blocking the gate, not just for tooz.\n\u003e \n\u003e I\u0027ve pushed this through for now, but this shouldn\u0027t be blocking the gate if you\u0027re using upper-constraints, right? That will cap the dependency at whatever\u0027s stored in there\n\nThank you. That is my understanding. But not all projects fropped using upper constraints, and https://review.opendev.org/c/openstack/tooz/+/778878 failed to pass until I added *this* here as dependency.","accounts_in_message":[],"_revision_number":2},{"id":"b477f2ce4da4e883f9c234e264da6b8349f94e84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 13:07:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5703707790f6411b8318ba13571ad29e : SUCCESS in 1h 46m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90bab76036aa4b52b1d0aa21a72f850f : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e3b1fed34a54b62b9e4b215f55ee5f5 : SUCCESS in 9m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d0ccf4b88714e35bd49aa5467b120a8 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fef8026e3874bc3ad8f4e036ebbe24d : SUCCESS in 9m 16s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/1ff3a2ae63ec48ab807abe12ab650993 : SUCCESS in 9m 18s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/78e803cb7a5544a0af9ddeaef633399a : SUCCESS in 8m 26s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/8732b50b38d243209aab60677c63756a : SUCCESS in 5m 13s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/6ea0b87d3e7e48958b1ac6fc633eb575 : SUCCESS in 9m 24s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7f04c5f5b33c4eecbb1a47c4bbd7b72d : SUCCESS in 9m 03s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/aa6c52894fb348b6ad1b604fbf2b905a : SUCCESS in 8m 41s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/13995150ceae4511a9f4bace78243e59 : SUCCESS in 6m 53s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/765cfae9f4b24e9785a26f0e6590f38b : SUCCESS in 5m 19s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/bbca81b63ccb48e2a75882dc1a27a7ca : SUCCESS in 5m 27s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3d9fd292c1664e7e989bffce301845e4 : SUCCESS in 8m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/ec16d5980a8e49adab63e79524e0d6fa : SUCCESS in 6m 41s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/7a5d4c5a6901446e8a8447266c2c83cf : SUCCESS in 6m 43s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/3f9a55fb9a3c44ae972d80d800003c70 : SUCCESS in 5m 29s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/e533be7a7e8649e280492e76884e4fcd : SUCCESS in 7m 49s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/91e42ef02ab3473cb38f5bcdf9afd204 : SUCCESS in 9m 50s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/08f7deb707c34d1fa3be86774ae7c986 : SUCCESS in 5m 14s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/5079840e6f604af3859112e03df2ff00 : SUCCESS in 9m 54s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a9545259ddbf48709ae6841ca3e12ea5 : SUCCESS in 7m 50s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/204bdbb7aafe413aa506d89e086da238 : SUCCESS in 6m 30s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/b718f4d3666b43bf880539904c3c1cea : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":2},{"id":"468c220fe345e46c920e3edffff0092f60c7939c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 13:08:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"df6fe5f76fcb27887344465f94a243cddc7b9dba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 13:09:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04242b13c0944e1ba96f8bbda9e05efc : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f2005df3d6565e050e4ca710df4eda518353b9c2","revisions":{"0a4c7e431c670d8e59db9ae0a46cdf9197f146c8":{"kind":"REWORK","_number":1,"created":"2021-03-08 07:24:33.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/61/779161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/779161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/779161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/779161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/779161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/779161/1"}}},"commit":{"parents":[{"commit":"fa2b6cc7aecd71677ba6c244abc6954aae48cd2a","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fa2b6cc7aecd71677ba6c244abc6954aae48cd2a"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-08 07:24:14.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-08 07:24:14.000000000","tz":60},"subject":"Cap tooz to unblock the gate","message":"Cap tooz to unblock the gate\n\nChange-Id: I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a4c7e431c670d8e59db9ae0a46cdf9197f146c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a4c7e431c670d8e59db9ae0a46cdf9197f146c8"}]},"branch":"refs/heads/master"},"f2005df3d6565e050e4ca710df4eda518353b9c2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-09 16:14:42.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/61/779161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/779161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/779161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/779161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/779161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/779161/2"}}},"commit":{"parents":[{"commit":"fa2b6cc7aecd71677ba6c244abc6954aae48cd2a","subject":"Merge \"Use py3 as the default runtime for tox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fa2b6cc7aecd71677ba6c244abc6954aae48cd2a"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-08 07:24:14.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-09 16:14:24.000000000","tz":0},"subject":"Cap tenacity to unblock the gate","message":"Cap tenacity to unblock the gate\n\nRecent version 7 changed retry and dropped the\nretry_state parameter.\n\nChange-Id: I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f2005df3d6565e050e4ca710df4eda518353b9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f2005df3d6565e050e4ca710df4eda518353b9c2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
