)]}'
{"id":"openstack%2Ftooz~784036","triplet_id":"openstack%2Ftooz~stable%2Fwallaby~I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f","project":"openstack/tooz","branch":"stable/wallaby","topic":"tenacity_wallaby","hashtags":[],"change_id":"I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f","subject":"Cap tenacity to unblock the gate","status":"MERGED","created":"2021-03-31 08:11:40.000000000","updated":"2021-04-16 13:02:47.000000000","submitted":"2021-04-16 12:56:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"784036-tenacity_wallaby","cherry_pick_of_change":779161,"cherry_pick_of_patch_set":2,"meta_rev_id":"67b7f1ee11316a8ec0a72c81379975010be96bed","_number":784036,"virtual_id_number":784036,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 12:56:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-04-01 14:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-04-16 11:10:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-04-12 07:06:39.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-04-16 11:10:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-31 08:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-31 13:47:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-04-12 07:06:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-16 11:10:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f1f9e9c36b5fe001d9b9ed3fcd2e64aad0abff70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-31 08:11:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f4d7d55b3bcc3d52d6de81a50797b5be7d8dd507","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-31 08:15:35.000000000","message":"Patch Set 1:\n\nThis is tooz failing the stable maint check for ussuri (and victoria, stein, train):\nhttps://zuul.opendev.org/t/openstack/build/95ae416b472b46cab4739b08a5eb2638\n\nupper constraints are probably not checked.","accounts_in_message":[],"_revision_number":1},{"id":"e1f3aa60ed7503757123d8235aa7ced9656471d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 08:48:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d422c8ef03944f3dae93b654d4118b4a : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38db261fa93740a3b007bd2fd38d328d : FAILURE in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e0cbe06872747bba499649379b24860 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa9f509392d041029c7068a0583cb8fd : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56eca93b71a0403bbb677d7bb2430032 : SUCCESS in 6m 24s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/9d1d92e096634226bdeab17d19e55e72 : SUCCESS in 6m 51s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/de21700406ec4c4e93fa5dfc37dbe9d8 : SUCCESS in 5m 41s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/e41c60f44748452ea6ccc1dee8cd1068 : SUCCESS in 4m 59s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/288fef1b7e5446369052348615085050 : SUCCESS in 5m 21s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/37c0b5d28a8749898e67888dcb85a611 : SUCCESS in 5m 24s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/816aefa47b6541e296fe5ec5c549c909 : SUCCESS in 6m 21s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/58f96904a2c9449781efe251e9cebc40 : SUCCESS in 6m 14s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/7c7ff4c5c4c946038d063f5ae67494d4 : SUCCESS in 6m 02s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/af1e9c69dd13445c98902f44904a2eb9 : SUCCESS in 6m 24s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/c4a346d186ad40879a940531f82db71c : SUCCESS in 6m 00s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/dc290b5e313140a78f604a307b496b5a : SUCCESS in 5m 24s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ad1d0af01e2847c2b35594efe8a39682 : SUCCESS in 5m 15s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/cdd6a44334f847679ce79f571ae7c321 : SUCCESS in 6m 50s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/cb7b340e4799440c8cef42fe205d29c7 : SUCCESS in 6m 11s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/4e22844b16e045938e86b050e9b32138 : SUCCESS in 5m 50s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/2f46d6045afb476a97adb28bf38d13c5 : SUCCESS in 5m 41s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/226e8930d8844a26b21126c2a81f1b2c : SUCCESS in 6m 03s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/ab4f4cc168504cc285aa81bada9a3aa0 : SUCCESS in 6m 10s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/c5f7ef9576d3411e8f8a691bc6432617 : SUCCESS in 7m 19s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/b38ebf4e44034b3fa6202fdfe71daf89 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cf75bc13bb414be0abdc066fa9c669818927999c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-31 09:27:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b1b0ec7d301f807b6c3b8b70b4a4a825267f414","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-31 09:30:40.000000000","message":"Patch Set 2:\n\nThis includes https://review.opendev.org/c/openstack/tooz/+/781424 , which failed because of tenacity not being capped.","accounts_in_message":[],"_revision_number":2},{"id":"dc93b01cfbcffa4a5009758fd0bed59a978e5ec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 09:49:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca867db6e9d045978cc733904a12f583 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63384b7b717b426baa2b864a075134be : FAILURE in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/beb22cb377024249ada9e2dd1f696373 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8021017334be495491fd8e1f16968c45 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e35106b0b83a4b7cb3ce940a2ef278bd : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c3f38ee6174aeea98862a27543719e : SUCCESS in 5m 05s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/78af5c1a56fc489d90bf010b6ba1484b : SUCCESS in 5m 44s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/216d667aeeb64a0bb8ddf97d24c1fca6 : SUCCESS in 7m 31s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/59f9b894550d4b968604114a2bcced41 : SUCCESS in 5m 36s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/024289c606664d17996769283211d1d8 : SUCCESS in 6m 02s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/88bb398e698148ea94098c61b81dbd21 : SUCCESS in 5m 23s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/60649a5d8f8b409fbbec427d78476edc : SUCCESS in 6m 02s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/2628bf69b7f54a60b997217f8cadf007 : SUCCESS in 6m 42s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/70b94539125742878a391b30c3cf2d95 : SUCCESS in 5m 30s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/44404dc2b1c54a12b7a0cef6ad904e59 : SUCCESS in 5m 14s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/4064d287944c4793a33de9ce12efdd18 : SUCCESS in 6m 26s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/464bc6e16af343b98b40a630d4d7fd26 : SUCCESS in 5m 55s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/16b3e04f6f0848a9b1d6c4f698d9da7d : SUCCESS in 6m 36s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/55c52b53a7f940758afe175f1632e6b3 : SUCCESS in 6m 14s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/10818a78926b4f519c8d9e13ad613911 : SUCCESS in 6m 00s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c9b25ca939664009b239e2f0bba62538 : SUCCESS in 7m 24s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/2d889b507cd44b9a8f41e6d527dc79a8 : SUCCESS in 7m 19s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0166c5ff86db441f8e344ab4c0d30aad : SUCCESS in 5m 06s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/cc880251e6be44ee85d895a3db06459c : SUCCESS in 4m 57s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/cfb72a149d394c80ac82bee37092749d : SUCCESS in 5m 11s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/3b4e5c8f5e744530992e6eaf1b13a7d8 : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":2},{"id":"9199788fecaf98ca2db0452886244be558aa489f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-31 13:47:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThere is a flake8 version conflict, see my comment inline.","accounts_in_message":[],"_revision_number":2},{"id":"23d16610a23a3cdfd36aa3f70188882bfbae15cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-31 14:09:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"75c4c551d34b34c1cd1b144066bf5ca3e9b91285","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-03-31 14:09:30.000000000","message":"Patch Set 3:\n\nthank you, will do.","accounts_in_message":[],"_revision_number":3},{"id":"88cdb6da164c318404808bdbf8b1fe2a564998ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 14:40:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4314e05f40394f52b769660634929cb7 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99164b9aaeab4c9cbbbdc63300b224b6 : FAILURE in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99991a99ecee475f803844da2ab7e17c : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf7dda3b33514fbba0c2c6683db589fd : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5300a993241c44cfbaef72092fee5891 : SUCCESS in 7m 19s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/71c8657652d3440084e234279992866e : SUCCESS in 5m 36s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/49d1dd7fb4074f5485117d68ba6e3a16 : SUCCESS in 5m 34s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/247199039c7040fd9e78cb92c4e9261d : SUCCESS in 6m 04s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/7fca23ba33784113a5ad239e58fda780 : SUCCESS in 6m 07s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/0595e48be6384fcfb7df57383302431f : SUCCESS in 5m 22s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/d4debf292c714375bc3aace5ab6e399d : SUCCESS in 7m 21s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/7c8d4356f0f74f9da378951396e58536 : SUCCESS in 6m 41s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/41a9e8e0531842de99574f4a8521f0b9 : SUCCESS in 6m 52s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f61aca9eb8f34c6cbf8fc45705ffff9d : SUCCESS in 5m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/4c9ccd11c0364280a775fac58cd65a2e : SUCCESS in 5m 43s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/cd8c76cfda8c41bca08b82dc6a00641e : SUCCESS in 5m 09s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/85977e91e6a14cb9a9843a4ed3f73243 : SUCCESS in 5m 37s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/d33a5858827f49bc9235d57c0fbe3118 : SUCCESS in 6m 31s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/bb10a27a0d2541d59d2d85db6724cbfa : SUCCESS in 5m 34s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/cc37241cff87414189b3ba1a773294ab : SUCCESS in 6m 11s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/8b123f636e8c48deb3a7f44dc50a8267 : SUCCESS in 6m 01s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/915c12b05c29493483d7b6a83eb10f5f : SUCCESS in 5m 31s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/d7b47fbca7cb4f45af088a3aa3c77ef4 : SUCCESS in 5m 26s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/5101cb46efd443edbc0fb249868105ba : SUCCESS in 5m 32s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/7fd63bc9329f441093468b028b9468dd : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":3},{"id":"01d3431ebf07c29e0d5f4837884736b4e490a6a8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-31 17:00:53.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nFor me the pep8 tox target locally passes if I keep the original hacking constraints (\u003c3.1,\u003e\u003d3.0). Is there a reason not to keep that?","accounts_in_message":[],"_revision_number":3},{"id":"ab01f262c0182cfed5c6433599b27f62eb83bc6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-04-01 10:04:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e743641a27bbee4162174860d74a1a537e9259b5","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-04-01 10:07:08.000000000","message":"Patch Set 3:\n\nPep8 passes for me as well, when running local. I just started to change it, since the used hacking version in tox.ini and in pre-commit conflict; this is fine in master, but not in wallaby branch.","accounts_in_message":[],"_revision_number":3},{"id":"62d05463881d985d0a4f2e1a8207d2bded64ed78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:23:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54609e9f1ec9409c9cfee9491e62b6c0 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e91b2b2f6d9a4e119fe7793cfc8a0fb8 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b4b57b3ff5f413a9f5cf1b3d3510670 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e0b3cfed0134551a61f8119ab1a8a30 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63357dcbab5b40408d73898ddddb8d86 : SUCCESS in 5m 33s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/f3e172714385472eb8adb41cd050c658 : SUCCESS in 6m 05s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/bb633a1a351d43acb5f10e917cf5c1e4 : SUCCESS in 6m 18s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/0c77e786b030487281669a92b875dacc : SUCCESS in 6m 36s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/4e4e1938bb19498494fa629ece5c6054 : SUCCESS in 6m 00s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/cedff23785004166b4f666672fd79d70 : SUCCESS in 5m 46s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/9ab5fb2106994e4581911c8c4f864ed2 : SUCCESS in 5m 58s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9416ad19b4334d45b3dc16ffa3a6f0d3 : SUCCESS in 6m 14s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/7016c465fdac44cb9924b14c757f8117 : SUCCESS in 7m 19s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/52fdd4809bf54a6ca91064cd447dc9cf : SUCCESS in 7m 37s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/f869f0fd546c4829a2bcd1bdf41b59c5 : SUCCESS in 7m 30s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/26d82653cad3419db3047b7911ab4a3c : SUCCESS in 6m 39s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/4d86d07053f24fe8a2cdd788d9bfb918 : SUCCESS in 5m 33s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/a5aa216f0c534aa782ca79701ec056e1 : SUCCESS in 5m 52s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/23799f2a02c448eeadffd7b47d9d6dd3 : SUCCESS in 5m 40s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/674d7e670cb94395a3bbac4859c67097 : SUCCESS in 8m 02s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/3b468685169c47cc87559558f137637e : SUCCESS in 5m 53s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/cc1a7865b5e6466982c5472feffde1a0 : SUCCESS in 9m 40s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/6a92b023f39a4481a13403b9aa48f5f3 : SUCCESS in 7m 21s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/4a439b85ac914806aa5abf04d7ea883d : SUCCESS in 6m 25s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/1fd0905e33f8438eb8fb0d828370ab7e : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":4},{"id":"d83aadbeaac84b68228418e912ba09292ed4fba4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-01 14:43:37.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks for the update. LGTM and Zuul also likes it :)","accounts_in_message":[],"_revision_number":4},{"id":"44a7eb56cfa904611b6097238a2df7e2452e699b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-12 07:06:39.000000000","message":"Patch Set 4: Code-Review+1\n\n+1 for the rationale behind these changes, however, I think that this patch (https://review.opendev.org/c/openstack/tooz/+/782457)  should pass first.\n\nIt will avoid to downgrade the flake8 version","accounts_in_message":[],"_revision_number":4},{"id":"387bdec6e76333ab329dbc58d7cafad951d6b3ae","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-04-12 16:03:47.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review+1\n\u003e \n\u003e +1 for the rationale behind these changes, however, I think that this patch (https://review.opendev.org/c/openstack/tooz/+/782457)  should pass first.\n\u003e \n\u003e It will avoid to downgrade the flake8 version\n\nActually, flake8 won\u0027t be downgraded as hacking\u003c3.1.0 requires flake8\u003c3.8 already, and hacking is constrained in tox.ini and in the precommit config too. I don\u0027t even know how precommit passed in the past with this contradicting constraints :S","accounts_in_message":[],"_revision_number":4},{"id":"ece9037a59117a7bccd758fc701ba36b1f929a31","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-16 11:10:24.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"2162da92264d687e85fe1bb89e01b700b35ce345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 11:11:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"82273697701cf04ef9549e5fce8fa2aa2f74acf7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:56:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df839b3bc19a48a9a7d2b831c6afd932 : SUCCESS in 1h 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/909577c6c1d84a8c9e51ed44bde114d0 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0aab6a581784b018670d6565399abbd : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ab1f3f71cb145279cc4f25822f05c74 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53a388f467148aa9f728f97502e09ec : SUCCESS in 5m 52s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/a457ed85352c487aaedf2ce0ed412ff3 : SUCCESS in 5m 38s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/11602690ce4944a185ec70c3e3f69d91 : SUCCESS in 5m 19s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/1dc1b8f6ea4b4fc5ae4db5e634ca9fb8 : SUCCESS in 7m 11s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/5a9b45c4b331459fbf521af4ad01ce05 : SUCCESS in 6m 14s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d23c76a074fc4bc0ab437ef799b2ac32 : SUCCESS in 5m 04s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/9abace6b3a0849ce8773be70ac8f770a : SUCCESS in 5m 23s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/92abd11454a24320bb9c49f51fcb3040 : SUCCESS in 5m 55s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/7fe34838ca634d3c806d9351c7798930 : SUCCESS in 4m 58s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/59c391f9cdd64828a58f7bb103aa64cf : SUCCESS in 5m 40s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/bfac0c81f3aa4b37a54606ad7c0e0af6 : SUCCESS in 5m 39s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/f6afd7718c3f44168aab6bd9f9f53e55 : SUCCESS in 5m 56s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ea1c85e1d9664348ba219e9da3b01f40 : SUCCESS in 5m 00s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/8341a414b842411b8f0cee2ca35b155b : SUCCESS in 6m 58s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/90c67bc230c6482cb2bbc23789ea7785 : SUCCESS in 5m 09s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7ca9ba6169354f48989bb0e269bd3b98 : SUCCESS in 4m 56s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/2aaf4ce1aa8e44afbfb7862ea574f75e : SUCCESS in 5m 37s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0f93beeaf8c44e7482641c93acc2e359 : SUCCESS in 5m 09s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/f0623bed089e49b5b2d46c15976c8f48 : SUCCESS in 5m 08s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/0a60d41885f8442dba4dcaf9e8bcec53 : SUCCESS in 5m 29s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/9b28ab5708784fa9b466bd8fe09a82ac : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":4},{"id":"d676de70cd57aeaf608c5b5e907eb647f10b74a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 12:56:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"67b7f1ee11316a8ec0a72c81379975010be96bed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 13:02:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1999e5d5e3a74a2a93968ef5cfa1ce9c : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fe9c4304551cce0ac85366965009ad102fd9610a","revisions":{"005eab82c477e07a0f81d59070edec62d68fdc48":{"kind":"REWORK","_number":1,"created":"2021-03-31 08:11:40.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/36/784036/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/36/784036/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/36/784036/1"}}},"commit":{"parents":[{"commit":"54448e9d8b968c6d537a4a3b648ed280174b4f2c","subject":"Replace md5 with oslo version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/54448e9d8b968c6d537a4a3b648ed280174b4f2c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-08 07:24:14.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-31 08:11:40.000000000","tz":0},"subject":"Cap tenacity to unblock the gate","message":"Cap tenacity to unblock the gate\n\nRecent version 7 changed retry and dropped the\nretry_state parameter.\n\nChange-Id: I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f\n(cherry picked from commit f2005df3d6565e050e4ca710df4eda518353b9c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/005eab82c477e07a0f81d59070edec62d68fdc48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/005eab82c477e07a0f81d59070edec62d68fdc48"}]},"branch":"refs/heads/stable/wallaby"},"39de93994885c6e6fb75e2c8732fa66e0c598c85":{"kind":"REWORK","_number":2,"created":"2021-03-31 09:27:43.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/36/784036/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/36/784036/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/36/784036/2"}}},"commit":{"parents":[{"commit":"54448e9d8b968c6d537a4a3b648ed280174b4f2c","subject":"Replace md5 with oslo version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/54448e9d8b968c6d537a4a3b648ed280174b4f2c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-08 07:24:14.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-31 09:26:15.000000000","tz":120},"subject":"Cap tenacity to unblock the gate","message":"Cap tenacity to unblock the gate\n\nRecent version 7 changed retry and dropped the\nretry_state parameter.\n\nAlso fixed a conflict between pre-commit hook and tox.ini where\nhacking was just conflicting\n\nAlso fixed .gitreview to point to stable/wallaby for wallaby branch.\n\nChange-Id: I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f\n(cherry picked from commit f2005df3d6565e050e4ca710df4eda518353b9c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/39de93994885c6e6fb75e2c8732fa66e0c598c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/39de93994885c6e6fb75e2c8732fa66e0c598c85"}]},"branch":"refs/heads/stable/wallaby"},"7a8a7fb7ca0dde8a0241b09ac9351a2d06bd2854":{"kind":"REWORK","_number":3,"created":"2021-03-31 14:09:11.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/36/784036/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/36/784036/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/36/784036/3"}}},"commit":{"parents":[{"commit":"54448e9d8b968c6d537a4a3b648ed280174b4f2c","subject":"Replace md5 with oslo version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/54448e9d8b968c6d537a4a3b648ed280174b4f2c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-08 07:24:14.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-31 14:09:03.000000000","tz":120},"subject":"Cap tenacity to unblock the gate","message":"Cap tenacity to unblock the gate\n\nRecent version 7 changed retry and dropped the\nretry_state parameter.\n\nAlso fixed a conflict between pre-commit hook and tox.ini where\nhacking was just conflicting\n\nAlso fixed .gitreview to point to stable/wallaby for wallaby branch.\n\nChange-Id: I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f\n(cherry picked from commit f2005df3d6565e050e4ca710df4eda518353b9c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7a8a7fb7ca0dde8a0241b09ac9351a2d06bd2854"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7a8a7fb7ca0dde8a0241b09ac9351a2d06bd2854"}]},"branch":"refs/heads/stable/wallaby"},"fe9c4304551cce0ac85366965009ad102fd9610a":{"kind":"REWORK","_number":4,"created":"2021-04-01 10:04:54.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/36/784036/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/36/784036/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/36/784036/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/36/784036/4"}}},"commit":{"parents":[{"commit":"54448e9d8b968c6d537a4a3b648ed280174b4f2c","subject":"Replace md5 with oslo version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/54448e9d8b968c6d537a4a3b648ed280174b4f2c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-08 07:24:14.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-04-01 10:03:59.000000000","tz":120},"subject":"Cap tenacity to unblock the gate","message":"Cap tenacity to unblock the gate\n\nRecent version 7 changed retry and dropped the\nretry_state parameter.\n\nAlso fixed a conflict between pre-commit hook and tox.ini where\nhacking was just conflicting\n\nAlso fixed .gitreview to point to stable/wallaby for wallaby branch.\n\nChange-Id: I109d3b5770c3cd8fe0bdeae78517f4d36ba3b65f\n(cherry picked from commit f2005df3d6565e050e4ca710df4eda518353b9c2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fe9c4304551cce0ac85366965009ad102fd9610a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fe9c4304551cce0ac85366965009ad102fd9610a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
