)]}'
{"id":"openstack%2Ftooz~792140","triplet_id":"openstack%2Ftooz~stable%2Ftrain~Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b","project":"openstack/tooz","branch":"stable/train","topic":"bug/1980079","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-05 14:54:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-07-05 16:32:26.000000000","reason":"Change was submitted"},"5241":{"account":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"last_update":"2022-07-05 16:32:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b","subject":"Retry on redis connection errors","status":"MERGED","created":"2021-05-19 13:04:23.000000000","updated":"2022-07-05 16:32:26.000000000","submitted":"2022-07-05 16:32:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"792140-bug/1980079","meta_rev_id":"a88c56aec009b792ee4ee6c1fa8b90e854e6272d","_number":792140,"virtual_id_number":792140,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 16:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-07-05 13:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-05 14:54:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 13:45:58.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-05-19 13:45:58.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"},{"updated":"2021-05-19 13:45:58.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-05-19 13:45:58.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-19 15:06:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-08 13:16:02.000000000","updated_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"CC"}],"messages":[{"id":"391fda1f02c6f586983a59678a56238eac2460e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-19 13:04:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bfbbc7548179760e4f39eb5db14ca200f9844ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 15:06:04.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/67fb70b3340d4580a7d8e33a685b7605 : SUCCESS in 1h 56m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9892c8cb470d4103be95098a7dd2ad5a : SUCCESS in 1h 47m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ff0ae88a7bb44359714befe35943c20 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/422e1cfb40bb43a19479c84161a22548 : FAILURE in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/18009db4da0f414e872d21d7165b8be5 : FAILURE in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1749187a33ae4ed284e019cee497a171 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/334e5c5b7f6b4f5187334aee8de19871 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781dc1bfb11944eba45bca2771c9d91e : SUCCESS in 4m 59s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/b67c808d5ded469aafc5db04deddcbbb : FAILURE in 4m 01s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/06d03ae4fe8c49fdb510ed5bfab6a898 : SUCCESS in 5m 10s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/4766afcaf0074fd5b3183f41264d40e8 : FAILURE in 4m 08s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/16f1e37dfa7f48288193a5c2ceb27326 : SUCCESS in 5m 03s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/f9dd22867a9d4e6e8b79c9d42643c290 : FAILURE in 3m 42s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8da8f60053ab4638b1f60ab5b72e151c : FAILURE in 4m 57s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/05eba538e2694b2c833a294295dc027c : FAILURE in 4m 12s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/f90be53e08584c52a1fdf34f2db3fc71 : SUCCESS in 4m 54s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/e062179d517b43008bc807f3dcb21816 : FAILURE in 4m 19s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/d386dedfa99344f889ac5de1212b9e7e : SUCCESS in 5m 13s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/9cc23ce4386e489ba3d8db39eda54121 : FAILURE in 4m 26s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/3116ef1bfd67479ab2e3e7990b5b238f : SUCCESS in 5m 00s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/14b3572bd93d499394710784a48e11ea : FAILURE in 7m 09s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/3b58d6b9cd46474fa0c77f5432d17d31 : SUCCESS in 5m 07s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/f5374b51a2c24008b377d37aa0173cdf : FAILURE in 4m 44s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/6ab130d30f594c5a81de5048cd1a5098 : SUCCESS in 4m 37s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/b2182b3bd12f49a299c5549c5e484a2c : FAILURE in 4m 22s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/3f4503440e5b41829bdee476d9e34368 : FAILURE in 5m 12s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/421d5d58e8e041038a0d958d221213a1 : FAILURE in 9m 28s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/1e134f8473824b08a5b6618552f643ac : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"95c47707cfa42bd40e0a055e6f62cf25e29ecc22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-19 15:38:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"231a18b8ea904bc72365f319ce5b58e8c09af9b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:23:59.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/15fa137ee1964cc6bf18773a4999a299 : SUCCESS in 2m 32s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ae7afd0cf53a4fddbe61e4eae844becb : SUCCESS in 1h 39m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f0864ad735b409e891c2aa36bc4eb2f : SUCCESS in 1h 30m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f8b6e976104bae8a6947ab3d600a22 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547a0559e61147448e9aaac7f9c9e7b6 : FAILURE in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6d493f35106422eb9a512c809a7153e : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24c1ae97956f410da271497549f96f12 : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a8dc1c128df4d9b9bcfa0c248f8f842 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2937c6e1e49342c7916441a66a504e16 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c6b09ddf09e4bde8d7a93d93f32f67a : SUCCESS in 6m 09s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/74bff88a4d8e4860a9906a018d5e83c2 : SUCCESS in 6m 03s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/9f59809ccdbd42b5967255e9713ceaac : SUCCESS in 5m 00s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/ab85f1bc31c742978d59185bb90efe07 : SUCCESS in 5m 30s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/1fa28c8dd0214324ac1d25c243e174fd : SUCCESS in 5m 24s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/63af58090a4a4d418b4ae1937603c2d6 : FAILURE in 4m 48s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/43d97cc038cc4c93b650423b305345ed : FAILURE in 5m 02s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/72788e6612224236868604661af99701 : SUCCESS in 5m 32s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/048232c0d7fc4711aa12b0c30bf815b3 : SUCCESS in 5m 24s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/7c8e359d5ace416cab4fe22314422b8f : SUCCESS in 5m 27s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/9ad68589d67745eaa70190c2df6b9313 : SUCCESS in 5m 26s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/0cd227603cea4a1f956908a60f068fa9 : SUCCESS in 5m 15s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/4a1872aaa8df45029de929d339add4d5 : SUCCESS in 5m 23s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/9cb9de660c784a2f886083172344a5c9 : SUCCESS in 4m 54s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/9b0d021632514db585800260f087f1d2 : SUCCESS in 6m 02s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/4d0b31d63b134e4597402c0172e2cb9c : SUCCESS in 4m 58s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/3bf0d0f5bce344c79ceebfc6ee9718d2 : SUCCESS in 5m 54s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/4107e8caad2a4f00aca960c698fdac52 : FAILURE in 6m 20s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/45354c485c874f4287047dc2cc8d025d : FAILURE in 5m 03s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/0df86ec2f9e04f68a49ea8fd54994a23 : FAILURE in 6m 58s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/dccc8af6818041138dc0176c494c5e42 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":2},{"id":"51e14124206e2b3d88d4b60009ef6dd824722c00","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2021-11-08 13:16:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c539d396aa3bdf617154194c01ebcbef84cf53d6","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-11-08 13:25:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c199ada410fad99bbe7be88b73e49fec42012ac9","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2021-11-08 15:33:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a67ab3c8cc980325d216e0881f57886cf3dcc480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 21:14:50.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/dc5fd1954861414595cbd9af06a34e19 : SUCCESS in 3m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/4e2a96898f6b4a9486569a7b91748dff : SUCCESS in 1h 42m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71a3f40b6f3e4f4d90b7fd39dce127d8 : SUCCESS in 1h 34m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f77fd113f8848fdba3fff4bd90111d7 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b92dd227455640c9b610ca9fb0b2b4b3 : FAILURE in 6m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8462533c6fa94880b6afe2f6ead84e6f : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd7585d400a94ecea18ddb905ee13dce : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85d2adc50ec54309a0dbdb48480b01a1 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99669e42eae437b8c07bd72df0eb844 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7eb48f6b203d453e804ea9d6a20ef8a8 : SUCCESS in 6m 12s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/a82fae70bf6d4108bfa57a567a2088fd : SUCCESS in 5m 42s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/5eb76ebe57a94b7189f8b9623bc4a66d : SUCCESS in 6m 24s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/a13eed8de7d5417e8c23e977c73c4208 : SUCCESS in 7m 09s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b4012aada6844b6b9e7b05dff895051c : SUCCESS in 6m 35s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/e60ac5c0b16a4e5c947d328703fd56c3 : FAILURE in 6m 27s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8cdf1809fcc6465d9f0aa29ed482ff77 : SUCCESS in 5m 48s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/3a27624d13fc47a09f14ac72e10a6e40 : SUCCESS in 6m 36s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/c3ab80bddd794d398ef9545200cf2a4a : SUCCESS in 6m 06s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/a57d2685e08040a1b591844cf2eccf17 : SUCCESS in 6m 36s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3191e6ca228d44f8b0eec73c688672f8 : SUCCESS in 7m 37s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/eea1ee9d84ff4a6da71493b245e3277d : SUCCESS in 7m 20s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b492cebdf09848448ff9a6be53cbe1c6 : SUCCESS in 7m 25s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/5579775a217d4800b2f13b5a8ecb00ed : SUCCESS in 7m 50s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b96adba2f22d4ee6aea7fc4833be6299 : SUCCESS in 7m 26s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/5557830e9b274c1d8ba211a89cee1a5d : SUCCESS in 7m 42s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/e53f1b0ced0b4cb0bef775470dac645e : SUCCESS in 7m 11s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/86ea037d03c34f3a8092ebc2674ead81 : FAILURE in 7m 36s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/77c7c12e56034c5899764da48042a964 : FAILURE in 6m 59s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/de663878146f4b9692a74860cffd50e3 : FAILURE in 6m 40s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/82d91f8fb48a451ebe749f35c32db618 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":2},{"id":"7f3f22d2f731bc503388654a559f71c6c1f90293","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2021-11-08 21:54:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a5d4cce0393c345bce8d6666ea3fa3d9985505a","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-11-17 08:56:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3566dbd422cb8cb91110df79e665d229d41a4ce6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 16:54:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"290350fb62bdfed66e8776778a627dcd67c2cbdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 17:09:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9760ff387ce0dfd61c4a43036758cb818cfe4237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:42:53.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/feb07c0bfecd491fa781e8ee69f1278e\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/9ff4c1891aeb4420bf48d63b5ecf7193 : SUCCESS in 1h 27m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3c0ff57424d47a89d5c9648324c76c7 : SUCCESS in 1h 28m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25838597ccf4497380c3b7430750ea24 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f84b13651e44b7bab2676c24aa2a7d : FAILURE in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ef326bbdaea427eb48bf60c6690564f : FAILURE in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b31cb7d5b8b492d85dce46e46bd327c : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bdc9a6b7ab1441ca865d25d0a43efc39 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab4b13dd61d4e45a5b1c77f9a8b56f5 : SUCCESS in 5m 14s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/c6e9f381e2bd4413a6db067ec6a1f662 : FAILURE in 3m 59s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/21ca13a6290c4d43bcefb7066c34294c : SUCCESS in 4m 50s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/294aca82b3514b999e6d4f53b3432fb8 : FAILURE in 3m 58s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/683af78773854274ad6c7e918e9edf46 : SUCCESS in 4m 25s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/bb350c67e7a44d27b6a3df3c0c100e08 : FAILURE in 3m 34s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/16f064446ad34a0c842a2e290f9f73a8 : SUCCESS in 4m 08s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/25eeabc7e32b4e47a8bedd4ffebff1db : FAILURE in 3m 57s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/e3b45349acaf44b6a973bf1bc35dca0e : SUCCESS in 4m 08s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/c44e86ca4b17445fb89b384df9c4c124 : FAILURE in 4m 28s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/7d7884699db34224b4b54c807767f39f : SUCCESS in 5m 23s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/ff36e755c34549668c75300e4e4f414c : FAILURE in 4m 47s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/4d5c926bd6ca472d846ec60c535c9ca5 : SUCCESS in 5m 18s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/315145c9810649cc81a9c770d9c93a64 : FAILURE in 4m 38s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/bbf8e03b22e74ef3bb8c75374df330eb : SUCCESS in 4m 52s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/a4823e0a3abc4f5c9b25e8beb789a83c : FAILURE in 4m 05s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/07bb6b4cbbeb4ab7af4a7ed766aecd72 : SUCCESS in 5m 22s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/e45cf2b2b7cc494ea524fe1a0cc0cf8c : FAILURE in 4m 23s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/3e848f8ab9a0497c9a4b6a934b57dcf7 : FAILURE in 4m 51s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/ee08ae54ba2a4348a90fff305f192438 : FAILURE in 4m 57s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/eb4b7c60319644ac9de945b7e0f6c37b : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":4},{"id":"484c79cd6cee64525bc8b04aa1fe7c7d5bbdc478","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:00:15.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e6c7fcaa96aecae040a213404493999dfa18f49e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:01:13.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e17798654ab407865d62fe8efb83c35910e5c300","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 11:49:46.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f84bcdf705b423b82c59e47029ca50a\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/c21d34c5feb1478cbcfb96f68f2b4072 : SUCCESS in 1h 31m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f30c2c8bf5747c2b1637f0c5636aa8e : SUCCESS in 1h 10m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abcca54912af4a4b85ff14dd2b990ceb : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94896ada55054e4bb89d33d0f57de640 : FAILURE in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fccd9085c4e4fecb3f8545544ec0fe8 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69a1ad28a0374a52a72adcc46ffcecf9 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bea4c9332eb47e3b0760db29e821415 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9b53365707c44999f4b6284ab6c6072 : SUCCESS in 4m 29s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/b442f2498c304e5ca99c2a3a9fdffbb6 : SUCCESS in 4m 25s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/567caf1ea1d344cfaa67a802c64c2427 : SUCCESS in 5m 21s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/f08aaf377ea7418da260ea95293ceb07 : SUCCESS in 4m 43s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/7ef3caa510f741a2947ae4ac6215565b : SUCCESS in 5m 24s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/b65b05d9dcc44632bfdd907aa2ff6aaa : SUCCESS in 4m 15s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/6d62c8d957d64d2d896677a2b28a7322 : SUCCESS in 4m 47s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/3388e24971a54097ad3988b473f43c7a : SUCCESS in 5m 23s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/5c980e9482e6436db3438f767f23aae8 : SUCCESS in 5m 36s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/2a0a6d571ab74c6182d01435ee8f52e2 : SUCCESS in 4m 45s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/2cc1e9e912504d6e8e4ec21a19e00da4 : SUCCESS in 5m 11s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/f2070da0efb0412fbc8932bff49774dd : SUCCESS in 4m 24s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/7e6f535a702e4c0396d66b94add8a3aa : SUCCESS in 5m 32s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/cf1217bae01449d48fdfd9d699c3a461 : SUCCESS in 4m 50s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b3e9f6f698454ff79e8440f8ca06fcd6 : SUCCESS in 5m 19s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/8b6691a3d20346f79b7b6c21c3588d66 : SUCCESS in 4m 20s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/f0aab67c40d1474ea18fe9bab40fc965 : SUCCESS in 4m 58s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/a155de6f423146799f537b5dc9e84fd3 : SUCCESS in 4m 25s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/47e3039df6884ed7bd30e5174d3d10c8 : SUCCESS in 5m 54s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/49698c606d55489482f5d5632c854a9b : SUCCESS in 4m 31s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/9cfa090d8766458280a1de6b7ede27a3 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":6},{"id":"bfb3114ed3f4877423d89050e4b5742b506b8e4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 12:05:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0ea7afd0c4787f37cfa83975e4284e4a2a0e82f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 13:51:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/413c7e89e0ac4ec0a459ef4094b76184\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/6704e79363d949fe97f2c350cba5e4b9 : SUCCESS in 1h 37m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4a54c68c65e74734976f6539d3a29293 : SUCCESS in 1h 31m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e0613d4e17a4d989bbe8c84902b501c : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa9dea052a1445c0858c5503f1fea1f3 : SUCCESS in 3m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ecba9d16e39c4885a117efb53b87a67f : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e2c7e5cc8c34f909c8da905d44c2b89 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30029220e19d448e969fa41a08b260c0 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ead0acad5b2a4c55a5ec065fff25b994 : SUCCESS in 4m 11s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/75f04f6fb7fd4cc2bac09c9dcd681858 : SUCCESS in 4m 25s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/7a12832afabd43419fc3032e7a8dfc2e : SUCCESS in 4m 15s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/e18f5911fae6474baad0c04481ceb995 : SUCCESS in 4m 14s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/a35e159db5c54a4786667cd3afce5204 : SUCCESS in 4m 13s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/ee248402ce04464b8fc4909f65b2f489 : SUCCESS in 4m 07s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/a14a62e4eee840608c5d49cfdd8a0ef5 : SUCCESS in 4m 16s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/b9471ee7dfb042d099ca7b2d4c33de86 : SUCCESS in 4m 23s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/12a1488f6e6e43ebbc4bedcc2e81e79e : SUCCESS in 4m 18s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/8fd6ffdf24fe4cf39f641180ca88d2a6 : SUCCESS in 4m 18s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/c9576aa782164dd4b714c62a0bc2315a : SUCCESS in 4m 30s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/82357eff5cd44e5689c39745b0a90b42 : SUCCESS in 4m 18s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/4bada41b25094e2eb472253c45198f0f : SUCCESS in 4m 31s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/7c8f7ae5b9ad4a08b6ef7910427f51ef : SUCCESS in 4m 33s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b4cb80c0aac246d1826fe737c2a9ac89 : SUCCESS in 4m 14s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/806c882846fc46fca9d0142b7ce55dcb : SUCCESS in 4m 04s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/7fff50cccd2a41dcb2836a7c761a3414 : SUCCESS in 4m 12s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/d44193fa697b4e3699ea25623e627c20 : SUCCESS in 4m 26s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/2ff6949635b247c981e679d53e50eaf0 : SUCCESS in 4m 06s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/6c8469384a17404ea8f28ec50d4afe08 : SUCCESS in 4m 18s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/babb9fd0ba424995b7e8b5a31bc2fc68 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":7},{"id":"fd843e09de814093f003d285a9e8dcd659d7e10c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-05 13:54:52.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"972f2086ae0faaa7b7ab031a0376a05326f0c1bd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 14:54:34.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b290253b6faea7c365d40c3bfef583951a59f9b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:55:26.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"040b003ebd900ea3ae6664c344facbc0cefe185e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:32:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd6508cb182843b4b9c9822f4eaa9bcd\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/07f3ebd36ce5448eb6199bb635b192ba : SUCCESS in 1h 22m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6676dc85428f44ffb0f1b4501e659810 : SUCCESS in 1h 19m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6691436231c04b6ab1598748917743fe : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61cf65a02c6b4fa494c9f858666a70d0 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75588d3da9bb4e16a8b87a3281787f03 : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65eb2ebc47a94bf1be5aeb169e1fb1a4 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63f4c3da9d5b4f12b3af11f2a9051e1b : SUCCESS in 4m 56s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/95fed53c9bd04915b034f1a3e559c486 : SUCCESS in 4m 25s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2975f39ad8d246139616ac7e5efa3a9d : SUCCESS in 4m 31s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/cdf0bbb47c0b46f89d53cabc504e6f08 : SUCCESS in 4m 18s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/596c83e309f8485aa12daadc9e1551ff : SUCCESS in 4m 41s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/ebc4213598cd4abfa0aa56a1f75cf37b : SUCCESS in 4m 23s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/1af2092d2bb84feeb5d1e732edb799a3 : SUCCESS in 4m 56s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/bdd8f15a9bdc4d8fba477cb7095a0a20 : SUCCESS in 4m 56s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/74caeaadb35d49b98886c81c1cf9e731 : SUCCESS in 4m 32s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/8397148cfe6041379542e4e0064887d4 : SUCCESS in 4m 14s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3d10cda7fbdd471d89c4e749807fe703 : SUCCESS in 4m 40s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/30d9c9718aac412c9d40267d12a1f949 : SUCCESS in 4m 08s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/99e34e451e234353a1bf9481f687c8e3 : SUCCESS in 4m 54s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/443c49618e124a289dd2b0faedc8f5e7 : SUCCESS in 4m 22s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/215ca29021d04133af48bdc95f695eaf : SUCCESS in 4m 37s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/05e2ea8414674b99802c3c5007e4eea3 : SUCCESS in 4m 34s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/1a6ef11d97044793af8256a44b225f93 : SUCCESS in 4m 28s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/ee30595a63674ec582d767e3cdad55d4 : SUCCESS in 4m 35s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/382478830fb24e2abbce14f7ed67ec57 : SUCCESS in 5m 05s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/cfcba3a0f12647fba7aebdd8bfde9cb4 : SUCCESS in 4m 41s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/3bc3b624006b46a5951bdd0ff8c0c63f : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":7},{"id":"a88c56aec009b792ee4ee6c1fa8b90e854e6272d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:32:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b8e1cc5ff89dd4b888e6fa8e84f6ce2c6594b1f3","revisions":{"8f9120c37cbfd532676673efa3fad0943ba90f3a":{"kind":"REWORK","_number":1,"created":"2021-05-19 13:04:23.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/40/792140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/792140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/792140/1"}}},"commit":{"parents":[{"commit":"13a6dffa32789ac91b724977ed1d73375baf0824","subject":"[stable-only] Cap stestr for python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13a6dffa32789ac91b724977ed1d73375baf0824"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-05 10:24:25.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-19 13:04:14.000000000","tz":120},"subject":"Retry on redis connection errors","message":"Retry on redis connection errors\n\nSometimes, connections get closed by the server. This change\nadds retrying to the code.\n\nConflict:\nrequirements.txt\n\nNOTE(hberaud): This conflict is due to the cap of tenacity during wallaby\nto unblock the gate. This requirement cap won\u0027t be an issue for victoria\nand for previous stable branches.\n\nhttps://opendev.org/openstack/tooz/commit/fe9c4304551cce0ac85366965009ad102fd9610a\n\nChange-Id: Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n(cherry picked from commit a025b4d4ee9fe1bd722960ea03445034580bec6f)\n(cherry picked from commit 62d5d17a5d2495f1f898163bbd83acbbc9873620)\n(cherry picked from commit 1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8f9120c37cbfd532676673efa3fad0943ba90f3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8f9120c37cbfd532676673efa3fad0943ba90f3a"}]},"branch":"refs/heads/stable/train"},"664568454e7bf0dd6762045115512ea38c21d6e1":{"kind":"REWORK","_number":2,"created":"2021-05-19 15:38:28.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/40/792140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/792140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/792140/2"}}},"commit":{"parents":[{"commit":"13a6dffa32789ac91b724977ed1d73375baf0824","subject":"[stable-only] Cap stestr for python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13a6dffa32789ac91b724977ed1d73375baf0824"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-03-05 10:24:25.000000000","tz":60},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-19 15:38:12.000000000","tz":120},"subject":"Retry on redis connection errors","message":"Retry on redis connection errors\n\nSometimes, connections get closed by the server. This change\nadds retrying to the code.\n\nConflict:\nrequirements.txt\n\nNOTE(hberaud): This conflict is due to the cap of tenacity during wallaby\nto unblock the gate. This requirement cap won\u0027t be an issue for victoria\nand for previous stable branches.\n\nhttps://opendev.org/openstack/tooz/commit/fe9c4304551cce0ac85366965009ad102fd9610a\n\nChange-Id: Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n(cherry picked from commit a025b4d4ee9fe1bd722960ea03445034580bec6f)\n(cherry picked from commit 62d5d17a5d2495f1f898163bbd83acbbc9873620)\n(cherry picked from commit 1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/664568454e7bf0dd6762045115512ea38c21d6e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/664568454e7bf0dd6762045115512ea38c21d6e1"}]},"branch":"refs/heads/stable/train"},"d18dab42409fa017066c28c9f232ebc53668cd75":{"kind":"REWORK","_number":3,"created":"2022-07-04 16:54:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/792140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/792140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/792140/3"}}},"commit":{"parents":[{"commit":"c00c5b81874563aa7c6f0b26a81092747d876afc","subject":"RedisLock release() should not check if the lock has been acquired","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c00c5b81874563aa7c6f0b26a81092747d876afc"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:43:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:54:25.000000000","tz":60},"subject":"Retry on redis connection errors","message":"Retry on redis connection errors\n\nSometimes, connections get closed by the server. This change\nadds retrying to the code.\n\nConflicts:\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): Conflicts are due to the absence of change\nI95809cf4035a6bc487b24f9e91aaa715921ad842 (\"Stop to use the __future__\nmodule.\") which we don\u0027t really need to backport.\n\nChanges:\n  requirements.txt\n\nNOTE(stephenfin): We don\u0027t need the cap to tenacity that was found in\nthe stable/victoria backport since that\u0027s already been done here by\nchange I8d6647118be22b0ce55e01b7e5451612ebe30e73 (\"Cap protobuf \u003c 4.x\").\n\nChange-Id: Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n(cherry picked from commit a025b4d4ee9fe1bd722960ea03445034580bec6f)\n(cherry picked from commit 62d5d17a5d2495f1f898163bbd83acbbc9873620)\n(cherry picked from commit 1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9)\n(cherry picked from commit 0734d86ae2322aeff92a2d44aa9b796bb7a14c26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d18dab42409fa017066c28c9f232ebc53668cd75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d18dab42409fa017066c28c9f232ebc53668cd75"}]},"branch":"refs/heads/stable/train"},"b398033b3306e6126af6516422fff268971a5bdb":{"kind":"REWORK","_number":4,"created":"2022-07-04 17:09:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/792140/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/792140/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/792140/4"}}},"commit":{"parents":[{"commit":"7e7f868893ab93a1fbca0ce2e7a606d692913b08","subject":"RedisLock release() should not check if the lock has been acquired","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7e7f868893ab93a1fbca0ce2e7a606d692913b08"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:43:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 17:08:15.000000000","tz":60},"subject":"Retry on redis connection errors","message":"Retry on redis connection errors\n\nSometimes, connections get closed by the server. This change\nadds retrying to the code.\n\nConflicts:\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): Conflicts are due to the modifications we had to make\nwhen backporting change I2437071bb575e95083b0d57a6a0a33aaf8356113 in\norder to comply with stable policy.\n\nChange-Id: Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n(cherry picked from commit a025b4d4ee9fe1bd722960ea03445034580bec6f)\n(cherry picked from commit 62d5d17a5d2495f1f898163bbd83acbbc9873620)\n(cherry picked from commit 1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9)\n(cherry picked from commit 0734d86ae2322aeff92a2d44aa9b796bb7a14c26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b398033b3306e6126af6516422fff268971a5bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b398033b3306e6126af6516422fff268971a5bdb"}]},"branch":"refs/heads/stable/train"},"a16af91e1dc80723bfb2fc1ae139d01b909905dd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-07-05 10:00:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/792140/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/792140/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/792140/5"}}},"commit":{"parents":[{"commit":"af87a977aa3a9ada4c4683605ce8c2a3c4054c11","subject":"RedisLock release() should not check if the lock has been acquired","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/af87a977aa3a9ada4c4683605ce8c2a3c4054c11"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:43:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 09:59:47.000000000","tz":60},"subject":"Retry on redis connection errors","message":"Retry on redis connection errors\n\nSometimes, connections get closed by the server. This change\nadds retrying to the code.\n\nConflicts:\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): Conflicts are due to the modifications we had to make\nwhen backporting change I2437071bb575e95083b0d57a6a0a33aaf8356113 in\norder to comply with stable policy.\n\nChange-Id: Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n(cherry picked from commit a025b4d4ee9fe1bd722960ea03445034580bec6f)\n(cherry picked from commit 62d5d17a5d2495f1f898163bbd83acbbc9873620)\n(cherry picked from commit 1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9)\n(cherry picked from commit 0734d86ae2322aeff92a2d44aa9b796bb7a14c26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a16af91e1dc80723bfb2fc1ae139d01b909905dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a16af91e1dc80723bfb2fc1ae139d01b909905dd"}]},"branch":"refs/heads/stable/train"},"b0ee28d12eb9805ac5a2158f673ff5d239763019":{"kind":"NO_CHANGE","_number":6,"created":"2022-07-05 10:01:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/792140/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/792140/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/792140/6"}}},"commit":{"parents":[{"commit":"85d0a8b2e09e37639ad29bb3bbcbbad55719dd8e","subject":"RedisLock release() should not check if the lock has been acquired","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85d0a8b2e09e37639ad29bb3bbcbbad55719dd8e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:43:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 10:01:06.000000000","tz":60},"subject":"Retry on redis connection errors","message":"Retry on redis connection errors\n\nSometimes, connections get closed by the server. This change\nadds retrying to the code.\n\nConflicts:\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): Conflicts are due to the modifications we had to make\nwhen backporting change I2437071bb575e95083b0d57a6a0a33aaf8356113 in\norder to comply with stable policy.\n\nChange-Id: Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n(cherry picked from commit a025b4d4ee9fe1bd722960ea03445034580bec6f)\n(cherry picked from commit 62d5d17a5d2495f1f898163bbd83acbbc9873620)\n(cherry picked from commit 1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9)\n(cherry picked from commit 0734d86ae2322aeff92a2d44aa9b796bb7a14c26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b0ee28d12eb9805ac5a2158f673ff5d239763019"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b0ee28d12eb9805ac5a2158f673ff5d239763019"}]},"branch":"refs/heads/stable/train"},"b8e1cc5ff89dd4b888e6fa8e84f6ce2c6594b1f3":{"kind":"REWORK","_number":7,"created":"2022-07-05 12:05:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/40/792140/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/40/792140/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/40/792140/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/40/792140/7"}}},"commit":{"parents":[{"commit":"85d0a8b2e09e37639ad29bb3bbcbbad55719dd8e","subject":"RedisLock release() should not check if the lock has been acquired","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/85d0a8b2e09e37639ad29bb3bbcbbad55719dd8e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:43:14.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 12:03:42.000000000","tz":60},"subject":"Retry on redis connection errors","message":"Retry on redis connection errors\n\nSometimes, connections get closed by the server. This change\nadds retrying to the code.\n\nConflicts:\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): Conflicts are due to the modifications we had to make\nwhen backporting change I2437071bb575e95083b0d57a6a0a33aaf8356113 in\norder to comply with stable policy.\n\nChanges:\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): We need to fix a style issue that occurs because the\nversion of hacking in use here didn\u0027t cap flake8, which means we\u0027re\npulling in a new version.\n\nChange-Id: Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n(cherry picked from commit a025b4d4ee9fe1bd722960ea03445034580bec6f)\n(cherry picked from commit 62d5d17a5d2495f1f898163bbd83acbbc9873620)\n(cherry picked from commit 1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9)\n(cherry picked from commit 0734d86ae2322aeff92a2d44aa9b796bb7a14c26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b8e1cc5ff89dd4b888e6fa8e84f6ce2c6594b1f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b8e1cc5ff89dd4b888e6fa8e84f6ce2c6594b1f3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
