)]}'
{"id":"openstack%2Ftooz~792194","triplet_id":"openstack%2Ftooz~stable%2Ftrain~If233680e6a00a105a7a77a48386bcea9aaba18de","project":"openstack/tooz","branch":"stable/train","topic":"fix-ci","hashtags":[],"change_id":"If233680e6a00a105a7a77a48386bcea9aaba18de","subject":"Fix requirements for python2.7","status":"ABANDONED","created":"2021-05-19 15:41:58.000000000","updated":"2022-07-05 10:01:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9d72d2074bbabd94a59e5cd46d57dfc12d0fda4","_number":792194,"virtual_id_number":792194,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-19 17:16:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 17:16:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6930292cbf9d31c5233d69d51c0729a5549424a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-19 15:41:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5881afe3aebd38845987bf180b2ff077f963a6dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:16:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5f74c98c675d4cbb82da84cff19b28a9 : SUCCESS in 2m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/8c11cd1e6a3b4cf8b4a3a8ff36aa898d : SUCCESS in 1h 29m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3fc38686118442692c76e7f858fd91e : SUCCESS in 1h 29m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5d2eb87fb5045f0b95b31300b50d556 : FAILURE in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df17de3151a46e9a7693dbf0eaa5bdc : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b674523b9cdf46fa9f648fa2f24e69ae : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20031eed75864b3aba4bfca9f4063cc4 : FAILURE in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b8a2ef986714339bbf010633b5ef0ee : FAILURE in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b8c2e433a3d4976a69370fd208edcde : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0771df342db0453389c6e653662209d2 : SUCCESS in 4m 55s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/2a700ec10473403995b62b39572f1801 : SUCCESS in 5m 07s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/c9be64edfd454632bc2fc01b6583ddc2 : FAILURE in 4m 33s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/4851c7be5e314bcd885d3b4eba77edab : SUCCESS in 5m 52s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/8719d728ab4d41d38e4aa8533ec17912 : FAILURE in 5m 01s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/127dedb43a0d41c68fefb37f7c5c4f5f : SUCCESS in 4m 55s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/e7969f28ccb94afe98b0767ea7ee6e3d : FAILURE in 4m 37s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/ac9ee5c88c0445ea8b0622cfe9e28b64 : SUCCESS in 4m 58s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/a67a6b1bd85e49aa963f10dc1d6b38ee : FAILURE in 5m 07s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/bb8933661cf048c8ace8b5c514aa6154 : SUCCESS in 4m 42s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/3d5c773642354ad29cec792256717761 : FAILURE in 4m 49s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/a24496faf47846beba11d87dc34f0a58 : SUCCESS in 4m 56s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/50510eabaee84f78a2e1599d80b77451 : FAILURE in 6m 15s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/df33dba496234f32a834c3a228ace959 : SUCCESS in 5m 00s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/ca32708e84824b68adb25803db389397 : FAILURE in 5m 39s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/a6f5abe893dd4f048d9df750988e8367 : SUCCESS in 6m 19s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/d1ca87c11d534d7c947b929275b5b7e5 : FAILURE in 4m 25s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/8af83884c5ed4fe7850f013baa537bbb : SUCCESS in 6m 13s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/844ee6276ed7499e8981a4c6939406d8 : FAILURE in 4m 47s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/e25c1961602a4d3da29a0a3aebaff1e8 : FAILURE in 5m 59s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/fe4e5a3dc034491e8bf38903b0bbd9e3 : FAILURE in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d9d72d2074bbabd94a59e5cd46d57dfc12d0fda4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:01:36.000000000","message":"Abandoned\n\nThis is done in https://review.opendev.org/c/openstack/tooz/+/848622/ now","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11e6d18a06d9b99147ad5fb640895044efeefc8c","revisions":{"11e6d18a06d9b99147ad5fb640895044efeefc8c":{"kind":"REWORK","_number":1,"created":"2021-05-19 15:41:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/94/792194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/94/792194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/94/792194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/94/792194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/94/792194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/94/792194/1"}}},"commit":{"parents":[{"commit":"9febe34244f2ab428aa4efebcd714ccdb6d5dbce","subject":"Fix broken CI and PyMySQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9febe34244f2ab428aa4efebcd714ccdb6d5dbce"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-05-19 15:39:56.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-05-19 15:39:56.000000000","tz":60},"subject":"Fix requirements for python2.7","message":"Fix requirements for python2.7\n\nChange-Id: If233680e6a00a105a7a77a48386bcea9aaba18de\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/11e6d18a06d9b99147ad5fb640895044efeefc8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/11e6d18a06d9b99147ad5fb640895044efeefc8c"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
