)]}'
{"id":"openstack%2Ftooz~795791","triplet_id":"openstack%2Ftooz~master~Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","project":"openstack/tooz","branch":"master","topic":"redis-retry-fix","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2021-11-07 12:28:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","subject":"Enable retries in redis driver","status":"MERGED","created":"2021-06-10 14:18:16.000000000","updated":"2021-11-08 10:36:41.000000000","submitted":"2021-11-07 12:28:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"795791-redis-retry-fix","meta_rev_id":"5e7db66ffada8cc2e1e8cbd09f41eb9cd3f132e3","_number":795791,"virtual_id_number":795791,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-07 12:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-11-07 11:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":1,"date":"2021-07-06 08:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-06-16 17:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-07 11:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":1,"date":"2021-07-06 08:18:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 14:18:45.000000000","updated_by":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2021-06-10 15:55:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 13:48:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-06-16 17:05:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-05 16:16:34.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"state":"CC"},{"updated":"2021-11-07 11:05:24.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"8c52e976066f2fe3d7fe8fe26380ed699107d748","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-10 14:18:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f4d0853f4a0dc1834b822425c5904a25c0ccc4f","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-10 14:31:32.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\ncough. Good catch! Embarrassing....","accounts_in_message":[],"_revision_number":1},{"id":"665e013298da2605df21eb1d5c18e8dfcbdacb1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:55:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a64afda46fce4b8eb4be80f87b83cf7d : SUCCESS in 1h 32m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea50279092ef435d95152416b291ce94 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e294f9723b44b64b19409cf18d3feac : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/771419d5780b4e87bb450ff16ac8d007 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/954d10c5b15d4284806566e5c1108592 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/064a5d3eafce46f3bea67a5f7ada7d10 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b12b6b494a4f249fc2debcd8f62462 : SUCCESS in 5m 30s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/bbac0e8065e0459895f34987fc8c0a81 : SUCCESS in 6m 00s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/be4f62f3406e48fbbb52295a38ecffc6 : SUCCESS in 6m 50s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/58023e92c43d4dcca86c23cb22695ee4 : SUCCESS in 5m 59s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/5ceb2c0cac3a4288b5f297181bbe7f2b : SUCCESS in 7m 03s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2be056a351044576bf3ac18ad2a9eb19 : SUCCESS in 6m 09s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/4f3530d4f0284776b2cf7c7fc3878d49 : SUCCESS in 6m 51s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/208c8cc596db4f8f945e1c1987391a21 : SUCCESS in 5m 11s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/6aa4e416d07e471abc01e7dd2170d60e : SUCCESS in 6m 20s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f7770a600e3247d3a317660acea04f99 : FAILURE in 8m 23s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/f656c269322b4abbb46aa2267eccca99 : FAILURE in 5m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/584664e1f48b481f99360719616cfa8f : FAILURE in 5m 12s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/197bb6a3ab2b4674b77e61f80745462a : FAILURE in 7m 31s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/2898cb9941714efe885045f020c8efba : SUCCESS in 5m 16s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5713cfb01c7b40e3a760942bf6347da2 : SUCCESS in 5m 51s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/334386671ee7447e9d428fd8519ceb43 : SUCCESS in 7m 35s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/5aada263c0d94c1d82095efcc009b159 : SUCCESS in 4m 50s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/16ec80022cec446fbf467cbb95aec9f0 : SUCCESS in 5m 39s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/239e937148814c06aa1ebf5819b7948f : SUCCESS in 6m 55s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/50ffc236adf94ce3bf6c5b67c6ae51a3 : SUCCESS in 5m 38s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/14f37a98d0ba402c9754fda113a194f0 : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":1},{"id":"41468b9088fd2e1d6fb04b14b865f7adc021a96e","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-10 16:20:37.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"a5a66e18dc5500eca2211697e5e0855f6afadbfd","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-11 09:48:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2eba46dfb79b728ee6f01e3cc1b7ebf4c050391f","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-11 09:57:36.000000000","message":"Patch Set 1:\n\nI get the same error in the same test when running locally, investigating...","accounts_in_message":[],"_revision_number":1},{"id":"34cc4cbf27c8b4528319f7189b6b44a1cadabc5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 11:10:38.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/448d02425a0f4e7099ffcd229dd1b6fc : SUCCESS in 1h 18m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d723cfa872d46fa87d9ba73e7dbc284 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7af54dfcd034239858813d806165c8c : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/605e0ce60a60487cbb365648bff264c5 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13556d36ba1f4fb58ee6e60098a27d11 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6eeb04a1d0f04e159a95baf4e9a42675 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4acffb370797433495dfaa5f76872c72 : SUCCESS in 6m 31s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/72403d634487488db2dd294d364f2e22 : SUCCESS in 5m 32s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/c82c6edbbc2d46c6b5a1576da6741257 : SUCCESS in 5m 42s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/487ef76bafde4a97bcef1c4499e8e3cf : SUCCESS in 5m 02s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/9049bfd9d8614ec59ffedc66665384e9 : SUCCESS in 5m 19s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/a9b5784e4cb54623b4b3303a331324aa : SUCCESS in 5m 25s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/b0addaa9efb7415b9d0d1b49de9b085d : SUCCESS in 4m 49s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/a7ce7ec201484f8d83ea5f28355ef3fe : SUCCESS in 4m 54s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/81c87249bdd4469d8da7d842f2decd85 : SUCCESS in 5m 05s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f6b9242d120141599d71e0c0f942dc61 : FAILURE in 6m 15s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/070f98a2f11b43d1bce587678e047ad0 : FAILURE in 5m 16s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/8c5ada8e62de4e1facfc452575cf3f03 : FAILURE in 5m 38s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/f4a594b9dd6a46c6ae4e89df089afb67 : FAILURE in 5m 29s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/11466669181344a0b77783645e90c9e5 : SUCCESS in 5m 09s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/62ec469ca7614b85997063ad814b4178 : SUCCESS in 5m 47s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/e8f93bcfe8f643559e1c15a0f07e1556 : SUCCESS in 6m 46s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/65c6a859f3ec448497555a3193b4b3fd : SUCCESS in 5m 58s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/3d07ad003df1424dac5b5168993fb370 : SUCCESS in 5m 21s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/accfc5ceb1a6441e9c97f22ba487748f : SUCCESS in 5m 21s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8a625f8af78d4864991cb3354c26c536 : SUCCESS in 5m 28s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/3b115e2fec1945b0815ea2ada1ae4576 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"251a33ce33e0b8107c5518a86f7e75508f4f94ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-11 13:51:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ff473a9e5ac9e6843c3a6052e764fd234a4aebf9","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-11 14:07:07.000000000","message":"Patch Set 2:\n\nafter closer inspection I realized I do not understand how the current version of _handle_failures decorator is expected to work and be used\n\nI took a liberty to re-write it in more standard \u0027decorator with arguments\u0027 way..\n\nand now it seems to work\n\ntest script I to showcase new implementation of the decorator http://paste.openstack.org/show/806567/","accounts_in_message":[],"_revision_number":2},{"id":"c6bc4db5c69e407deee24ad9c016708ee231966e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 15:22:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db9a9ada8366489e85b21d1caf154d96 : SUCCESS in 1h 25m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e285e4dbb094cda84e667d01c360968 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95b0fe2cec6e4030bb8aea1fd53a6172 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a3392aa8c0a418cb4d57a60f8414b99 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a6752c5671e4f8e81553d8008bf6fd9 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e93f4e71dd4a4d7db8f681b36dd1d1a8 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d602405c03b049efb63a0d187fd73f21 : SUCCESS in 6m 02s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/ae6057bf07ab4c0085560be07720dd5e : SUCCESS in 6m 03s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/a8fce0836b7a4dd7b82169ebd82aee5a : SUCCESS in 5m 40s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/b248ba47264242c59f43c50dbd65ee43 : SUCCESS in 5m 13s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/57fa17d3341c4547a7ca4ee3a4852093 : SUCCESS in 5m 55s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/ab3cfcd56a044f368fa3716b787f2786 : SUCCESS in 6m 13s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/856e374c70404a5b843f200425c8d1b3 : SUCCESS in 5m 12s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/1d2a928a4e6045afbb8cd51329da4e3f : SUCCESS in 5m 23s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/7105fbfce83c462792142d8d88790814 : SUCCESS in 5m 04s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d05ee50181e8413db3df936628784396 : SUCCESS in 5m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/9c2e4ec1d43a41efb14944b615174479 : SUCCESS in 5m 52s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9ee1f1b2448c4f3f9a36d3720f90f820 : SUCCESS in 6m 04s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/05b2f39dfd144d348e8f7ce2a837050f : SUCCESS in 6m 32s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/d8041554c64f409393a583236b7e473e : SUCCESS in 7m 05s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/768d9547b4ea471283562b1d4635c5c5 : SUCCESS in 6m 21s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/1112ad363ba74b3d839cbec6bb0693e6 : SUCCESS in 6m 43s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/953fbc073a4443c291037091990e2ad1 : SUCCESS in 5m 36s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/93662540e86241898dac2b7e81c1ee78 : SUCCESS in 6m 47s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/9eaba5b62e5245a687575d2c59da2877 : SUCCESS in 6m 33s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/b80687e593934b2099b7345600d51c05 : SUCCESS in 5m 22s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/524ab946ef1f406097a6377f4626823c : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"1e3867f3c717f3f717fed5bc69ecd422c52b102b","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-14 06:42:02.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\nThank you for working on this, this patch looks better than my version.","accounts_in_message":[],"_revision_number":2},{"id":"2a2d404612b065f8a0f669b15c225f5371d8b450","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 13:48:19.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nHello,\n\nThis version LGTM.\nParenthesis added make sense in this case because this is a callable that will return a decorator.\n\nI wonder if we can simply handle this kind of case with tenacity:\n\n```\n@retry(retry\u003dretry_if_exception_type(exceptions.ConnectionError), stop\u003dstop_after_attempt(15)))\ndef \u003cfunctions\u003e():\n```\n\nAlso I\u0027ve a minor concern about a log debug, please see my inline comment.\n\nPlease can you add a related release note to high light this fix.","accounts_in_message":[],"_revision_number":2},{"id":"311eded2ae3b08970b34def3a7bf451382da3e0b","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-16 16:37:25.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n@Herve,\n\nwe can\u0027t simply replace @_handle_failures with @tenacity.retry since we are also translating the exceptions from the redis ones to tooz natives.\n\nor you meant have a simple translating decorator as it was before, but decorate _that_ with tenacity?","accounts_in_message":[],"_revision_number":2},{"id":"1dd913443859d7c0187431c02ffc8545cb76afe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-16 16:44:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc7565bbe53b00f7dadf8c88ace0fdc69cc893d5","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-06-16 16:45:34.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nrelease note added","accounts_in_message":[],"_revision_number":3},{"id":"68d83f1e392442b84205c6a16004fc18b3fbda7d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-16 17:05:39.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nAh yes I forgot the translation part of this problem. Ignore the tenacity part of my previous comment.","accounts_in_message":[],"_revision_number":3},{"id":"6296f305f4bfb2022ae4dfc6e33a6e71d55d313b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 18:10:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15bae0065ee84b85a03b7ddf30201e4e : SUCCESS in 1h 20m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8485339e2bbd4e24afca0ecc11638c06 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/458ee0ab7090436ab464c5ce992b039c : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/937eeed82393490aad6cfe9a40c356d1 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c53b17d45c02410689652068f2af500c : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57a1a4de14744e04b03e824f7201d9dc : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1407afaf2fb0447a99e07dc6b4a09448 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d692534b4a9341a4aeb55c9b7eaec287 : SUCCESS in 6m 06s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/d75195a997a94224b88cf65fb0c1212f : SUCCESS in 6m 47s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/37e966f30f2544e3a8aafb3bde4acd52 : SUCCESS in 6m 06s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/4a2215a53bd640efbf4914cbf1d54e7b : SUCCESS in 5m 38s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/c1b978df2a6e4a4da06776ac018a7fd5 : SUCCESS in 6m 34s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/40bea4f5c8c743f18bd05737c9833c67 : SUCCESS in 5m 12s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/614ca0c48e3a41dda356b728fb62ff34 : SUCCESS in 6m 08s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/da8a6395c5934a5c835cd101f0a8dce1 : SUCCESS in 5m 01s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/60bd1f73268344f5a02c32563b7bde66 : SUCCESS in 6m 15s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/616e18a2bd4546da8da588da2fa799c4 : SUCCESS in 5m 21s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/0f001247f43447798bae4f4461fcd2f3 : SUCCESS in 5m 48s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/c780aac73021437f8edeb4b3ca8cc855 : SUCCESS in 6m 21s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/1e8e522fb65f4f3cb8cef470ef298a6b : SUCCESS in 7m 47s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/61f6f91043524d309b047c57cd5d0190 : SUCCESS in 6m 02s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/ec18e043e13f4677ac88074cee357421 : SUCCESS in 7m 11s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a8eb00ae58d148a59734605494ace58e : SUCCESS in 5m 18s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/eac3b4d41d99431a8bfc6f5613198fcb : SUCCESS in 6m 28s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/dbffad0f3bcd4e8e86d37f7f84a6e43b : SUCCESS in 5m 44s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/61ea6dbe999542a589b7446d10337bfa : SUCCESS in 6m 21s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/347fb5f7d8df4e67ac1fdb4ccbe512e6 : SUCCESS in 5m 51s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/69f14dc784f545fea7c6600b33e9f628 : SUCCESS in 6m 43s","accounts_in_message":[],"_revision_number":3},{"id":"cd9e760c86e3a8404da2b090cb924b774a1104de","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-07-06 08:18:38.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1\n\nThank you, can we proceed with this?","accounts_in_message":[],"_revision_number":3},{"id":"6f3b9d583c6b0716893995e175a48ae42e475734","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2021-11-05 16:16:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6c1d60c6a06283a6a33f0ce1d5e006d8415df259","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-07 11:05:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"05b1e4fecc97c63db617477147a9dfe11c426ac6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-07 11:05:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0734a530343ee632ef3045ce748df6464f8c8a6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 11:05:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0d757ac4828d5d0fa7f10f50b50a86813e8778d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 12:28:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c2a2ded558449e799ad91e210b8f899 : SUCCESS in 1h 21m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e7b446f5384abfbb0328f40a85adf7 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21bce61d44b841ed892482a59daea970 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b816e6d51474b23bea478558619058b : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ceb9ab67df4022a1dadf3c36f3e1dc : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7eea0f5ec384015be402b310abefc6d : SUCCESS in 6m 15s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/9a9dab2c17ce4291a9d40fb95bd067cd : SUCCESS in 5m 43s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/894962d4bc3e4b7cabbab387a133e397 : SUCCESS in 5m 46s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/91f4bb41b940442cbf0bc403124ac5ee : SUCCESS in 5m 46s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/1c67e0d90df542b79cc716dbb43b386c : SUCCESS in 5m 19s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/fcf3410b79894e63b1f86e23fc13edcf : SUCCESS in 5m 32s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/ce88c66e6abe4492baefc91e404b4a9b : SUCCESS in 5m 35s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/09ed1d01d71742da838a101cbc2cddad : SUCCESS in 5m 42s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/10bb0ab0f69d400abdb69f2c0dab6aab : SUCCESS in 5m 29s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a63010bac4be41aead325336d1bc09ab : SUCCESS in 5m 47s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/25c6e66ab76c42c1a790a69e5b656f63 : SUCCESS in 5m 45s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/edf47ab522dc451cafe9a4d9582a90ef : SUCCESS in 5m 29s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2fc1a5423be744f8a1a4d8695613128c : SUCCESS in 5m 45s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/6cca1c99f2294c4e99e696010a2655b9 : SUCCESS in 5m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/cae95d4088ba45d493584027448dcfb6 : SUCCESS in 5m 36s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/85740d437f2b42379bb8b7c312ff1ad9 : SUCCESS in 5m 08s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/85e8c1dd8ad9436493719a9de5886ce1 : SUCCESS in 5m 21s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/5bdb83f5eb864e6aaac15c63820cd663 : SUCCESS in 5m 34s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/fdab1e02e9034deba116fe5b830ef7d2 : SUCCESS in 5m 41s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/db029236baea493784d3122ceb0d4136 : SUCCESS in 5m 19s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/175b23367e734eeba609524f80a85c5c : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":3},{"id":"bd58cc5233b84c2f7dd9505efd1ece3698d1a948","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 12:28:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2ea2ea11e564e0dd8e6fc3e008dc99990cdbcea8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-07 12:29:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139f69b0da0c44a681d33de6e439370b : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4877d9c24e4346afa076f9bff763fed3 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3},{"id":"5e7db66ffada8cc2e1e8cbd09f41eb9cd3f132e3","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2021-11-08 10:36:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"47c4d56e446a49726e7145c6cbc3bb7620a431f7","revisions":{"4b7a28258675ccab10104b5d3abcbfc59b1d34bc":{"kind":"REWORK","_number":1,"created":"2021-06-10 14:18:16.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/91/795791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/91/795791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/91/795791/1"}}},"commit":{"parents":[{"commit":"292d768b8019abed2dfe97c9cdd8e491f4b8167f","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/292d768b8019abed2dfe97c9cdd8e491f4b8167f"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4b7a28258675ccab10104b5d3abcbfc59b1d34bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4b7a28258675ccab10104b5d3abcbfc59b1d34bc"}]},"branch":"refs/heads/master"},"e667547acb8f29d3f172aab12050b8013b4af2a9":{"kind":"REWORK","_number":2,"created":"2021-06-11 13:51:11.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/91/795791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/91/795791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/91/795791/2"}}},"commit":{"parents":[{"commit":"292d768b8019abed2dfe97c9cdd8e491f4b8167f","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/292d768b8019abed2dfe97c9cdd8e491f4b8167f"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-11 13:47:48.000000000","tz":180},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e667547acb8f29d3f172aab12050b8013b4af2a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e667547acb8f29d3f172aab12050b8013b4af2a9"}]},"branch":"refs/heads/master"},"47c4d56e446a49726e7145c6cbc3bb7620a431f7":{"kind":"REWORK","_number":3,"created":"2021-06-16 16:44:51.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/91/795791/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/91/795791/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/91/795791/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/91/795791/3"}}},"commit":{"parents":[{"commit":"292d768b8019abed2dfe97c9cdd8e491f4b8167f","subject":"Merge \"setup.cfg: Replace dashes with underscores\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/292d768b8019abed2dfe97c9cdd8e491f4b8167f"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-16 16:44:38.000000000","tz":180},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/47c4d56e446a49726e7145c6cbc3bb7620a431f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/47c4d56e446a49726e7145c6cbc3bb7620a431f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
