)]}'
{"id":"openstack%2Ftooz~816962","triplet_id":"openstack%2Ftooz~stable%2Fwallaby~Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","project":"openstack/tooz","branch":"stable/wallaby","topic":"redis-retry-fix-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2021-11-16 17:09:55.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-11-16 15:31:35.000000000","reason":"stephenfin replied on the change"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2021-11-16 13:45:44.000000000","reason":"Matthias Runge replied on the change"},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2021-11-16 17:09:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","subject":"Enable retries in redis driver","status":"MERGED","created":"2021-11-08 15:10:30.000000000","updated":"2021-11-16 17:11:43.000000000","submitted":"2021-11-16 17:09:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816962-redis-retry-fix-stable/wallaby","cherry_pick_of_change":795791,"cherry_pick_of_patch_set":3,"meta_rev_id":"c80664e0301aaf78e1f1508459092aed856fad68","_number":816962,"virtual_id_number":816962,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-16 17:09:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-11-16 15:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-11-16 13:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":2,"date":"2021-11-15 12:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-11-16 15:31:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-11-16 13:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-08 15:10:30.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2021-11-08 15:11:02.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2021-11-08 16:48:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-15 12:54:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-16 13:07:34.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"ec280baf96c5a27875491506da9bb79aed59f41d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2021-11-08 15:10:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2b1fbed61b7737be36839315d9693b31524c76e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 16:48:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/23f3ac8d725c4013bef1941d57c77bbe : SUCCESS in 1h 33m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed6239f2d2c84e70adb1468765a288a9 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd049dc4d5354c3a8dac331236647a07 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9aa325a9b134020aa9f1f1b6a915552 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c4299e77431460782d6db569a684a3a : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d90c765063dc457e920664464f82535b : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf4a8588f5f84c198ff34f0abfa9aeae : SUCCESS in 6m 09s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/f78405f5b2944d759ab82c3be26db7c0 : SUCCESS in 6m 39s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/28c79d2843094939a36daa4f5acc7819 : SUCCESS in 5m 59s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/afaef935e16c4974a6db3976e62f2918 : SUCCESS in 6m 18s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/9dfd709b18574b1fbc580d415a9bc0b4 : SUCCESS in 6m 01s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2704e28a70ff4af282c2dc462787507e : SUCCESS in 5m 40s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/b625777e77ec419bbc09a0c312c45327 : SUCCESS in 6m 03s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c58444a41673485ea50d03c5dd7efd7c : SUCCESS in 6m 46s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/6d947b58301745e7a9859067bbccfe12 : SUCCESS in 6m 28s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/661c7ec22dc24ae49430030809812b59 : SUCCESS in 5m 47s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/cc710bc7feb54a8aa5282a5e75c8569c : SUCCESS in 5m 37s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/63f1460f911248a6b84b85ba21efdeed : SUCCESS in 7m 54s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/e9c79ec003f341958a2dad86ba4822ae : SUCCESS in 5m 40s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/4307f6a332bd4fa5a98aa2ada766642e : SUCCESS in 7m 02s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/9ba02df710ad467ea6b04f7cf23a7b34 : SUCCESS in 5m 27s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/f59a3f3c11e04aba8a766aa0773a787a : SUCCESS in 6m 31s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/eb28c185cb4d428fa58d5a0364e01bef : SUCCESS in 5m 26s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/671fd7fe854c4290814e2d834dd15e4b : SUCCESS in 5m 31s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/36fbbf9fa34d47d1a70364b9a2d8f831 : SUCCESS in 7m 44s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/e5a9e871144a44329b68ebe0fa335cac : SUCCESS in 5m 38s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/f4c6610a12884322b1dcf1070d3c2b60 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"6b296858fc3a8e47a15ce77d5f6647a7c6fe58c9","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-11-15 12:54:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13eb376c34d880b37a1298d47fa977b617fc75c1","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-11-16 13:45:44.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7f0a1ef13db93dc50b19fdc9750cb1abe757bcb5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-16 15:31:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c784f56c11de23703eaaf686903ad379c3590b9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 15:32:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9a9b7c76af7daac11bb83282b695350f4c70d0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 17:09:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1cb953fdf0524bda869ec61544827040 : SUCCESS in 1h 36m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34eb7ccfe9584a62adb4678841ddabc9 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72cb6ed638ae4636a4eba7c2548d1f26 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82062627ecb44d538ad1513f05d6ce4d : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f18a58274ef499ba2ac638a9d3e0943 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2182fa8d476468ca1c416c49b0ce985 : SUCCESS in 5m 40s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/b903205ef1e14a2c9587943e56da1cf3 : SUCCESS in 6m 50s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/dece4abd06964967889ddb157e7ed193 : SUCCESS in 6m 06s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/aab0e14301174010894c3dd34d5333b1 : SUCCESS in 6m 42s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/5de5746725984650bcb40c871e771d36 : SUCCESS in 7m 49s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/1b77e5fdf256437bbe0224ba5a6f43a0 : SUCCESS in 7m 50s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/abbbbb5faf9d4991b47951308502c6cd : SUCCESS in 5m 50s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/7342b9f18a2b4f739af70ea472574909 : SUCCESS in 7m 50s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/fb7e3e6753e84ce08136af51729a6bd7 : SUCCESS in 7m 46s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ba21e0bd952e410792d58dc063ca998f : SUCCESS in 7m 04s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/23be790013944442b738b55e4d0c994c : SUCCESS in 8m 54s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6dd3ec477f7f4a44b34631702c9ea9cb : SUCCESS in 7m 19s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/eadfe5ae1b1b4fd781be5edbb38e758a : SUCCESS in 6m 53s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/d154f4618b2141538a38536f8f02d235 : SUCCESS in 6m 27s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/c435ea73eb354e21a2a2a5341264c3a0 : SUCCESS in 7m 46s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/eeed9ffc7e1f44de9eb2b80c12167fa0 : SUCCESS in 7m 30s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/b84961241935432781abf04f23dd7fec : SUCCESS in 7m 17s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/5eb85ce3de9a4630a96072fa0bf01e86 : SUCCESS in 7m 00s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/da343d6e410f473bb689fe89286222d3 : SUCCESS in 9m 29s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/f14c63a0859c45b8b04688f0a07b2eb5 : SUCCESS in 7m 05s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/ece7300082e246269609350ee58a4503 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":1},{"id":"af9bae4c66050b866e9a40c46c46a2173be26d86","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 17:09:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c80664e0301aaf78e1f1508459092aed856fad68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 17:11:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f694caad61ac4245a1888a16bc395163 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a59ffd4ca714ea9b98f32be0e866257 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11526e594c7e7cef8b5a4a4220f100d76a418036","revisions":{"11526e594c7e7cef8b5a4a4220f100d76a418036":{"kind":"REWORK","_number":1,"created":"2021-11-08 15:10:30.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/62/816962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/62/816962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/62/816962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/62/816962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/62/816962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/62/816962/1"}}},"commit":{"parents":[{"commit":"62d5d17a5d2495f1f898163bbd83acbbc9873620","subject":"Retry on redis connection errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/62d5d17a5d2495f1f898163bbd83acbbc9873620"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2021-11-08 15:10:30.000000000","tz":0},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n(cherry picked from commit 47c4d56e446a49726e7145c6cbc3bb7620a431f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/11526e594c7e7cef8b5a4a4220f100d76a418036"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/11526e594c7e7cef8b5a4a4220f100d76a418036"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
