)]}'
{"id":"openstack%2Ftooz~817985","triplet_id":"openstack%2Ftooz~stable%2Fvictoria~Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","project":"openstack/tooz","branch":"stable/victoria","topic":"redis-retry-fix-stable/wallaby-stable/victoria","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2022-04-15 13:09:19.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-11-16 17:47:01.000000000","reason":"stephenfin replied on the change"},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-04-15 13:09:19.000000000","reason":"Change was submitted"},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-04-15 13:09:19.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-04-15 11:21:44.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","subject":"Enable retries in redis driver","status":"MERGED","created":"2021-11-16 17:17:29.000000000","updated":"2022-04-15 13:10:44.000000000","submitted":"2022-04-15 13:09:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"817985-redis-retry-fix-stable/wallaby-stable/victoria","cherry_pick_of_change":816962,"cherry_pick_of_patch_set":1,"meta_rev_id":"d101211b55ca4377199a1342daea7d18827dcc66","_number":817985,"virtual_id_number":817985,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-15 13:09:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-11-16 17:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-11-16 17:18:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":2,"date":"2022-04-15 11:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":0,"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":1,"date":"2022-04-15 11:23:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-11-16 17:18:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"value":1,"date":"2021-11-16 19:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-16 17:17:29.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"REVIEWER"},{"updated":"2021-11-16 17:18:05.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-11-16 17:18:05.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-11-16 17:18:05.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2021-11-16 18:32:24.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"CC"},{"updated":"2021-11-16 19:10:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-16 19:23:59.000000000","updated_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"reviewer":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"state":"REVIEWER"}],"messages":[{"id":"62b67b573e0b1ff367594a6fe0ce69805c405060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2021-11-16 17:17:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"f74e7171b3ed44c5de7f990ba5b29ddb02addd15","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2021-11-16 17:18:25.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d87d33e380c5e7e049840e976482e5c0a14992fe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-11-16 17:47:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10ad3eaee692fd1717db1384915688c3313057e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 19:10:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8c1e227c6614567893e5f222b7537ff : SUCCESS in 1h 49m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df9aa1a9fe894548b0ba6b1019911a05 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed8fda04179540efa741b4487344c230 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f1a96b2adc54aeeba2724563f7972b9 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/724e7bcd5b3c445ba4d6d26a6ee62f08 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17b2dca26e94fe7a03552f7da26b269 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bade2b25b4624cae8650e8acbb6a491d : SUCCESS in 5m 10s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/f3ba4ef8e9d44ad5ad8932d23138d9f9 : SUCCESS in 6m 15s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2fb0704aa7104eda99adb2aa66788761 : SUCCESS in 5m 36s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/b4e40289b4b84b078a41bf78f0998caf : SUCCESS in 6m 47s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b99704acf1dc434ca2bb1b9d03bc22e1 : SUCCESS in 5m 05s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d85623fca9d94eff8eb07945c3479b19 : SUCCESS in 6m 22s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/5039ece9be954d23b152f8b86a833df0 : SUCCESS in 6m 15s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/23bd13380f73440f87406828df1f1e1e : SUCCESS in 6m 35s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/b4fa3ed436cd43a2b799f0b7a8e2d89a : SUCCESS in 7m 55s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a25c19c5ba8d4d8d952f07806d082b4a : SUCCESS in 8m 07s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/ed3c141eb451499fa1a1fe842733a50c : SUCCESS in 6m 52s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/12afa2abfb33424890b39e8689518077 : SUCCESS in 6m 50s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/5567e6b549d540baba81dd8f7a1cb647 : SUCCESS in 6m 01s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/51b81e3041bd46029af84d2f974cfc88 : SUCCESS in 6m 22s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/e51874001f04401fa3fc3d6334c6d3e4 : SUCCESS in 5m 50s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/0da1ed3a817a459c9676c578351193ef : SUCCESS in 6m 56s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/01c4a0bb26ab4b038dbbfeb6aa507f64 : SUCCESS in 6m 38s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/48fb9ca8caa343ff8377464bf225ffbc : SUCCESS in 7m 15s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/3f43e0621e35424ab9fb689e12a76aa7 : SUCCESS in 5m 41s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/fc3b09be88634069bc1db542b904d670 : SUCCESS in 5m 56s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/3b3813fcfe6a4811a1166906d3d0c3e3 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0021d715246842f48c50ccbf4415e3d9c92e1a31","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2021-11-16 19:23:59.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"40a2fbb73a7efafc1a82faf3af2eacf90669d80b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-15 11:21:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ab154d89b2e727d9a7f304d9bb9e7d9dffbab40","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-15 11:23:58.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52416f35f768bb9ee94df3268d4fba250fbff8f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 11:24:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e79572a2d770f164c3283519f3f4993f6dcf7c32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:09:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/998f3f58f3fa4f8a922a2768ee8e9bd7 : SUCCESS in 1h 40m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/302cdf095a1c461489020e8325399984 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca82eb9160a344548fb0988d27bb85f7 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b0517757a6140129a8a4a895b82e0c5 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e4abd022c54f2386a4e1bcb02c5ee3 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78166df201ef4cd49abd38ef3c7eee57 : SUCCESS in 5m 25s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/09af92bd3e7242e6a2344aa1c13a48b6 : SUCCESS in 5m 17s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/215ab5763ff7496082e68c10bb9a4ebd : SUCCESS in 5m 56s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/3192d53a8d6b45d4afd664576eee5cd6 : SUCCESS in 4m 57s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/098ec0d41a5949d2ae0adc529f55fd52 : SUCCESS in 4m 52s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/38ef95d2d64e4079a97d967a9f262307 : SUCCESS in 5m 03s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/d87a0ff1dfa94568b08e8604f453a3f2 : SUCCESS in 4m 37s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/de84e0c9e14348089888a2d848dbb378 : SUCCESS in 4m 57s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/c974da07ca344f40bf3e772ebe6d536c : SUCCESS in 4m 42s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/2223f976d47b4b5890a2c8356828850a : SUCCESS in 5m 23s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/cb616ab71bf54d3b93e2d63a0bf08945 : SUCCESS in 5m 02s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/e43ca3de7db142fcbd4d42aa796de752 : SUCCESS in 5m 32s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b53dc724c40f4ba092e1f76da52fe94b : SUCCESS in 4m 41s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/3dd98c4c2ec8456ebb42343fe47ac5d6 : SUCCESS in 4m 53s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/a9d935783a984e42907ec1b8f637fbc8 : SUCCESS in 4m 43s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/6b8a96d3938c484e9a2eb431ccbf5a0e : SUCCESS in 5m 04s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/ddbd3f7d2f9e43a39b82f143aaa2424a : SUCCESS in 4m 56s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/3a8337fb2a8740adb8ac6602bb6f8de1 : SUCCESS in 5m 43s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/fb204d7565394f60a4e6c39aaa15908a : SUCCESS in 5m 05s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/6bfe7525a0f844d2a9a3cd856ac22ac1 : SUCCESS in 5m 27s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/841fca0408c1451580be8cac0914efc9 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"3c0f0ac49fa45c4f094a6c1d6bac15056d5ecea6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:09:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d101211b55ca4377199a1342daea7d18827dcc66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-15 13:10:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a3427b7deb54321b7b8b7755711174f : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c3a4800126e41a3a2138560781bf1c8 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1ae1649d22b356f2bbb2293cb421eb155fdee14","revisions":{"c1ae1649d22b356f2bbb2293cb421eb155fdee14":{"kind":"REWORK","_number":1,"created":"2021-11-16 17:17:29.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/85/817985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/817985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/817985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/817985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/817985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/817985/1"}}},"commit":{"parents":[{"commit":"1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9","subject":"Retry on redis connection errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1ddddbba734e5eb4cbba82ae9951a8eb09f94bb9"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2021-11-16 17:17:29.000000000","tz":0},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n(cherry picked from commit 47c4d56e446a49726e7145c6cbc3bb7620a431f7)\n(cherry picked from commit 11526e594c7e7cef8b5a4a4220f100d76a418036)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c1ae1649d22b356f2bbb2293cb421eb155fdee14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c1ae1649d22b356f2bbb2293cb421eb155fdee14"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
