)]}'
{"id":"openstack%2Ftooz~821166","triplet_id":"openstack%2Ftooz~master~Ib11562fe3398fe5faeba0477994513f674ee5bd6","project":"openstack/tooz","branch":"master","topic":"doc_get_members","attention_set":{},"removed_from_attention_set":{"11292":{"account":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"last_update":"2021-12-21 12:38:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib11562fe3398fe5faeba0477994513f674ee5bd6","subject":"Fix docstring for get_members()","status":"MERGED","created":"2021-12-09 09:03:53.000000000","updated":"2021-12-21 12:39:43.000000000","submitted":"2021-12-21 12:38:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"821166-doc_get_members","meta_rev_id":"ff132376223dfac89ee86aa4b253fb972e70d4fb","_number":821166,"virtual_id_number":821166,"owner":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-21 12:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-12-21 11:07:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2021-12-21 11:07:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 10:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-21 11:07:35.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"f28cca0aceacb9f544ea8ebd97c88e038b2e59d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-12-09 09:03:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b3c451d59d1f1817cd0faa6a8134cbf0feef522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 10:38:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9a715c7f87414a2586f89abe434e2378 : SUCCESS in 1h 31m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/420c66c41aa643a38fe7ce4ec962821b : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7d3056edb4b4a118b3ebf9d65241120 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83a7ccf27d0d4a078c491d585e0710f3 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9448122302aa413980700f33cd980a86 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29fd19e875ce43668839b7c4ecb8ab2e : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/836f843640ac448c8b4516868065fa7a : SUCCESS in 4m 45s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/e351ebc8f39d48b2bcaef50925c17a87 : SUCCESS in 5m 37s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/5c929f1253b84820b1b49b946f6191b0 : SUCCESS in 4m 35s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/f7cc2c794d544e74995306aff5d15ca4 : SUCCESS in 5m 37s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/2844b7ecd3b8478f8dc4c5bae9a5873a : SUCCESS in 5m 23s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/873793d0dea843edbf94e8c3cd3b3287 : SUCCESS in 4m 41s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/71348b092b234e8fb059128ca5bf04ba : SUCCESS in 5m 03s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/bf3a13251b624957897dbc710b123519 : SUCCESS in 4m 57s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/1e840ed6788e4026aa26402af6acd3c9 : SUCCESS in 4m 59s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/9c058aa3907d4cb288a91f2dac455b11 : SUCCESS in 6m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/46a59124feec415a9b509493c6679146 : SUCCESS in 4m 36s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/d7bc2c1b142c403d95408185e417a4da : SUCCESS in 5m 19s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/074f3a1ff128401296bfb1ececb58a41 : SUCCESS in 4m 46s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/51c7c9cbbde141b396ba1365bfdb6fd3 : SUCCESS in 5m 19s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/10faf02b752840569b3ee4ad4e69d7ce : SUCCESS in 6m 22s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/995d354f6e9a4ba1ba7cf3cc1591ab9c : SUCCESS in 5m 14s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/b32696c2821e474c94b70c0b8783d7a9 : SUCCESS in 5m 31s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0ea3145b93614e7ab9d307d5219c1ef8 : SUCCESS in 4m 47s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/94844810e9894556b44e7875555688fc : SUCCESS in 4m 58s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/d47abd6c12234cf4af2a9e1d969d0702 : SUCCESS in 5m 49s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/bdf92c838608450fb1fed6f5de1b3ee0 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0fb24d2f960d39b583663474366009fe84e3e821","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-12-21 11:07:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d96fd22cea5fcb94c845d3b9ef26bc06f8594178","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 11:08:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85ebe7cb501a856fcd817c3f38d2c1433800e0ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:38:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14b3262eb2334468ba2f5a56514b682c : SUCCESS in 1h 27m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e116d9da97f4b4eba3ff3d5c95d1fd5 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73773e10af78472492500c4c79c873ea : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3870650390b44651acf9ebe92b3c8586 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f72d8078b74190a01fbfdc029038a0 : SUCCESS in 5m 28s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/f8eb88b306f847908c7461cf1eeb2f0e : SUCCESS in 6m 19s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/f06cfd38951146b68008270c3d9991bc : SUCCESS in 5m 25s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/be92d7f8beab43d29b9a7c16f5faf221 : SUCCESS in 6m 10s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/7e1fff2561164ffba0644b38a55d549a : SUCCESS in 6m 00s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/35ef9879566440a6a783aa9f3d2c61c1 : SUCCESS in 5m 49s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/eadcab61529e4120a7b1ce9cdedd3075 : SUCCESS in 6m 18s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9b4376c984dc4be894e75654e04bc33b : SUCCESS in 5m 56s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/112cb6da33fb41cebee7ab587b9dc1c3 : SUCCESS in 6m 23s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/589a31391ed54400b1e55e3481ea194b : SUCCESS in 6m 06s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/ee68dc8519214aaaa11a7676624ed74b : SUCCESS in 7m 06s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/d09d963987e24456ba91caea84f5d878 : SUCCESS in 6m 36s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2fe05d67d4864f708f3b7097a69d89a6 : SUCCESS in 5m 46s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ed1ca5241ad54161a117a0c7afba48ba : SUCCESS in 7m 09s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/85287ad9dd314ccc8ff7d6acf44cf72e : SUCCESS in 5m 45s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/cfe566e4c5324766a06c03023193a966 : SUCCESS in 6m 12s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/cda6ae1f25264ba48c5165a095b43b6c : SUCCESS in 5m 47s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/b8a5f01b22804707a06f619d74736df5 : SUCCESS in 6m 12s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/b2ed07d3f87046b98d2f997de9f6e9be : SUCCESS in 5m 58s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/a059b16a3d1d408e859ef596daaf103a : SUCCESS in 5m 53s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/8f2f00f2905643e8b2bc6c3a94696779 : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8c59d5de37e0a2e7828ef40accc2637778ed2a05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:38:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff132376223dfac89ee86aa4b253fb972e70d4fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 12:39:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/485ce796af66486489a663599a769eaa : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e88e601f925677d3faeadc6851dd28a0b8ec630f","revisions":{"e88e601f925677d3faeadc6851dd28a0b8ec630f":{"kind":"REWORK","_number":1,"created":"2021-12-09 09:03:53.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/66/821166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/66/821166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/66/821166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/66/821166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/66/821166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/66/821166/1"}}},"commit":{"parents":[{"commit":"47c4d56e446a49726e7145c6cbc3bb7620a431f7","subject":"Enable retries in redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/47c4d56e446a49726e7145c6cbc3bb7620a431f7"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-12-09 09:01:22.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-12-09 09:01:22.000000000","tz":60},"subject":"Fix docstring for get_members()","message":"Fix docstring for get_members()\n\nAdd \u0027asynchronously\u0027, fix a typo, and correct a copy and paste error.\n\nChange-Id: Ib11562fe3398fe5faeba0477994513f674ee5bd6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e88e601f925677d3faeadc6851dd28a0b8ec630f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e88e601f925677d3faeadc6851dd28a0b8ec630f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
