)]}'
{"id":"openstack%2Ftooz~829412","triplet_id":"openstack%2Ftooz~master~I756b44443d8b5892c04bd1be9faa010e39c763fc","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-03-22 14:03:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I756b44443d8b5892c04bd1be9faa010e39c763fc","subject":"Support later tenacity versions","status":"MERGED","created":"2022-02-15 21:37:48.000000000","updated":"2022-03-22 14:05:26.000000000","submitted":"2022-03-22 14:03:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"829412","meta_rev_id":"746de657d47cc3e49fa49e3e50f183d0edbfcc35","_number":829412,"virtual_id_number":829412,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-22 14:03:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"all":[{"value":2,"date":"2022-03-22 14:05:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":2,"date":"2022-03-22 12:04:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":1,"date":"2022-03-22 12:04:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-15 22:39:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-15 22:39:35.000000000","updated_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"reviewer":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"state":"REVIEWER"},{"updated":"2022-03-22 12:04:19.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"99c6f93ffc94906b70ddc927112abc5424070755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-15 21:37:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e9d5050193d3efdf8d4b5b66a70d0238a10194d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-15 22:39:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a392686e7214a1f903168b14a556121 : SUCCESS in 58m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f0bd8cc6dbc4700904c1282b7539aed : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caa00891379c4e5381747e412e9d8574 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c4b6e2cf82545f99266ac7be52c6b3a : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52540b4d1f7e44ad9062cd90de69f5c2 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4781b6488f9a454fa8b2c3c5958b4e77 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15a174cb901540b285282f5b80fd9875 : SUCCESS in 5m 11s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/d5403dbde0fb4728842321636b9cd79e : SUCCESS in 5m 09s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/36bda80287d046e788b4c616b5491000 : SUCCESS in 5m 51s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/35eed3b8b8bf48bea22f4c80f6148269 : FAILURE in 8m 06s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/33a12c80052e4cb0b2bc79205065fd07 : FAILURE in 5m 14s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/414d9d90720b4008a6d867850378e374 : SUCCESS in 6m 44s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/035445610438407d8fc8abf2fdbc9dcc : SUCCESS in 5m 07s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/e13ace8d7f464befa06fa6de6681fa2a : SUCCESS in 6m 08s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/40f360f467e94544978d79a17e3aaa61 : SUCCESS in 4m 33s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/1903bad6db2a4030b5dab4d765f599da : SUCCESS in 6m 07s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a25972e1e61c4fd9949a26c74b634af4 : SUCCESS in 5m 34s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/30224b6b829a4170a4fb68b6c04fc188 : SUCCESS in 6m 00s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b63567ccaeaf455e856aeebf165edfad : SUCCESS in 5m 19s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/849cc36622fc458d894b76ff1bb2d1c1 : SUCCESS in 6m 14s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5cba6b4ae4c74c96b678311f5bfe20f9 : SUCCESS in 5m 25s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/0bfea903019541df9a9874d731173f1b : SUCCESS in 5m 13s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/f94947e2c2b143d1aa816678b329466a : SUCCESS in 5m 35s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/a07127cac17841bcb901971bedb6f8a3 : SUCCESS in 4m 43s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/ea7c82ee2e30471ebe1a636250bd8ebb : SUCCESS in 7m 03s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/66005123f3c8470bbcaa8af85932d71c : SUCCESS in 5m 55s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/43b0c327c2fd4dfa9b6ec883c1bd6063 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"be012eb3517d5a0e0b35181788d8f98d27b19a08","author":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"},"date":"2022-02-15 22:39:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e8a5e81af5f1c8e4379c7badd3aedd0f1a6ed879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-16 00:10:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb2f5b43971c458cb098e07bc918fb20 : SUCCESS in 1h 27m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d904c40e0b5549c0a61c287656f0ba5a : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60af2674f4148698c984019878f5070 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c68f81ac92a495f92cd34856d4e7eb4 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1a3bc2a563b4501a5b03d9130867e14 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aae7cbcd995471f876fb7c26c213598 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1719e3312c084ae6912729faa22e8562 : SUCCESS in 4m 40s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/0a32d2561e744f8b98ee98217c94e14a : SUCCESS in 6m 59s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/87feab78340141e49d016d421169b323 : SUCCESS in 5m 05s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/ee446f7cf9f04b509456ae5d6ca2f2c1 : FAILURE in 6m 56s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/670c1301c1a141508adfcaecc2b95ff6 : FAILURE in 4m 41s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/708135e2aa3949e9926078758ae01b80 : SUCCESS in 5m 05s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/5d96a66bc576474fbfd9bdb6d1a217ac : SUCCESS in 4m 59s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/129758a84f474ca6b7e174115e800331 : SUCCESS in 4m 56s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/63821b7fdec440dd96f3842dfdd6e51a : SUCCESS in 5m 04s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d0f94758c5d54ff299b86c8fa8751acd : SUCCESS in 5m 05s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/fe9898ea4f4a411b8c9d4229d52d4ccf : SUCCESS in 6m 26s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/f86bc2889d454a59b950ae0a2f5bc0f9 : SUCCESS in 4m 55s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/23ab9366e6e348a7839e051bae9e2e93 : SUCCESS in 4m 49s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e0421848b9154da69cd7a458322ef281 : SUCCESS in 5m 09s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/b87c64f28886475ca133bb720d79142d : SUCCESS in 4m 45s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/8d090a7b75f74426871ce3c099cd3f90 : SUCCESS in 5m 10s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/5400a08e0169425b8f7290c85e644b31 : SUCCESS in 5m 01s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/802fbb42b218415b940070165ea294e1 : SUCCESS in 4m 20s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/9b95adfcc99e48de9ac14967bedf9b60 : SUCCESS in 5m 16s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/1ce0a4001cbe48d59634cf38909b9983 : SUCCESS in 5m 14s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/2d90698938d5420c804460f45b9be834 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"804a832106d1e8bdb01cb9788cb8666350cb54f6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-16 12:46:43.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"65211ac2369349cc7dae73714dde11a66e6c0a08","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-02-16 12:47:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"47eb5fddf516685fcd65658704030374615b6199","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-10 15:19:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"df65c462743ea47fd93170116493db72e39b1eb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:00:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/094e903e0b194331a09630f677e38e10 : SUCCESS in 1h 35m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5980e2e05c3f4535b08b4cbf2181c238 : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46451551d57e4fbba9a3b3a5da6f1eb2 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cf6ee8c420e49469006c626ba801369 : SUCCESS in 8m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ec2b162715e4383ba8f9647a70c6f3d : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4adb5b7ce8454e368368d93709a67425 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2daf7bc0824016b69d8873f1e30166 : SUCCESS in 8m 06s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/373449a1221c4031a849cd9d18967bce : SUCCESS in 8m 22s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/1c80fc0abd514f0b892e08a87bcafbbe : SUCCESS in 7m 13s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4a4502669e5b42ca89b61c23f8067c64 : SUCCESS in 7m 23s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/cc2b8200df354418ab8acab94779be20 : SUCCESS in 8m 58s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/94fca294313d40cf8f54f14b42b96601 : SUCCESS in 6m 56s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/c9179c3054104258bab6124f77d52ed2 : SUCCESS in 6m 28s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a2c6dcf4425a4b7f9f276706c2d44ab7 : SUCCESS in 7m 13s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/cef63e4ab6ff40d1a1ae8e58fdcc5382 : SUCCESS in 7m 54s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/889ca19bba9e4b2782bc6a5964734576 : SUCCESS in 6m 56s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/a3c09369bf3b4e128da53ae4b08ec9cd : SUCCESS in 7m 42s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ad5f8c162bae4ff69db783dfc6139463 : SUCCESS in 8m 55s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/9d7f27c9230148b69981dfbb1f4abed3 : SUCCESS in 7m 16s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/f6931ed53abb4d36b05cbc750b2d34f6 : SUCCESS in 7m 18s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/ee703ca8cf0149f480ccbd454539c971 : SUCCESS in 7m 27s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/97710cf14ed84677965ea74dee39b97c : SUCCESS in 7m 26s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/aeafda38cc7d4e78bec4d22150a8ecca : SUCCESS in 7m 26s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/1b30d31d731741b585b0166b5b6c301f : SUCCESS in 8m 26s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/4497426f826c49ca86e22259672962cd : SUCCESS in 7m 20s","accounts_in_message":[],"_revision_number":2},{"id":"ff211dfc21f198aac6114c9948e27ea0faec28b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-15 10:35:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"316135534dc1a7152e1ebc8f4629c5c30ef44e5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 12:12:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d772798d69149398b1e523b781deb08 : SUCCESS in 1h 33m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03ae41e4fa9040fd92ced442355a26b5 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5d4bd1982274c599e67d6a28a98c6d5 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08d340bc88564edd83fe500b587b5bc2 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e00439c6f5af47dab9612588e9a81693 : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6204da88b3f4dd084ff84c3c0b88ec7 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be112a6984c409dbb0f02b3b71e4e70 : SUCCESS in 7m 49s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/76d5f4d183fa4d96b906ab50a9adc74b : SUCCESS in 7m 20s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/8322cb0e2f854b1ba4ecb83d1259ec94 : SUCCESS in 7m 19s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/d43dc422fac64745b3a2a5eaf49bd3dd : FAILURE in 7m 04s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/36437736252548929adfd16286982ea7 : FAILURE in 6m 16s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7c22505c86d54263af7fe283f3f1cc12 : SUCCESS in 7m 02s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/d5f78d1eb6114b458ffeef1566a93cef : SUCCESS in 6m 19s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/02d7bb35f5214305af25473bc9d2aa3d : SUCCESS in 7m 44s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/b9ee9fffcac94213ac2c75011e4535c7 : SUCCESS in 7m 14s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/1e3a989d06244cfa933c335fc4a68bae : SUCCESS in 7m 44s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/42f049acf8d74cf9a55c0102d5e1ad65 : SUCCESS in 7m 37s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/d568ecee604f42b0900a323bed1dc04e : SUCCESS in 6m 57s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/64143cf651e342358e05a60bc1f0bda5 : SUCCESS in 7m 06s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/844b4d9ba8fb48359218de1addb0ed59 : SUCCESS in 6m 45s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/2b6227f1371a42fc992fbd1e9f88551d : SUCCESS in 6m 25s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/91403110daa347fd98693545b5eb3577 : SUCCESS in 7m 05s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/db449017c2e647aa80654901a06016b3 : SUCCESS in 6m 54s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/2a0f8702cc76447fb56efc9c8d64426a : SUCCESS in 6m 33s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/eef85707b43044b58b8afe76f466cd12 : SUCCESS in 6m 38s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/4d45217bcf2141118be176fa60e63572 : SUCCESS in 6m 30s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/a2e1d863d6054936a41983ba546b47ac : SUCCESS in 6m 59s","accounts_in_message":[],"_revision_number":3},{"id":"e2195be77995918efb39a8154310fb0e699e2ed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-03-21 12:08:19.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6a5af188d4390eb5a058eaafa14da90a9ec15366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-21 13:49:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d48f22976b34d15b483b51a155c1b90 : SUCCESS in 1h 38m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b56b3feb24f049ec9f8de5906cdf7c65 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6c88d38d1d436692d4652cca39e8b0 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a382d4f4d644b2495e358e1a834cd28 : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8aba6d49ab34e078ee4dadca8f38d53 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0417e1d18ce340f29764c5cbed7fcead : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140014819e9f481da389a5cdb75aaff4 : SUCCESS in 7m 09s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/ad81af2be3f14d339878672dc0d4561d : SUCCESS in 7m 58s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/0279c21e47ed418fb08cc5ddeeb26246 : SUCCESS in 7m 13s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/b16902060f504433b5984d7e7eb7304c : SUCCESS in 7m 27s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/141692d9ad254abe821cd648a1cd23d3 : SUCCESS in 7m 30s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b7a801771579489d9590b86f102f930a : SUCCESS in 7m 41s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/119802db53934b04a008ebc5e4f26c7e : SUCCESS in 7m 26s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/5c6e001de0ec4f78b4fabaab9dd7d5f6 : SUCCESS in 7m 47s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/96354ed3b66b473dab933c9b20880588 : SUCCESS in 7m 38s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6335c86d9d474cc7a259a4a657f9f2e2 : SUCCESS in 7m 43s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2f4ec9286455431f974313285c00d4d2 : SUCCESS in 7m 42s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/2c9b068f2b82461e826d5652c6de947e : SUCCESS in 7m 38s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/91c5b8e70b014b9482f43cf51a42cea5 : SUCCESS in 6m 54s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/bf7d7ac9de384ec39473d844cd55531f : SUCCESS in 7m 06s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/02b72cd8fb344eb99920afcc565f6dd2 : SUCCESS in 7m 18s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0c82ecd394224b84992fe10e21b3e435 : SUCCESS in 7m 22s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/8a73cbe4dbd448eaadb3d30f747514d9 : SUCCESS in 7m 37s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/026a6f7f1ca64285aa89b76df0ada13a : SUCCESS in 7m 27s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/e10047076420488e86ce1e40c41a8b9c : SUCCESS in 6m 41s","accounts_in_message":[],"_revision_number":4},{"id":"5bc4105307c781d4062e4134b031248ed24e79db","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-03-22 12:04:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6475797c8212440847f83fcfcee9217ce7c444be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 12:05:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b0e3099ce1c1d6cc0f8c1e312ac22de04dba92e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 14:03:49.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20fcdfb2149c4fef901013227c5511e1 : SUCCESS in 1h 52m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97cdaa8a8bf14ffa814fd37cc6d617e5 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a29c035600d424487c90a66a4b9cfc6 : SUCCESS in 7m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b9af4ba5d0046679d14141bffd24d20 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484794dcef2c436587520d5bf137ef5c : SUCCESS in 6m 49s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/b6e91cd176934ac382a8548bc0724e22 : SUCCESS in 7m 45s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/ca50d80c598c401b93e98be57b3bd502 : SUCCESS in 6m 39s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/351ac77127364dca8110da3ca853a1e5 : SUCCESS in 7m 38s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/58f2ae8aff124647bf3993e7ae514be9 : SUCCESS in 6m 23s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/82aa9c60545a40b08b19f192205f7be6 : SUCCESS in 6m 45s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/b59b6e80e0fb4e72b45e4a2c770777f2 : SUCCESS in 7m 56s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/2244cf3fd54845638a1e5904dd386961 : SUCCESS in 6m 35s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/cb21b65587d7411098cca84de5213401 : SUCCESS in 6m 18s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/49052bcd4bd94d0184468a4636626c1a : SUCCESS in 7m 50s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/cd295b7dd40743c8a7b30d98d6178ff2 : SUCCESS in 6m 34s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/63452f8370044eb4bd22a03ef3d5c4a5 : SUCCESS in 7m 38s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/405601aca0f147fa9575453f2ef00fec : SUCCESS in 7m 38s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/d078b10fc941420facd79f537862084c : SUCCESS in 6m 47s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/1429fb171bea47e682f2089b0c81f325 : SUCCESS in 7m 19s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/320c28d3b24d413787303ee68b15fbc1 : SUCCESS in 6m 46s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/6cda67b034134dcea2506d646ab46afe : SUCCESS in 6m 56s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/daca07637dfe489abffb6ac10d487d77 : SUCCESS in 8m 14s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/10010115b49b4540b6b6f80b29d8ab24 : SUCCESS in 7m 52s","accounts_in_message":[],"_revision_number":4},{"id":"f3cdb2bff66efa469703944e65637ea846784655","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 14:03:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ffa1e32f8f2d7d460f32802c577b362493a75266","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-22 14:05:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d742a4694844759a943891c7dd9f4c : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5fbb63fa4ff399624fdadd0adc63b35dfb527116","revisions":{"1bd565bf6a9d799057617b0fe87a6352c4260a05":{"kind":"REWORK","_number":1,"created":"2022-02-15 21:37:48.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/12/829412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/12/829412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/12/829412/1"}}},"commit":{"parents":[{"commit":"e88e601f925677d3faeadc6851dd28a0b8ec630f","subject":"Fix docstring for get_members()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e88e601f925677d3faeadc6851dd28a0b8ec630f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-02-15 21:37:06.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-02-15 21:37:06.000000000","tz":0},"subject":"Support later tenacity versions","message":"Support later tenacity versions\n\nInstead of lambda func use the defined\nretry_never Retrying class.\n\nChange-Id: I756b44443d8b5892c04bd1be9faa010e39c763fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1bd565bf6a9d799057617b0fe87a6352c4260a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1bd565bf6a9d799057617b0fe87a6352c4260a05"}]},"branch":"refs/heads/master"},"a08beb9bfe324d00c69a70a5d52d11028cf6b46f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-03-10 15:19:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/12/829412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/12/829412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/12/829412/2"}}},"commit":{"parents":[{"commit":"b250863edaacd39c21c3e5aba7cf18a217ad690d","subject":"Deprecate the etcd3 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b250863edaacd39c21c3e5aba7cf18a217ad690d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-02-15 21:37:06.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-03-10 15:19:13.000000000","tz":0},"subject":"Support later tenacity versions","message":"Support later tenacity versions\n\nInstead of lambda func use the defined\nretry_never Retrying class.\n\nChange-Id: I756b44443d8b5892c04bd1be9faa010e39c763fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a08beb9bfe324d00c69a70a5d52d11028cf6b46f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a08beb9bfe324d00c69a70a5d52d11028cf6b46f"}]},"branch":"refs/heads/master","description":"Rebase"},"beb9b32d3f9ca9c974c047a0ffed3c714e053fe0":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-03-15 10:35:21.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/12/829412/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/12/829412/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/12/829412/3"}}},"commit":{"parents":[{"commit":"e88e601f925677d3faeadc6851dd28a0b8ec630f","subject":"Fix docstring for get_members()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e88e601f925677d3faeadc6851dd28a0b8ec630f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-02-15 21:37:06.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-03-15 10:35:21.000000000","tz":0},"subject":"Support later tenacity versions","message":"Support later tenacity versions\n\nInstead of lambda func use the defined\nretry_never Retrying class.\n\nChange-Id: I756b44443d8b5892c04bd1be9faa010e39c763fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/beb9b32d3f9ca9c974c047a0ffed3c714e053fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/beb9b32d3f9ca9c974c047a0ffed3c714e053fe0"}]},"branch":"refs/heads/master","description":"Rebase"},"5fbb63fa4ff399624fdadd0adc63b35dfb527116":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-03-21 12:08:19.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/12/829412/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/12/829412/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/12/829412/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/12/829412/4"}}},"commit":{"parents":[{"commit":"7ee2780af87a013c83fb5a8a5d0c7ab0767b3db8","subject":"Deprecate the etcd3 driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7ee2780af87a013c83fb5a8a5d0c7ab0767b3db8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2022-02-15 21:37:06.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-03-21 12:08:19.000000000","tz":0},"subject":"Support later tenacity versions","message":"Support later tenacity versions\n\nInstead of lambda func use the defined\nretry_never Retrying class.\n\nChange-Id: I756b44443d8b5892c04bd1be9faa010e39c763fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5fbb63fa4ff399624fdadd0adc63b35dfb527116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5fbb63fa4ff399624fdadd0adc63b35dfb527116"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":1669,"name":"Julien Danjou","display_name":"jd","email":"julien@danjou.info","username":"jdanjou"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
