)]}'
{"id":"openstack%2Ftooz~837119","triplet_id":"openstack%2Ftooz~master~Ibd46a3697b59cd22f386e60d29649acc5c5cd88a","project":"openstack/tooz","branch":"master","topic":"bug/1968323","attention_set":{},"removed_from_attention_set":{"6928":{"account":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"last_update":"2022-10-12 08:35:27.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-10-12 08:35:27.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-04-14 14:05:52.000000000","reason":"Hervé Beraud replied on the change"},"7549":{"account":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"last_update":"2022-10-12 08:35:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibd46a3697b59cd22f386e60d29649acc5c5cd88a","subject":"Enable watch functionality for Etcd3Driver","status":"MERGED","created":"2022-04-08 12:03:16.000000000","updated":"2022-10-12 08:36:29.000000000","submitted":"2022-10-12 08:35:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"837119-bug/1968323","meta_rev_id":"320bf6a7c146b81826c347e6e5caaf77551629e1","_number":837119,"virtual_id_number":837119,"owner":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 08:35:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-07-15 19:03:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-04-14 14:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2022-10-12 06:39:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-08 13:30:06.000000000","updated_by":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2022-04-08 13:30:06.000000000","updated_by":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-04-08 14:10:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-11 12:32:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-04-14 14:05:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 19:03:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"f3575cf50fc58998749e36584fe49d4a26f15c64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-08 12:03:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3dea886f554a56e03c10f42585069274fe1fd8e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 14:10:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa9ed9559dbe4970ab7f18ce46091de6 : SUCCESS in 1h 57m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1295e3e7151748ff82718a51f65246cc : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77cb8ed8cb814044bfc31d78eb53fec4 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73395af3cd654bcd86e62675af0ea900 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1e9ffc20acd4888bd6b5990f31ab419 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a73f86056ae4cd9a17cecd743f001d6 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34476f3ed36444679b84820a3b0cbbd9 : SUCCESS in 5m 21s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/98d14610d8c044f4a1dfd81634ad4a64 : SUCCESS in 5m 47s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/5983dec6c658418da11f9fd564ed4335 : SUCCESS in 5m 08s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2710d8dfcabb43aba5fcd8b3ade3e409 : FAILURE in 4m 58s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/899ce92b685f44edbee498a59a6ceff6 : FAILURE in 5m 02s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/97085ddedbe9438dba91deb0e7069e66 : SUCCESS in 4m 50s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/39f5d499f5df4c7491d1dbf53d44b130 : SUCCESS in 4m 54s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/5a20b20188ca48b580b7e791810375b0 : SUCCESS in 4m 57s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/018f11ef6f2c4e3bb3ae395ba5ad8d2e : SUCCESS in 5m 11s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/c655c67f767c4212a91c38ad2e55dfba : SUCCESS in 5m 05s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/055db806ccdb4b75918f8ddecddac4cb : SUCCESS in 4m 46s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/f13d7ade89b345859dab5dd577eaf824 : SUCCESS in 4m 39s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/81d2d97f847442b28308bbc486474b50 : SUCCESS in 5m 24s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/bf924eca26a746d388ebddc6872b0c9c : SUCCESS in 4m 28s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/9a387097260948878cc9a162ff4187b4 : SUCCESS in 4m 53s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f48eab576f234f9f9f26b28db92f69bb : SUCCESS in 4m 32s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/27636610e6b2442eac77a3bec055d1da : SUCCESS in 5m 15s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/35406ca4e520479f96f1280bbbf58fda : SUCCESS in 5m 21s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/d05c4bf57d814808b0c4cfc9230e46d8 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ac8d164329716887e774ac07576cfba3bce21949","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-08 14:39:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"31c6d60562690c0dd968e3348e9e9f81c89ecf91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 16:04:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29b9c3dada0942cda755dc762e817246 : SUCCESS in 1h 23m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f85f2b26414251b431cada817f94c3 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159cc7ca8d6c426586d73b7bb4d5c828 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/830c66e040e040698faf2abdc7f151ea : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51d2d006524a4af4aa80fdf68a3551bb : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c848ca3c8094ab5bb4c5ac7d142fb70 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3c9d1d815594f3c9f4a398c597c7e96 : SUCCESS in 4m 57s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/41a2f4f945394810b3aa34a78d65b474 : SUCCESS in 5m 06s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/6c9674902a2046dd885910d2b888714f : SUCCESS in 6m 06s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2268404e903d402da842058b903aa1df : FAILURE in 5m 35s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/d04997d784b14fbeb7d456f6a30986d6 : FAILURE in 5m 47s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/71c8088414f54af1974ac45eeeb21d20 : SUCCESS in 6m 14s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/45f24d25a26b4a85b8baf9597cd0690a : SUCCESS in 5m 05s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ff55959dc67a41beaf797bcffb1e8ac3 : SUCCESS in 6m 08s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/9b496e15e69447dcb1c7da22a596ed2c : SUCCESS in 4m 44s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/167b3985646f4173b472b2ed042ecbdb : SUCCESS in 6m 48s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/697c6dff0a6646908e02d9425461dd21 : SUCCESS in 4m 59s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/d2cb1cddbd054e70b0de5fc4d30ee630 : SUCCESS in 6m 09s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/bcbb90e0c39b4e6da37e0cfd8bb9f48e : SUCCESS in 5m 37s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/08b3bfac588348e2ad6ba78d3a143227 : SUCCESS in 5m 01s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/ee083350516d4b558600b865cb9bf099 : SUCCESS in 5m 41s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/24b6f129e5a04ab2b673e9ec9922535f : SUCCESS in 5m 11s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/45cd2cb6ffde40ffa64f7f1c5e5ada15 : SUCCESS in 4m 46s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/454161c901234f4bb82fd2e2a70b3068 : SUCCESS in 6m 16s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/7ec8829022654dbfb1d6a25cea038c56 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c43dba64fa212e4b3ff435849905ca0ac0d86b16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-11 10:55:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fe7daf18f4692010778b12956b10b57432927b2c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-11 12:32:50.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"5b9203b2f5828e5e08093850c018dfe9c0ac2312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 12:46:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/88c7ae356c964bae9042c8a40e961d98 : SUCCESS in 1h 43m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a10fc97d9c0c4e40a2f8cd307a11ad11 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace5777f078546cbb99d92a79682101c : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19198651e0f84e38ba17c5a0824f3887 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f8493662636471284a5372fa221ca0d : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/647837690b6e449bab3c244a094cfcef : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0276796bda18424fa45384d3aa211fe5 : SUCCESS in 5m 04s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/887c313eb80e475c8207a966a68fced5 : SUCCESS in 8m 49s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/f4a72b4159a84d58addd25a6c6d83b15 : SUCCESS in 5m 20s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/1e586ff822c6406184cd58b0a599c652 : SUCCESS in 6m 56s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/79c1bc96c28042aa9ff540718dc3c58a : SUCCESS in 5m 45s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/97e5285db21e44d18894f070b1254f3f : SUCCESS in 5m 41s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9287f359ba734f7381b4d3d2e00fc452 : SUCCESS in 5m 42s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/899bf8e459e44b88a80969d9adae69e1 : SUCCESS in 5m 19s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/6f6f2c82e6ea4e658834f6685728cb8e : SUCCESS in 5m 13s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/94ea142cfb5b4e7fa6a43271e58828f9 : SUCCESS in 6m 59s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/655197645cb948758972c97ec9411926 : SUCCESS in 6m 19s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ec351218ec5d4c34acafdf4dd282066a : SUCCESS in 5m 37s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/686997af63bc49ceb18d36a418356f54 : SUCCESS in 5m 17s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a24feac2d26546f29d0dd64880c004fb : SUCCESS in 5m 31s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/a955f7c32c554b648933c1a169d45c21 : SUCCESS in 4m 42s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c91317228885449db33dd80929e42846 : SUCCESS in 7m 02s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/9d9a2e3a3e2346d8afd662e3624e117c : SUCCESS in 5m 16s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/3109b715be1c4e41b3fdf96c7f6d3d5c : SUCCESS in 5m 11s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/18988fe52abc40e89797ac861b34f4d7 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":3},{"id":"b050cfccb49a1ffcfc76fd315bcd7c6a71aeac17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-11 14:13:07.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f7614b0c81da037944df1817f0ef75a2e5ebfe94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-11 14:19:08.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"54d5c07d76d72ff5b1493eb89d4fd5f2bc5ba84d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-11 15:25:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"688ef96a680fe9749c2cf773020b619aa0c1ea7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-11 16:08:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2432517a99341638cad356e3f7e1e91 : SUCCESS in 1h 39m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aed9d3c8e7f4fff9c4796203020fb8a : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed5c1b8ab4946d4892d43fc2256e36f : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f1fac7e45064522a67cb8381f8fe415 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b0f4f7cfa246b68551db0969f143c2 : SUCCESS in 6m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3a8c381aa1b491b9ef87f2b30298d10 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ed36fafd29424b8aa160b77e410b83 : SUCCESS in 5m 13s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/a007b3b74b1f45c69571d8b6867cc3f1 : SUCCESS in 5m 41s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/7b16a5eaa54c4c38a0259268c8592fdb : SUCCESS in 5m 09s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/e62289eaee344603b7ee44814c46a74d : SUCCESS in 6m 59s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/c4fdeeadb77548cfa5472cec4b8a95c7 : SUCCESS in 6m 21s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/4ea7468fb327438c8fb51891fe5bce63 : SUCCESS in 5m 38s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/eae2f04277f941399e8fe679dd6e9368 : SUCCESS in 6m 03s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/be867cf29a664ac0986f7f4de7a5d01f : SUCCESS in 5m 55s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/572e2ce4aecb43cc9ef791c96cc8acd5 : SUCCESS in 5m 11s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/3ec2ae0a96324ddab0c5a979c9bd535f : SUCCESS in 6m 00s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/a2dc733eaa5c4ee1bb2288f6691fb0f2 : SUCCESS in 5m 13s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/5fbbca575f574f589949efc7b9d4d8b7 : SUCCESS in 5m 26s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/a09fcbddcbff49e885da6ef5912ae7b5 : SUCCESS in 5m 02s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/bae693cc76bc440298657c811164ad6d : SUCCESS in 5m 33s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/49ec140212a0497682cd9365fbd0c27c : POST_FAILURE in 6m 35s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c0fd88a67f964c198ab8b23e53c76198 : SUCCESS in 5m 54s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/cc678fbeaf0e4fc0bd65706eb18c9724 : SUCCESS in 4m 59s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/0ffd9e3ea2dd4bb8bc835105f58b227c : SUCCESS in 5m 31s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/0778ece11878401e9c7b536a052c2e4a : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":5},{"id":"4b1e89df28447f5e7389ad2078afabd531d9d304","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-12 06:42:09.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c9bfa1ace8122f60dabe586dcbd740a05246f892","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-12 06:43:58.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ee585a36ba1eb5b37036ed836c20c66074a48d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 08:07:24.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41a35b31f6de40eebc28ccae9fb6a8a0 : SUCCESS in 1h 21m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc90a152bad34b92b561948596cc707a : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc42544dc3042a3882cc2192e6be887 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96bc531df8ed459abde4fdea7567109a : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daf20969dcb5487894e96bf950e734b8 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/279925ba054245d799a38546e4f2135c : RETRY_LIMIT in 2m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86c1b2d7c494d7699690d4fe3d9ab91 : SUCCESS in 4m 40s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/7bb22d88eb1941c096c831bc77e3baf3 : SUCCESS in 5m 29s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/e226b10e8beb4c1d83c9045783a4733e : SUCCESS in 4m 40s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/bdd60653c31d4005b4bf322879c60a46 : SUCCESS in 5m 17s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/53e603c7c3274d0e9416370f6a576cc3 : SUCCESS in 5m 10s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/5f66a97f366f45bc83215fed2b96b1ca : SUCCESS in 5m 20s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/2707f4ffa74f4e74b73e26d4621a5910 : SUCCESS in 4m 49s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/e028bb3a24dc42a49dad4e5efd9e1a22 : SUCCESS in 5m 21s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/7a8cb812531341cdacd4839d9e8b54ab : SUCCESS in 4m 30s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/cacfaaea77754d8895c1ce018db6c0b5 : SUCCESS in 5m 11s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/8d7cba71c02c463bb0f9f0299bd65435 : SUCCESS in 4m 38s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/9fdf63ebd38f4084996a069fe9443397 : SUCCESS in 4m 57s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/11d6d4b9c0624ae4b4f190b938c44c5d : SUCCESS in 5m 06s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/4e9820a7b449406ab7bfd796f6b27919 : POST_FAILURE in 4m 47s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/0df45586dbbc48149e980ff9af54c5aa : SUCCESS in 5m 07s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/31fead23525c495e913e878587edce3a : SUCCESS in 4m 44s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/806621e6df9e45baa705d67b18527745 : SUCCESS in 5m 16s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/ad9a23ab032748f2ac1df5a1f32ba61a : SUCCESS in 4m 48s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/4a34bf38b6d545f2b264f660d17fdd22 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":5},{"id":"bc715026d9cefe02ab3e2af87c1aa28f82a76f28","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-12 10:27:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"80053fd01212dfd3be8a392665b83883292d1543","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-12 12:39:31.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13346d63f1384ae2a309db50a6594b1f : SUCCESS in 1h 46m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cafb1e0944f45dfb7934f46a5120c6f : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57d1c94267047c9833793bea400ebf0 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d8246e42f6647cea8582c29e75e9339 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bdcfbe113794ee0b0166fe4aefa68d4 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a90a3f767754eea8dba90b1477c97d7 : RETRY_LIMIT in 2m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50f2ac29b77f471a98bcbc9f43f321e4 : SUCCESS in 5m 42s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/b011e87a88604ca79dd52d95a7687aae : SUCCESS in 6m 15s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/3a1034ebd4f843edb61d372283f6adb8 : SUCCESS in 5m 27s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/cc0de4391a4e4255a0fcfa713fabd80d : SUCCESS in 6m 44s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7a79096b6e1b4ebf9521aefd031bedf5 : SUCCESS in 5m 52s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/89f62d96fb924f938d30b9c35e0a064a : SUCCESS in 5m 54s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/04cf707a91314827a699f7ecd926ef44 : SUCCESS in 6m 03s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/4936036ce6d14302b6aa5e6c3086efd7 : SUCCESS in 5m 35s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/b3dfb3aafd8b4b25af2d54ef169501b2 : SUCCESS in 5m 06s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/914e0c7d0527474c9c2cfca500eeb6ab : SUCCESS in 5m 42s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b2ae12586caa47a880ca7f716b042eaa : SUCCESS in 5m 46s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/242084fba1ea479e87e8f0bd58781642 : SUCCESS in 5m 37s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/f4aa40f34f934a159190f9c913789f62 : SUCCESS in 5m 52s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c1ef7530d52b4aeb9db77e5475f5ae46 : SUCCESS in 5m 54s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/78259758763c402ab4a1cc27a91cca06 : SUCCESS in 5m 35s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/9c1b6114ed4241f5993bd91404dfbe35 : SUCCESS in 5m 19s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/c44e7fb80812490aaf8e4cf9643b0fb2 : SUCCESS in 6m 21s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/3878e3547495490eb52f692873ab5632 : SUCCESS in 5m 33s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/e046487ffe374c3eb45a7e8162cc12f6 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":5},{"id":"d4efe27087bb9c71ef2c6e3363a9c46425014890","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-04-14 14:05:52.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4d130689808de09134e09d2c6780766405651066","author":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"date":"2022-04-19 09:34:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"11f7d16f8744722f1c265935df233050c807521d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 19:03:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"364bd2037237c90104c76b515be0a4f6b233eac2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-12 06:39:53.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"fd913e8008463d7a4df8cf3f8d14895859e9d78a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 06:40:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"6d524410304c05380deba6c3c153f9b3fa66126f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:35:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb61afc78a654f70813538513d484c32\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/508dcfeb350d4a73bff061eda80bfc1f : SUCCESS in 1h 50m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1aa7d4a5818441e9c7fb38049a749ce : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6df7ee975e454f3c83c3271980b936aa : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8972df2188a54020a93dbb83fa4fc04f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3190f0138f44c1a9e80464fa3bbb32b : SUCCESS in 4m 46s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/ac6801a998e34993af41042054c14383 : SUCCESS in 5m 17s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/e655d747201645e4a1bd0553d4e5f7d1 : SUCCESS in 5m 42s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/74f433e4bd7a423a931eb9191e45d644 : SUCCESS in 5m 40s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/947601e2b1294f8987136589810e5ba4 : SUCCESS in 5m 57s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9dc9ebd2596847ac8cdbbef778e2287f : SUCCESS in 4m 17s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/b6a53dec25c54027aa6dc77fb2f03a62 : SUCCESS in 5m 37s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/6d7d63d684ee4b2a80231d16efc1bd1d : SUCCESS in 4m 22s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/31182b1a2e814bf8be334466129e2f79 : SUCCESS in 5m 29s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/008e61b994f44106b7e31c4b96ccaa6e : SUCCESS in 4m 44s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/87e80d9788bc49a5b1627ee4a8f3931d : SUCCESS in 5m 28s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/57052366271f4ecba61d946cc547f537 : SUCCESS in 4m 27s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/515551a66f0e4f2cbff9119b3219ea73 : SUCCESS in 6m 06s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/cb2f50de0e574de291c8167ce0ec1e10 : SUCCESS in 5m 27s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/abaa5dd37e94414998bf1fd2f8136433 : SUCCESS in 4m 32s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/1d0e6d66aeb84d148724ee0131c2277e : SUCCESS in 5m 36s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/b221b89d8b0a4cee8c4af3bab3125625 : SUCCESS in 4m 30s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/ecb22ef9e27f4431989bb793c40421f3 : SUCCESS in 6m 03s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/37ba8aae88ae479daf3b364cdbc35a8a : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":5},{"id":"8b950e451557790debf49c3b68102ee0105cd357","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:35:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"320bf6a7c146b81826c347e6e5caaf77551629e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 08:36:29.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d7be4c220a1476a83fcd10693327b47\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb66619e8e84b8896a970a4fe3d5188 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c0f6cc0fe5a0ebf43368aab3e8897a0eac0ab489","revisions":{"797a290a8cc37f702699e8095709fbf3d5bdd62e":{"kind":"REWORK","_number":1,"created":"2022-04-08 12:03:16.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/19/837119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/837119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/837119/1"}}},"commit":{"parents":[{"commit":"96f91b961ff593406a7cf9155e309d9988f8e2ae","subject":"Fix getting group with prefix in etcd3gw driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96f91b961ff593406a7cf9155e309d9988f8e2ae"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-08 11:47:10.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-08 11:55:55.000000000","tz":180},"subject":"Enable run_watchers for Etcd3Driver","message":"Enable run_watchers for Etcd3Driver\n\nAdd CoordinationDriverCachedRunWatchers as parent for Etcd3Driver\n\nCloses-bug: #1968323\nChange-Id: Ibd46a3697b59cd22f386e60d29649acc5c5cd88a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/797a290a8cc37f702699e8095709fbf3d5bdd62e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/797a290a8cc37f702699e8095709fbf3d5bdd62e"}]},"branch":"refs/heads/master"},"6e6bda0fb08d6d9b27cfbf01f9c088e9e5bb49fd":{"kind":"REWORK","_number":2,"created":"2022-04-08 14:39:53.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/19/837119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/837119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/837119/2"}}},"commit":{"parents":[{"commit":"96f91b961ff593406a7cf9155e309d9988f8e2ae","subject":"Fix getting group with prefix in etcd3gw driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96f91b961ff593406a7cf9155e309d9988f8e2ae"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-08 11:47:10.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-08 14:39:03.000000000","tz":180},"subject":"Enable run_watchers for Etcd3Driver","message":"Enable run_watchers for Etcd3Driver\n\nAdd CoordinationDriverCachedRunWatchers as parent for Etcd3Driver\n\nDelete raising NotImplement error\n\nCloses-bug: #1968323\nChange-Id: Ibd46a3697b59cd22f386e60d29649acc5c5cd88a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6e6bda0fb08d6d9b27cfbf01f9c088e9e5bb49fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6e6bda0fb08d6d9b27cfbf01f9c088e9e5bb49fd"}]},"branch":"refs/heads/master"},"d57953c7b31a27727b0f7f8e3e4ac6d103e46230":{"kind":"REWORK","_number":3,"created":"2022-04-11 10:55:31.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/19/837119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/837119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/837119/3"}}},"commit":{"parents":[{"commit":"96f91b961ff593406a7cf9155e309d9988f8e2ae","subject":"Fix getting group with prefix in etcd3gw driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96f91b961ff593406a7cf9155e309d9988f8e2ae"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-08 11:47:10.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-11 10:53:43.000000000","tz":180},"subject":"Enable watch functionality for Etcd3Driver","message":"Enable watch functionality for Etcd3Driver\n\n* Add CoordinationDriverCachedRunWatchers as parent for Etcd3Driver\n* Update supported and unsupported methods in driver\n\nCloses-bug: #1968323\nChange-Id: Ibd46a3697b59cd22f386e60d29649acc5c5cd88a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d57953c7b31a27727b0f7f8e3e4ac6d103e46230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d57953c7b31a27727b0f7f8e3e4ac6d103e46230"}]},"branch":"refs/heads/master"},"dc68adcc2461cffd7fde47865e03b000da33bb94":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-04-11 14:13:07.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/19/837119/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/837119/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/837119/4"}}},"commit":{"parents":[{"commit":"96f91b961ff593406a7cf9155e309d9988f8e2ae","subject":"Fix getting group with prefix in etcd3gw driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96f91b961ff593406a7cf9155e309d9988f8e2ae"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-08 11:47:10.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-11 13:57:51.000000000","tz":180},"subject":"Enable watch functionality for Etcd3Driver","message":"Enable watch functionality for Etcd3Driver\n\nAdd run_watchers support to Etcd3Driver.\n\nRaise NotImplemented during leader change watch methods\n since just like in any other drivers to let corresponding\n tests pass\n\nCloses-bug: #1968323\nChange-Id: Ibd46a3697b59cd22f386e60d29649acc5c5cd88a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dc68adcc2461cffd7fde47865e03b000da33bb94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dc68adcc2461cffd7fde47865e03b000da33bb94"}]},"branch":"refs/heads/master"},"c0f6cc0fe5a0ebf43368aab3e8897a0eac0ab489":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-04-11 14:19:08.000000000","uploader":{"_account_id":7549,"name":"Ekaterina Chernova","email":"katyafervent@yandex.ru","username":"efedorova"},"ref":"refs/changes/19/837119/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/837119/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/837119/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/837119/5"}}},"commit":{"parents":[{"commit":"96f91b961ff593406a7cf9155e309d9988f8e2ae","subject":"Fix getting group with prefix in etcd3gw driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96f91b961ff593406a7cf9155e309d9988f8e2ae"}]}],"author":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-08 11:47:10.000000000","tz":180},"committer":{"name":"Ekaterina Chernova","email":"efedorova@mirantis.com","date":"2022-04-11 14:17:22.000000000","tz":180},"subject":"Enable watch functionality for Etcd3Driver","message":"Enable watch functionality for Etcd3Driver\n\nAdd run_watchers support to Etcd3Driver.\n\nRaise NotImplemented during leader change watch methods\nto let corresponding tests pass since only group\nmembership is supported\n\nCloses-bug: #1968323\nChange-Id: Ibd46a3697b59cd22f386e60d29649acc5c5cd88a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c0f6cc0fe5a0ebf43368aab3e8897a0eac0ab489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c0f6cc0fe5a0ebf43368aab3e8897a0eac0ab489"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
