)]}'
{"id":"openstack%2Ftooz~840641","triplet_id":"openstack%2Ftooz~master~I9f9e63fcea02a90a08bfb381755ed76cdd35ddef","project":"openstack/tooz","branch":"master","topic":"drop-py36","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-12 13:54:38.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"I9f9e63fcea02a90a08bfb381755ed76cdd35ddef","subject":"Drop python3.6/3.7 support in testing runtime","status":"MERGED","created":"2022-05-05 10:13:58.000000000","updated":"2022-05-12 17:40:14.000000000","submitted":"2022-05-12 17:37:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840641-drop-py36","meta_rev_id":"d23a89f054bd1ed76541917f025d75faf695e8d3","_number":840641,"virtual_id_number":840641,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-12 17:37:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-05-05 17:04:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-05-11 14:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-05 22:58:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-05-11 14:23:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 11:31:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 17:04:42.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-05-05 22:58:33.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"683198a96e4af22cbd85b07bc7fac8734fa4a0d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 10:13:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44e1c86a7aa84935e4910e7ecd34d1ee190d56b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 11:31:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/677356ed682e4c6c95eaf5c9e4c01714 : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2627b8deb47346238a91464e8ec2ca7b : SUCCESS in 1h 13m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cc318c9d96e4598a81fc1294faad1f4 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/133a54b216264132b4614c7b1a19f2d4 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11911b4ff8974a2ca757bc53df98385a : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79d8e820d74a43b18018e1a8dad382b1 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1088548945624db4abb24477e1006ec8 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a10648c36396458bb5158f55ca95190b : SUCCESS in 5m 29s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/a8dd59b7584e426bb24d63357389764d : SUCCESS in 5m 05s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/cb3f5e6cd77c43698b0dee9310ccf04c : SUCCESS in 5m 32s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/8d58425e96f6448fb6bac03789200566 : SUCCESS in 4m 39s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/4bcdc221ac09491ca26901a79456adcc : SUCCESS in 5m 31s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/45a12e1052f54964af85d836dd242162 : SUCCESS in 4m 53s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/c10e97c6c2884b3599cc2a574da4570b : SUCCESS in 5m 00s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/286855bdf17a41a5966dba747410ee12 : SUCCESS in 4m 31s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/2c70f24fbe564c588a519468394a9481 : SUCCESS in 6m 51s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/0b252fb49d634bf6b3112228b4126dba : SUCCESS in 6m 43s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/07ecdb102aab440f876bbbcb4f7bfdfb : SUCCESS in 6m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e02439c81d584babb95baf86b6882f32 : SUCCESS in 5m 02s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/669b544be52147c0b354d285b840fc13 : SUCCESS in 4m 34s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/6dcc0c40f8434906ae266653a9b26e3e : SUCCESS in 5m 05s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/3e5b6602d6d44b2ab2bd5ba91b266755 : SUCCESS in 4m 51s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/61c83b06314d4e3c96cd292a727f071a : SUCCESS in 4m 25s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/e87092bcd2c0481abe90cbf13cd68dd4 : SUCCESS in 5m 17s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/d715aad90a7d470193fd6a9a5469ef12 : SUCCESS in 4m 37s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/9a640637bdf347479f18f5f97b9abc22 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0dbbb879ea2ea7163307468b34ddc0f2712483a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-05 14:01:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"04768d188d7d8c3556f21dac54b615e4d8dcb8f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 15:23:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/88fed0df214b4c7f821852cfb701d4cf : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/032a3771827640f79592abf11ea0b4a5 : SUCCESS in 1h 20m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0893c516c7d46da93f3de1c04f94731 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5006eeac28349b2953f115d3fd6bf13 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad71fe1dfb314f2ea8eab0d895e81234 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9cfc56c1c66b4a6cae022f0db181601e : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64029f4c6fd542749e0d88d6da296481 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570e5e7a2aea42ed8a7d63b8442ed96b : SUCCESS in 5m 15s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/fa5217b1a8d642299b4bf333e62e692b : SUCCESS in 5m 35s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/2293b39fa2f7488ab7eaaf8f783e14cb : SUCCESS in 5m 45s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/816179cfe65d41469524900db51aa5bb : SUCCESS in 5m 07s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/7e19c562f8224c86a9a01befb81f8e70 : SUCCESS in 5m 59s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/d52788abd6de4e578c479bc4f3e8b1a0 : SUCCESS in 5m 59s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/e442cad8e064406db034ececa37a0d15 : SUCCESS in 6m 04s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/2fdc912a0aa948519fd8159f7b258467 : SUCCESS in 5m 33s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/4e240bcdbe06448083daa8f3cf5cbce5 : SUCCESS in 5m 50s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/41b730dd9a374ec7a10e598bad084568 : SUCCESS in 6m 35s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/c85119e9a72948f8938410d1e95e1514 : SUCCESS in 5m 28s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/efbb039d994d464582fc87dd4ac806d9 : SUCCESS in 5m 27s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/78a86f6c74354c199699aab7099cf66e : SUCCESS in 4m 54s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/07284e3a829a4feaa90c184394d5b618 : SUCCESS in 5m 20s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/78a939c1abb846ce8049a4949eeacf32 : SUCCESS in 4m 52s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/6b7cea9ea08643188a00cc79f36f5ef1 : SUCCESS in 6m 04s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/5ee882b18ece42a4a0ce66164e026d05 : SUCCESS in 7m 15s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/74709682a31b4e1fad5aaa5b9b4b36bc : SUCCESS in 4m 39s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/3b2de34145f74a8882c3da5f3744c190 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b1bc3accde6cc71c5f3e513cf1afea5e45a7d313","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-05-05 17:04:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"027a9af383d6079c9e1bc31f61c364a00ffc9fbc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-05 22:58:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"27ce3d47d32445378dba2c78f7f7a3b4042dc197","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-11 14:23:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"454dd5239f2fbaec7cd13dcba50286d8e5d7892e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 14:24:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ea96a8a1db86b5b6cbe02c261c6ba0841c6efc70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-11 15:53:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72a9c61ac5724ca3ae1d1485b7310274 : SUCCESS in 2m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e30a5d243bcf4cfc8090d597dc79774f : SUCCESS in 1h 28m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b35883456d542d695f5fce788f6b08c : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3b03b0311ec4aabae29e8ee8af3d629 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8e436085e5644f6bdccb39db94e4813 : FAILURE in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d373959e51e9466e83c231d343c0ca95 : SUCCESS in 6m 10s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/1d3a489d6e354004a423b867f9b79e1e : SUCCESS in 5m 29s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/a5fa812908424a0baf4119be655caaec : SUCCESS in 5m 35s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/9554b1f6215449c79ff0a648a5b6463a : SUCCESS in 5m 10s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/e074058a3d84437a9f5bbf677a9f4710 : SUCCESS in 5m 37s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9d44ce48c5d54da69db316ce8a901836 : SUCCESS in 5m 26s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/995025f627ac4e11bf99c81ee7169858 : SUCCESS in 5m 57s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/992de75bc01744ef9e7cfbf753e61df4 : SUCCESS in 4m 39s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/f25b810893484dee91b6439badbf937c : SUCCESS in 5m 37s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/5a6931cc6b8945cea9a5f471321abf70 : SUCCESS in 6m 07s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/4401ecc42f5e4b6c9bc99a930d89ee85 : SUCCESS in 5m 33s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/cbdc34cfc5ed4d39b442c000e5024bed : SUCCESS in 5m 09s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/966870ed5f0c42968e19c797fa459ce1 : SUCCESS in 5m 27s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/07532115513c463298f8678086ddd3ff : SUCCESS in 5m 11s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/b4478e6da9894c8bb84df3178b345679 : SUCCESS in 5m 30s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c533676670514a7fab0a908d3a10826a : SUCCESS in 5m 23s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/ae8d589b09b44d54a01572c8f1babf03 : SUCCESS in 9m 01s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/6dbc8a8ec13246fab751fba290bd8995 : SUCCESS in 5m 30s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/07a71c9e5c464951a8107a4ebcba8414 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"ba08f744124507b3e5af4f6e7e8e1d4e77d1b549","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-12 13:54:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c35d1f877ec9a7ff7a82b4e9ef4c2679f9372ad2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 16:01:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d942a6044b1a43be901ccb476abcf021 : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3798b5faf1c4de4be3f57420ff4de00 : SUCCESS in 2h 00m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a34e01a90dd405994665c7a7464d192 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f38dc11bf40c4663ba9cea158cee699b : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a60b513300844293b366d80c2e70be42 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00ab4fe4a318478b8e43f180d871290a : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6152ce32e734488951dce6aa1bda9d9 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd3fb10be054273b27da7cdc322f83d : SUCCESS in 4m 48s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/2ad2ea5aed68438090ad7cc8844cf7fc : SUCCESS in 5m 53s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/c15ba9afa1fe4a0e82993c9921482584 : SUCCESS in 5m 16s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7ebfa218a0c94eb59b3da4558d33bc0f : SUCCESS in 5m 53s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/820d27b1ac764df4b6b4a4a513508e33 : SUCCESS in 5m 25s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c7d76be5a6324e93aabc40384a2a3381 : SUCCESS in 5m 05s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/637439e2c0fb4151a68900a6eea9f719 : SUCCESS in 7m 28s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/efe6202be09a4bb394ae1015c6f16a41 : SUCCESS in 5m 09s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/ccaa96352173490da9552f7815c4b867 : SUCCESS in 6m 01s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/4439504cf9e44283a82a81ce46fe8855 : SUCCESS in 5m 43s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/66a5a5a38e4a4c1b8d4fbe74d624c130 : SUCCESS in 6m 00s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/9e3747ae4c9a402a9f518581d8bbd15a : SUCCESS in 5m 51s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/d3275c6a9d2941a3b2d6ba82baac8308 : SUCCESS in 5m 08s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/dde283c6959e440e831d39c673bf5b14 : SUCCESS in 4m 45s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/85c60887c0af4d17b6cb3fd3f3155562 : SUCCESS in 5m 48s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/251cfa25ed8b447597aa6772a60c7f2b : SUCCESS in 4m 52s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/621f3b48e6404eeebf5dd87d58dcf6b0 : SUCCESS in 5m 41s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/e49c95aaa6744130937719a6538c36b7 : SUCCESS in 5m 23s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/75089761844c4f6a98359f44e2a5962c : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"091c8d8168b415beb1d565acd9715feddd97eb67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 16:01:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2a7b0f0d8f02c86c1cc5b1a0a187d988afd382bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:37:36.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/24c39550188a42c0ac54725aa3470d08 : SUCCESS in 2m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e28e9412ae241d982c7667b42df4cff : SUCCESS in 1h 31m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be29b2a523f46249246138e5641081b : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31d059eaf7cf4d0da8cf61a2e4626ad4 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74efabe84a314d7ebb2b718c1c0be87b : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5d81fc28af146768f20ead552baea1e : SUCCESS in 4m 50s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/91217f70c00e4557895ed69f4dc527ab : SUCCESS in 5m 13s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/3379bf3f9ca44a76931803cc4eea6e4a : SUCCESS in 5m 29s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/081a2767a6184894a39bcf6235db9b45 : SUCCESS in 4m 42s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/087db56695ad46e68908ef212e627e83 : SUCCESS in 4m 57s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/2b74ccd360884457b547f53b9db1c4ba : SUCCESS in 4m 50s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/b3311aaebfc2480fbfe506c0ca616d93 : SUCCESS in 4m 59s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/8b9f829a3321425180ee44c67915dc50 : SUCCESS in 4m 58s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/6e402d53f7214fc68f3898e74c0492b5 : SUCCESS in 5m 14s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/c826d13ede6e4ea6adff0a08a1281454 : SUCCESS in 4m 53s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/3ba8fd8aaa4b4f73a55879657018b079 : SUCCESS in 5m 22s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/4e203ad670f844dd9a343db560bc9fcf : SUCCESS in 4m 46s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/a6794330a4a9458f815ab289ac27bfcb : SUCCESS in 5m 03s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/be82f55912304aacaf7997e05cf764e8 : SUCCESS in 5m 04s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/65e15977c1a946cbbec9d80fcaff61c4 : SUCCESS in 5m 23s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/ab1fe6f8bbaa4791b53569a624ac22c1 : SUCCESS in 4m 54s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/0d48f824ed5e475c8aa11f86ed89c5ae : SUCCESS in 5m 03s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/c8975817489f4a788e091e4b151d7adb : SUCCESS in 4m 54s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a94396389bd449718c3bd4c5218f2f8d : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":2},{"id":"c51b339ee20b518becc12dd133d1fb4a56a89c0a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:37:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d23a89f054bd1ed76541917f025d75faf695e8d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:40:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f1772253c394ca1b7a90e7e594fce37 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b21c20f79fbcf81b79706a5910ff802b405ca733","revisions":{"fb4b69414909606f7ad8ea3a279aa3cac3d10f6f":{"kind":"REWORK","_number":1,"created":"2022-05-05 10:13:58.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/840641/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/41/840641/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/41/840641/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/41/840641/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/41/840641/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/41/840641/1"}}},"commit":{"parents":[{"commit":"96f91b961ff593406a7cf9155e309d9988f8e2ae","subject":"Fix getting group with prefix in etcd3gw driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96f91b961ff593406a7cf9155e309d9988f8e2ae"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 10:13:39.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 10:13:39.000000000","tz":120},"subject":"Drop python3.6 support in testing runtime","message":"Drop python3.6 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6 support, project started adding python 3.8 as minimum,\n\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nChange-Id: I9f9e63fcea02a90a08bfb381755ed76cdd35ddef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fb4b69414909606f7ad8ea3a279aa3cac3d10f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fb4b69414909606f7ad8ea3a279aa3cac3d10f6f"}]},"branch":"refs/heads/master"},"b21c20f79fbcf81b79706a5910ff802b405ca733":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-05-05 14:01:57.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/41/840641/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/41/840641/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/41/840641/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/41/840641/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/41/840641/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/41/840641/2"}}},"commit":{"parents":[{"commit":"96f91b961ff593406a7cf9155e309d9988f8e2ae","subject":"Fix getting group with prefix in etcd3gw driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/96f91b961ff593406a7cf9155e309d9988f8e2ae"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 10:13:39.000000000","tz":120},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2022-05-05 14:01:30.000000000","tz":120},"subject":"Drop python3.6/3.7 support in testing runtime","message":"Drop python3.6/3.7 support in testing runtime\n\nIn Zed cycle testing runtime, we are targetting to drop the\npython 3.6/3.7 support, project started adding python 3.8 as minimum,\n\nexample nova:\n- https://github.com/openstack/nova/blob/56b5aed08c6a3ed81b78dc216f0165ebfe3c3350/setup.cfg#L13\n\nAlso indicates that we support python 3.9.\n\nChange-Id: I9f9e63fcea02a90a08bfb381755ed76cdd35ddef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
