)]}'
{"id":"openstack%2Ftooz~842685","triplet_id":"openstack%2Ftooz~master~I6fc33b8e0a88510027bcfc30d1504489d2a91b4e","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-15 20:17:58.000000000","reason":"Change was submitted"},"30858":{"account":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"last_update":"2022-07-15 20:17:58.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-05-31 09:03:37.000000000","reason":"Hervé Beraud replied on the change"}},"hashtags":[],"change_id":"I6fc33b8e0a88510027bcfc30d1504489d2a91b4e","subject":"Fix inappropriate logic in memcachedlock.release()","status":"MERGED","created":"2022-05-20 08:10:54.000000000","updated":"2022-07-15 20:18:58.000000000","submitted":"2022-07-15 20:17:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"842685","meta_rev_id":"3a7c0131a6fe448c3582375e9870b66057a982dd","_number":842685,"virtual_id_number":842685,"owner":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 20:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-07-15 18:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-05-31 09:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2022-07-15 18:54:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-20 09:31:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-23 06:13:03.000000000","updated_by":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-05-24 12:55:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-05-31 09:03:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 18:54:57.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"01e854b9aa59d0e53d30683e9b479728db671663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-20 08:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"637b6624366e44ccc3b3c85b46f6c10c812c4fc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-20 09:31:56.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae58cafcc6294217aa81b3269ea6ea0a : SUCCESS in 1h 15m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b19546a7d8704997ab6de3d8d02038c6 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/146300e0c68646f6849cb48201012426 : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d220106ec0e0449493680c22d125109d : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e22f028ff8654ef198f6e1262f8e6bcf : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/18e954f91646492f895b5a149c752413 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a2281838d94895ac92b567c5fd8844 : SUCCESS in 5m 27s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/2b65cf2ec0984b5bb4afbbba2db7ed0d : SUCCESS in 4m 58s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/5a17654ef77a438383c22a3d54928a69 : SUCCESS in 4m 55s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/34d25d149db744c9b24416aadab474b7 : SUCCESS in 4m 38s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/40707c85976548e3949e36cd92e11500 : SUCCESS in 5m 10s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/6a31e106a7494f3b9be29a130ca486a0 : SUCCESS in 4m 48s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/05b2fe2133fa4257aff35b10efd8b478 : SUCCESS in 5m 16s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/dbd9c0c1917b4ff59e352c7297531a55 : SUCCESS in 5m 16s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/3b2049b9c3734cd68fb35eeecb028330 : SUCCESS in 6m 11s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9dec0abd94a9447980bd8d197ea63a09 : SUCCESS in 4m 58s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/46478001c52048ac8a0f3e297d808d3c : SUCCESS in 5m 23s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/c0f86fc3e2f94d58b099096e37069080 : SUCCESS in 4m 49s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/3eb877c057b143f78a65db7db57a5e91 : SUCCESS in 4m 44s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5b8e7b984ef448239fb1d659d9c41eba : SUCCESS in 4m 23s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/bfb7269bbbcd477fad683da7851edb91 : SUCCESS in 4m 49s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/5a8b74f13a754f0c80f8cd3d98d45806 : SUCCESS in 4m 26s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/c432a0293a8c4308878bda2936509b1d : SUCCESS in 5m 12s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/87922f0910084574aedc43f808526fdb : SUCCESS in 4m 49s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/c372be7f91c34e8298bdf16323190435 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ac1f723b72e9e94d62998f4e2e168b28593d70bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-21 07:35:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc05cdd23e2b7239b97d5584a613f42e2da4c53c","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-21 07:36:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2478f54c9bd19624da9e5697ce653daec013cf84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-21 09:21:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bba74f09ad3c411892b718223676db00 : SUCCESS in 1h 41m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72765d4d700f45cdaee657db1d2a0c84 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27e6b42ae73c4b1bb28c674e8febd15a : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29193b0e04e5493f80b17e38477793e8 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12d88cf0a3284d50a81fcc984dffa352 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21b25b55055443d3b026943e0db8179c : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3a83749ff7b454caf7162ee58b7921a : SUCCESS in 5m 17s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/5a6b0b75fd0c43808c5663cc905e28e4 : SUCCESS in 6m 17s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/35d9fa7d84da4a35b44b9c12b162998c : SUCCESS in 5m 27s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/0a5a0580ac494d868ffca628c5e5f62d : SUCCESS in 4m 24s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/054d745d9da945edaaccf9ed66d230f3 : SUCCESS in 5m 56s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/df8275c28b7a48a589272aa52524858a : SUCCESS in 5m 13s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/202c2e38322349869035c9c485bf6211 : SUCCESS in 4m 40s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d493dfe984d3401fbcaa28201e9dc282 : SUCCESS in 5m 19s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/724ef6b72569456d8353e6772d05f671 : SUCCESS in 4m 36s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9424470c15e74ce880ab70cfa08e1900 : SUCCESS in 5m 22s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/dfe1395421654b328831ef7a8bf4544b : SUCCESS in 4m 28s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/c9428ca802fc46378efe4771c87e5c2b : SUCCESS in 5m 14s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/c3eef98dbfe8454eb43fa4c7b940556c : SUCCESS in 4m 37s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7bfc9848e78743d18ad068e5a08ad4f7 : SUCCESS in 5m 08s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/48883d56cd94461a84ad592f332d8552 : SUCCESS in 4m 27s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/9ec1177e266643359c597cea325cc166 : SUCCESS in 4m 19s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/1e7b08ccc9cc49f2b50ea2b138e4dffd : SUCCESS in 5m 24s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/18f39fa30a7e45abb7eeb9058ce72d6e : SUCCESS in 4m 20s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/d7dbe19d700d451fa9b67feedf8f3357 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"82d0307b8397e3c3348393d26232c10c4e5b204f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-23 06:09:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cfec311bb2c2ce4a6c658272c11c493f02a845c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-23 07:41:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/919cbfcb52dc40519e6eeba613d81856 : SUCCESS in 1h 30m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0519f9f3064e4fb1a5870e3a45975772 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf229f26d55412baaa2111c5a66c2b8 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d655b1b3f3564942a996f7b983b98598 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e96f53456a544d88b0fe7d428116f9d : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7877873c329416abaafe65058c3d976 : FAILURE in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/effdceb72e5540fc96e69e9b3c426868 : SUCCESS in 4m 51s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/2ac4e99e295d4f63bfb8a748fb2b7a0f : SUCCESS in 5m 04s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/47e9759b05ed48c79c38e445fb3495e7 : SUCCESS in 5m 39s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/363558e8133b476889fe58ab26436f8e : SUCCESS in 4m 15s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/8df8702b02244e9eb018f23e7c20fac5 : SUCCESS in 5m 05s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/3cd7e3572bae4096b272df5fa66f058a : SUCCESS in 5m 29s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/948e3b428c934a329033a4e16f6ade03 : SUCCESS in 5m 43s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/138e771dc17f446bb66ad015c1523951 : SUCCESS in 5m 20s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/4c4af2dcb2a94583ae4967ecf39653be : SUCCESS in 6m 01s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/38f99c195ea1475b990aef61d4b7fc25 : SUCCESS in 5m 12s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/12dccf67c0764708acca0927e5d4cfb7 : SUCCESS in 5m 41s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/03db88737d6b4ac78adcc123c7b62a01 : SUCCESS in 5m 32s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/ce28ef88f5f04073991af02a235ac0df : SUCCESS in 5m 30s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/8a739477968f48f886d2ca158374ddc4 : SUCCESS in 5m 12s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/188073e67f004ac694ff473802e051c6 : SUCCESS in 5m 36s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/61b394c358d74dc2b3e0bd2fc8ec5b7b : SUCCESS in 5m 14s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/1078c22d8cf844d198bfed07a873135e : SUCCESS in 5m 23s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/26191a9de63c4911a55094a26f9182be : SUCCESS in 5m 23s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/79c36dfaeb7f4ed9b240d2f0afbecea2 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":3},{"id":"02e038944bb3b0d3770a8923894b51d6ecefb81a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-24 12:55:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"405ec8353511291ec0d7576607ab48e9355c3789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-25 09:06:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bd603082175672586da2f7243b2c865f44fdf90c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 09:34:29.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70aaeb24aa2248de866d4644b25a88df : FAILURE in 22m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78bfd4c98fe64eada17d6201ccec7ec8 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e1c3cfe8de347d4a8adca56b08aaf48 : FAILURE in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d7a7a7c5fdd49b3885fca7ab0410737 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e963dcc028742f788b385b36ba57c1c : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2402dc7739c946069b36e3f4b1b88b27 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10cf695207554d168dfd681f36f0599e : SUCCESS in 4m 58s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/5fd4360b316740ae93dcb8afba746d75 : SUCCESS in 4m 56s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/b286757723ed408dac9c3a6577f46d10 : SUCCESS in 4m 57s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7cff963a7adb4640bc76f71a29065fb9 : SUCCESS in 5m 40s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/f94ae1aef66b4a65823b8f7ab31b744c : SUCCESS in 5m 25s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/53848b16a0c54e75b65b132657a1ca3c : SUCCESS in 4m 51s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/5165633a439949d59e201666fc3be66e : SUCCESS in 5m 02s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/81d353d98cd54a7aaadab0c8cc24586d : SUCCESS in 4m 58s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/604dfc9a6c564e3ba6caad1c4cede642 : SUCCESS in 5m 35s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/1862154a031c4ff8907ce100b55bc564 : SUCCESS in 4m 51s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/91e07de6b40341bba379d7a051cf6d1c : SUCCESS in 5m 53s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e4f283a108824b21b8bd25889c1c6517 : SUCCESS in 5m 08s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/5f9e6d1a89314177953cf871642f4245 : SUCCESS in 5m 28s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a09ce2a8628f441d8033bb2fe5a66959 : SUCCESS in 5m 05s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/bc158b30cabc4acca38d98ca008a5c10 : SUCCESS in 5m 31s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4223ad158c174cb297f1669825afd927 : SUCCESS in 5m 52s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/6dd4b1c9497f4c7a8a41445b7235ccc2 : SUCCESS in 5m 57s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/488aaa278eac4897a78442d5eae31a75 : SUCCESS in 5m 01s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/1018d8b17fbb426988513ddef32237b3 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":4},{"id":"cb9265291b42014ee5783f7c15ce815a170bfe07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-25 09:53:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6d3b8d548fc022649d7b8104dc533bd4b87ec52c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-25 11:39:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a7f8cb7f50e40acab93dc1f85b2fdf7 : SUCCESS in 1h 44m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb1e9aa6078d43508a7c8e5acb8e94d9 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597f756d18fd42358c9feefb88a4d1b1 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba6c86d91897459ea7cffe0fae9739ab : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6609388ca5a7488caace2881374035b5 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/760ef38f83e84cee87981cc36c9366e8 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc0bd30c855b407ea2241c2bbbe8ac8b : SUCCESS in 5m 56s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/6c42188484574e68aba0609eaca381b9 : SUCCESS in 6m 14s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/b04d9badbb964425a2bcc78911dc421b : SUCCESS in 5m 48s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/00c7fa06cbc6496a8cb01ffe1bd4285c : SUCCESS in 4m 35s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/fccd58f641b14c5d8c25d9662e94a0ba : SUCCESS in 4m 30s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b58f9c37f6d84fd1ab373f5deeb07177 : SUCCESS in 5m 19s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/561314a397cf4819ade71d398bd53002 : SUCCESS in 5m 00s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/cc07ff799a55463ab0a563a578447385 : SUCCESS in 4m 44s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/1b0116c96892415f97e84832df058c9b : SUCCESS in 4m 38s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/5b449c61458e43388d665ad93011a150 : SUCCESS in 5m 05s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/be311687ed574198b8ecbd5d8f3231a8 : SUCCESS in 6m 05s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/b0a87317bd4e430382e68153981d0929 : SUCCESS in 5m 06s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/ad9277174c9646d3b7acc6d4432b1a46 : SUCCESS in 4m 45s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/1b2cd3a9d9254cc48f42ba97ca7c7457 : SUCCESS in 5m 26s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/8e0fbc41e4a64450b293fc9d6c452699 : SUCCESS in 4m 40s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/a73a68c307b04d728d8fefdbce5828a0 : SUCCESS in 5m 24s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/cbdc88c82943413d8b869ab4c1fbecf8 : SUCCESS in 4m 28s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/d237829a79ca433b89781914ee37d847 : SUCCESS in 4m 26s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/33cda9ec66a8443ab896d597fc7aca02 : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":5},{"id":"d76bcd6ac65de9d496e4f26d1a871cadcb9cb1dd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-25 13:53:10.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"29cee73e164471b750d46b6228c73f523a97c021","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-26 09:09:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"23fa1c8b405c9b95d62609168296efda25b462f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-26 09:24:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d494ec0ccd98504fcaf50c6e62943c11d34d833a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-26 10:42:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2339c08ace81437394f4d65866e05642 : SUCCESS in 1h 16m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dc50b22f9d14ab7bdfec68533fa778b : FAILURE in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adac19996cd46baa7aca4495c083a5e : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/128883f7626044979995b5494cc4e6e9 : FAILURE in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40335c070aeb4f5ea40f44c71aded507 : FAILURE in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11a905f8f54247328c2a22b49c704ec8 : FAILURE in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddbcbab9ae1a47aaad1e483a3be34556 : FAILURE in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bca937ef716e45eeab641386443274a8 : SUCCESS in 4m 53s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/2fff0824b3b748e3afd902d4b97b9238 : FAILURE in 3m 46s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/298dac015ff5419aace11110c04275fd : FAILURE in 3m 45s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/cf173eafc58e4e2a84fe43618b1ad3de : FAILURE in 3m 46s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/562f5b61b3264a17834bb4f12ea52e63 : FAILURE in 4m 18s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c25dbd9a45794c098e573eda3f1ab27f : FAILURE in 3m 46s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/829ca00132ee4a9a88d42cfda2134dc2 : FAILURE in 3m 53s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/8e971a0c8b464df4aa28bd37e699a926 : FAILURE in 3m 54s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/15d602272385456f94887631f3bd42b6 : FAILURE in 4m 06s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7cac4ab23dce4a8f9ace3ac8c3f695c0 : FAILURE in 4m 18s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/5d8b02a9acad4ccb92bfa905805d9171 : FAILURE in 5m 32s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/91f9e2a1e1604e97a399483d045c5bbd : FAILURE in 4m 36s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/a6831d9be3974633ac6b10226be4949d : FAILURE in 5m 45s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c151624efe0141feac115554a7062a59 : FAILURE in 4m 26s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/f40b999a67dd43868909e15cd66a6741 : FAILURE in 4m 11s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4ae0d76f442e447f9140ae61e84acf97 : FAILURE in 4m 37s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/3c3a8057ca654e2ea5b2f73533fac360 : FAILURE in 4m 24s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/fb057a618ce940129759185eda73b487 : FAILURE in 6m 12s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a028451887674aeb99b553e166336f40 : FAILURE in 6m 10s","accounts_in_message":[],"_revision_number":6},{"id":"e4768972e150bf2d8b04bdc2838b490b03bb20bc","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-05-27 01:16:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e3871063f8ff8c09614dc8d3191d092f5c209b86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-27 03:02:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e826e1742b54f55912ca92a536fd88f : SUCCESS in 1h 40m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fe8ad060b5347b39be35c990e436929 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe25812c46604fd7bd0b401e5c300caa : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f97c4e1a0e364d0183dfe0e5d7f7382d : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7cac1368c29468787590dafcf676d4a : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2520420c56d42bbb79d605503356433 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c28021fdfb54608a5c3ad76f3c34cac : SUCCESS in 5m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5d7c0d9bf35434681a41d2fc788f033 : SUCCESS in 4m 36s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/40b0e701300142b29165f0a29de59c16 : SUCCESS in 4m 43s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/d2ae9ca53da4429f957f35778368bd7a : SUCCESS in 6m 14s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2f48b57d2bfd4dcba094be0ec97fb936 : SUCCESS in 4m 19s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/f0e914d9b9794b64bcbda373c1383d4d : SUCCESS in 6m 00s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/124f06a06bcc416d8d8ff91636e39190 : SUCCESS in 5m 04s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/f86ba84dc4af4af7ba133751b4a193ae : SUCCESS in 4m 44s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d07f1a160cf54a45bf6d53b9249d975e : SUCCESS in 5m 54s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/875b173061634a55a2fde373a7e860f7 : SUCCESS in 5m 41s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/5005de7cf1c942449b6c90d058727e51 : SUCCESS in 5m 23s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/c94927dbb8944b53aecc1bcc08621434 : SUCCESS in 4m 49s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/8b9ef02f9790459783599392d546f8a9 : SUCCESS in 5m 52s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/95216c9752d34488a37925d35cb6c606 : SUCCESS in 4m 34s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5fb41f7188ec4f2dae8893c99255fc66 : SUCCESS in 5m 34s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/80a6d65219c247949f6ed3feff2f6f21 : SUCCESS in 5m 38s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/53805cd285784582b1723b9fc4268e0a : SUCCESS in 4m 38s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/89c3819b104a4b9ab97c039bd419dd28 : SUCCESS in 5m 34s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/2afe2dcf02364455b8cebe1fcc535e68 : SUCCESS in 4m 15s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/31673b49abc140e783d5f47d6120fbc3 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":6},{"id":"cbc0d707a7fd72e6c47bee05afc3e2b0fc8b13ed","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-05-31 09:03:37.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"60e657c23ea1c93c377e764e3a61002e6aa1a637","author":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"date":"2022-07-15 06:20:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f9861827ec95054afaa6fd44ba1a2573c5fc5cdc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-07-15 18:54:57.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f885ac6aca5b6f709bf108e9283cc838e2357e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:55:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"8682c2fd9b1f4ae44e819f470676d92a2f75b711","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:17:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2314c6df8cae4a7182d5ee2c76b68b85\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ea9f9cd0e0c74ab4866da7afaeda59ce : SUCCESS in 1h 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ab8bf2dd27c4ce48c7f01f5da3ecd21 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0071afe6700c48b4873c2e17fbac632e : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36c20c109b3643feadb45fa912502d3d : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3c69ab362f4423be2a3602984062cb : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c7f0d83886f4b2581345e1cc67d21c4 : SUCCESS in 5m 29s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/e183c3627a014f46b347bae5d34a9f3f : SUCCESS in 6m 22s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/b82598efb2bb4ed4bcf15895bc81bbcf : SUCCESS in 6m 43s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/41ddc70f745a4b498a7fd5ab7ab9a146 : SUCCESS in 6m 13s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/3582eb6e1ab24a75b336f889b1969cb3 : SUCCESS in 8m 16s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/023116f362734cccab7e093c327ec3aa : SUCCESS in 6m 04s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/5c18ff2898a64328ae4df7e6b9048769 : SUCCESS in 6m 33s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/bafae83696d64956a182ce363450995f : SUCCESS in 6m 06s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/aa8970d2aa034529ae8c8c58984c028f : SUCCESS in 5m 16s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7b4c1a1420ad4cb2a52e2d4646bbf7e6 : SUCCESS in 6m 19s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/70111876902842c2a1d7dee19a042ed2 : SUCCESS in 8m 16s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/a80ad0da4eaa4a0089a9b6e32dfc2722 : SUCCESS in 6m 16s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/74328e585f2e4183a0311ce92ce30f84 : SUCCESS in 5m 29s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/2417a1d0816048139f98448aee55cd6e : SUCCESS in 5m 50s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/6c6da2d4b5c84954ad823c3c1d6b71f4 : SUCCESS in 8m 10s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/5e5cbb73f2ff4d05ae4cff41c132155d : SUCCESS in 6m 05s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/f3031cde67234ad7abb6b36b69ea7fa9 : SUCCESS in 5m 41s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/d276351b9a834df6ab8717d7b645d8cd : SUCCESS in 6m 20s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/752d9da4c2f94155bd5d073a2e773cf4 : SUCCESS in 8m 10s","accounts_in_message":[],"_revision_number":6},{"id":"48def1f051b6d176163b4d4cd235c12544a6fe0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:17:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"3a7c0131a6fe448c3582375e9870b66057a982dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:18:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6a6f06895e446b9b4d88f221bd454a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6c339d99c24f7b8375087dde15c7f7 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39c48fc0a4de4937beaf622735649921 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6b2216ebcce5f699b51d7d10b4100eef2dcbd5d0","revisions":{"57422baf1787710122c9523170b17b8aa978b224":{"kind":"REWORK","_number":1,"created":"2022-05-20 08:10:54.000000000","uploader":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"ref":"refs/changes/85/842685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/842685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/842685/1"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-20 07:57:53.000000000","tz":480},"committer":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-20 07:57:53.000000000","tz":480},"subject":"Fix inappropriate logic in memcachedlock.release()","message":"Fix inappropriate logic in memcachedlock.release()\n\nWhether \u0027was_deleted\u0027 was \u0027TRUE\u0027 or not, eventually we have to remove\nself from \u0027_acquired_locks\u0027.\nFor example:\n1. App #1 with coordinator \u0027A\u0027 wants to release lock \"b\"\n2. \u0027self.coord.client.delete()\u0027 failed for some reason(.e.g,\nBrokenPipeError,MemcacheUnexpectedCloseError)\n3. According to the former logic,lock \"b\" will not remove\nfrom \"_acquired_locks\", so \"self.heartbeat()\" will make it alive\nforever until App #1 was down or lock \"b\" turned expired.\n4. Now App #1 with coordinator \u0027A\u0027 wants to acquire lock \"c\", who\nhave the same lock-name with lock \"b\",It is clear that this will\nfail and prevent the locked program from continuing to execute.\n\nChange-Id: I6fc33b8e0a88510027bcfc30d1504489d2a91b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/57422baf1787710122c9523170b17b8aa978b224"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/57422baf1787710122c9523170b17b8aa978b224"}]},"branch":"refs/heads/master"},"c4fc1be88b1337b11579081b87cc7d7ea77ca44f":{"kind":"REWORK","_number":2,"created":"2022-05-21 07:35:36.000000000","uploader":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"ref":"refs/changes/85/842685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/842685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/842685/2"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-20 07:57:53.000000000","tz":480},"committer":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-21 07:34:47.000000000","tz":480},"subject":"Fix inappropriate logic in memcachedlock.release()","message":"Fix inappropriate logic in memcachedlock.release()\n\nWhether \u0027was_deleted\u0027 was \u0027TRUE\u0027 or not, eventually we have to remove\nself from \u0027_acquired_locks\u0027.\nFor example:\n1. App #1 with coordinator \u0027A\u0027 wants to release lock \"b\"\n2. \u0027self.coord.client.delete()\u0027 failed for some reason(.e.g,\nBrokenPipeError,MemcacheUnexpectedCloseError)\n3. According to the former logic,lock \"b\" will not remove\nfrom \"_acquired_locks\", so \"self.heartbeat()\" will make it alive\nforever until App #1 was down or lock \"b\" turned expired.\n4. Now App #1 with coordinator \u0027A\u0027 wants to acquire lock \"c\", who\nhave the same lock-name with lock \"b\",It is clear that this will\nfail and prevent the locked program from continuing to execute.\n\nChange-Id: I6fc33b8e0a88510027bcfc30d1504489d2a91b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c4fc1be88b1337b11579081b87cc7d7ea77ca44f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c4fc1be88b1337b11579081b87cc7d7ea77ca44f"}]},"branch":"refs/heads/master"},"dce2613aa154c7156f004d10be24415c38944eec":{"kind":"REWORK","_number":3,"created":"2022-05-23 06:09:14.000000000","uploader":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"ref":"refs/changes/85/842685/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/842685/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/842685/3"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-20 07:57:53.000000000","tz":480},"committer":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-23 06:08:37.000000000","tz":480},"subject":"Fix inappropriate logic in memcachedlock.release()","message":"Fix inappropriate logic in memcachedlock.release()\n\nWhether \u0027was_deleted\u0027 was \u0027TRUE\u0027 or not, eventually we have to remove\nself from \u0027_acquired_locks\u0027.\nFor example:\n1. App #1 with coordinator \u0027A\u0027 wants to release lock \"b\"\n2. \u0027self.coord.client.delete()\u0027 failed for some reason(.e.g,\nBrokenPipeError,MemcacheUnexpectedCloseError)\n3. According to the former logic,lock \"b\" will not remove\nfrom \"_acquired_locks\", so \"self.heartbeat()\" will make it alive\nforever until App #1 was down or lock \"b\" turned expired.\n4. Now App #1 with coordinator \u0027A\u0027 wants to acquire lock \"c\", who\nhave the same lock-name with lock \"b\",It is clear that this will\nfail and prevent the locked program from continuing to execute.\n\nChange-Id: I6fc33b8e0a88510027bcfc30d1504489d2a91b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dce2613aa154c7156f004d10be24415c38944eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dce2613aa154c7156f004d10be24415c38944eec"}]},"branch":"refs/heads/master"},"6b5abfc2f07a91f5d340a7672775474f3fe9bef7":{"kind":"REWORK","_number":4,"created":"2022-05-25 09:06:43.000000000","uploader":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"ref":"refs/changes/85/842685/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/842685/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/842685/4"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-20 07:57:53.000000000","tz":480},"committer":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-25 09:05:09.000000000","tz":480},"subject":"Fix inappropriate logic in memcachedlock.release()","message":"Fix inappropriate logic in memcachedlock.release()\n\nWhether \u0027was_deleted\u0027 was \u0027TRUE\u0027 or not, eventually we have to remove\nself from \u0027_acquired_locks\u0027.\nFor example:\n1. App #1 with coordinator \u0027A\u0027 wants to release lock \"b\"\n2. \u0027self.coord.client.delete()\u0027 failed for some reason(.e.g,\nBrokenPipeError,MemcacheUnexpectedCloseError)\n3. According to the former logic,lock \"b\" will not remove\nfrom \"_acquired_locks\", so \"self.heartbeat()\" will make it alive\nforever until App #1 was down or lock \"b\" turned expired.\n4. Now App #1 with coordinator \u0027A\u0027 wants to acquire lock \"c\", who\nhave the same lock-name with lock \"b\",It is clear that this will\nfail and prevent the locked program from continuing to execute.\n\nChange-Id: I6fc33b8e0a88510027bcfc30d1504489d2a91b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6b5abfc2f07a91f5d340a7672775474f3fe9bef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6b5abfc2f07a91f5d340a7672775474f3fe9bef7"}]},"branch":"refs/heads/master"},"c0471fe71a4c7a6e77d3ce211eaac8a485794f34":{"kind":"REWORK","_number":5,"created":"2022-05-25 09:53:23.000000000","uploader":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"ref":"refs/changes/85/842685/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/842685/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/842685/5"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-20 07:57:53.000000000","tz":480},"committer":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-25 09:52:25.000000000","tz":480},"subject":"Fix inappropriate logic in memcachedlock.release()","message":"Fix inappropriate logic in memcachedlock.release()\n\nWhether \u0027was_deleted\u0027 was \u0027TRUE\u0027 or not, eventually we have to remove\nself from \u0027_acquired_locks\u0027.\nFor example:\n1. App #1 with coordinator \u0027A\u0027 wants to release lock \"b\"\n2. \u0027self.coord.client.delete()\u0027 failed for some reason(.e.g,\nBrokenPipeError,MemcacheUnexpectedCloseError)\n3. According to the former logic,lock \"b\" will not remove\nfrom \"_acquired_locks\", so \"self.heartbeat()\" will make it alive\nforever until App #1 was down or lock \"b\" turned expired.\n4. Now App #1 with coordinator \u0027A\u0027 wants to acquire lock \"c\", who\nhave the same lock-name with lock \"b\",It is clear that this will\nfail and prevent the locked program from continuing to execute.\n\nChange-Id: I6fc33b8e0a88510027bcfc30d1504489d2a91b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c0471fe71a4c7a6e77d3ce211eaac8a485794f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c0471fe71a4c7a6e77d3ce211eaac8a485794f34"}]},"branch":"refs/heads/master"},"6b2216ebcce5f699b51d7d10b4100eef2dcbd5d0":{"kind":"REWORK","_number":6,"created":"2022-05-26 09:24:36.000000000","uploader":{"_account_id":30858,"name":"zhen","email":"zhen00fa@gmail.com","username":"zhen"},"ref":"refs/changes/85/842685/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/842685/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/842685/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/842685/6"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-20 07:57:53.000000000","tz":480},"committer":{"name":"zhen","email":"zhen00fa@gmail.com","date":"2022-05-26 09:23:41.000000000","tz":480},"subject":"Fix inappropriate logic in memcachedlock.release()","message":"Fix inappropriate logic in memcachedlock.release()\n\nWhether \u0027was_deleted\u0027 was \u0027TRUE\u0027 or not, eventually we have to remove\nself from \u0027_acquired_locks\u0027.\nFor example:\n1. App #1 with coordinator \u0027A\u0027 wants to release lock \"b\"\n2. \u0027self.coord.client.delete()\u0027 failed for some reason(.e.g,\nBrokenPipeError,MemcacheUnexpectedCloseError)\n3. According to the former logic,lock \"b\" will not remove\nfrom \"_acquired_locks\", so \"self.heartbeat()\" will make it alive\nforever until App #1 was down or lock \"b\" turned expired.\n4. Now App #1 with coordinator \u0027A\u0027 wants to acquire lock \"c\", who\nhave the same lock-name with lock \"b\",It is clear that this will\nfail and prevent the locked program from continuing to execute.\n\nChange-Id: I6fc33b8e0a88510027bcfc30d1504489d2a91b4e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6b2216ebcce5f699b51d7d10b4100eef2dcbd5d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6b2216ebcce5f699b51d7d10b4100eef2dcbd5d0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
