)]}'
{"id":"openstack%2Ftooz~847152","triplet_id":"openstack%2Ftooz~stable%2Fussuri~Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","project":"openstack/tooz","branch":"stable/ussuri","topic":"redis-retry-fix","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-04 11:51:10.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"4264":{"account":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"last_update":"2022-07-04 20:01:09.000000000","reason":"Change was submitted"},"13177":{"account":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"last_update":"2022-07-04 20:01:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1","subject":"Enable retries in redis driver","status":"MERGED","created":"2022-06-22 13:10:23.000000000","updated":"2022-07-04 20:03:10.000000000","submitted":"2022-07-04 20:01:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"847152-redis-retry-fix","cherry_pick_of_change":817985,"cherry_pick_of_patch_set":1,"meta_rev_id":"7eb26efbfe91c36898b8d33392966d28e7cc57c4","_number":847152,"virtual_id_number":847152,"owner":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 20:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-07-04 16:46:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-07-04 16:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-04 16:46:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"}],"reviewers":{"REVIEWER":[{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-22 13:10:23.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"},{"updated":"2022-06-22 13:14:37.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"state":"REVIEWER"},{"updated":"2022-06-22 13:14:37.000000000","updated_by":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-06-22 14:30:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-06-22 15:06:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5935c7e79adfed8fbaa4df7e92b5435ea2082c23","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-06-22 13:10:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.\n\nThe following files contain Git conflicts:\n* tooz/drivers/redis.py","accounts_in_message":[],"_revision_number":1},{"id":"aa7099278b58d7a153ff90bba7e1778486000497","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-06-22 13:14:37.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"907f75a396e8ae8acd25cfb4f2c49d5b8540dc46","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-06-22 13:17:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"722e8481888e7ae7ed4c6ef2e18aaa14331fbbe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-22 13:25:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0b2cbc3a478c6bbfbeae829221055a2f7f560e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-22 13:25:42.000000000","message":"Topic redis-retry-fix-stable/wallaby-stable/victoria-stable/ussuri removed","accounts_in_message":[],"_revision_number":2},{"id":"9a4ecccb21385ca6c45dbe117439ec2f96968105","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-22 13:25:44.000000000","message":"Topic set to redis-retry-fix-stable/wallaby-stable/victoria","accounts_in_message":[],"_revision_number":2},{"id":"05c55f9cb3e7c890edcc02264711e7dca2d31ee5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-22 13:25:45.000000000","message":"Topic redis-retry-fix-stable/wallaby-stable/victoria removed","accounts_in_message":[],"_revision_number":2},{"id":"6a8827cbcff2376b209479ee9ba04e165742017b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-22 13:25:51.000000000","message":"Topic set to redis-retry-fix","accounts_in_message":[],"_revision_number":2},{"id":"3ee4855923a86699dd5a5463e294904b753e0ed3","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-22 14:30:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d12d3818c7bef9b9641c4a5cb6f528209e3ca7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 15:06:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/619b40e906254f22aee6aee8633098bb : SUCCESS in 1h 33m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7350ae0267334ba9a2ab36023324769b : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd00ce397fa14a5f9a6d02c4071e5531 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1af5ced2b2af4f62a5edd2bab482fa7c : FAILURE in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/441edd6e297141238aa88e7cc1336317 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/817fd214e7ec4cae8b615c5d5b790413 : SUCCESS in 6m 58s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/60cc817d6b1a4462bed6583308a7104e : FAILURE in 5m 18s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/6d2dd38608cd4af6a77ff1a9f57d0a0d : SUCCESS in 5m 58s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/1a1c9350488e40d3a186846fa8112735 : FAILURE in 5m 52s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/d1ed0d0d221243e58a3bf2993778fde9 : SUCCESS in 4m 18s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/1b5187aa86234f97bf640190b6bbe291 : FAILURE in 5m 06s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/0d24085a1a9640629538e7a265975efe : SUCCESS in 6m 13s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/28fa2cee12f14b92b156bf32e1c1aced : FAILURE in 5m 31s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/7d9724d4734e4e71b053b57bcdd24846 : SUCCESS in 6m 46s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/f5e34f0608014f26be38e21fc70d1735 : FAILURE in 6m 17s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/b09714fd2b744f1e9563407693ecb1a5 : SUCCESS in 4m 52s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/5155085a49af4422bf1686a9ddc0d5c9 : FAILURE in 5m 26s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/c7f825ee74884e27ab1d4c1779924506 : SUCCESS in 5m 27s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/7722527bf4214b5c88e6c90b284724a7 : FAILURE in 5m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/d7344d1c72534b6fa3da48d3bfe89fbf : SUCCESS in 5m 58s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/bd9d6abdbf3c4982a590a1ffdc99f4cc : FAILURE in 5m 51s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/c7f05be61b464c8188aa84bc7053ea02 : SUCCESS in 4m 33s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/828527ee09ae42d98a8195b65ab466d6 : FAILURE in 4m 36s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/bbab517be9b842d98a391957edcfeaef : SUCCESS in 6m 22s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/e5b150031ab54a71babdf7a0be027e7a : SUCCESS in 4m 26s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/edde9594f77d4e1f8f8aaff8ecc2d709 : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"20924caafd0b09a77543177fbd7ab01fc5cddaca","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-06-28 10:09:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"79e63ef3e1c96e7a30ef928c50b16113f4db9c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 11:58:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c8d5bdc16114750bb01aa25ca2ff0fb : SUCCESS in 1h 47m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/623a9c8e6a43454fae1a454dc5ad208c : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a27708151464706b0dc64d40d1909d3 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8cf40a23446e4957a38b90d0d7e6d4ca : FAILURE in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a61630687154fb188762bc86f03d83c : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1103ff5e0c464ecc95092288e9aa36c9 : SUCCESS in 4m 33s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/cca104af958d4b318bd9c843bf79e1c2 : FAILURE in 4m 46s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/26fc0b65a0754fc1870fab087415bc04 : SUCCESS in 4m 45s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/acbabd13dab8403ca64f4d2014b25cf2 : FAILURE in 4m 27s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/cc58c3dc22ed477798f3b3bf12b91c2a : SUCCESS in 4m 20s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/ebebcf7df3f64a4a956f7d4540bba55e : FAILURE in 4m 31s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/ac415c50a8764bb8929166deeb6c274b : SUCCESS in 4m 19s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/8c75770e918a4f71901b94b233baa513 : FAILURE in 4m 44s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/2dc36eb4ea9f4154a54b7e5764b8f710 : SUCCESS in 4m 21s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/69a929d5e57f489dad02ce76377cb44c : FAILURE in 4m 33s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/5c3a5b8257bc436eb1ad5d2830192885 : SUCCESS in 4m 22s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/1e2768fc38624109a77d1fc320dd21c0 : FAILURE in 4m 29s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ca230514429a4b2096df9460a85497b5 : SUCCESS in 4m 37s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/a3ff63f86bae49cc9e947eafa27f13a5 : FAILURE in 4m 18s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/76a26f9875d449d6bbd90c063430208b : SUCCESS in 4m 19s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/e79bc96e082a428893b3945790b56e5f : FAILURE in 4m 53s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/bd925de37e9f4bf4be870ae003fac648 : SUCCESS in 4m 11s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/95c4ce055a7e420abd24c7fef196d8a4 : FAILURE in 4m 32s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/290d4bb1b12a44b1b9fd8895322b1c7b : SUCCESS in 4m 20s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/d295ead747f14cdea45bcff71e2ab231 : SUCCESS in 4m 38s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/b029ca9f0fed4c4c918e2321fc7d682c : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"9a178d74b9359b849a4d685c4ee54123fa45400d","author":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"date":"2022-06-28 13:26:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0726140a70eed394dc0ba8d18b2ac9a624012459","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:51:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9662144b161507546944d3f6dcafe9087c32f2cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 16:44:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ebf422e8d5e63174cdb1c6ac8ff0ba7423226ac6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 16:46:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e9f5552487f653d50defeb9fd253ae3b92cd18c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 17:48:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24cfedda566a45d5a902d46736997f9a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b46f81bfc034c5ea985ecfb1b4d2dd5 : SUCCESS in 1h 00m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f6c49c331d74af68937c7d766e5645d : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dd83402981249379dc782341116373b : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4375b1c4204405ca318ac9982fdb89d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d0cebed00740f2acf736ed51d8f64b : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35db8d27593d4e35973ef52278329efd : SUCCESS in 4m 01s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/b4720ae894fd49f8a4f0ddfde32e2cd0 : SUCCESS in 5m 45s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/e36a780e63d6434cbb425df47e48a45d : SUCCESS in 4m 31s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/3ab604efa5c44d0a8ef0e73c6b9cd820 : SUCCESS in 4m 55s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/f2319ccccff24e1cbf3078f7b6b4514a : SUCCESS in 4m 19s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/c9fad9b468344be6a6fcc37ce51677dd : SUCCESS in 5m 10s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/05a699b7980e438c9ac47f04f38ae1d6 : SUCCESS in 5m 00s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/7fab9920b6b842d2a478cf4d8af58339 : SUCCESS in 5m 24s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/aa5cd291d1bd4342a445f6b04de5e160 : SUCCESS in 4m 44s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/07ec60fb500a48e8a04ad446e7268eb7 : SUCCESS in 5m 13s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/89e437e338b5492c98a5376b9623ec91 : SUCCESS in 4m 37s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/e01720cca60c42e4a5d1f6905f176b9a : SUCCESS in 5m 02s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ed59e2acada2487abc7f03cb0bdd7126 : SUCCESS in 4m 42s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/c7ef05217d0242c491d5befe2defb85b : SUCCESS in 4m 52s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5dc557c1acef4543b7d3ee8313c6819a : SUCCESS in 4m 37s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/698daeb0a7f04998908a39385da1ffb6 : SUCCESS in 5m 13s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/f0eeb4a5d390460fb7c54ca825aa9cd5 : SUCCESS in 4m 39s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/9ea81254807e40a49807c68a2f9483a9 : SUCCESS in 5m 22s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/db130781a0a74bea83cd7257022d6bcc : SUCCESS in 4m 16s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/5b86d59ec7da4472975ddd56b6040fa7 : SUCCESS in 5m 01s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/6b86caf9c2b74fda83b2e6551a637800 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"b9d076ff4aedcfd81894ebaebe04ac3416b8a821","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:16:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d55f0ea1e7de6f628e1659c0f397040e66407cfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:01:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e972ded233844697a5c979d284d2cb9a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c40c3cc56632494fb94f2857716f37a6 : SUCCESS in 1h 26m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/281a9366623341beab84fa3fda1c91d1 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c5d87a5f20a45dbad70c600ca460ef1 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aae2eab7ccd48d494ec2631bed464c2 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf0e881bad844dd28cb9fb2bcf2a52e2 : SUCCESS in 5m 22s\n- tooz-tox-py37-etcd https://zuul.opendev.org/t/openstack/build/0754f6bb2005474a84725e8ea46c02b0 : SUCCESS in 6m 10s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/608db7e8a1294caa9767a94e2cf16cbb : SUCCESS in 5m 48s\n- tooz-tox-py37-etcd3 https://zuul.opendev.org/t/openstack/build/501552a916a1471c9c5a1467ac87090d : SUCCESS in 5m 55s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/20983dd624344c2abd591c4c430d9e51 : SUCCESS in 4m 53s\n- tooz-tox-py37-etcd3gw https://zuul.opendev.org/t/openstack/build/22e8125e35544ae4970605f1adbdc666 : SUCCESS in 4m 48s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/4cacacec456545188035d9a5c4ec55a1 : SUCCESS in 4m 40s\n- tooz-tox-py37-zookeeper https://zuul.opendev.org/t/openstack/build/1a4c2171374048c684b064478a8509c1 : SUCCESS in 5m 16s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/e717ce06fd5b4845bdf00e12cbbbd305 : SUCCESS in 4m 39s\n- tooz-tox-py37-redis https://zuul.opendev.org/t/openstack/build/a4fb6870aed2426e82a66938fac20ba8 : SUCCESS in 5m 12s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a67615b13c854ece9e9394d9ef53eeec : SUCCESS in 4m 45s\n- tooz-tox-py37-sentinel https://zuul.opendev.org/t/openstack/build/9edd9246e3f94e8cb20a9585f81d2539 : SUCCESS in 5m 23s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/50c1bc95470b4518b5d086908a714516 : SUCCESS in 4m 38s\n- tooz-tox-py37-memcached https://zuul.opendev.org/t/openstack/build/5cd5e761bfd84fff92eb0fb658026712 : SUCCESS in 5m 18s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/d407ee863431404a8a2ce5294b02c638 : SUCCESS in 4m 49s\n- tooz-tox-py37-postgresql https://zuul.opendev.org/t/openstack/build/e60ab92cdd63468a9a361f1e4a5dac4c : SUCCESS in 5m 06s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/61c5037bfdc24e5c8584b3b085181b0f : SUCCESS in 4m 22s\n- tooz-tox-py37-mysql https://zuul.opendev.org/t/openstack/build/073b54e7a30047b696980b5befcf9243 : SUCCESS in 5m 15s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/2df01434d0a84a448232208baa7295b2 : SUCCESS in 4m 31s\n- tooz-tox-py37-consul https://zuul.opendev.org/t/openstack/build/1c26fc63c45a49e39d1cb8df1486221c : SUCCESS in 4m 35s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/48a0cfb539804ba3a35cb9209572aa9d : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"ea2707d057042300c7f085f88126f976397edddc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:01:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7eb26efbfe91c36898b8d33392966d28e7cc57c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:03:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee1098dafd2b4ee6872cad6552d1ff62\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c86e797838ac42d1810e6872cc6d8617 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e37f8d1e1194f81a582dc0263752709 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"816ea504c92b141b34759faaf814698499fa0c75","revisions":{"b7a5b073f898dcbc5aa79eaa9cfebe07cf0484aa":{"kind":"REWORK","_number":1,"created":"2022-06-22 13:10:23.000000000","uploader":{"_account_id":13177,"name":"Emma Foley","email":"efoley@redhat.com","username":"emma-l-foley"},"ref":"refs/changes/52/847152/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/847152/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/847152/1"}}},"commit":{"parents":[{"commit":"667e2ec1f4d179c1ec9bd5f89ed12a50cad71e24","subject":"Fix broken CI and PyMySQL support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/667e2ec1f4d179c1ec9bd5f89ed12a50cad71e24"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Emma Foley","email":"efoley@redhat.com","date":"2022-06-22 13:10:23.000000000","tz":0},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n(cherry picked from commit 47c4d56e446a49726e7145c6cbc3bb7620a431f7)\n(cherry picked from commit 11526e594c7e7cef8b5a4a4220f100d76a418036)\n(cherry picked from commit c1ae1649d22b356f2bbb2293cb421eb155fdee14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b7a5b073f898dcbc5aa79eaa9cfebe07cf0484aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b7a5b073f898dcbc5aa79eaa9cfebe07cf0484aa"}]},"branch":"refs/heads/stable/ussuri"},"ea55824be2ae5eb4d282f3a1ea9d4bc2fd855a15":{"kind":"REWORK","_number":2,"created":"2022-06-22 13:25:22.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/847152/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/847152/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/847152/2"}}},"commit":{"parents":[{"commit":"b8e0d819d406857f431723de47f9f2ceb700d1c8","subject":"Retry on redis connection errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b8e0d819d406857f431723de47f9f2ceb700d1c8"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-22 13:25:06.000000000","tz":60},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n(cherry picked from commit 47c4d56e446a49726e7145c6cbc3bb7620a431f7)\n(cherry picked from commit 11526e594c7e7cef8b5a4a4220f100d76a418036)\n(cherry picked from commit c1ae1649d22b356f2bbb2293cb421eb155fdee14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ea55824be2ae5eb4d282f3a1ea9d4bc2fd855a15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ea55824be2ae5eb4d282f3a1ea9d4bc2fd855a15"}]},"branch":"refs/heads/stable/ussuri"},"816ea504c92b141b34759faaf814698499fa0c75":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-04 16:44:27.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/847152/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/847152/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/847152/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/847152/3"}}},"commit":{"parents":[{"commit":"0734d86ae2322aeff92a2d44aa9b796bb7a14c26","subject":"Retry on redis connection errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0734d86ae2322aeff92a2d44aa9b796bb7a14c26"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-06-10 13:32:32.000000000","tz":180},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:44:18.000000000","tz":60},"subject":"Enable retries in redis driver","message":"Enable retries in redis driver\n\nthis is followup to Iaab5ce609c0dcf7085f5dd43efbd37eb4b88f17b\n\nactually retry for specified number of retries instead of raising\nerror on first ConnectionError\n\nChange-Id: Ibca3f568b65dfea252da4b67f6d5105ba7f1ecb1\n(cherry picked from commit 47c4d56e446a49726e7145c6cbc3bb7620a431f7)\n(cherry picked from commit 11526e594c7e7cef8b5a4a4220f100d76a418036)\n(cherry picked from commit c1ae1649d22b356f2bbb2293cb421eb155fdee14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/816ea504c92b141b34759faaf814698499fa0c75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/816ea504c92b141b34759faaf814698499fa0c75"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
