)]}'
{"id":"openstack%2Ftooz~847949","triplet_id":"openstack%2Ftooz~master~I8d6647118be22b0ce55e01b7e5451612ebe30e73","project":"openstack/tooz","branch":"master","topic":"bug/1980079","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-06-28 15:25:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I8d6647118be22b0ce55e01b7e5451612ebe30e73","subject":"Cap protobuf \u003c 4.x","status":"MERGED","created":"2022-06-28 10:56:54.000000000","updated":"2022-07-04 13:31:21.000000000","submitted":"2022-07-04 13:29:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"847949-bug/1980079","meta_rev_id":"55082e075c29601e751748cc84a4c2baeac6d0a3","_number":847949,"virtual_id_number":847949,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 13:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"recommended":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-06-28 14:35:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-06-28 14:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-04 11:44:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-28 12:27:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 14:32:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-06-28 14:35:27.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"d34c104e49fa2a9c36d8ed8a996f26524c8ef7d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-28 10:56:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b42107abcce07c7d0bc5a42398610978a83504a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 12:27:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/614ec31b09ea4a7eb5b68f4ab6ca313f : SUCCESS in 2m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/795d2d0605d243ffaf7c441d36333cf9 : SUCCESS in 1h 27m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa86fad6961546498d9c155dccfacf0d : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c265bd9e9d66448492b02287baf7d6de : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29c2d5d95d77455a9f490255b55caaa5 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9385a3d401c4809a3129dce68ec6332 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6167d780a67b402fad8d1fc23f1fedf9 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b3b85eb907433b9ff67cbc4eb51343 : FAILURE in 4m 31s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/3ba23cc839e148b0b34f0c972d45c4a5 : SUCCESS in 4m 36s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/b0c3c9dc3e6049218411ecbc9a7167ee : SUCCESS in 4m 46s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/40ef25f687034eb8b4c16360e416c45e : SUCCESS in 4m 56s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/8ab9fe0403234e1a9dad54fed014ec0c : SUCCESS in 4m 52s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/2546f5d848d54efea254b25a7210d8df : SUCCESS in 4m 52s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/2aba13b3026f4b42a06cb32c092c4a41 : SUCCESS in 5m 32s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/2c6fa364e94f42e3bb560d0fa85e1bec : SUCCESS in 5m 56s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/568b4f7dae1b428d923941c63ffc7ec7 : SUCCESS in 4m 56s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b3e370020ecf4006b34953ee603df27c : SUCCESS in 5m 33s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/ca69ad5aaede4e6389bbba298ae51b37 : SUCCESS in 4m 53s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/a78d8e6e725d4ca7b19fedfbdf9c49dc : SUCCESS in 4m 34s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/fedab4f8a0034d5c8bb9eb842936dda5 : SUCCESS in 4m 44s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/47b3b81729a54a819bfa50399c6e1084 : SUCCESS in 6m 24s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/424d01d730184122b8e393065d882203 : SUCCESS in 5m 05s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c570573ec3f143939ebddfbf77189010 : SUCCESS in 5m 28s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/c6d1a5aca33e41c398d3edd650fbee1e : SUCCESS in 4m 53s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/f4f76b654fd147fea575c6183950c2ed : SUCCESS in 4m 25s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a5dfe17fd8394516b4b419dcc08e409e : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6bb56b90ca92ed202e3afd2bd0128555f9918bc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-28 14:12:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16f02ffba0b931ae7295fe1820499ff51383349c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-28 14:32:53.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a76c9d635ef1335524ab871d1f6d92d93fab010","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-06-28 14:35:27.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fcd64db10bb0088beda893cc4a1c28f2a1284800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 15:23:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bdf4f8565829481ab444f11ead1ed103 : SUCCESS in 3m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1dcd6b2dbdb4c52ab3a0720147d7460 : FAILURE in 1h 08m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5af5df0ce984ae8943fdaec01f02f72 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b32eef4a8f426b8785ece393d762a3 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59724a072b6545018263331205ffe75b : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4a1725aebe44968b895cd597064d1a1 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4782aeee51254e4d9d717dc752db2ea6 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9181ace6b83142bbae5e3553ff32f754 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3725100bb1f8441fb02c74e3f4a1f3f2 : SUCCESS in 4m 57s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/efd5edcaec2f4295aa3c00623844eaeb : SUCCESS in 6m 13s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/2fdc0cf4d48d41959e4253db08237eac : SUCCESS in 6m 46s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/fa937ce321e944d8869f90d8324cc2f7 : SUCCESS in 5m 50s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/98bda581889246c3b49ced1e01fa5774 : SUCCESS in 6m 42s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/637db3af73f44aeebd926a84136b9f45 : SUCCESS in 5m 39s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/adc8cf91ce2548a5b8a70e37cfc52d73 : SUCCESS in 6m 15s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/c3c056cedcb944008cf89b4cb78886a4 : SUCCESS in 7m 44s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/27f92700df534d5689ef2d844450c8be : SUCCESS in 6m 02s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/dee6b351bdc64f3aae8a50cbd364378b : SUCCESS in 5m 36s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/38cd05c66d5346eb827ffc74c3a4fdc4 : SUCCESS in 5m 58s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/24e6eb3f6a894c3d8aae4c16f57dc1b9 : SUCCESS in 5m 52s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/5a85d8e3034849819e96a3c653b517f5 : SUCCESS in 5m 41s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/fac3426bc3854f42b9d3bfae8391e1c5 : SUCCESS in 5m 21s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/788d40d0fab14dd092cd93efb4e246c3 : SUCCESS in 5m 52s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/dace2b2405814d9b8c44d6441f875ec3 : SUCCESS in 6m 33s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/0e2560f60f674af1969e60322f7d7154 : SUCCESS in 6m 01s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/48975ebc2c094937b033973429849811 : SUCCESS in 5m 15s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/6e4e0d6afa01444bb44ccd30d8e6673d : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":2},{"id":"30b6dd18b3a2081d8ee30a276d917054ac5db634","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-06-28 15:25:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8039001a9f438e05fbf50d62aa95a78eb43a2919","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 17:23:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04d96dd3be08476795b60ba927cfaad3 : SUCCESS in 3m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f46e8b1a897a4d4a950f89802f7f91dd : SUCCESS in 1h 50m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/299936087af94e06917ad8e14f51ba88 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a579b6cf534509958bc26d4c5c8843 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51ff1d3e88df4aa2a1f0d71b9bb1158e : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a893659551a24994a6acce5bf3d6b925 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f66b58a909db4655a74ae1666aed2504 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9324427304a491d8185323a41c108a6 : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ce462b5ef144933acd058871009ddac : SUCCESS in 4m 40s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/c9c52d865ec647f9acaea67f53d81257 : SUCCESS in 4m 57s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/defb541075c94667b1208dbaf8c1c27a : SUCCESS in 5m 30s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/34366bba35a741ad8d2db8a3330ae8fa : SUCCESS in 5m 16s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/c2a10541bdcd489fb9c550dd9960fa04 : SUCCESS in 5m 33s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/5e534d76cff64a7387bb0451679dc1c2 : SUCCESS in 4m 50s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/9dc59530b269490e85f64ab27d867dc7 : SUCCESS in 5m 15s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/78479892b7574167aafad46babfd57d4 : SUCCESS in 4m 45s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/d179f476baad4b93abcb9f90909d56d0 : SUCCESS in 5m 30s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/38448c1fb5734a7db78e3c2434cd0730 : SUCCESS in 5m 18s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/11a9fd32dd944e5792d341784595ab68 : SUCCESS in 5m 19s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/58f93bef5e2c43de952c920256528143 : SUCCESS in 4m 59s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/403fb347bb404d0592b730f186ce0ddc : SUCCESS in 5m 11s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7515ed217aa449d5b379c3d52c9b234a : SUCCESS in 4m 42s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/111d4311272d42a5b0b2a97ce7a38349 : SUCCESS in 5m 17s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/cb39eaf1fec4417ea8001532d3d5a80b : SUCCESS in 5m 03s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/3ab7c2656445435e9d1662911b438547 : SUCCESS in 5m 07s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/dfc30b5484b54a429f3656d9feb82116 : SUCCESS in 4m 41s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a1de195bf45e4464887a0f59c8c8da03 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":2},{"id":"513b24b5acbc0f4132c02823267274da6182f075","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:44:58.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"dac8d8ab8a5e1a6987bdd5d4ac39fead6008304e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 11:45:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"df6c05cc77ebe1e9c46da43cb9c6f76b9294780b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:29:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0adbda00959d4b379eeb8973f8fa7619\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7e7abe0fa79c49e8b7097ff0a05ceea1 : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74f13a2134a04fc5ae3323a8779d76f7 : SUCCESS in 1h 36m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d857f050318f4b8d9cdf1c09fd2bb7a5 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/558ee53344704448bc3eabef069a87e5 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68a7dff3972f486d8b311699d82ecee6 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf2be8c3bf2947c49c62c9b9cf1c926d : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6ddedaa5fe4472f94c95dc969cf67b4 : SUCCESS in 4m 15s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/dec2798afa5e4e0ea391186bb03b0e54 : SUCCESS in 4m 58s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/bd55ae71e1564af8b60858b5f429b087 : SUCCESS in 5m 31s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/503c15ca85264d44a6640c147cf673c7 : SUCCESS in 4m 59s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/cca28ac9fb56439ca99ffd66f92f7213 : SUCCESS in 5m 51s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/01653ec14ae94f0f99ae0ff5e5b2c278 : SUCCESS in 5m 07s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/0666d5f7f4d44fa984d08a0473cb5435 : SUCCESS in 5m 23s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/8f27aed87af44543aacd8ea59117bda5 : SUCCESS in 5m 15s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/1279c10c72eb486cbcf9de99d58d2b63 : SUCCESS in 5m 43s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/fb6718344f4548d98210e09f53d9eb64 : SUCCESS in 5m 48s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/7eeaeaee711048fa9109e62c8c0d345f : SUCCESS in 5m 52s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/f9f4c68df0054c9f874030cd782bf76d : SUCCESS in 5m 55s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/6016fb2ece61409bbf4ba7fe7bf30f4e : SUCCESS in 5m 41s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a5b1320180944836a24dcc06190036f4 : SUCCESS in 5m 12s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/c9a86ef30c684117941c322d98090ba7 : SUCCESS in 6m 15s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/76d939b4be684e63b5fd48ce52dbf7ee : SUCCESS in 5m 31s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/f93aa1d4ae9b4684b03cf54e55e041c1 : SUCCESS in 5m 47s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/090dcfbc67054b1997442cba83a872bb : SUCCESS in 5m 34s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a1a3b49a734a4a8f90a14560ab85c0f3 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"b5ff306b6136acd40b9ebdaed68e3fc84edbd646","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:29:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"55082e075c29601e751748cc84a4c2baeac6d0a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:31:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e488b90f5a47f69fab9832a122c858\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1905c8ee71ea4463afbd023318318aaf : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3735f36e355c45ebab947e798798a2ce : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e3de0e634af62e46162eb8d2354416ce4f2ba77c","revisions":{"bdfc2646251fc3914f1afefee085dd916c20bc6a":{"kind":"REWORK","_number":1,"created":"2022-06-28 10:56:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/847949/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/49/847949/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/49/847949/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/49/847949/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/49/847949/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/49/847949/1"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-28 10:53:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-28 10:53:19.000000000","tz":60},"subject":"Cap protobuf \u003c 4.x","message":"Cap protobuf \u003c 4.x\n\ntooz doesn\u0027t use upper-constraints so we get the newest version of all\ndependencies. It seems one of these, etcd3, is incompatible with recent\nversions of protobuf.\n\n  Failed to import test module: tooz.tests.drivers.test_etcd3\n  Traceback (most recent call last):\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 436, in _find_test_path\n      module \u003d self._get_module_from_name(name)\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 377, in _get_module_from_name\n      __import__(name)\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/tests/drivers/test_etcd3.py\", line 22, in \u003cmodule\u003e\n      import tooz.drivers.etcd3 as etcd3_driver\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/drivers/etcd3.py\", line 18, in \u003cmodule\u003e\n      import etcd3\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/__init__.py\", line 3, in \u003cmodule\u003e\n      import etcd3.etcdrpc as etcdrpc\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/__init__.py\", line 1, in \u003cmodule\u003e\n      from .rpc_pb2 import *\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/rpc_pb2.py\", line 16, in \u003cmodule\u003e\n      from etcd3.etcdrpc import kv_pb2 as kv__pb2\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/kv_pb2.py\", line 36, in \u003cmodule\u003e\n      type\u003dNone),\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/google/protobuf/descriptor.py\", line 755, in __new__\n      _message.Message._CheckCalledFromGeneratedFile()\n  TypeError: Descriptors cannot not be created directly.\n  If this call came from a _pb2.py file, your generated code is out of date and must be regenerated with protoc \u003e\u003d 3.19.0.\n  If you cannot immediately regenerate your protos, some other possible workarounds are:\n   1. Downgrade the protobuf package to 3.20.x or lower.\n   2. Set PROTOCOL_BUFFERS_PYTHON_IMPLEMENTATION\u003dpython (but this will use pure-Python parsing and will be much slower).\n\nThe etcd3 dependency is only used for the etcd3 driver which was\ndeprecated in change Iff0cd3b51cfc45ecbddc831c55267e80e9c79cac. The\nprotobuf dependency is only used for etcd3. We can\u0027t outright remove the\netcd3 driver yet as it hasn\u0027t been long enough since we deprecated it.\nThat will happen soon enough though and until then we can simply cap\nprotobuf to an older version that is compatible with the most recent\netcd3 release.\n\nChange-Id: I8d6647118be22b0ce55e01b7e5451612ebe30e73\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bdfc2646251fc3914f1afefee085dd916c20bc6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/bdfc2646251fc3914f1afefee085dd916c20bc6a"}]},"branch":"refs/heads/master"},"e3de0e634af62e46162eb8d2354416ce4f2ba77c":{"kind":"REWORK","_number":2,"created":"2022-06-28 14:12:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/49/847949/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/49/847949/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/49/847949/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/49/847949/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/49/847949/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/49/847949/2"}}},"commit":{"parents":[{"commit":"b21c20f79fbcf81b79706a5910ff802b405ca733","subject":"Drop python3.6/3.7 support in testing runtime","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/b21c20f79fbcf81b79706a5910ff802b405ca733"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-28 10:53:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-28 14:11:35.000000000","tz":60},"subject":"Cap protobuf \u003c 4.x","message":"Cap protobuf \u003c 4.x\n\ntooz doesn\u0027t use upper-constraints so we get the newest version of all\ndependencies. It seems one of these, etcd3, is incompatible with recent\nversions of protobuf.\n\n  Failed to import test module: tooz.tests.drivers.test_etcd3\n  Traceback (most recent call last):\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 436, in _find_test_path\n      module \u003d self._get_module_from_name(name)\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 377, in _get_module_from_name\n      __import__(name)\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/tests/drivers/test_etcd3.py\", line 22, in \u003cmodule\u003e\n      import tooz.drivers.etcd3 as etcd3_driver\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/drivers/etcd3.py\", line 18, in \u003cmodule\u003e\n      import etcd3\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/__init__.py\", line 3, in \u003cmodule\u003e\n      import etcd3.etcdrpc as etcdrpc\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/__init__.py\", line 1, in \u003cmodule\u003e\n      from .rpc_pb2 import *\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/rpc_pb2.py\", line 16, in \u003cmodule\u003e\n      from etcd3.etcdrpc import kv_pb2 as kv__pb2\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/kv_pb2.py\", line 36, in \u003cmodule\u003e\n      type\u003dNone),\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/google/protobuf/descriptor.py\", line 755, in __new__\n      _message.Message._CheckCalledFromGeneratedFile()\n  TypeError: Descriptors cannot not be created directly.\n  If this call came from a _pb2.py file, your generated code is out of date and must be regenerated with protoc \u003e\u003d 3.19.0.\n  If you cannot immediately regenerate your protos, some other possible workarounds are:\n   1. Downgrade the protobuf package to 3.20.x or lower.\n   2. Set PROTOCOL_BUFFERS_PYTHON_IMPLEMENTATION\u003dpython (but this will use pure-Python parsing and will be much slower).\n\nThe etcd3 dependency is only used for the etcd3 driver which was\ndeprecated in change Iff0cd3b51cfc45ecbddc831c55267e80e9c79cac. The\nprotobuf dependency is only used for etcd3. We can\u0027t outright remove the\netcd3 driver yet as it hasn\u0027t been long enough since we deprecated it.\nThat will happen soon enough though and until then we can simply cap\nprotobuf to an older version that is compatible with the most recent\netcd3 release.\n\nWhile we\u0027re here, we also sync the lower boundaries in the docs\nrequirments and add some comments indicating where they come from/what\nthey\u0027re for.\n\nChange-Id: I8d6647118be22b0ce55e01b7e5451612ebe30e73\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e3de0e634af62e46162eb8d2354416ce4f2ba77c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e3de0e634af62e46162eb8d2354416ce4f2ba77c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
