)]}'
{"id":"openstack%2Ftooz~848619","triplet_id":"openstack%2Ftooz~stable%2Fwallaby~I8d6647118be22b0ce55e01b7e5451612ebe30e73","project":"openstack/tooz","branch":"stable/wallaby","topic":"bug/1980079","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-04 15:00:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d6647118be22b0ce55e01b7e5451612ebe30e73","subject":"Cap protobuf \u003c 4.x","status":"MERGED","created":"2022-07-04 11:46:26.000000000","updated":"2022-07-04 15:02:59.000000000","submitted":"2022-07-04 15:00:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848619-bug/1980079","meta_rev_id":"2fca717880a383bb0533724f5cbfcafcafd35fe7","_number":848619,"virtual_id_number":848619,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 15:00:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-07-04 11:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-07-04 11:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-04 11:51:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 11:56:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-04 13:01:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a6485579f68a7c15596a21c709f7c64544f777d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:46:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dbb3c647aa97e2b7263e9f9bc2f31a7a624437ee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:51:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"286ec58ec86509013637a5bbe141e998e8993ca7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-04 11:56:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1105e448f3bc85e1191be192fd18118449e53f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:01:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceb0b5e49313455b9e9910373dee0bd9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8107bc790fb74e8c8ba4cc38b35edbdd : SUCCESS in 2m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b1cbf34d445449ac8404cba955f40494 : SUCCESS in 1h 12m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce92ee345b724d15ba8d0f6be3f1d89a : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1075c71bdb8d4338bcb752bb24d74b6b : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32e99dcbfaef410582961aacdccd6219 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bd651b4cdb44ceb835c655a6d97760f : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d599232bf2345e7b7cb242e598d04b2 : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3fe595c25174460b38954d1b7099fbd : SUCCESS in 4m 29s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/8e92689efaa44fa981e9b157faabf187 : SUCCESS in 4m 17s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/2390e94eebe14d40ac3d2267cf48d137 : SUCCESS in 4m 35s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/72917a0b4a744d8ebf3fd4df52131f1a : SUCCESS in 4m 28s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/56b384ec123444feb4cdb014cd21af57 : SUCCESS in 3m 58s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d33390a83c7b4a968824464180c893da : SUCCESS in 4m 10s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/c6f9e805c9584aac9658434e512a35de : SUCCESS in 3m 59s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/13b9ef0e77d949ac81b805960f04cdd3 : SUCCESS in 5m 03s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/97c667f1416a43d3a62b7e870c4c9461 : SUCCESS in 4m 25s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/97881effeeaa487f99a796b0f6c13a9c : SUCCESS in 5m 14s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/2547f45c18df4b59ae1285c442eeaec0 : SUCCESS in 4m 46s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/a2f77c370bf0481081f4c8aa54b24b5e : SUCCESS in 4m 12s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/b6d55cae4f514d9eb2237f95768fcc10 : SUCCESS in 4m 09s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/f516fccf2b68412cba4d934535c17013 : SUCCESS in 4m 12s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/571fb580d826479098c483663e33016f : SUCCESS in 4m 23s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/e678df8ca5a2437887e4575843f7f2be : SUCCESS in 5m 29s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/46c52433621746fc9a1f58c9647ef5fa : SUCCESS in 3m 52s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/5e61df533d374f5591553701f8c58c20 : SUCCESS in 5m 26s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/9095e08e29434525b42b84da662823af : SUCCESS in 5m 46s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/f16d97e5365944e08613ce6323525406 : SUCCESS in 5m 12s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/631daadec5784491884b00d8f0f723d2 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"ad70f986fd733a6a6fa7159439d32b2388f2b3c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:01:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a1947172dea65be8ff4240f16bd29094ff0be773","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:00:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ad3a3c6a14f4fbca55fe88878c493bd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3c0cedf9147445859817aace46f6176a : SUCCESS in 2m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f4e76b3da5e4fd1b99aae3bf35a6ee0 : SUCCESS in 1h 58m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59e36cbd05c4487ea7313f9d6cc0e02d : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/133e24fbd318402b99cafc61be538cb5 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aed33c733ba34e6fb0456ce5e32cd4a1 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd19723c7634461192d11a04682055b5 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0254772449fc48abb8d9b7d5e9813db7 : SUCCESS in 4m 05s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/17183fc6770f4d0994035ddd9d2c9200 : SUCCESS in 5m 16s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/a31a43cf0e24479db2af53dfc77b74e9 : SUCCESS in 5m 10s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/c4652e38eb7c436db2de08bd599bea76 : SUCCESS in 4m 38s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/b032d9c060f54040be862bff84276dde : SUCCESS in 4m 49s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/efe7e3ee679749078283f2680c64a775 : SUCCESS in 4m 56s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/cc3385e335ff4a8fa1b6802ac044569d : SUCCESS in 5m 35s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/d2286c982ec34fb48d711c33c7af250c : SUCCESS in 5m 07s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/dddfa910c453404b8af409724915c913 : SUCCESS in 5m 10s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/14c2c9c5125c48a1a034ecb0100057bc : SUCCESS in 5m 29s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/1b427133671f4ad998319f065b5628bb : SUCCESS in 5m 42s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/f8065b5477fa4148b4bf11bfd5918fd0 : SUCCESS in 4m 47s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/0471fd0450d2406794ea44d9229a5e99 : SUCCESS in 5m 15s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/39e22189ca4b4880a30768169edc07f1 : SUCCESS in 5m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/7211d0dffb8e4dc09fdf3a18d70c1aa1 : SUCCESS in 5m 28s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/235d7fd29fd240e0b8b2b8747718f059 : SUCCESS in 6m 06s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/9e8433b9def74ed59a2171e12273ea01 : SUCCESS in 5m 20s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bdd9bc04fa1148d78f894e43e3d07172 : SUCCESS in 5m 12s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/705cc68f13784889ac467cd098a6ba26 : SUCCESS in 5m 24s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/6e02e99cb88e4cc689b0d7cb4ed8a5dd : SUCCESS in 4m 49s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/7653d1a6d87741d695f16ce23709c966 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"45de6565cc5d9bda0a0530aa49d9b5b24f4898b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:00:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2fca717880a383bb0533724f5cbfcafcafd35fe7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:02:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64e7a8dfb4974b928b0cc7018a5a69c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f528fb14704d5c9a38f40a55c19424 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd26d8389b434a5d89170e8861f4a469 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73dbe0e6fc1403cbb121985781d7b6ec996b099a","revisions":{"73dbe0e6fc1403cbb121985781d7b6ec996b099a":{"kind":"REWORK","_number":1,"created":"2022-07-04 11:46:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/19/848619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/19/848619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/848619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/848619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/19/848619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/19/848619/1"}}},"commit":{"parents":[{"commit":"11526e594c7e7cef8b5a4a4220f100d76a418036","subject":"Enable retries in redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/11526e594c7e7cef8b5a4a4220f100d76a418036"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-28 10:53:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 11:46:16.000000000","tz":60},"subject":"Cap protobuf \u003c 4.x","message":"Cap protobuf \u003c 4.x\n\ntooz doesn\u0027t use upper-constraints so we get the newest version of all\ndependencies. It seems one of these, etcd3, is incompatible with recent\nversions of protobuf.\n\n  Failed to import test module: tooz.tests.drivers.test_etcd3\n  Traceback (most recent call last):\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 436, in _find_test_path\n      module \u003d self._get_module_from_name(name)\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 377, in _get_module_from_name\n      __import__(name)\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/tests/drivers/test_etcd3.py\", line 22, in \u003cmodule\u003e\n      import tooz.drivers.etcd3 as etcd3_driver\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/drivers/etcd3.py\", line 18, in \u003cmodule\u003e\n      import etcd3\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/__init__.py\", line 3, in \u003cmodule\u003e\n      import etcd3.etcdrpc as etcdrpc\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/__init__.py\", line 1, in \u003cmodule\u003e\n      from .rpc_pb2 import *\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/rpc_pb2.py\", line 16, in \u003cmodule\u003e\n      from etcd3.etcdrpc import kv_pb2 as kv__pb2\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/kv_pb2.py\", line 36, in \u003cmodule\u003e\n      type\u003dNone),\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/google/protobuf/descriptor.py\", line 755, in __new__\n      _message.Message._CheckCalledFromGeneratedFile()\n  TypeError: Descriptors cannot not be created directly.\n  If this call came from a _pb2.py file, your generated code is out of date and must be regenerated with protoc \u003e\u003d 3.19.0.\n  If you cannot immediately regenerate your protos, some other possible workarounds are:\n   1. Downgrade the protobuf package to 3.20.x or lower.\n   2. Set PROTOCOL_BUFFERS_PYTHON_IMPLEMENTATION\u003dpython (but this will use pure-Python parsing and will be much slower).\n\nThe etcd3 dependency is only used for the etcd3 driver which was\ndeprecated in change Iff0cd3b51cfc45ecbddc831c55267e80e9c79cac. The\nprotobuf dependency is only used for etcd3. We can\u0027t outright remove the\netcd3 driver yet as it hasn\u0027t been long enough since we deprecated it.\nThat will happen soon enough though and until then we can simply cap\nprotobuf to an older version that is compatible with the most recent\netcd3 release.\n\nWhile we\u0027re here, we also sync the lower boundaries in the docs\nrequirments and add some comments indicating where they come from/what\nthey\u0027re for.\n\nChange-Id: I8d6647118be22b0ce55e01b7e5451612ebe30e73\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit e3de0e634af62e46162eb8d2354416ce4f2ba77c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/73dbe0e6fc1403cbb121985781d7b6ec996b099a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/73dbe0e6fc1403cbb121985781d7b6ec996b099a"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
