)]}'
{"id":"openstack%2Ftooz~848620","triplet_id":"openstack%2Ftooz~stable%2Fvictoria~I8d6647118be22b0ce55e01b7e5451612ebe30e73","project":"openstack/tooz","branch":"stable/victoria","topic":"bug/1980079","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-04 16:36:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I8d6647118be22b0ce55e01b7e5451612ebe30e73","subject":"Cap protobuf \u003c 4.x","status":"MERGED","created":"2022-07-04 11:46:49.000000000","updated":"2022-07-04 20:02:09.000000000","submitted":"2022-07-04 20:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848620-bug/1980079","meta_rev_id":"90e3080f792adac0ac39dd38a1aceb9ca2ae34fd","_number":848620,"virtual_id_number":848620,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 20:01:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-07-04 11:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-07-04 11:55:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-04 11:52:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 11:55:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-04 13:12:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1386d3b838ff1a778a0b7424dc910fba6aabcc8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:46:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cf456ab1581bf847ef695a199cb70f8524cc4dc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:52:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83df0e6dcb7b288a0d82a792cb0edb6e31abada1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-04 11:55:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab40779ed22d3cc3c68cbbb725ca1d96b70ec5d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:12:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3fb9f31b029450983365e68fb1ef2ec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1bfcc62c43ca442e848c52d6134cd576 : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/242c201d27ca41d4bb4440d060f8d401 : SUCCESS in 1h 24m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e14d1c94a7924ec2b9e7a37f9a57d331 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ee5b55ccc1543429dfa84946c69d60c : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a724ca59eb2450fb9b23274b04118cb : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39408eef98df4d00ba50da0aa68b9936 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3799389dc25b480dae7eeaf95b845041 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df4ba2d2c4ba463d9a458d33d9d3e584 : SUCCESS in 4m 41s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/2a294d7a8cce4409a2f723bcb295b4d3 : SUCCESS in 4m 24s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/62e8cd443ce64d988ea09bad48b8e0e5 : SUCCESS in 4m 18s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/e73bf971afaf43eeb2f39ca072bd51b6 : SUCCESS in 4m 08s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/e4a0856dd36e4a0db89ec478c050b2d0 : SUCCESS in 4m 03s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/97f2b4f8e38748d5ba5c6f84f1b9a39b : SUCCESS in 4m 42s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/5dc4dc12f8ae49b3b66ce39467ffb476 : SUCCESS in 3m 51s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/bd89c39274004bf4ae846a0433d1bbc9 : SUCCESS in 4m 16s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/2ac9b55d1c3a47a08e22c3dcd0638b34 : SUCCESS in 4m 21s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/c9219ca057964230b60c0de16cf1268e : SUCCESS in 4m 16s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/91261452b28946ebbd445337933727c1 : SUCCESS in 3m 59s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6ccde4dbc36b451493ee3c23bfab07a9 : SUCCESS in 4m 17s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/0ba0d4a192444b3d9c0b2ccc5022d5d9 : SUCCESS in 4m 57s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/8ad3ffef7faa4e39bf67ed5e658d9504 : SUCCESS in 4m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/5e4e4a03fa82442dbf773ee74c06d81e : SUCCESS in 4m 16s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/2b099202444b4e84a7ec980419aef55a : SUCCESS in 4m 56s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/a0cfc50724bd4666acbe8edb299f650f : SUCCESS in 4m 23s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/9262a18d125c4614b922846ce94ab718 : SUCCESS in 4m 24s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/3e0c754346694d97b5469ce8e08620ae : SUCCESS in 5m 45s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/6ac759b04aa44e1991d01fc25522ee4c : SUCCESS in 5m 22s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/b0a90080a9754e5d867174c015b80bbb : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"34c73c10aa78c0f0d9cc91e15a9b9fc219b387a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:13:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d76b952d42aec55349a32b12691f2080e165be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 15:07:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5385ebe9bb9461097d04d2762f69000\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3802fc33c0b34256abb2d265b4f92252 : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1547f3d09d4c4e5caa241a297c2fd684 : SUCCESS in 1h 47m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41f934ca03b249d7bc9387ff758d325f : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dedd0b1e33b5485cb57c2ebe207a4e8e : FAILURE in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64888f816aa24926b17d69bd375a1d23 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9daf06c4efe491eb876dd8966c9842b : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/801ca125b805440794a00d8efd733435 : SUCCESS in 4m 12s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/450786b47cd448d7b65851cdb0141806 : SUCCESS in 4m 44s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/0c085e59f785446190ee28aa68761580 : SUCCESS in 4m 31s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/19f023684f474270a4c3ba400896d9d5 : SUCCESS in 5m 12s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/30d626c265e54a039b7f469949239fe8 : SUCCESS in 4m 30s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/f708318aed3b4395b3dc85b151cdd386 : SUCCESS in 4m 52s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/1e421f16c61f475eb52f1913667dc496 : SUCCESS in 4m 18s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9cf76811aaf2424b9b68bd24cfac360c : SUCCESS in 4m 38s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/d63141b845e44b37b3be90d2e41336eb : SUCCESS in 4m 49s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f16efff5d9c54f6091997b2acdf9fbd9 : SUCCESS in 4m 42s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/ac936175b066413884c02eb384c37eb7 : SUCCESS in 4m 27s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/eee65418e95b494483b5d653d058c1a5 : SUCCESS in 4m 51s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/06998c2efcf14a64954e809685a19fe3 : SUCCESS in 4m 29s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/3c40054d21c7404baf87bf815c82676e : SUCCESS in 4m 15s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/64774bc501d64244836f93038219d52d : SUCCESS in 4m 33s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5113a3dcb5644b99a02f088e1fcf84d5 : SUCCESS in 4m 14s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/60a366ec1be34198b841f35b5048754f : SUCCESS in 4m 25s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/b9805d9c5e9b406485ff065353c8e923 : SUCCESS in 4m 25s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/998b7e86a4374873a1971202d3d328be : SUCCESS in 4m 33s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/fab45936803d4f0495ef1fb20dc04f8e : SUCCESS in 4m 38s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/61516a63016640ef934c034f074aa312 : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":1},{"id":"53119122d5525dd16fa338958cfdcc490f739bc5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 16:36:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03b4d37c05bcee09c41dc299a4377b04b9f7ab7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:13:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16fa926f5b8e447aace7ca0231f12e0f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7d770ecacb1d461e8a9a94d64022998e : SUCCESS in 2m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1ee6fde276142c09e65863be8eb83b9 : SUCCESS in 1h 31m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/842596a78429419ba86a33ddf22d342a : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a934b25cabb4c849e6065222be27206 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/880c165fb0e64bbf9b549e4f3e6e6344 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/775cf0dc29a34ffba170bd6355fbc761 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2bd091bf86d4858bd7dd3d0cced8445 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b7cf32ccc6b4cb2811c28577c9e5653 : SUCCESS in 4m 50s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/d9f43d1d702842e2895e93b48712efab : SUCCESS in 5m 38s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/c4168b4b64024a89aa89ea8122a6c236 : SUCCESS in 5m 05s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/4862e9d66dfc4cecb333dc04fbed2661 : SUCCESS in 5m 42s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/30c7ee84d34b4b9b8d0d5d2a4d38aa1b : SUCCESS in 4m 57s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/9a5965bad5194bb1a10ea1d89eff5dc5 : SUCCESS in 5m 02s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/75d4cc7694394dcc8c4bc479d512e948 : SUCCESS in 4m 37s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/502b2688b68a45749df1925ac66901ec : SUCCESS in 4m 49s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/cb6994ab02ce496fa31b2c6a20dd2400 : SUCCESS in 4m 29s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/e293a88318414213b62d8aad22477efa : SUCCESS in 6m 12s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/dc17578134504ef98577b1931ebe2e8c : SUCCESS in 4m 52s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/a285a0d7bcb34ec1a6978b46cfbd888a : SUCCESS in 4m 58s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/1c07f3f2c9a14872bf0a0c07d8ee60c6 : SUCCESS in 4m 53s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/a24a65960b904554926384f265e33a4f : SUCCESS in 5m 48s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/a2703e90bd7040f8bcd52c9db135e3a9 : SUCCESS in 4m 53s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/d37061a899c24ba68361ecb1ff252e3b : SUCCESS in 4m 51s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/348d3a911e1646248b7b198c0be8268b : SUCCESS in 4m 34s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/6e388075ae0d4d0c81c899b9c40b3bfd : SUCCESS in 4m 39s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/13f3bc0f55a5489099cb4d518f411fea : SUCCESS in 5m 01s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/2ecc62deceae4ac5a799d45d075376cc : SUCCESS in 4m 44s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/1c363d315d1245978c1f54a92105a50c : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"18567cf2bc5679d05aa219ca2c0282c6bbd61e7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:14:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76dd9d2732df3fc1d887aa9fc45b2bd589ec6015","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:01:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3931b60ded5741db85724cc4d4039f83\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac0bf2b5446d47e182fc5af842f1617d : SUCCESS in 2m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68af29cef0014c039d44c838f235b765 : SUCCESS in 1h 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e5ebecd38774309bf59d6b90bb8edd3 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29d4eb57737a4bd88cd8505310ed1fee : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d68d4b6f5bd45b8849be7893896f10d : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f004a57c783f4cada4702975ca54d0d2 : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c49a26d14da44f798f1c843ffb9ad675 : SUCCESS in 4m 10s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/43b8c722f6f248078d2435e15b96eed8 : SUCCESS in 4m 46s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/666fc4910db64ff1a00ea72b85b30589 : SUCCESS in 4m 31s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/972f13df5f6b43cbb90913028c42b7ff : SUCCESS in 4m 43s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/9db0212ae16d454d92844bb733d45c65 : SUCCESS in 4m 59s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7a05e8a5034a4ca5ae2df0d823f6cfc9 : SUCCESS in 4m 34s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7f1321544c994125abd4ef00ee3b3abe : SUCCESS in 4m 05s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/437b59fbff4540b7adbb1512240e407b : SUCCESS in 4m 47s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/c90094f11df64bdebdcd475d2f316942 : SUCCESS in 4m 14s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a448d28a8e904952a1f11930c6e3d1ee : SUCCESS in 4m 32s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/87518df192cb4563bdd48aef94b5aa0d : SUCCESS in 4m 34s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/c25a8bb31d84454b9f2fa2e8992720c4 : SUCCESS in 4m 29s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/ff2157495fbe4644919c5fe8dc67e801 : SUCCESS in 4m 45s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/18ddd4a3c2324b4a888cb65d47f29306 : SUCCESS in 4m 24s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/a6d4cdb3f4494e14aef21af240504500 : SUCCESS in 4m 29s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/e20dc3afddb94a0db4ef485cf91b42ea : SUCCESS in 4m 17s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/1eda2d10d5e748afb07cc52ea1f2bef0 : SUCCESS in 4m 12s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/342179cd27d84104b12061e67c1e1151 : SUCCESS in 4m 59s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/3fb78277aafa430ba1df0b320e31470e : SUCCESS in 4m 23s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/210ed05c855c4a1181d05ceabf96d84e : SUCCESS in 4m 35s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/4401c62dc26d4dd090dfc446b3995123 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":1},{"id":"06778b287ae4ec9166d7ed3e52f926ee4175fd7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:01:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"90e3080f792adac0ac39dd38a1aceb9ca2ae34fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:02:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae10234d9d7b4a978498c218ffcb17e4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6698830ccc81469db7a9dc4e66f23cc1 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e51c0833dd345e9be236509e1b6b02d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ca40da4bbe813a76cd48d3e8e54b78d10f25765","revisions":{"8ca40da4bbe813a76cd48d3e8e54b78d10f25765":{"kind":"REWORK","_number":1,"created":"2022-07-04 11:46:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/848620/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/20/848620/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/20/848620/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/20/848620/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/20/848620/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/20/848620/1"}}},"commit":{"parents":[{"commit":"c1ae1649d22b356f2bbb2293cb421eb155fdee14","subject":"Enable retries in redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c1ae1649d22b356f2bbb2293cb421eb155fdee14"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-06-28 10:53:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 11:46:43.000000000","tz":60},"subject":"Cap protobuf \u003c 4.x","message":"Cap protobuf \u003c 4.x\n\ntooz doesn\u0027t use upper-constraints so we get the newest version of all\ndependencies. It seems one of these, etcd3, is incompatible with recent\nversions of protobuf.\n\n  Failed to import test module: tooz.tests.drivers.test_etcd3\n  Traceback (most recent call last):\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 436, in _find_test_path\n      module \u003d self._get_module_from_name(name)\n    File \"/usr/lib/python3.7/unittest/loader.py\", line 377, in _get_module_from_name\n      __import__(name)\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/tests/drivers/test_etcd3.py\", line 22, in \u003cmodule\u003e\n      import tooz.drivers.etcd3 as etcd3_driver\n    File \"/home/zuul/src/opendev.org/openstack/tooz/tooz/drivers/etcd3.py\", line 18, in \u003cmodule\u003e\n      import etcd3\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/__init__.py\", line 3, in \u003cmodule\u003e\n      import etcd3.etcdrpc as etcdrpc\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/__init__.py\", line 1, in \u003cmodule\u003e\n      from .rpc_pb2 import *\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/rpc_pb2.py\", line 16, in \u003cmodule\u003e\n      from etcd3.etcdrpc import kv_pb2 as kv__pb2\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/etcd3/etcdrpc/kv_pb2.py\", line 36, in \u003cmodule\u003e\n      type\u003dNone),\n    File \"/home/zuul/src/opendev.org/openstack/tooz/.tox/py37/lib/python3.7/site-packages/google/protobuf/descriptor.py\", line 755, in __new__\n      _message.Message._CheckCalledFromGeneratedFile()\n  TypeError: Descriptors cannot not be created directly.\n  If this call came from a _pb2.py file, your generated code is out of date and must be regenerated with protoc \u003e\u003d 3.19.0.\n  If you cannot immediately regenerate your protos, some other possible workarounds are:\n   1. Downgrade the protobuf package to 3.20.x or lower.\n   2. Set PROTOCOL_BUFFERS_PYTHON_IMPLEMENTATION\u003dpython (but this will use pure-Python parsing and will be much slower).\n\nThe etcd3 dependency is only used for the etcd3 driver which was\ndeprecated in change Iff0cd3b51cfc45ecbddc831c55267e80e9c79cac. The\nprotobuf dependency is only used for etcd3. We can\u0027t outright remove the\netcd3 driver yet as it hasn\u0027t been long enough since we deprecated it.\nThat will happen soon enough though and until then we can simply cap\nprotobuf to an older version that is compatible with the most recent\netcd3 release.\n\nWhile we\u0027re here, we also sync the lower boundaries in the docs\nrequirments and add some comments indicating where they come from/what\nthey\u0027re for.\n\nChange-Id: I8d6647118be22b0ce55e01b7e5451612ebe30e73\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit e3de0e634af62e46162eb8d2354416ce4f2ba77c)\n(cherry picked from commit 73dbe0e6fc1403cbb121985781d7b6ec996b099a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8ca40da4bbe813a76cd48d3e8e54b78d10f25765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8ca40da4bbe813a76cd48d3e8e54b78d10f25765"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
