)]}'
{"id":"openstack%2Ftooz~848624","triplet_id":"openstack%2Ftooz~stable%2Ftrain~I4593085e60930f039d13db2a4c7799b16a1f055a","project":"openstack/tooz","branch":"stable/train","topic":"bug/1980079","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-05 16:32:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4593085e60930f039d13db2a4c7799b16a1f055a","subject":"ignore reno generated artifacts","status":"MERGED","created":"2022-07-04 11:49:52.000000000","updated":"2022-07-05 16:34:33.000000000","submitted":"2022-07-05 16:32:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848624-bug/1980079","meta_rev_id":"81a70c331b0b9857a72d38c9898379b1b0af88f2","_number":848624,"virtual_id_number":848624,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 16:32:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-07-05 12:05:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-07-05 13:55:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2022-07-05 13:55:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 11:49:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-07-04 13:26:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-05 13:55:48.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"0b8afb8f8e522f88744ec695ae3cb70ade1476a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"027cb1d582dadd4ad55e60cdd3e26974575562dd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 11:54:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ade94a7f2111526f32d3b781a467288ab8228c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 13:26:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/de487b7f9444425eadca675e3e1623fc\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/f48150c8eb6e42fd84a032a4ad6192e9 : SUCCESS in 1h 31m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8247808bab73465ea449b7eab64424c4 : SUCCESS in 1h 26m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5aafb271cd0642a7abe5f190af70c393 : FAILURE in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e45afda236134245b3eef4445bd871a1 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72fc0e7ca91d407f8c534a10b8ff2594 : FAILURE in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e450138e42d47ac8654c05545bcac66 : FAILURE in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6bf0658b7d34d2cb473ca57691dc39c : FAILURE in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831ae28328b54c59a90e96842e071f7b : SUCCESS in 4m 47s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/fecbac3653ec4bc4a16a20eee089b47f : FAILURE in 3m 44s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/3e3aebb31b55449d920bcd5ec0a06f03 : FAILURE in 4m 20s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/92745f30a76b49a799df9844df2a151e : FAILURE in 4m 06s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/2005ee40bca2441f85264175cb6d081b : FAILURE in 4m 29s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/a8815e0f18f049b09737f9fa4229184f : FAILURE in 3m 59s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8e3695e03d604b8fb090b52aab67edcb : FAILURE in 4m 14s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/a583bc891f604aa7a7e2e1eb3c389f17 : FAILURE in 3m 48s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9f206b25b5174243964ed4ecd0399401 : FAILURE in 4m 50s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/2aec708c57da4609bd8b008162fda2a3 : FAILURE in 4m 18s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/a47c1713a8cc4420bbe929a2cd29b322 : FAILURE in 4m 38s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/5277505a33d645fdb026774402816c56 : FAILURE in 4m 05s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/e01cae2a0a9c42d9b47f43982f4011b4 : FAILURE in 4m 24s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/3260e920d6ec4df7ab43cc8d6f5165ae : FAILURE in 3m 12s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/0d7bc3f79fd54c32a88f3f95049a3e5b : FAILURE in 4m 24s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/4c7e68b7d7364c0d9c4f71cc23f70665 : FAILURE in 3m 50s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/4d0388d87ae44ff2b91bf109d71956fc : FAILURE in 4m 25s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/5a062c106e304d1ea02e30683d05fad3 : FAILURE in 3m 47s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/cb9bbccb3e524a7ab1ef63e88bbd54e1 : FAILURE in 4m 28s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/afdb507289c8457e82d9390aa552c8af : FAILURE in 4m 53s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/3724a6a043c54804818c5b3de92ab779 : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":1},{"id":"845676cc5bb9178ada47670c7f55cfdaf72ea9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:00:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c757d02e1a25503b964da1fd4a6209faa907fb5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:01:13.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"d93eac9cc55a132ff554661ea16e4a68286dc3db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 11:53:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd14deea1e3d42a7bab094e1179c806e\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5024205cf796420cb97fe47b20bec719 : SUCCESS in 1h 36m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b8c13160cef4faba8526020226db997 : SUCCESS in 1h 13m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e0894ca73e64796a2a4fb5fe95cba0f : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/627a30838339431c9348e96f3405ac21 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2a907d34a9944fe7955c8ed55c5f633b : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04da1796eda1431688cdc7b3ba8925ca : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f10fca4ee72f4108acaf1bcbf5e51421 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae687e86794e4f5ab99da7b53db39a79 : SUCCESS in 4m 24s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/99692c82ef744dfbbbadc189fdd5ec5e : SUCCESS in 5m 28s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/f0b707fdbe16415492aad85b6e4ecaf1 : SUCCESS in 4m 47s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/924708e2b315470b9870848a00aa7e93 : SUCCESS in 5m 49s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/0670ecdbc1d6434d992b9df7f700c8ad : SUCCESS in 4m 45s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/08e8a228a359423498e3dd86627a51b1 : SUCCESS in 4m 47s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/27cd6c0ef0eb490e9d5df316061cfc6d : SUCCESS in 5m 17s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/58e7ba867e9b413087612b607a76bd69 : SUCCESS in 4m 55s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/c4d7af40ad334fcfa6c8d27362d7f924 : SUCCESS in 5m 13s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/f74b83d3a8914445a4cc6f0244cdfc97 : SUCCESS in 5m 04s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/63e029ce96314b27b52d3247ddb5cdd8 : SUCCESS in 4m 31s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/eb819f5b0d8c4d29a55bbbf422fdbfd5 : SUCCESS in 5m 24s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/2d939b0b9bdd43dcac4198946832ac21 : SUCCESS in 4m 39s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/6782c1845a7c43d2bc5dd6fe62162091 : SUCCESS in 5m 53s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/55db24b38b3b466f85766be9f572c18f : SUCCESS in 5m 21s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/9c68073e197949c08625cee1fc971a59 : SUCCESS in 5m 06s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/816ebadd52ac4f709e226ad1727fd944 : SUCCESS in 4m 04s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/2ceced5de022459392b15ee2cdde753a : SUCCESS in 5m 01s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/4816e30444da4c28b8e04293fee18db1 : SUCCESS in 5m 05s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/085f737d34c6422daf490708cd5e655f : SUCCESS in 4m 15s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/9e7be3475a9340998db59922233b66f4 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":3},{"id":"cd123d47e812f8a9fe1d6dcd8959af816e8bbcb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 12:05:49.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"16e1ad4127c1dfba9b31127bafbd8ed2f9c45d6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 13:38:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ec7b7c111e4082b8112e4d1e95df2d\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/5b765b3c675449aca41520809ce42e21 : SUCCESS in 1h 30m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ddcf75b679034e14a01527df15e3223b : SUCCESS in 1h 12m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/144bd8e494d946ef85e7616b0eab536a : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bae07bed2e1417282f2b965a0053ec8 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c2427abf6274e3fbae0299b547e6814 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32271be2b4f416abedb64be93a3fb9b : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5edb35c60f7543d49803ef208fa815ab : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c931ddc7a90c4d13ba884f0182c31167 : SUCCESS in 6m 11s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/ec3d035095984c979242d8e7c3f61d83 : SUCCESS in 4m 21s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/0d40eeec1c0f44b3959b738eca1be03e : SUCCESS in 5m 10s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/fad8ed2112a149ffa57d2375df05b2ed : SUCCESS in 4m 13s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/dda5efd2eb8e452091c5b4e4772e20c9 : SUCCESS in 5m 20s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/5158b23b8a6347ab8d356b3499db0680 : SUCCESS in 4m 54s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/8c5798db23f54dcaadd18396ae2d8e58 : SUCCESS in 4m 41s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/898c915c98134e09ab503e3cca90fa4c : SUCCESS in 5m 25s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/0f72363b346b46c8a1744d77846aad67 : SUCCESS in 5m 31s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/c266127008b249ed89678f05d8c0a348 : SUCCESS in 4m 44s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/d5b06b8b3a6f4308b1d57394b6c41ad0 : SUCCESS in 4m 13s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/9acdb993cf1a450999486710a70e3f62 : SUCCESS in 4m 37s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/93baffe79bbb4703ad44168640701f99 : SUCCESS in 5m 30s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/0d21453c04bf4e7f9e84b6261e65c1da : SUCCESS in 5m 30s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/7050cd4a454c4b57b945ecbe8cc81da9 : SUCCESS in 4m 03s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/4ba2a7269715468cbd6a4b0791668196 : SUCCESS in 5m 11s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/85cf91ced35f4e8ca0b3c7999dbe2eec : SUCCESS in 4m 54s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/45d2c4d11f19435196257ebb9d523f81 : SUCCESS in 4m 18s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a5f5c80aa6c44c38b22513f9c12fcbd5 : SUCCESS in 5m 14s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/7ac0623d14ac41888818d0131fb17b25 : SUCCESS in 5m 05s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/2a4884f48c654f2698946447f88c42ca : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":4},{"id":"3614156df29d7d3cec22c4e39963be023b8ecf7f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-05 13:55:48.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"1cb6f9b92b1c2aa3975d0feaddd6b8171c3d8f42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:55:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"952ba4bf1cb4e4359269756967ad1b58deb796c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:32:32.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9aa389dc561c4f1cb7c329562a60d464\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/68c86adc8a204711aa62d05a7693cf33 : SUCCESS in 1h 25m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/542a542b43924240af19dfe1da7dde9a : SUCCESS in 1h 23m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a42078529ad4224b82b17a270a0acbf : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad4a03b9a8cd45cd9b32379d48383b4d : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4310b9bf0bee43a5a48dd898e3c0e485 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d057a6eb34d542b1a2982ae6064247f3 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69ef7594b2ce4810923dfc03ff99d1e7 : SUCCESS in 4m 40s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/effc051a25b6448fbd4c4d09a947bd85 : SUCCESS in 5m 26s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/5303e52065604f31a31b6f8adfc74a58 : SUCCESS in 4m 42s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/414bca4a5c0740bdb481b441e6b0e317 : SUCCESS in 5m 04s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/021aebd3063d450e95a1c080d3bf67af : SUCCESS in 5m 10s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/9ca17bebb714441eb0b7a45e29b099e0 : SUCCESS in 4m 35s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/afb84b5e3acf4b78afee1fe779cf428f : SUCCESS in 4m 30s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/ed740d9a5e734e6491140505538dc77d : SUCCESS in 4m 40s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/369edf542ae74aca9f0914940ebd4564 : SUCCESS in 5m 02s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/b4a1821fb9c0439cae20568ea7200737 : SUCCESS in 4m 29s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/aff4e880412b47b1a5771263acd36e62 : SUCCESS in 4m 50s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/cf98bb89a7344ec3bee8b3aed31be0b5 : SUCCESS in 4m 45s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/dd000aeace874b89aeb9541ca18c68e6 : SUCCESS in 5m 56s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/422d87e6b8044973b12abaca1f17f59f : SUCCESS in 4m 27s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/484d18bb6e644ed888c34d0ce8dfb437 : SUCCESS in 4m 22s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/d69837442a884a65838ad14e5e7765a8 : SUCCESS in 5m 12s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/bcc527f074264b5fbfd9048fe895bd46 : SUCCESS in 4m 43s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/83427e6fcd5449e58f3ce846f0bfce04 : SUCCESS in 4m 39s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/a8b39caf865144149a918ffb2b6717b4 : SUCCESS in 5m 05s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/1a5ad9b5fa6046f4a59bdef4e21deaca : SUCCESS in 5m 08s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/de9eb1b15d40446fb6b1a834f4775328 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":4},{"id":"1033fb37eaeb0ccde9114ffdd5ce046c79fa96f9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:32:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"81a70c331b0b9857a72d38c9898379b1b0af88f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:34:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec0124c6a1c4203b94fb3ca8b742591\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed396f5f126d44f3abb529fc002d7078 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dc572af710f97bd47ade7e3262f9653a0fe52f23","revisions":{"018dd16c7442476c72fd42d9d4972357921b0b29":{"kind":"REWORK","_number":1,"created":"2022-07-04 11:49:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/848624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/24/848624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/24/848624/1"}}},"commit":{"parents":[{"commit":"13a6dffa32789ac91b724977ed1d73375baf0824","subject":"[stable-only] Cap stestr for python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/13a6dffa32789ac91b724977ed1d73375baf0824"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-25 12:17:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 11:49:46.000000000","tz":60},"subject":"ignore reno generated artifacts","message":"ignore reno generated artifacts\n\nChange-Id: I4593085e60930f039d13db2a4c7799b16a1f055a\n(cherry picked from commit 4be7780adbd9235233143880044a56e8586b5f68)\n(cherry picked from commit 44f05cddc5ca11d69c87c375d8242deb8bace92f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/018dd16c7442476c72fd42d9d4972357921b0b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/018dd16c7442476c72fd42d9d4972357921b0b29"}]},"branch":"refs/heads/stable/train"},"2a3072e0e2e152c1f86999846f577639bcca0cbf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-07-05 10:00:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/848624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/24/848624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/24/848624/2"}}},"commit":{"parents":[{"commit":"acceb797dce3c1acd6da8d4862c51b3eaf07cc5f","subject":"Enable retries in redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/acceb797dce3c1acd6da8d4862c51b3eaf07cc5f"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-25 12:17:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 10:00:03.000000000","tz":60},"subject":"ignore reno generated artifacts","message":"ignore reno generated artifacts\n\nChange-Id: I4593085e60930f039d13db2a4c7799b16a1f055a\n(cherry picked from commit 4be7780adbd9235233143880044a56e8586b5f68)\n(cherry picked from commit 44f05cddc5ca11d69c87c375d8242deb8bace92f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2a3072e0e2e152c1f86999846f577639bcca0cbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2a3072e0e2e152c1f86999846f577639bcca0cbf"}]},"branch":"refs/heads/stable/train"},"161ea3c03bcf3e1f3b0cbc8ef3a0a6d55b229fc0":{"kind":"NO_CHANGE","_number":3,"created":"2022-07-05 10:01:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/848624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/24/848624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/24/848624/3"}}},"commit":{"parents":[{"commit":"3df3f477bb3aec5aea73d68cae51a5e9c36dc780","subject":"Enable retries in redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3df3f477bb3aec5aea73d68cae51a5e9c36dc780"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-25 12:17:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 10:01:06.000000000","tz":60},"subject":"ignore reno generated artifacts","message":"ignore reno generated artifacts\n\nChange-Id: I4593085e60930f039d13db2a4c7799b16a1f055a\n(cherry picked from commit 4be7780adbd9235233143880044a56e8586b5f68)\n(cherry picked from commit 44f05cddc5ca11d69c87c375d8242deb8bace92f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/161ea3c03bcf3e1f3b0cbc8ef3a0a6d55b229fc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/161ea3c03bcf3e1f3b0cbc8ef3a0a6d55b229fc0"}]},"branch":"refs/heads/stable/train"},"dc572af710f97bd47ade7e3262f9653a0fe52f23":{"kind":"NO_CHANGE","_number":4,"created":"2022-07-05 12:05:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/848624/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/24/848624/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/24/848624/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/24/848624/4"}}},"commit":{"parents":[{"commit":"883bfb68381623c06c21b493171c319c820e718a","subject":"Enable retries in redis driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/883bfb68381623c06c21b493171c319c820e718a"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-03-25 12:17:59.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 12:05:43.000000000","tz":60},"subject":"ignore reno generated artifacts","message":"ignore reno generated artifacts\n\nChange-Id: I4593085e60930f039d13db2a4c7799b16a1f055a\n(cherry picked from commit 4be7780adbd9235233143880044a56e8586b5f68)\n(cherry picked from commit 44f05cddc5ca11d69c87c375d8242deb8bace92f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dc572af710f97bd47ade7e3262f9653a0fe52f23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/dc572af710f97bd47ade7e3262f9653a0fe52f23"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
