)]}'
{"id":"openstack%2Ftooz~848652","triplet_id":"openstack%2Ftooz~stable%2Ftrain~I2437071bb575e95083b0d57a6a0a33aaf8356113","project":"openstack/tooz","branch":"stable/train","topic":"bug/1980079","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-05 14:54:21.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I2437071bb575e95083b0d57a6a0a33aaf8356113","subject":"Avoid redis lock\u0027s expire_time exceeding timeout.","status":"MERGED","created":"2022-07-04 16:54:40.000000000","updated":"2022-07-05 16:32:21.000000000","submitted":"2022-07-05 16:32:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848652-bug/1980079","meta_rev_id":"d7010b985ee5c883f0a45bc94a409069ae42abd4","_number":848652,"virtual_id_number":848652,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 16:32:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-07-05 13:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-07-05 14:54:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26505,"name":"wayen","email":"1412950785@qq.com","username":"chengkang"}],"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 16:54:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":26505,"name":"wayen","email":"1412950785@qq.com","username":"chengkang"},"state":"CC"},{"updated":"2022-07-04 18:28:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-05 13:52:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"c3becd42f1866319f91694350ab392b2d0d413b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 16:54:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54b0fd441c1d3ad25c2cff9056ee06ac0d9eaeaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-04 17:09:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d111d8d456190ea767fd5d7ecc1472f089e1c566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:28:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a286c730a6944e2ba57699d40207753\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/d4e8ee1c9ade476dbf9465a6d2e72d4d : SUCCESS in 1h 16m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed3910c12bda453b9583110debb9133b : SUCCESS in 1h 16m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe8a8b9085bf4875aa7f7a97a2ae39df : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1443868e0cfa42979d5b32efece75e1a : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5cdfae0e9daa44ffb1ac508502e920b6 : FAILURE in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a162b2867c7d4527be61b73d88d09f50 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/13c6c26cbae94d16baf268c4940d575e : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae5e66b65b60487eadfc2d0aeab85395 : SUCCESS in 5m 14s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/14a83b6f4b674019a38eea3671aabebd : FAILURE in 3m 28s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/551b5f12c6bc4ec0953b3ce8167540d4 : SUCCESS in 4m 07s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/2a3d9cd0c548471e8da8fb7d99e4344d : FAILURE in 3m 25s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/4c89917bd6a048de87fbb2a7698d7160 : SUCCESS in 4m 16s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/57d40c1c49a943bfad42edd371074bfc : FAILURE in 3m 27s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/320e9285536f4eb58ba226037cca8cc6 : SUCCESS in 3m 58s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/7fec9c4b28704a9a988be6c882f13608 : FAILURE in 3m 29s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/cfb1bfe32ba14f1f8c295bbae0e3b7a3 : SUCCESS in 5m 27s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/5f8d4f8b390b42a28ef274f5d8067ed4 : FAILURE in 3m 55s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/10051b058d18419697fb4af05f280f30 : SUCCESS in 4m 15s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/62ca27a0707d433f95785d00c66e168d : FAILURE in 3m 36s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/57ef964284aa474ab1b2d4a584b1bfe3 : SUCCESS in 5m 15s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/b464790f07f04fd38045f92a3dd8ff82 : FAILURE in 3m 43s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/c3bf40aa0ed5487fa66f79c6c86cb849 : SUCCESS in 4m 22s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/797767da7120479192d220e32f8f18ef : FAILURE in 4m 18s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/437fa095515c4a6dba467a862a5f9eb7 : SUCCESS in 4m 14s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/b60de9f387fd4951ab90485ff5e5d48c : FAILURE in 3m 35s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/62c485027fe74add9279ed5b390f87a4 : FAILURE in 4m 16s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/907f9c76630342dba6f494ecb848f160 : FAILURE in 5m 13s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/d4f2373b57874d22acd040fb45ca9426 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":2},{"id":"cc9f0b115869ebebe84d183f39056828e7171c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:00:15.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"10aca90eee11aa2c4f60f12cae2187c13bebe16a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 10:01:13.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"beb18f06292e266ee3d6067589e144d400217e17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 11:18:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/106cb194df17478a8447777a4838eab4\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/8d400c7a6cfd47739a793330ba89d72b : SUCCESS in 1h 13m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/707247479ec94e318145bc88305d4c79 : SUCCESS in 53m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59414892eb7849dcbaee4287fa84faa0 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fcd0b7de3934a34b2484fe38f59ea03 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/412a25b57a90413581773527421110c7 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee1b52b215174f959753efdd89a31bf4 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/78eb2b2c515246dc901aaa56c62bece5 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71039ebc1d88444dbf5ce97e0557e2a5 : SUCCESS in 3m 58s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/262c1f8447d44d148bb1870928153832 : SUCCESS in 4m 11s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/1b506cb12bd340c593425b407b511637 : SUCCESS in 4m 24s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/4bbef3b21f8d4e019e42c5092329accc : SUCCESS in 5m 34s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/8e06321537494cd081f1b7038dc591cd : SUCCESS in 4m 00s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/6fb4f5eb659f42f083ba864298208666 : SUCCESS in 3m 58s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/fe35964d669e490997d2ae560742edbd : SUCCESS in 5m 08s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/63d3d17ef3284429ac403f8ea792e140 : SUCCESS in 4m 17s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/7cb4b6170f514b0a8390e4b8cf2f5f9a : SUCCESS in 5m 29s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/764389d92fef40d58fec2f1f46887349 : SUCCESS in 4m 13s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/b32d4c6092a7416eb7ab5649081719ae : SUCCESS in 5m 18s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/0c55f66b9a0143a89a1a8a87c187bd41 : SUCCESS in 4m 41s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/5d0357e3bb504446a783d7b418f68d6d : SUCCESS in 5m 16s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/8da2ee91d52a41ceaf49a4aaf6a666da : SUCCESS in 5m 10s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/65fd91ee7bad40d390b5c1441da89268 : SUCCESS in 5m 28s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/b995bee0d9f14d8f8341d6ac9d402780 : SUCCESS in 4m 35s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/fb1e8137e0b94636a6bf6534df65bede : SUCCESS in 4m 00s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/126dcce816734ca4a16a991ba454075b : SUCCESS in 4m 01s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/5820304000ac43cb9899b7b965f23006 : SUCCESS in 5m 00s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/9916931ddca4404a837d00e1df058623 : SUCCESS in 4m 38s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/41c297da189e4bceb52bc5cf70b687c9 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":4},{"id":"7f3045bcf3cbb4cde2b2acabd28900f71b813e34","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-05 13:52:07.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9a1235a16afade04c97f5fde1dc93096a5db7642","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-05 14:54:21.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b6d7703a926921ab7ea185f531e61c0eeb99b5c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 14:55:14.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"263da38d6044f38e859bcf2a60ea0fbfb8707b57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:32:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb8a38e448264d61a415bf40dcc3d68c\n\n- tempest-full https://zuul.opendev.org/t/openstack/build/742c8c44a49f4316a9b5268a8c3a5276 : SUCCESS in 1h 26m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47ba75d8f3204ee29223af71149a558a : SUCCESS in 1h 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ea4140746a49208cbc364dbad978e6 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e39cea3d29524ef9b7ef51668d57c01a : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88bdd001ba704223ba37663e04f0dc2b : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d0676425d534346a73a40aaf25ff5c3 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab61087ed0f4054ac7ad78fcac6b6c7 : SUCCESS in 4m 27s\n- tooz-tox-py27-etcd https://zuul.opendev.org/t/openstack/build/cc747e1acd3043e083cf90739f742fe7 : SUCCESS in 4m 24s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/00fbac20e5514b708a97cddaaaa29793 : SUCCESS in 4m 48s\n- tooz-tox-py27-etcd3 https://zuul.opendev.org/t/openstack/build/cb9e19e9ab5d4d71b42d6edb63743cb0 : SUCCESS in 4m 44s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/d5f8e71a44804a8697c52d59e2c9d398 : SUCCESS in 4m 26s\n- tooz-tox-py27-etcd3gw https://zuul.opendev.org/t/openstack/build/02d244df51d14a12ae1c2f2f422f8488 : SUCCESS in 4m 35s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/7e736c59c57842d087b3b40059a19070 : SUCCESS in 5m 13s\n- tooz-tox-py27-zookeeper https://zuul.opendev.org/t/openstack/build/7ad0a0d5850a47a9aa0e29b428cec212 : SUCCESS in 4m 10s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/c2090f0f16094e17b769d5793aeaa938 : SUCCESS in 5m 39s\n- tooz-tox-py27-redis https://zuul.opendev.org/t/openstack/build/9b0144245f6644f79b978436bd7e7e9c : SUCCESS in 5m 11s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/2728ec6a1c1344bea630f656b0f6f7b7 : SUCCESS in 4m 50s\n- tooz-tox-py27-sentinel https://zuul.opendev.org/t/openstack/build/85cf188a1a804bcb814d5218545663b3 : SUCCESS in 4m 49s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/873c58671ab3401d9c98e969ff694864 : SUCCESS in 5m 00s\n- tooz-tox-py27-memcached https://zuul.opendev.org/t/openstack/build/70eca8c380554e4bbbad5bfd269627d3 : SUCCESS in 4m 02s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/608cc780657343e59c1f6f9bace60ac2 : SUCCESS in 5m 02s\n- tooz-tox-py27-postgresql https://zuul.opendev.org/t/openstack/build/f916521d56fa4ff59493f1b708330609 : SUCCESS in 4m 44s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/306b531be9c24e069d800c4d1716aeea : SUCCESS in 4m 43s\n- tooz-tox-py27-mysql https://zuul.opendev.org/t/openstack/build/51c03a7fd76b42c48f7da7a7a01d49bc : SUCCESS in 4m 55s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/43ceb06c12304fbc9ea8a2beb607dcb3 : SUCCESS in 4m 51s\n- tooz-tox-py27-consul https://zuul.opendev.org/t/openstack/build/2f70f8718b344197aaa05a06af7a8598 : SUCCESS in 5m 10s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/41b59d8d900740a38aed1cfbbd6c5ea5 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d7010b985ee5c883f0a45bc94a409069ae42abd4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 16:32:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"462c30cc07e3ac04b84b153162334a2802f59efe","revisions":{"7451cf31fd8113ca04568e7a933560be0004eb85":{"kind":"REWORK","_number":1,"created":"2022-07-04 16:54:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/848652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/848652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/848652/1"}}},"commit":{"parents":[{"commit":"6100806dca26fe95899b044fdf71d69868cd78c2","subject":"Cap protobuf \u003c 4.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6100806dca26fe95899b044fdf71d69868cd78c2"}]}],"author":{"name":"garenchan","email":"garenchan23@gmail.com","date":"2019-04-03 13:58:56.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 16:54:11.000000000","tz":60},"subject":"Avoid redis lock\u0027s expire_time exceeding timeout.","message":"Avoid redis lock\u0027s expire_time exceeding timeout.\n\nAfter n heartbeats, the expireation time of the lock key becomes\n(1 + 0.5n) * timeout! Therefore, when the lock holder performs a\nlong task and exits abnormally, the lock will not be automatically\nreleased in time.\n\nChange-Id: I2437071bb575e95083b0d57a6a0a33aaf8356113\n(cherry picked from commit 20bda6ed999e71823b6eac90f01d1a29898b63b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7451cf31fd8113ca04568e7a933560be0004eb85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7451cf31fd8113ca04568e7a933560be0004eb85"}]},"branch":"refs/heads/stable/train"},"fc77e72f2bd6a94d245afd99a3fa37e4baaae4b3":{"kind":"REWORK","_number":2,"created":"2022-07-04 17:09:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/848652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/848652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/848652/2"}}},"commit":{"parents":[{"commit":"6100806dca26fe95899b044fdf71d69868cd78c2","subject":"Cap protobuf \u003c 4.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/6100806dca26fe95899b044fdf71d69868cd78c2"}]}],"author":{"name":"garenchan","email":"garenchan23@gmail.com","date":"2019-04-03 13:58:56.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-04 17:05:32.000000000","tz":60},"subject":"Avoid redis lock\u0027s expire_time exceeding timeout.","message":"Avoid redis lock\u0027s expire_time exceeding timeout.\n\nAfter n heartbeats, the expireation time of the lock key becomes\n(1 + 0.5n) * timeout! Therefore, when the lock holder performs a\nlong task and exits abnormally, the lock will not be automatically\nreleased in time.\n\nChanges:\n  setup.cfg\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): We have to remove the changes to requirements in\nsetup.cfg since it\u0027s not possible to raise the minimum version of a\nrequirement on a stable branch. Instead we check to see if the attribute\nadded in the new version is present and use it only if so.\n\nChange-Id: I2437071bb575e95083b0d57a6a0a33aaf8356113\n(cherry picked from commit 20bda6ed999e71823b6eac90f01d1a29898b63b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fc77e72f2bd6a94d245afd99a3fa37e4baaae4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fc77e72f2bd6a94d245afd99a3fa37e4baaae4b3"}]},"branch":"refs/heads/stable/train"},"8e74fbb096012a98eab4dc0fefc29a4e74a8e831":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-07-05 10:00:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/848652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/848652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/848652/3"}}},"commit":{"parents":[{"commit":"8567cc924cec318144115b563155ae6cf6125254","subject":"TODO","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8567cc924cec318144115b563155ae6cf6125254"}]}],"author":{"name":"garenchan","email":"garenchan23@gmail.com","date":"2019-04-03 13:58:56.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 09:59:47.000000000","tz":60},"subject":"Avoid redis lock\u0027s expire_time exceeding timeout.","message":"Avoid redis lock\u0027s expire_time exceeding timeout.\n\nAfter n heartbeats, the expireation time of the lock key becomes\n(1 + 0.5n) * timeout! Therefore, when the lock holder performs a\nlong task and exits abnormally, the lock will not be automatically\nreleased in time.\n\nChanges:\n  setup.cfg\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): We have to remove the changes to requirements in\nsetup.cfg since it\u0027s not possible to raise the minimum version of a\nrequirement on a stable branch. Instead we check to see if the attribute\nadded in the new version is present and use it only if so.\n\nChange-Id: I2437071bb575e95083b0d57a6a0a33aaf8356113\n(cherry picked from commit 20bda6ed999e71823b6eac90f01d1a29898b63b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8e74fbb096012a98eab4dc0fefc29a4e74a8e831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8e74fbb096012a98eab4dc0fefc29a4e74a8e831"}]},"branch":"refs/heads/stable/train"},"462c30cc07e3ac04b84b153162334a2802f59efe":{"kind":"NO_CHANGE","_number":4,"created":"2022-07-05 10:01:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/52/848652/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/848652/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/848652/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/848652/4"}}},"commit":{"parents":[{"commit":"ad58cba1798397fca1e7e698f07fad585d50d35d","subject":"Cap protobuf \u003c 4.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ad58cba1798397fca1e7e698f07fad585d50d35d"}]}],"author":{"name":"garenchan","email":"garenchan23@gmail.com","date":"2019-04-03 13:58:56.000000000","tz":480},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-05 10:01:06.000000000","tz":60},"subject":"Avoid redis lock\u0027s expire_time exceeding timeout.","message":"Avoid redis lock\u0027s expire_time exceeding timeout.\n\nAfter n heartbeats, the expireation time of the lock key becomes\n(1 + 0.5n) * timeout! Therefore, when the lock holder performs a\nlong task and exits abnormally, the lock will not be automatically\nreleased in time.\n\nChanges:\n  setup.cfg\n  tooz/drivers/redis.py\n\nNOTE(stephenfin): We have to remove the changes to requirements in\nsetup.cfg since it\u0027s not possible to raise the minimum version of a\nrequirement on a stable branch. Instead we check to see if the attribute\nadded in the new version is present and use it only if so.\n\nChange-Id: I2437071bb575e95083b0d57a6a0a33aaf8356113\n(cherry picked from commit 20bda6ed999e71823b6eac90f01d1a29898b63b2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/462c30cc07e3ac04b84b153162334a2802f59efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/462c30cc07e3ac04b84b153162334a2802f59efe"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
