)]}'
{"id":"openstack%2Ftooz~854585","triplet_id":"openstack%2Ftooz~master~I4f7b8da623df0ba76ac6dbb2a9a1ffaecfe19590","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-08-29 05:31:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4f7b8da623df0ba76ac6dbb2a9a1ffaecfe19590","subject":"remove unicode prefix from code","status":"ABANDONED","created":"2022-08-29 00:39:19.000000000","updated":"2022-08-29 05:31:55.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"01591d547b1d39e4bac069c0d952e4843364aa11","_number":854585,"virtual_id_number":854585,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-29 02:18:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-08-29 01:21:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-29 01:21:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2022-08-29 02:18:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8849f08e30b6c3cd74ac70445241ae4906aed0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-29 00:39:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfff0173728276b98e89312f6effc7b84d690201","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2022-08-29 01:21:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf9903d9a3f8c7545c4616f0860d9b25a0bcbde9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 02:18:27.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0bb355065494933a32c1c5a1b03fbb2\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/148b3d58497b445192a4dfe65067ca21 : SUCCESS in 1h 37m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2e3e9092623446797f9e01dd2bf8bd2 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e675097f1664f7b8903f23d38275592 : FAILURE in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1ae9f9ee00d4f1fb6a61c13bd70b433 : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9a2218d37c442e69fe3bdf4f5132f22 : SUCCESS in 8m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/748a627672fc4475adeb13256c40e946 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b87584f0c294d84a72b1ce608e62d62 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8beccc944094ed28ae461067fd0cf1b : RETRY_LIMIT in 3m 41s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/77bb91a1fe354a34be3f6d95beed16b2 : SUCCESS in 7m 09s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/600bb55029e540e292cb8792548f622d : SUCCESS in 5m 22s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/bdaef420b747427f87d1a372218c8863 : SUCCESS in 4m 03s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/a06076ae07944e54882339e86e145a0f : SUCCESS in 7m 28s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b4bedbf665d84b458a7098f521fabe36 : SUCCESS in 7m 27s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/c4ce9d79843d41a1a5a4ed989be33159 : SUCCESS in 4m 23s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a6d2c62a99244a8eb887b9f9ac95381d : SUCCESS in 7m 26s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/c518ebfc97924c799d5f248472a699f6 : SUCCESS in 5m 17s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9779e52ed16f4f958849209e5f37690e : SUCCESS in 7m 58s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/66e42f089b2b46eebc9907e0fb403cad : SUCCESS in 4m 23s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/9186d51cbaa94732a53a593591454a64 : SUCCESS in 8m 08s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/6be784ec59cd4954bb49a985c29fea76 : SUCCESS in 5m 15s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/8956c2d5f8dd4cda9489d334f9a0a89c : SUCCESS in 7m 31s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/6af148d31400476da4b9374830527b52 : SUCCESS in 4m 27s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c6c2131e7866485f9f135c6a087a5139 : SUCCESS in 8m 55s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/64acb2e29df84432a44d38658bf88a19 : SUCCESS in 9m 03s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/f3309f10ed7148dd89e21369be7f402e : SUCCESS in 8m 19s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/bfc18b4873c44d85a0d65e18db500700 : SUCCESS in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3401f8424d39a02a9d338d02dbb77a0c10bd64e6","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-29 02:21:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87afdacaa4a898f6ce13993dac52723a74050217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 04:25:51.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/98eaf895d8ea4726ab246e455257fc61\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/383f7b54db964f7d96624611471e6098 : SUCCESS in 1h 56m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd86b46f49f0407987273fc7f9eed058 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c01f2293c2644b89a1b888e906c37eb : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88484412fcb1443eadfbb43cc5dccc69 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/294af5f9ebff4367995c6fe734f219e2 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0abc6729dcf141b5900f4444646b36b0 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08cee6b6409c4e8fb5ca8b4afcf06253 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9020d63a4de5417f9cc29b20d6156d06 : RETRY_LIMIT in 3m 37s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/b0a55ab8092343d8b5592886d6f34deb : SUCCESS in 4m 12s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/68f337e45f054778970ae50368329b1c : SUCCESS in 4m 55s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/f6bb78f6699b4b089b75ea33adaed5c7 : SUCCESS in 5m 34s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/6acd8d578af04abcb3a02b5e2c64c91e : SUCCESS in 5m 36s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/4249c751796049789975f68a1ecf402b : SUCCESS in 5m 21s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/6f86e2113def4d2b8ec7385539bc23d0 : SUCCESS in 5m 32s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/45d2e87d3b3040a4a6eef5e72be9ccc7 : SUCCESS in 5m 26s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/07b27622cc1344f1952b662dbc61f970 : SUCCESS in 5m 24s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/ef966410392d4cea95cac2351b4001ca : SUCCESS in 5m 20s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/5a5d00720fe64cf08d74c9b28c434cc8 : SUCCESS in 5m 21s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/2ad6b231c0b6469ea902fdbf52880575 : SUCCESS in 5m 19s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/f731113c029b4b0ebd1d786ffd2738c9 : SUCCESS in 4m 41s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/de3adc097a414975b70f08a3f298bb77 : SUCCESS in 5m 43s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/953b60d8ab91431a9d379329f8c246c9 : SUCCESS in 5m 05s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bba5bf3169bc4a34a5b407368f85f370 : SUCCESS in 5m 30s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/c0a47ef79df14c1d8a05925d1bda86a5 : SUCCESS in 5m 43s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/99fec0d7d25640f69e918309b7fa432d : SUCCESS in 5m 24s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/96d48c4fcbc8447d86c1b67308827c14 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":1},{"id":"01591d547b1d39e4bac069c0d952e4843364aa11","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-29 05:31:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18ce7a43c722216cbaef1009cbbd829c7269d716","revisions":{"18ce7a43c722216cbaef1009cbbd829c7269d716":{"kind":"REWORK","_number":1,"created":"2022-08-29 00:39:19.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/85/854585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/85/854585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/854585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/854585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/85/854585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/85/854585/1"}}},"commit":{"parents":[{"commit":"27e5e506eb8ed6b17c1effcf5fbd282a50c97902","subject":"remove unicode prefix from code","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/27e5e506eb8ed6b17c1effcf5fbd282a50c97902"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-29 00:35:56.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-29 00:36:08.000000000","tz":480},"subject":"remove unicode prefix from code","message":"remove unicode prefix from code\n\nChange-Id: I4f7b8da623df0ba76ac6dbb2a9a1ffaecfe19590\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/18ce7a43c722216cbaef1009cbbd829c7269d716"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/18ce7a43c722216cbaef1009cbbd829c7269d716"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
