)]}'
{"id":"openstack%2Ftooz~859544","triplet_id":"openstack%2Ftooz~master~I89146dcf63cf984410f29b31ac2492f5ccaf3468","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-08-11 09:36:23.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I89146dcf63cf984410f29b31ac2492f5ccaf3468","subject":"Change StrictRedis usage to Redis","status":"MERGED","created":"2022-09-28 10:33:08.000000000","updated":"2023-08-11 11:11:35.000000000","submitted":"2023-08-11 11:10:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"859544","meta_rev_id":"03cd8903e97081f9b549c9d17d44db4de99ffb15","_number":859544,"virtual_id_number":859544,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-11 11:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-04-11 05:43:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-08-11 09:36:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2023-08-11 09:36:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-28 12:28:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-09 09:26:28.000000000","updated_by":{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-11 05:43:41.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"e8bd57eb632b64ee5375a20394f91c22d45f3745","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-28 10:33:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e775a93473d38a09518961fa8f56c4f9b8da3e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 12:28:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5a58db035734e75bbd18029b9609e0c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dda83a73c7c41dd9031f0e3bfd58988 : SUCCESS in 1h 49m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff14125f89e041709a0345b83f982c68 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea342efc2ea422b95bf2f1a895ff49f : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/738a4647cf874437b9f94365b4ec8a25 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d268ab8bba9482ba727157ce2c34de5 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0543e59a610c41f0802062fc822063ba : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c481c5cce524c8a83b735eb8a279fa4 : RETRY_LIMIT in 4m 38s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/3aae90c583114a8a95b3740890d7810f : SUCCESS in 5m 15s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/e5eff0baed7e46e885f48df83c14e7f4 : SUCCESS in 5m 23s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/efe8ce413d4f4197b5bb329579ede905 : SUCCESS in 4m 04s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/f26be91dc24b4803a9c39462fe3d7ce1 : SUCCESS in 5m 00s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/fa880195626540bd8020e8cda97c4d35 : SUCCESS in 5m 07s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/be6ef9f822644510bf534e15c11a9927 : SUCCESS in 5m 32s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/244a9fdacdcd4478a92d29ef10d40689 : SUCCESS in 5m 18s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/105940e9a84b40bbb0fb9c850d20dcf6 : SUCCESS in 5m 21s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/a2112b4897aa413a89639defb18bdc82 : SUCCESS in 5m 09s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/3da4e2d06a9d4c55b83b97478e188712 : SUCCESS in 4m 22s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/06276259cb9846a2aa7f76b902cf7282 : SUCCESS in 5m 05s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/5f112ada88fb410c952861ca871d6592 : SUCCESS in 5m 24s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c9ce1451fffa43c7a2a6b3cbfaa9760f : SUCCESS in 5m 05s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/7a6b4cdfd69d489b90ff5ce80ac359b8 : SUCCESS in 6m 07s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/be238c20e43f47f89dd7fab893006846 : SUCCESS in 6m 01s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/e9709599022940d0978947cebeb227ff : SUCCESS in 6m 32s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/bcf392cfcf0e4270aff5a2cce64ffa17 : SUCCESS in 4m 18s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/dd62fb56f12341afb02512f59420f3d4 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":1},{"id":"9e558a64f4e9d0b45bd5817f70a960f1e9644cff","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-28 17:15:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cc8702c331bba828bc666d84b175945e7dc644b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 18:53:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec740bfdd3164207a5c2b84685f970d4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f70fe8854df44d0eb3709dab4d90c123 : SUCCESS in 1h 31m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00fe3e21dbcd4ded8468f7cd6b0363b8 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f91a327be6dc4ac7a93e8a89ae9da5f9 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e96b74ac354436ab7be63b3d5d41f42 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aeb15de60ff3496896f11dbf7dd57f26 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b52821ac79447f6b40775f5e28dbe94 : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88773446c0db4823bb2baf4ad078ebf1 : RETRY_LIMIT in 3m 34s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/95ac8cdee0164123aa51d0b0d1502c53 : SUCCESS in 5m 43s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/b4c7a8a617e74503a5c8728108b8c6f4 : SUCCESS in 5m 02s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/599e6b87c8c547e6bc3be5f49fd41add : SUCCESS in 7m 29s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/df86270c5ef146d781b021904d8977a2 : SUCCESS in 6m 14s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/89ec7d0f11224b0c824a14b2c50d667e : SUCCESS in 5m 01s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/b925841dd6cc4424abbfd9ba2c8f6e14 : SUCCESS in 6m 30s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/5dc02a39211842b09fe6fffa70dbc704 : SUCCESS in 5m 02s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/5dec7395926e40b7849d5d38b1ef25bc : SUCCESS in 6m 21s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9984f761b2b04b6aacdec9521a59860e : SUCCESS in 4m 56s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/fe9e6f8192904b78a9714aae2ac7b09d : SUCCESS in 6m 17s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/b3cb7239bbb7488d88105f2706644da5 : SUCCESS in 7m 41s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/b0dc47131f9b4eceabc7536214092bba : SUCCESS in 5m 19s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/202d3275bf0643c6b723ca623faa69ca : SUCCESS in 5m 04s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/701b3d0bd19048f1b4443fd61257c24d : SUCCESS in 6m 31s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/9c9aa0ee5f104bcb9594af595046441c : SUCCESS in 7m 34s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/555fcdddff0b4d8db89730d846cc4293 : SUCCESS in 5m 51s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/a9b28fdc201f4295b05be9ca65beaaf6 : SUCCESS in 5m 10s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/42047b300b944e1f936420a2318907da : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":1},{"id":"51cceaa559ac9b727089ed64cc3c18e3260b00d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-09-28 20:27:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"d21313443f3a53a89200aae609ddec5d47291fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 22:07:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17298717f6d24785a99050501d7cad2c\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/557519e0802a4321995cc7b6cad8c476 : SUCCESS in 1h 37m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95fb78da4d4d463f9d94e2acb3c60d54 : FAILURE in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ea0c24f614447fb8314cc310244d81 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be756f4e56474bff8a29b4793cf1ec17 : FAILURE in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b1fe1e24f418460fbe8fb31c763356e8 : FAILURE in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b51e7aca53e416186c319609c9b9e0c : FAILURE in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7e9ba5123174d28b74cd946731d8c6b : SUCCESS in 6m 08s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/6544e927a911465198a22304451ba7c3 : FAILURE in 3m 47s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/e3bd50fef30041d58d1f3d68deb46130 : FAILURE in 5m 19s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/c96fd0ff80b54d5eb972cf3a169f5b8e : FAILURE in 5m 17s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/8d6c87126a99470e9579b4af8cbde595 : FAILURE in 5m 29s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/38881ae1e77d46d5a9b2ce8c55bbc421 : FAILURE in 4m 49s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/61ed612f029a44109fe2010e22a4348e : FAILURE in 5m 08s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/89a07278ae024ca6ac432e17c3463ea7 : FAILURE in 5m 52s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/9989f876e7a6469ba7858e7a1c2a534f : FAILURE in 7m 05s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/d37a78ef355d4a77af34b5daa23205ea : FAILURE in 4m 49s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/9796b0d9f2e841399c25e8e3ef28af3d : FAILURE in 5m 30s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/7d0d01ee8725462b8389eb80857186c7 : FAILURE in 5m 41s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/5549cb17c3984364ba0b33adb69bbd43 : FAILURE in 6m 39s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/6d06545addda473ab9ef5a901cdfff09 : FAILURE in 4m 50s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/2f08fb70b41f451cab5525c39794627b : FAILURE in 6m 06s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0b4b34ff02ff464590231a8c37e3ba82 : FAILURE in 6m 35s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/e74b3e3f7442452c847a16dd4c960ac5 : FAILURE in 4m 51s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/68617d2ae9864e35a955432f1a466a9d : FAILURE in 6m 02s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/0a05d40740db483da8ff20a995857a1a : FAILURE in 6m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3e07635234a765d4a6289cd46050e2d9e9bf8e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-17 13:45:04.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"26d6a4eba219c8e0c22d3ddea6cafed3b99f7d74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-17 15:18:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02734fb36ccc482dab9007c9bdda78bc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2e60fe83e4004914b4c307d5fa51295a : SUCCESS in 1h 32m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a788639035f494392f9c3ab8f867d3a : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e831d03ffdb9422ca217613117cdd113 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1907ff7cf6b147c19c76b80f66396497 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99dec5b620e84bfbb35284ff59c3c1df : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389a3bd83b034313bcca9725b296b258 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c713163e1ca7481f9ca9f735e56d1eb0 : RETRY_LIMIT in 3m 10s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/7599e24d93344f22a5fcb00d25322494 : SUCCESS in 5m 11s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/76b6b21e845549ccafcbcc8e97926d4e : SUCCESS in 5m 32s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/9187deefe52e42cbb58577c7ac16f50c : SUCCESS in 5m 10s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/fa6e1f2f71904cb4b6577a5bc2a51485 : SUCCESS in 5m 38s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/96a702b533d244a6974da045fc965f0b : SUCCESS in 5m 16s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/453dab9e778a42acab2da1ab52b29c13 : SUCCESS in 5m 16s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/02a1b27ae97a4c1bbb99b2fb70f3157f : SUCCESS in 5m 22s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/f35ca286ee9c4c3e9a1d9e4c5e96e24a : SUCCESS in 4m 54s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7bc31bfae2e446b98d1319c6288f80c5 : SUCCESS in 5m 01s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/b5dafe59f8764516b2a5432bc38ccf50 : SUCCESS in 5m 22s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/20254dde101e41c597cf30edfa4a2dd9 : SUCCESS in 5m 23s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/b0130772d1d64b929fac4bf31b4ca43b : SUCCESS in 5m 20s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7209514432fc4494a111de847f728a3f : SUCCESS in 4m 49s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/9169bfce5f524b9ba5d073355b18a560 : SUCCESS in 5m 38s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/cb47eb6805394136b8e82c4ffe75380d : SUCCESS in 5m 16s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/d7f9271da5914bb1802af66a4e51e302 : SUCCESS in 4m 55s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/421ac8c00b594650a9ebfd16923369a8 : SUCCESS in 5m 10s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/d6c73a2eaff0485ba815a7bc27a1fbb5 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"f42f85222f12f94d7fe9c53f4cbcc15d550b35b1","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-09 09:26:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cebef68653ef8ac752aab698011af3b12d2a1c94","author":{"_account_id":22474,"name":"OPNFV CI","email":"opnfvci@gmail.com","username":"opnfvci","tags":["SERVICE_USER"]},"date":"2022-12-09 09:26:28.000000000","message":"Patch Set 3:\n\nBuild Started https://build.opnfv.org/ci/job/releng-tox/22/","accounts_in_message":[],"_revision_number":3},{"id":"243979b5a256832c2051377dbb318c4b9dcd9bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 10:44:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47e860ce68834d38be07b876621a49ac\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/847f368149854edcbc8f0cc33516dba6 : SUCCESS in 1h 10m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b86045ad11741c3809cf7a41e5bc87d : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9e86f24830b487e87cfc5770795431c : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5fb4d50c552474ebd97e6ebcf70b027 : FAILURE in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ed7854eee334aa0a33de110742494a5 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0460a39b14d747088f5974d0efc26229 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4ca36e9d0384da3a61ac41ae075a6bc : RETRY_LIMIT in 3m 03s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/2c84ceedc3bd4fb8a5b1a1b6f2115983 : SUCCESS in 4m 48s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/c2f91a491c844b3f8f7210cf99cf13a1 : SUCCESS in 4m 47s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/efe6e8795eef447294cebe9f452377ae : SUCCESS in 6m 01s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/877850c1f5dc434eafcbc4deacf9a3f9 : SUCCESS in 5m 19s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c5c9a3611b5643b59e26a0a00dab5e6e : SUCCESS in 4m 22s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/b41b48a757d84a90b4e0da9e40a39d30 : SUCCESS in 4m 55s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/87d32698eeb44f0e831b5495d2bbfa2b : SUCCESS in 5m 07s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/28851fac01274c44bd2b4cd23aa1707d : SUCCESS in 4m 53s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/30297da7f59b4350b7d1b0aa2b4d8f8a : SUCCESS in 4m 14s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/51448a1edc52456fa542f5d57194abd2 : SUCCESS in 4m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/a509e8ed63244e69a6ec9d2b3a78917f : SUCCESS in 5m 24s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/9d0d366524c24d3887f66ec99e7462d8 : SUCCESS in 3m 59s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/6865a7d93ebd41c89e3020c82b7a7f97 : SUCCESS in 4m 55s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/b4218436987a4fdbafcbdc56b07cebef : SUCCESS in 4m 25s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/e0d8a5a3fe4c490984aa2575170e7530 : SUCCESS in 4m 46s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/c926a2893b014684a417c7c720592db3 : SUCCESS in 5m 45s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8f4cffd3cc594d17b802f45fc32e3fc9 : SUCCESS in 4m 46s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/bf5c62536daa41178aca5557e8aa623a : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":3},{"id":"c54d3624bded0da3f33e35ac222975a2904e0d11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-09 15:09:13.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"c6797b7294c7dc14117f0c65a14cbf509bb9029c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 16:49:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1789aa5c7e1417781099931de5b29fa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfaf7d39131440c48f96b285a2c4cda7 : SUCCESS in 1h 38m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78a1ac0f1ea94d29bb321aadf4563c9c : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a54c1ee3ce4f75ac7bd7fbd3c41edf : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a0b017ae3c549dbbf07bdf4957de1e0 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/366bababf7fe4e7dbcff0b40f3149d3c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa6d93c09fd4703a349bae1e77e91b9 : SUCCESS in 5m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/822af7197981408a9dbd2d98b87d3d5f : SUCCESS in 4m 57s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/7298f122fc5443218b80b816bc844b3a : SUCCESS in 5m 16s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/ce131d51199e4ed98d5d356b475e19d3 : SUCCESS in 4m 23s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/b71ebb9afa2a4e4a9762fc9c2d2696de : SUCCESS in 4m 46s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/03e36a4b6d274040b1d1800b133ceb8b : SUCCESS in 5m 48s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9f020daee0b04fdc94d17bdb5cbc2e3a : SUCCESS in 4m 02s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/a9c28a0edb8a4fada8c5884f651127ac : SUCCESS in 4m 17s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/bf8821cc5ce3486fa9f8c75df86fd4e8 : SUCCESS in 4m 44s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/769ec8469cb6470781359661a1406136 : SUCCESS in 4m 12s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/0aa3d3a352d5469e891302c6f10db371 : SUCCESS in 3m 59s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/c399fccc56d04f7c84285aa152d68dd3 : SUCCESS in 4m 54s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/3bf34fbb314b45a68decb5969f5a209d : SUCCESS in 5m 37s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/ed314d9bffba4b41baec69b0d603ba71 : SUCCESS in 4m 27s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5e196adf2edb4c48a73b1cfb38ec5e0f : SUCCESS in 4m 46s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/a2160fd1212643a38e9a4e44c2c38e5c : SUCCESS in 4m 13s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/2f391fd2f5c749c3a9e7bb2aa96fcb8d : SUCCESS in 4m 55s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/9d17b4d1de904636b56203919c21a649 : SUCCESS in 4m 10s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/1a93fb41f3ff43329ab18fb2f9c00e17 : SUCCESS in 5m 39s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/8c20a6362b4f4069acfae49484556c5e : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":4},{"id":"7eeb785430cd245617f193c40d75630496632dea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-16 06:59:24.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"43671b1355083b9efe2c814504c81e0d4316327c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-16 08:21:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8282db42ad8b4607aa43eccb20f60c8b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/980f433d807944a789b5f0ad1cb0f418 : SUCCESS in 1h 19m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5554790ec5548ccb4a492030a5e13c4 : FAILURE in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c8a650e234d40cc8f344528b8dc89e9 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ea45d27daa0435db35f19d4827c41cb : FAILURE in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efdccfc5cc4e467a8588035092565d59 : FAILURE in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c321451aac864013a01915047662d710 : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e503850945440679bcd15b865bbad4f : SUCCESS in 4m 56s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/de537dff39f343b3bad3674227d21f95 : FAILURE in 3m 36s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/652661ab95e8485b918d0151f5ce76a4 : FAILURE in 5m 10s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4ca36bc309884be498996713e87003ea : FAILURE in 5m 42s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/9d3877be9b994dc68b87a32700e472e9 : FAILURE in 4m 58s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/8d10ddcfd40740829566816802d627ea : FAILURE in 4m 02s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/aa94cf5985b84a4c9baa9c534c10d0bb : FAILURE in 5m 16s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f06507b992f94aaeb9bed6c08c26a668 : FAILURE in 3m 38s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/a4155f284c904b52a932374d2a151ea8 : FAILURE in 4m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/51f58dbbe6a84bedb91ffee9b043a538 : FAILURE in 3m 55s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/429a421203694b8098aeb38470bd97e5 : FAILURE in 5m 36s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/39c0644b3df045e9bff948c223e1dcb2 : FAILURE in 4m 25s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/f919b5c393b049c9b88f536b73109639 : FAILURE in 4m 19s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5d45633e3530464c9ea93f0c42672a90 : FAILURE in 4m 00s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/dd2ca9608e0f4e5c8fda24428d602762 : FAILURE in 4m 05s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/940326859b794a298c61f6e646541e39 : FAILURE in 4m 07s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/fd4e6ac16f5942bd8abfe939e6fae231 : FAILURE in 3m 50s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/b3ce47581bc643b599184cfe87a365be : FAILURE in 4m 24s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/2993b851948c48d684bc08298e184791 : FAILURE in 4m 15s","accounts_in_message":[],"_revision_number":5},{"id":"d856c3be9b50175b468d12ab9212c9a2483aad72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-18 09:55:44.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"c8c8551fb212b81a9a966a54ef84799b2c2453fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 11:36:32.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dc87de031454476a24085d36143c921\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9ef4819077840849ff6369d11ac3a56 : SUCCESS in 1h 32m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b2d36c778a444a8808c3de51ba2913e : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc45bbf5ce6d4c2281274c814d0bf546 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cdabc011fcd4a4eacb8b8a112b2246d : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdd74081d565491889add4cfcc0e0ae0 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ac919e8f6304327b8b40a5def754b3d : SUCCESS in 4m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/562da906aec745f393ad67c5554e9095 : SUCCESS in 3m 37s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/27d474b0ae104537ac496f02937b627a : SUCCESS in 4m 55s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/0cb7f30612f746abbaaca380f0977830 : SUCCESS in 4m 37s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/73e773146e7b4b0a80b6acb022c6d362 : SUCCESS in 4m 24s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/0f7921e9f41b4381a77d1b02d2a8764c : SUCCESS in 5m 07s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/37a69fa560094eb6b63331ebf05b3484 : SUCCESS in 5m 14s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/53d29c031bf54f40849f340513415e58 : SUCCESS in 4m 15s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/fc5018fa15cc4a53a16f10e396d4f51d : SUCCESS in 4m 27s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/b9360c8f4019441fa785d2d929274303 : SUCCESS in 4m 14s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/75d263ccd5564be0bc42d01154d4d444 : SUCCESS in 4m 40s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/e4a6bae05cb14fcf835b0f15e5bf04ba : SUCCESS in 4m 22s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/de13178ed28a4ee89591b0ee3b88ebc0 : SUCCESS in 4m 35s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/09b67b78a78d4d4eb5cbad7ea2fbd6fb : SUCCESS in 4m 27s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/6220385c48e94ffda4d3fb745c0071c3 : SUCCESS in 4m 05s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/66ee018e1e0449928bfb786dfa6bb59b : SUCCESS in 4m 43s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/1e01ea0b995d4a98b5adb508aae90b70 : SUCCESS in 4m 19s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/d43d20d6065c41b59333a03d4b4bac53 : SUCCESS in 4m 18s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/b71f8dfa65ff4663a1e3425a9255e187 : SUCCESS in 4m 28s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a0d424d847ea4ceab6f590734461e375 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":6},{"id":"3d1b82792d434fd26d67147292b8553b84a7514e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-11 05:43:41.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"32347c9f2faee048b23c2a7e1eb4e7e4f641f524","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-08-11 09:36:23.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"aeadcc9d51a42c93dae5f6a7c3790cf8d6927e2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 09:36:51.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"eddf6851577375f7dc4d8fdde98d0fa5890d94f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 11:10:40.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c8c7a126db74ffa8cb80f1a707558ca\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/66e1f1ad884f4ebcb174f5c964600aae : SUCCESS in 1h 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f21016d2678e4ad4a415fa1976548ecf : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00258e6a19bb42119ba532ac9e78c128 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7674bcd26b3f498c9e4669ba910f4aeb : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1509dc9f20ac470ea7d4805fcb784ce8 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cb7812b09fa485c93d20f408c756a0a : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a9a2ca36881411aa96f29393a0bd0c1 : SUCCESS in 6m 13s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/d5cde37d00fc44c79b89afd72e6800a5 : SUCCESS in 4m 38s\n- tooz-tox-py310-etcd https://zuul.opendev.org/t/openstack/build/2a07c233c1c3481d9b11cab40f928aaf : SUCCESS in 4m 48s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d4ebe5e447474e958725ebc4353ed8e7 : SUCCESS in 5m 11s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/ff1fd55b2c8249faa24e8e62887eaec5 : SUCCESS in 5m 04s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/66f529b8e1d541fa82920274c30df471 : SUCCESS in 4m 23s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/31575d2d5949435d82570e3f80e7f81d : SUCCESS in 4m 21s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/23e646239cb3404ab1ba335f509955b9 : SUCCESS in 5m 10s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/71bde9f52dcc4043afc3697603607fde : SUCCESS in 4m 39s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b3b949024dc846509a6359b7f8ff2b24 : SUCCESS in 4m 27s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/43e9251d87654cc3862f73f04b6ff577 : SUCCESS in 5m 51s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/26622132530943f0b134de07514c8a22 : SUCCESS in 4m 35s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/caad93dd27d545e69e999d53725045a0 : SUCCESS in 4m 00s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/13815301c85d408fa635bcdc60ce4800 : SUCCESS in 5m 56s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/91d1e4a827754a53b6953aeb91f87968 : SUCCESS in 4m 21s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/3e107e09d26e4ebd87fdb76c8d27a2b6 : SUCCESS in 4m 41s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/696715c7f080436aa005c1c7aec7a0e2 : SUCCESS in 5m 32s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/ccfe034785624992bd5d7a69bcd03e7f : SUCCESS in 4m 45s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/eb0bae73c2724ac1bad280d47b7bcc74 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":6},{"id":"afa0538272c097d54708cd8f56bcdc7d003aa94a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 11:10:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"03cd8903e97081f9b549c9d17d44db4de99ffb15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 11:11:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/219faa5631004c138f3b908729916747\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825d858a0ee94d46a00411ae5477b6fe : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83df0a829357428cb3229ddd6926a32c : SUCCESS in 31s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"378566dab8d82f2c318fe568392e4083a0fa7c96","revisions":{"5014547511069b080598fdaae46216611b6ca96a":{"kind":"REWORK","_number":1,"created":"2022-09-28 10:33:08.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/859544/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/44/859544/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/44/859544/1"}}},"commit":{"parents":[{"commit":"cb0392ca6b91999c093f458da01c17bd4e4820f8","subject":"Merge \"Support etcd3gw api version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/cb0392ca6b91999c093f458da01c17bd4e4820f8"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 10:32:33.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 10:33:06.000000000","tz":120},"subject":"Change StrictRedis usage to Redis","message":"Change StrictRedis usage to Redis\n\nThe StrictRedis class is only an alias for\nRedis in \u003e\u003d 3.0.0\n\nChange-Id: I89146dcf63cf984410f29b31ac2492f5ccaf3468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5014547511069b080598fdaae46216611b6ca96a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/5014547511069b080598fdaae46216611b6ca96a"}]},"branch":"refs/heads/master"},"71b48eec11b6b8d8db54bf2fae3e26a9a91f889f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-09-28 20:27:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/859544/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/44/859544/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/44/859544/2"}}},"commit":{"parents":[{"commit":"67335453a975b91dca10982acf1f793fa0b94e52","subject":"Unpin protobuf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/67335453a975b91dca10982acf1f793fa0b94e52"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 10:32:33.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 20:27:01.000000000","tz":0},"subject":"Change StrictRedis usage to Redis","message":"Change StrictRedis usage to Redis\n\nThe StrictRedis class is only an alias for\nRedis in \u003e\u003d 3.0.0\n\nChange-Id: I89146dcf63cf984410f29b31ac2492f5ccaf3468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/71b48eec11b6b8d8db54bf2fae3e26a9a91f889f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/71b48eec11b6b8d8db54bf2fae3e26a9a91f889f"}]},"branch":"refs/heads/master","description":"Rebase"},"06dca4e23e19b3c18c39842c6ae0481b70e8c511":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-11-17 13:45:04.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/859544/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/44/859544/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/44/859544/3"}}},"commit":{"parents":[{"commit":"0c145b4567d3b97334ce0a8a2c289403e77b5bbf","subject":"Add grouping support in etcd to compatibility matrix.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c145b4567d3b97334ce0a8a2c289403e77b5bbf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 10:32:33.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-11-17 13:45:04.000000000","tz":0},"subject":"Change StrictRedis usage to Redis","message":"Change StrictRedis usage to Redis\n\nThe StrictRedis class is only an alias for\nRedis in \u003e\u003d 3.0.0\n\nChange-Id: I89146dcf63cf984410f29b31ac2492f5ccaf3468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/06dca4e23e19b3c18c39842c6ae0481b70e8c511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/06dca4e23e19b3c18c39842c6ae0481b70e8c511"}]},"branch":"refs/heads/master","description":"Rebase"},"3cb4f6f87092178876b10c6fff8a7c9eec36fafc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-12-09 15:09:13.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/859544/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/44/859544/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/44/859544/4"}}},"commit":{"parents":[{"commit":"7d50893625bd88d31c181a1c1c6f730ed64c4b3d","subject":"Loosen protobuf version that\u0027s used for docs/renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7d50893625bd88d31c181a1c1c6f730ed64c4b3d"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 10:32:33.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-12-09 15:09:13.000000000","tz":0},"subject":"Change StrictRedis usage to Redis","message":"Change StrictRedis usage to Redis\n\nThe StrictRedis class is only an alias for\nRedis in \u003e\u003d 3.0.0\n\nChange-Id: I89146dcf63cf984410f29b31ac2492f5ccaf3468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3cb4f6f87092178876b10c6fff8a7c9eec36fafc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/3cb4f6f87092178876b10c6fff8a7c9eec36fafc"}]},"branch":"refs/heads/master","description":"Rebase"},"1a367496a2d6d2b2e3a8dc55b6348481f8ef136d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-01-16 06:59:24.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/859544/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/44/859544/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/44/859544/5"}}},"commit":{"parents":[{"commit":"f11666b66c668dd731b59369ad3a62a6d16536c6","subject":"Allow to pass ssl-related args for zookeeper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f11666b66c668dd731b59369ad3a62a6d16536c6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 10:32:33.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-01-16 06:59:24.000000000","tz":0},"subject":"Change StrictRedis usage to Redis","message":"Change StrictRedis usage to Redis\n\nThe StrictRedis class is only an alias for\nRedis in \u003e\u003d 3.0.0\n\nChange-Id: I89146dcf63cf984410f29b31ac2492f5ccaf3468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a367496a2d6d2b2e3a8dc55b6348481f8ef136d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a367496a2d6d2b2e3a8dc55b6348481f8ef136d"}]},"branch":"refs/heads/master","description":"Rebase"},"378566dab8d82f2c318fe568392e4083a0fa7c96":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-01-18 09:55:44.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/44/859544/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/44/859544/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/44/859544/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/44/859544/6"}}},"commit":{"parents":[{"commit":"7698c582f31f78a4fab558420b18ac88e38284a1","subject":"tox: set allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7698c582f31f78a4fab558420b18ac88e38284a1"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2022-09-28 10:32:33.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2023-01-18 09:55:44.000000000","tz":0},"subject":"Change StrictRedis usage to Redis","message":"Change StrictRedis usage to Redis\n\nThe StrictRedis class is only an alias for\nRedis in \u003e\u003d 3.0.0\n\nChange-Id: I89146dcf63cf984410f29b31ac2492f5ccaf3468\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/378566dab8d82f2c318fe568392e4083a0fa7c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/378566dab8d82f2c318fe568392e4083a0fa7c96"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
