)]}'
{"id":"openstack%2Ftooz~862929","triplet_id":"openstack%2Ftooz~master~I507301618c2bb32fa989db3e5634016c8e15f305","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"32755":{"account":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"last_update":"2022-11-16 08:39:35.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-11-16 08:39:35.000000000","reason":"Change was submitted"},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2022-10-30 20:37:34.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-16 08:39:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I507301618c2bb32fa989db3e5634016c8e15f305","subject":"Add grouping support in etcd to compatibility matrix.","status":"MERGED","created":"2022-10-28 15:28:22.000000000","updated":"2022-11-16 08:40:40.000000000","submitted":"2022-11-16 08:39:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862929","meta_rev_id":"e8a1329ac57343ef443a6c95d39678f1580d2d7b","_number":862929,"virtual_id_number":862929,"owner":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 08:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2022-11-16 07:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2022-10-30 20:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2022-11-16 07:27:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-28 15:30:20.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-10-28 15:30:20.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-10-28 16:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 08:29:38.000000000","updated_by":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-11-16 07:27:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"623cd8b9d7e16208d9db48ae8ffb1884d55c9ca2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"date":"2022-10-28 15:28:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9bf23fddaefc5472d1f63eeda5966f9740c7e47a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 16:51:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6de7bdbe6c84474bd59dd30038a2fed\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d95a3082dfe4652804a0f368352d2a9 : SUCCESS in 1h 22m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1792fae57584f41b3ba74d1bb2d70f4 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/227b03caa9d64a9b98d735eb05ce0e3f : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf7ad825eafd4360939ad878d8645b2a : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8708e55c17254ce2843478aff55ae7f8 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78661a6538fc40eba6e58344288ea0a2 : SUCCESS in 4m 58s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/a02314c406464d03b7e17847b29fea4e : SUCCESS in 5m 41s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/94352a7cbdfb478b9bf28d6824aadeb0 : SUCCESS in 4m 28s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7eca0002f3a9464aa73e316b1d1c9b63 : SUCCESS in 5m 30s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/a4c2c0868b144dc69a18d9b8e1ac62fd : SUCCESS in 6m 01s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/a61aaebebea64b98aa6f6d500a85df76 : SUCCESS in 4m 58s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/627393dccfe3427dbd8b671456139ded : SUCCESS in 5m 39s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/c0bf1c561b2148d9a065227756b692ee : SUCCESS in 4m 00s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/bf27ae1e36d54d8091c673155ac26b69 : SUCCESS in 4m 53s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/2d7d9dc7d7a144b78620c3e005b8107e : SUCCESS in 4m 10s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/d94c9e8c4fa64786bbe3a8560a5fa5e7 : SUCCESS in 4m 41s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/28b65127387243e59fcd21aac7116c74 : SUCCESS in 5m 03s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/71fc024bc5184231a6012945d216becd : SUCCESS in 4m 47s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7167c1de2cfe43929fbc6799d89dfbee : SUCCESS in 5m 55s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/55c1edf93d394bd2b6f8a3f2b886823f : SUCCESS in 4m 04s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c10fe08adab24fc0aafc9f780c074230 : SUCCESS in 5m 14s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/cd27b7b7443947f197bb8a241d29ddc5 : SUCCESS in 5m 14s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/1e9e3ca072384dbdb5d4ff96edb33a15 : SUCCESS in 5m 23s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/095482ae0f5f4df8b84324b6b3a7d06c : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a6b11b59be0cf4516bbeb07aa8ca5b1b17ce23a3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-10-30 20:37:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"059d11ec7cc14a6024aed051e66d3f972ecdff3b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-16 07:27:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ace12835706b3a2efeb3bf92be97cb1457dc957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 07:27:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70d230243882c0fb75e430701148ab57d048f21b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 08:39:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04a67020a7e148afb3670052f25c50a6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0eb2bbc7ded7469cbc2b233296cf3129 : SUCCESS in 1h 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9c029eb51e428a9c29f83e9040af9c : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/880ed042b5794a5683220f6550b24d3e : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ce17cd692c64290b47a8f005ec61f14 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d643eb8a04e4df3a4fce51636226704 : SUCCESS in 5m 01s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/27adc2e7eac04121b87097aebdedf759 : SUCCESS in 4m 23s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/93891446f412404aa1270ac7b6ebb61b : SUCCESS in 5m 03s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/30b1ac635f5f40eda1efd4c95f5e9d1c : SUCCESS in 4m 35s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/416e19e847e4432fb3c2752e6f205d42 : SUCCESS in 5m 16s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/02b3bc50252d41a19c038bfc89246d33 : SUCCESS in 4m 40s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/0ebeebff73a14486aa28cf3b3766f947 : SUCCESS in 4m 55s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/23e7fdbc333d4167930c9c0991aec3f5 : SUCCESS in 4m 15s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/5a3bbff6eaa740e98c06183d66cadf66 : SUCCESS in 4m 46s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/443d950dc7aa479489fea54561d7dd66 : SUCCESS in 4m 27s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/bc1bdc3f626242f58d0a79bcd5340530 : SUCCESS in 5m 38s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/cdabac3720cb40d0a8231c3676c4be58 : SUCCESS in 4m 26s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/83c79cb1bf2c43e28214f71327c8b543 : SUCCESS in 5m 37s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/d6c8fbd9aabe418ebfa2ab832fb453de : SUCCESS in 4m 43s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/efefdeb033dd412d9ab2559591457a74 : SUCCESS in 4m 05s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f2e895c8fcdc410b8f8c858baac8229e : SUCCESS in 5m 12s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/8728c7b5b5954fdca5b2c82b0295911a : SUCCESS in 4m 51s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8ac3e16b70fc413fbaeb93138edc2df2 : SUCCESS in 4m 46s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a409db8ce9ff4d3785694a0f74c3fe6a : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"c425a6238913efb8bc1085d61543fcfab5111534","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 08:39:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8a1329ac57343ef443a6c95d39678f1580d2d7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 08:40:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4daf3f2931d440dbbb8a1d62f5a86625\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/775ce27be94b423daceaee9de4f4f5b4 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c145b4567d3b97334ce0a8a2c289403e77b5bbf","revisions":{"0c145b4567d3b97334ce0a8a2c289403e77b5bbf":{"kind":"REWORK","_number":1,"created":"2022-10-28 15:28:22.000000000","uploader":{"_account_id":32755,"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","username":"frittentheke"},"ref":"refs/changes/29/862929/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/29/862929/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/862929/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/862929/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/862929/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/29/862929/1"}}},"commit":{"parents":[{"commit":"1a76dd61999f9341ed6e7fbf601c030a5ea85c0d","subject":"Merge \"Enable watch functionality for Etcd3Driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1a76dd61999f9341ed6e7fbf601c030a5ea85c0d"}]}],"author":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-10-28 14:27:03.000000000","tz":120},"committer":{"name":"Christian Rohmann","email":"christian.rohmann@inovex.de","date":"2022-10-28 15:28:21.000000000","tz":120},"subject":"Add grouping support in etcd to compatibility matrix.","message":"Add grouping support in etcd to compatibility matrix.\n\nSupport has been added with\nhttps://opendev.org/openstack/tooz/commit/6ab8c380c8d6a2e15611b225da7594e820cc773e\n\nCloses-Bug: 1995125\nChange-Id: I507301618c2bb32fa989db3e5634016c8e15f305\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c145b4567d3b97334ce0a8a2c289403e77b5bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c145b4567d3b97334ce0a8a2c289403e77b5bbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
