)]}'
{"id":"openstack%2Ftooz~865401","triplet_id":"openstack%2Ftooz~stable%2Fwallaby~Ic3f3d05ce7bb10a7d29815444dac00dea4f9c9d1","project":"openstack/tooz","branch":"stable/wallaby","topic":"move_flake8_local-stable/wallaby","attention_set":{},"removed_from_attention_set":{"32240":{"account":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"last_update":"2022-11-25 15:17:38.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-11-25 15:17:38.000000000","reason":"Change was submitted"},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-11-24 09:46:59.000000000","reason":"\u003cGERRIT_ACCOUNT_28522\u003e replied on the change","reason_account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}}},"hashtags":[],"change_id":"Ic3f3d05ce7bb10a7d29815444dac00dea4f9c9d1","subject":"Move flake8 as a pre-commit local target.","status":"MERGED","created":"2022-11-24 05:00:52.000000000","updated":"2022-11-25 15:18:45.000000000","submitted":"2022-11-25 15:17:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865401-move_flake8_local-stable/wallaby","cherry_pick_of_change":782457,"cherry_pick_of_patch_set":2,"meta_rev_id":"a63aa804997bbd192e4c26c343f65717d236f98a","_number":865401,"virtual_id_number":865401,"owner":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-25 15:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-11-25 13:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-11-24 09:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-24 19:48:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2022-11-25 13:48:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"reviewers":{"REVIEWER":[{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 05:00:52.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2022-11-24 06:55:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-24 06:56:28.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-11-24 06:56:28.000000000","updated_by":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-11-24 19:48:44.000000000","updated_by":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"reviewer":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"state":"REVIEWER"},{"updated":"2022-11-25 13:48:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"599a9c53698d724ae44cb1c85c8db723cd3cf047","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-24 05:00:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* .pre-commit-config.yaml\n* tox.ini","accounts_in_message":[],"_revision_number":1},{"id":"f316fed8d09f7daeda74579cbc846deda9023cb0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-24 05:22:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a1160fc09560d0f68e15cc495f0897d03456130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 06:55:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c50f67c4f11641f7a758906498f19732\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2e7496048734e6eae34872047e74b66 : SUCCESS in 1h 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40d3cc75dc304a7393b98e08f54411a1 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/388f4023233943a389517d6e18f22c91 : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1bcf786431d4fe28a3637888878e96c : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd435fd8f45d4e8684002517c42f62c5 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0000ed06505841a3b691a8af034aed15 : SUCCESS in 3m 56s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/bce15ee52ebb46dc8523c0f0cd69a477 : SUCCESS in 4m 18s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/04c08f47a5ab44dbaa4c13a87cebc0e5 : SUCCESS in 4m 56s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/a990f2bee06047e1b4a3046f7e6defe1 : SUCCESS in 4m 06s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/7695b7ea4a134935a96192e67a6fdcd2 : SUCCESS in 4m 21s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/00f7f8c59a71439da80a62d7cff40690 : SUCCESS in 4m 04s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/1a6bc12206c841efb9d521212159623c : SUCCESS in 3m 52s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/12905afc31e547f5bd0eb4a847392620 : SUCCESS in 4m 15s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/2cf5d938cfb048ef89117c0a897a9319 : SUCCESS in 4m 39s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/4212ffff2c6943fd8894e0cbcca65343 : SUCCESS in 3m 56s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/66ba467531ab4925a214d52e9094cf32 : SUCCESS in 4m 12s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/fa3764966f194ee1bebcc489e172f24e : SUCCESS in 4m 22s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/fbc25db488de4c2a9e3ef14887ea51bd : SUCCESS in 4m 10s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/36e4a46947cc47c491dfcd293c5c745b : SUCCESS in 4m 18s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/83911fd1c37c4b978212d46a01e39fba : SUCCESS in 4m 22s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/118c7757b94742a394405f45cbff7892 : SUCCESS in 4m 07s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/a769d1a4c633479fb942f86bd97f627d : SUCCESS in 4m 09s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/474d5fd6cde44ac6b7531d3ba32ea8e5 : SUCCESS in 4m 24s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/e07cb257f74c4054b5347b27762a6bba : SUCCESS in 3m 52s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/9337550c3c5a4f9a89a4ccbb1b784a40 : SUCCESS in 4m 30s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/0686d0279386456281e8b7bc0bb83ba1 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":2},{"id":"965b92742898cec738e5ce2b1038c815105a359f","author":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"date":"2022-11-24 06:56:28.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"1b8e9e34a2f374463bc6e3bcaba080c23ac72c2a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-24 09:46:59.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"792752e891543e35f30df7ea40db5dd8f587058d","author":{"_account_id":7294,"name":"Leif Madsen","email":"lmadsen@redhat.com","username":"leifmadsen"},"date":"2022-11-24 19:48:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"de5a40b853f41c8ecee6ef372e93442274efaa9d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-25 13:48:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e0c9f069407ae01fd84cb5d4d0f6685294029cd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 13:48:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"756afa20eabe7b0e2aaa21ecda361a0d9319d41c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:17:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2d9af922d354eb28f744e770b50ac5b\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc3911ff361b411ab17847c26bcee0d3 : SUCCESS in 1h 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd48b0df330547bb995b59b33c81eac5 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/344d980abf42428b93fef1022fe7ebb1 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a53d516a7694a7d9424a60d08812d50 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b609d3bb7854ed489fc3d25f63833c0 : SUCCESS in 4m 45s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/7dd5c69bf9c8493799559a09e8bb8d59 : SUCCESS in 5m 37s\n- tooz-tox-py36-etcd https://zuul.opendev.org/t/openstack/build/4813ae12d2cd4f13993b463c7b9f6b43 : SUCCESS in 3m 58s\n- tooz-tox-py38-etcd3 https://zuul.opendev.org/t/openstack/build/2149b1a82f994e59b9a782064b85c4f6 : SUCCESS in 4m 54s\n- tooz-tox-py36-etcd3 https://zuul.opendev.org/t/openstack/build/8bbdb31b0e4f4fa98477be8304d1c21e : SUCCESS in 4m 55s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/8a14779b94114c4ca2655a0df6974495 : SUCCESS in 4m 59s\n- tooz-tox-py36-etcd3gw https://zuul.opendev.org/t/openstack/build/d43eb83ffb37427a8b0b63f74464f36a : SUCCESS in 5m 06s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/05d242242b4f4c728ad8f2a3386ee438 : SUCCESS in 3m 54s\n- tooz-tox-py36-zookeeper https://zuul.opendev.org/t/openstack/build/9ec4ed7bfdb64d89abaff3459e430b15 : SUCCESS in 6m 20s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/9e16aea6c0584323a4aa9d7ceb824e1e : SUCCESS in 5m 32s\n- tooz-tox-py36-redis https://zuul.opendev.org/t/openstack/build/9307f08f37294f8c9086951c90c2fe50 : SUCCESS in 6m 10s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7eaef9af7d284416bc505ff1fe727d7d : SUCCESS in 5m 03s\n- tooz-tox-py36-sentinel https://zuul.opendev.org/t/openstack/build/6af72fafdffa436e8d167d4a2300a1f7 : SUCCESS in 6m 12s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/b307ec19a8214533afb9d8f8795f453c : SUCCESS in 4m 41s\n- tooz-tox-py36-memcached https://zuul.opendev.org/t/openstack/build/ee107896302f47acb48f593c43d0ea35 : SUCCESS in 4m 30s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/0172e3cd80c14a25bf800782541636c6 : SUCCESS in 5m 40s\n- tooz-tox-py36-postgresql https://zuul.opendev.org/t/openstack/build/9c5e24db2466475cb42d521290fdabd4 : SUCCESS in 4m 08s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/3a0e28ffd6ca4f9794dec600cc02d53a : SUCCESS in 5m 08s\n- tooz-tox-py36-mysql https://zuul.opendev.org/t/openstack/build/7f7ddbe0a54345c28e1286daf0d73792 : SUCCESS in 5m 03s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/92b8ba7933334048bb0b67ebd3440395 : SUCCESS in 5m 24s\n- tooz-tox-py36-consul https://zuul.opendev.org/t/openstack/build/ff66d1e8cd1140a79c66a2f7ed242f4b : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"0102b06277706edc89952e3fa4f4fbcad16bd47b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:17:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a63aa804997bbd192e4c26c343f65717d236f98a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-25 15:18:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600cbd2d046247ff92459b4f0d47724d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e3952347c26428a9486521886de1251 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8611c15308925621260f573c0f275a1dea9f80e0","revisions":{"e8f32ae47922519c2fdb8ee2c7443086c5717154":{"kind":"REWORK","_number":1,"created":"2022-11-24 05:00:52.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/01/865401/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/01/865401/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/865401/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/865401/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/865401/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/01/865401/1"}}},"commit":{"parents":[{"commit":"73dbe0e6fc1403cbb121985781d7b6ec996b099a","subject":"Cap protobuf \u003c 4.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/73dbe0e6fc1403cbb121985781d7b6ec996b099a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-23 12:12:06.000000000","tz":60},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-24 05:00:52.000000000","tz":0},"subject":"Move flake8 as a pre-commit local target.","message":"Move flake8 as a pre-commit local target.\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach[1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements[2] will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Ic3f3d05ce7bb10a7d29815444dac00dea4f9c9d1\n(cherry picked from commit 714e77e71f69a223e47afd9d129fc50f76ecd1fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e8f32ae47922519c2fdb8ee2c7443086c5717154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/e8f32ae47922519c2fdb8ee2c7443086c5717154"}]},"branch":"refs/heads/stable/wallaby"},"8611c15308925621260f573c0f275a1dea9f80e0":{"kind":"REWORK","_number":2,"created":"2022-11-24 05:22:06.000000000","uploader":{"_account_id":32240,"name":"Yadnesh Kulkarni","email":"yadnesh_kulkarni@proton.me","username":"ykulkarn"},"ref":"refs/changes/01/865401/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/01/865401/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/865401/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/865401/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/865401/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/01/865401/2"}}},"commit":{"parents":[{"commit":"73dbe0e6fc1403cbb121985781d7b6ec996b099a","subject":"Cap protobuf \u003c 4.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/73dbe0e6fc1403cbb121985781d7b6ec996b099a"}]}],"author":{"name":"Daniel Bengtsson","email":"dbengt@redhat.com","date":"2021-03-23 12:12:06.000000000","tz":60},"committer":{"name":"Yadnesh Kulkarni","email":"ykulkarn@redhat.com","date":"2022-11-24 05:21:31.000000000","tz":330},"subject":"Move flake8 as a pre-commit local target.","message":"Move flake8 as a pre-commit local target.\n\nThe goal here is to avoid conflicts between flake8 and hacking version each\n2 days.\n\nInspired from nova\u0027s approach[1].\n\nThe flake8 version to install will be determined by hacking and\nrequirements[2] will stay aligned instead of relying on different versions.\n\n[1] https://opendev.org/openstack/nova/src/branch/master/.pre-commit-config.yaml#L26-L35\n[2] https://opendev.org/openstack/hacking/src/branch/master/requirements.txt#L1\n\nChange-Id: Ic3f3d05ce7bb10a7d29815444dac00dea4f9c9d1\n(cherry picked from commit 714e77e71f69a223e47afd9d129fc50f76ecd1fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8611c15308925621260f573c0f275a1dea9f80e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8611c15308925621260f573c0f275a1dea9f80e0"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
