)]}'
{"id":"openstack%2Ftooz~865532","triplet_id":"openstack%2Ftooz~master~Ied29512989f477a19753afcb789e5588877fd688","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2022-12-12 10:23:19.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2022-12-12 10:23:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ied29512989f477a19753afcb789e5588877fd688","subject":"Allow to pass ssl-related args for zookeeper","status":"MERGED","created":"2022-11-24 11:41:29.000000000","updated":"2022-12-12 10:24:21.000000000","submitted":"2022-12-12 10:23:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865532","meta_rev_id":"e8bb16230307ecda4b768e7c32ee62a2d96e4933","_number":865532,"virtual_id_number":865532,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-12 10:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2022-12-09 10:49:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2022-12-08 20:07:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2022-12-09 10:49:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-24 13:42:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-30 09:42:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-12-08 19:41:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-12-09 10:49:54.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"90905a3c2630dd1003cb1132cb262efdb417e01c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 11:41:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa91e4950c85bdf6ac929447f2f2b66c1078a903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 11:42:28.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8565b3a0aab1ea04aeb5efa7bc7a1745f0a13f00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-24 11:53:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d81f6d729e07a269cec97271bf4df06e579e498c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-24 13:42:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/284067290c574a7690ab648844db9322\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d0a21a3acb094e5081192505d835411c : SUCCESS in 1h 46m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ad60450ceb949abb9e08c27d3c584a5 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39b709a127db43b29134a598d6e9bc86 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d908303d981c470d8a4fa2aac70bfc59 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0976ce9c96c747b9b13407e76d955b0e : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b2bac4cf4848e992d539bb07ca9f08 : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27c4c523d61242158692644fa4759972 : RETRY_LIMIT in 3m 36s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/35f3abb2ff624a36b35901c1e7f4069b : SUCCESS in 4m 12s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/c162f93298c94c449f8d9ae4e72f1b53 : SUCCESS in 4m 28s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/02b4331995294c108fa300c0cefbec77 : SUCCESS in 4m 29s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/839ba2777d264310a28d9bd516776a6b : SUCCESS in 4m 42s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/60ac592bb03e46d694699076aa77a985 : SUCCESS in 4m 36s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/759e8117792641b6a054fe813dde388b : SUCCESS in 5m 08s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/e1b3c085439a4a8fb38e93dc62150c5b : SUCCESS in 4m 03s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/df3136356c4a429d84be0058c92b2b34 : SUCCESS in 4m 35s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6cc6d9f2f4e44ed4ad6b8fdae212e383 : SUCCESS in 4m 12s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/eac97bf2a8124e85bd89397fb1e08a6f : SUCCESS in 4m 26s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/b94fce56dc8a4cddad8ace9a312d6543 : SUCCESS in 4m 17s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/4283ed415dd74d61a8162e23c5639daa : SUCCESS in 4m 19s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7efef80d69bb4414a25b2d92f389f4bc : SUCCESS in 4m 34s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/9da691c728fe4714b4dd420c581a924d : SUCCESS in 5m 04s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/aaa8715101fd4cb7bb4972370f59e2c5 : SUCCESS in 4m 13s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/b7433d19b43e42efa2d3b3e1f56e635e : SUCCESS in 4m 22s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/3b6cd995d104467a9b7c0fb5a4bbe2be : SUCCESS in 4m 23s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/33b827e8842b4bbd889302410968df09 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":3},{"id":"5681dae2a2246e1b0a562d0bb13a96e0fd6dcc3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-29 17:31:37.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"39dc866676b937d55cbd7283a7491335d94e5185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-29 19:04:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5861a0e104934cee94d012cc69e758ea\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa64133219d5495ebd5516738dc1babe : SUCCESS in 1h 31m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f028ab0d2d1748eeb115915e869f0e5a : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec2c06ec2b641878fdc3eddde337de1 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0cb8b9529dc4c12adef2c06953ee563 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0c98ccd63494579a592912910d629c7 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c555eec524ed457ab649af51de67e830 : FAILURE in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c5ea4a3deb544bcad04186e092b3a53 : SUCCESS in 4m 02s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/f766d7f450ce46008332ef2897ab6e2b : SUCCESS in 4m 31s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/63db0d983dac463d8313093142f58874 : SUCCESS in 5m 13s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/f88b888807ca485e842b8e36d2584215 : SUCCESS in 4m 47s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/7afd9509ceeb4fe6b510642ed37c1c3f : SUCCESS in 4m 59s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/a7ca8bf962d34ccaaa8bfbe18cedbe31 : SUCCESS in 4m 23s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/989186d3e9a74acb9edb0755741a1a1d : SUCCESS in 4m 21s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/adc1418831c3470e82c6ca7ad72c538e : SUCCESS in 4m 35s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/0b44638690214d36a3e748e906bb50ad : SUCCESS in 5m 11s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/782378c285c2410192c690c208a553a2 : SUCCESS in 5m 04s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/9a8232635c8c49edb8e91d6d937ee086 : SUCCESS in 5m 24s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/1ef0da81510d480fa190bdddeeb2bc72 : SUCCESS in 4m 22s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/09c1677ae4194296a5d6c011a87051fc : SUCCESS in 4m 40s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5499c49d7f0c4d8e95927d616ab9d757 : SUCCESS in 4m 09s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/48bd1c529fb541e397d6f977d65e31cb : SUCCESS in 4m 25s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/8d1cf2a9106b413ba45c19e95cbc6756 : SUCCESS in 4m 55s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/724260df013a4290b87d9486a7671e93 : SUCCESS in 5m 37s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/f13dfabadd3f4721937cedc5afd8e770 : SUCCESS in 4m 52s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/f44577235b834d26ab6a368316dfc0e5 : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a033bc59d6d64e528b0a664b31ed2f4a6a90f3e7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-11-30 09:42:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5db3e91e0c994a0ac13940f1fd2b7ef0971c1a62","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-30 16:58:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11abd189ddc1ff8b066c8ef1fbeb657c1d157250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-11-30 17:18:30.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"f0cc13716717284c86409a128cbbec2c6b76fa2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 19:22:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3214af4d2f643f591ba6edb68998065\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fece1d0fde34489858a8a3d05e44d15 : SUCCESS in 1h 55m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f155bd23c2884f1f8a6b28cc43afe911 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9ab33a55c84373a18cae7f2e7e6663 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7c1328f3dfc4e49909f85fcd5600b3c : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5a74779aae14fe388e608b6d2d77601 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57dd02358b41426f903d03426117131b : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49053a7852b44144986e130e1f994829 : SUCCESS in 4m 50s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/c4026abea81b48d0896a9be06dea0678 : SUCCESS in 6m 30s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/706c42b9f8894c349194a31b2fc054c7 : SUCCESS in 5m 06s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d4e366f7d34f4c11aef72ff5419b8be7 : SUCCESS in 6m 58s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/e21903b243d74107b739d561aca62fea : SUCCESS in 5m 36s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/0641d43c5c06480582e60cc668164b01 : SUCCESS in 6m 06s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/9192225846c44c9780ea83dc2ec527c5 : SUCCESS in 5m 14s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/0078172835d1411592370b572e1455bc : SUCCESS in 6m 00s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/1b8ff51fa4554fa48f9f9b399daf62b3 : SUCCESS in 5m 04s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b5652a234640407d909d77382c4becd7 : SUCCESS in 6m 39s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/d01a44b8ccfa41e2a5d228baab0e52dc : SUCCESS in 5m 02s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/352637c403044b359bc9a2e7103cd275 : SUCCESS in 5m 10s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/5a97fac8e0b147e2bc12d15e3dfeeee4 : SUCCESS in 6m 23s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/48144b8bf22f4629b0264ce7427dd093 : SUCCESS in 5m 43s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/398fce8d69e94b13bc8c6cc809c40fe3 : SUCCESS in 5m 01s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/6d239516d91443e9983b5e731b0cb950 : SUCCESS in 6m 32s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/b2c69a265f8b4743b1de02120ceea77a : SUCCESS in 4m 42s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/901cb7d8004a430da89cf7a3cff25e2a : SUCCESS in 6m 40s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/ea97458b822e455294da5819935bd62e : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":5},{"id":"c1b9f5638db5145422251634e4c16508423b0c56","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-08 19:41:33.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"77ea6780953d87b7542564a401bd414b445e15fc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-08 19:43:17.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b93fb09c599ca35814106c17dd1e38229c20fe63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-12-08 19:54:11.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"086ba643ec79961299627ffe51821dec3f106049","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-08 20:07:32.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"917985cdf3eef5f1a3babca5254de01b2a9aad36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-08 21:19:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba90ef6762394b9988ee37abaae0c355\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/edc3fcf63b6141e3ab5b6d527689af8e : SUCCESS in 2m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bb029cfb79347718bce6b8b29b84991 : SUCCESS in 1h 23m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12cf772ad7584214a612dbc15084f447 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da1c69277207462b8647e9dcddb81094 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8114ecc39449d4820feead24f6ea3f : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa2570f662d34f0ea1176f0e861f7b4f : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f5ae2506cac4afc8a0aab4809dd9e75 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1713468148944ac966806c830b68f0e : SUCCESS in 3m 27s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/03cb68b69ca245d884a988d81aa7181c : SUCCESS in 4m 38s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/9fa33db057994d07b4e3187419b468e7 : SUCCESS in 4m 00s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/98ca407a7f47406c95bbe74c684b4980 : SUCCESS in 5m 28s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/eb8b6b0663704e84a9824de6e7aa92f0 : SUCCESS in 6m 14s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/a2eb51ca3b65494a95b130dc08d6c906 : SUCCESS in 4m 34s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/1e2539e860304e05be4d1bf9ccbe3aa4 : SUCCESS in 4m 18s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/02799ba3751c4ec49d76271ce73e7696 : SUCCESS in 4m 30s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/eb496f67e3aa4198a4269a2187108eff : SUCCESS in 5m 14s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/59cbc4eb931243a6b01226484739756e : SUCCESS in 4m 42s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/34d9f65a267e48dc8fbaf6561d2a4a52 : SUCCESS in 3m 52s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/6fd265fe130b4462a403c948da7213fb : SUCCESS in 5m 28s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/a0f11e6687ae4c5e9e882c3c747f7fb0 : SUCCESS in 4m 50s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/55928a253fc646c2bf1dc76e9d65f3ba : SUCCESS in 4m 08s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/4eb90976226d473a932da737c344722e : SUCCESS in 4m 53s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f6cd9017d1764b78808f2e873ffe9be4 : SUCCESS in 4m 12s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/e6c47fd040654abf8e60a424e5896730 : SUCCESS in 4m 56s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/73fcd25d575b405b958f55a3bde5e86a : SUCCESS in 5m 02s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/8d838280eeaa4f77a89bef693d76e592 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":6},{"id":"eb04d856f0ed350161758fb06761d27301e69716","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-12-09 10:49:54.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3f8f7f4c5ac3c828917bd82457c5ea8ac0920691","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 10:50:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"754e6a25819b8acca5387e6e6ccc4691f79487bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-09 12:45:18.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e641d6b539d4a3083183b0a6f915c16\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c331665daa584759a945ff1111d69321 : SUCCESS in 2m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b465c2f394984b59a385fa876d1585e9 : SUCCESS in 1h 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7da5c924e4ad4b8087affab8c610a0e4 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1c8369f3dc3443c938a93e5c87c15bc : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6897145f183944a9b8302486be0dc11a : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2964aec7bd4877b9fef899233be09d : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a23ed41935a3424191d785ca71919f52 : SUCCESS in 3m 39s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/401ca50948bc4898b5aa7858e0cc0151 : SUCCESS in 4m 47s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/fca7b3ccd89645afb8dfb6660dacf878 : SUCCESS in 5m 15s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/45783d02ae2548828a133727a82b2c91 : SUCCESS in 4m 01s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/0778faece19d4ea9926d045b3ed578f7 : SUCCESS in 5m 27s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/33f7c2dc2b664ac9ab4694adde02241d : SUCCESS in 4m 03s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/e86a47d66c6142c6ba3745879d632b67 : SUCCESS in 5m 03s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/0bd32c9a147f4a238755ff792b10c38c : SUCCESS in 3m 59s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/a76e6dbc5984453b95e67e1a6abb60ac : SUCCESS in 4m 14s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/cd47ee83fb374e3bacbdf8e5bc27156a : TIMED_OUT in 40m 51s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/8af6f9d9662f473c9f41a704b8e50fd4 : SUCCESS in 5m 27s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/2700472ab38e4d22b448721289fe3327 : SUCCESS in 5m 10s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/8adc062054bb44c0b521c4e4c67fc7c3 : SUCCESS in 5m 27s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/36d085a157ca4ac8a2c3c6b1a26d0127 : SUCCESS in 4m 23s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/83a5bb248ed74afab901977e8ed866d0 : SUCCESS in 5m 34s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/864fed8c20d54d3bb4bfbcccc06f7569 : SUCCESS in 4m 58s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/425cad499ab7471f9de22def0aff4307 : SUCCESS in 5m 36s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/b334b41ef03847d0bd23d9fa3f452f25 : SUCCESS in 5m 36s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/04bb35ce76b84ce5a3a48a8f9a526d70 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":6},{"id":"7936ae6efa580aa9ad40121243572c9d0376bc7a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-11 22:02:27.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f620a903df77883f67297d1a1355617756b97c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-11 23:04:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc11dfecfd544c098fa0b037a7485315\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/99e792ccf5ac4ed3a626efe06cd49651 : SUCCESS in 2m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92355cf3a36845ddb511452f04be6e6f : FAILURE in 1h 00m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ce306c2290e48d691bb2e293fcf5fea : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7bb75f9eb448159f26f717dfb301f3 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efb2e38a1401463595f0423dc4073c82 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa521dba20694bf39daa8240cb5739b0 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7735bc86da0463ca3914a0db335606e : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66dc6a68f83c4ba68af7c7cfa6edf34a : SUCCESS in 3m 31s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/5b5495e6220e4639b59bd0e5122f6e52 : SUCCESS in 4m 34s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/f02d7e8f3d48460a9668d1489937d172 : SUCCESS in 4m 33s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/1705e2e3629f4db8bbe9a1129bd46bb2 : SUCCESS in 5m 13s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/e24a5516e3b74ed7b0e7ff0da30f1fb1 : SUCCESS in 4m 44s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/89986b1f0c234dbdaa37dcafc83d1d76 : SUCCESS in 4m 16s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/4ad527e60f974d5187c8d1e89d82d222 : SUCCESS in 4m 52s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/97384a1222c04e30a674113087e08419 : SUCCESS in 4m 02s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/e7c68268089b42c7b4ecd6da27c8e2bc : SUCCESS in 5m 00s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/4a43e132d0b4442dafc354a8cb307608 : SUCCESS in 3m 48s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/eba7eb889dd644358a2ba87735508828 : SUCCESS in 5m 02s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ddbac7e6175c4033911098d996eeea96 : SUCCESS in 3m 57s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/ff71b243b0e741769090876e11ca8496 : SUCCESS in 4m 20s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/7ee36c43952c4ed9b12e984b94a11a3a : SUCCESS in 4m 37s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/10db23c6033340518d9ba1495234be88 : SUCCESS in 4m 52s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c310e97aa1ab4d9db2ffb2c7e1953aa8 : SUCCESS in 3m 42s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/1f5bf61ce598437793fec4d7221fe246 : SUCCESS in 4m 47s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/abf48ae5366d4a1ab0c8ac92d089ad51 : SUCCESS in 4m 41s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/d3144d8465a4436196d573b31a5b6f10 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":6},{"id":"beb83d45b0256c2040976c2145a24325f7369fce","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-12-12 06:49:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"885b18088e73ca6b2b6d8ca42aed12cafe456f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 08:54:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e12ea6c53734ac88039a0e45d409726\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9052344152b6440cac2f12bffc2ef454 : SUCCESS in 1m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8df50f8bfaf947518124d86bfceddfff : SUCCESS in 1h 47m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6250c6e4857940feb308efd9562da99a : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11e531ded56d4628ac83c124aa62d957 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64ad7b1310614bb4a84b7a48cf15815d : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73bd4a6620ca45f182d082b5f630e6c5 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56086e0339cc4e26ae0cf45df39a8c48 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59f5144bfe614b15982816ab1cb56aaa : SUCCESS in 3m 49s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/d58a52ea17c346918e52846ec8c38014 : SUCCESS in 4m 25s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/321373bef3be4a2b8a7356206f63e4c7 : SUCCESS in 5m 11s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/703cad8efdc04a148c8c995ccb4de829 : SUCCESS in 4m 27s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/82aecfbe07104873b72ca3ac193ca9bd : SUCCESS in 4m 48s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/848c701e8a454f049bb43064bfdd2c5a : SUCCESS in 4m 20s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/efae1dd247a945f390760321808c8857 : SUCCESS in 4m 04s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/e07e689a45914ecea6e6dc4ff213569c : SUCCESS in 4m 04s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/3e7d30265cd14193a503c8e0cf25218b : SUCCESS in 3m 51s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/b36d1b88450b41af81e516571b86defd : SUCCESS in 4m 31s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/db29b1a2d44f487990946ec3ead75f82 : SUCCESS in 4m 29s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/c56da9057abe4731b36869dbd2373969 : SUCCESS in 4m 38s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/59fd8b368c92419ba1552c07f9ccf397 : SUCCESS in 4m 02s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/8166dcbcd6f3419c95d384851a5744cf : SUCCESS in 3m 47s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/28077328f49f46489538a60dd5dda971 : SUCCESS in 6m 44s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f6c0e08af9254465b5af052156a6d016 : SUCCESS in 6m 36s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/1d9a10bacef447f2b91710cf895fbb41 : SUCCESS in 5m 19s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/076f8e1d1f16495baad55c58ddbeda6a : SUCCESS in 4m 55s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/57b68a132b2342e49cbc4cbb9bab55b2 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":6},{"id":"b626af011140b099c6e4dedc9fc17d5562b25349","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 08:54:44.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"75a090274adcdbec1ece223d91dcb53752481956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 10:23:19.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8666dd71a1e04d76af82de959882bf3f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/49c2914a4759408e8fd2c5f7eba66819 : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36c59629c5604fc79b184d32cd7acec9 : SUCCESS in 1h 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61347f869e454816b7f3214a0a3168e9 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/296f0b07bdee4386ae072efffe8a866f : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4677bd4c17c149858212d7b3a8eccf13 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34321ef503ec4562ad46ced4dece3835 : SUCCESS in 4m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1de8cd4c7f7348259186c6892841d82a : SUCCESS in 6m 17s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/0fe9b44d85ea42978a8f2b09286646d0 : SUCCESS in 5m 47s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/fda38b3308b84659a7b9a073fd6d3b82 : SUCCESS in 4m 26s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/13d4564e787144c8a6050b127965ee6b : SUCCESS in 5m 45s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/c2c317b2634a40fb89cc043b3baa643a : SUCCESS in 6m 50s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b1f5e46d40e9432b8aebda68102b1694 : SUCCESS in 5m 47s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/71bc4f787bbf4eddbaf340d492687ece : SUCCESS in 5m 51s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d8ff94d0d6cb4959b10587b0b34e769c : SUCCESS in 4m 54s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/de9856c2dbcb42d89f11d3b4c861a2b3 : SUCCESS in 5m 53s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/971da21116d84efb9556050983a3f2cb : SUCCESS in 6m 12s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/726065384fc54c7ca83c1e8a847d9895 : SUCCESS in 5m 45s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/05eec80782aa421c97ddfb3c73491983 : SUCCESS in 4m 26s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/35750bd066a94eb5924f62a181da80f0 : SUCCESS in 4m 54s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/10dd12ae828b45cf88fff70a4b782c7b : SUCCESS in 5m 26s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/63c35c66b6a54d9c8adea8a83aca73d1 : SUCCESS in 5m 01s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0825e9a4647042089ecaddb7984b9c49 : SUCCESS in 5m 07s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/8f31539109f645008ddcf93443dee740 : SUCCESS in 5m 17s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/a454e1b99c884500b1313e1a68786bce : SUCCESS in 5m 31s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/b82d694eb81d4fa3acf6999ec758b2ba : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":6},{"id":"2f8d5ea599e3c869787975344cbfca2a6c4ab60d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 10:23:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e8bb16230307ecda4b768e7c32ee62a2d96e4933","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 10:24:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af97d241997a43089b96f1f9899cdceb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e551c69050f24e93ad4e26cd1b765c2c : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1623bbcaa663452fbe52a611cad2dc50 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f11666b66c668dd731b59369ad3a62a6d16536c6","revisions":{"0a23f880da268cf49d076fa1ecd00d19944baadb":{"kind":"REWORK","_number":1,"created":"2022-11-24 11:41:29.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/865532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/865532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/865532/1"}}},"commit":{"parents":[{"commit":"0c145b4567d3b97334ce0a8a2c289403e77b5bbf","subject":"Add grouping support in etcd to compatibility matrix.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c145b4567d3b97334ce0a8a2c289403e77b5bbf"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-11-24 11:40:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-24 11:41:16.000000000","tz":60},"subject":"Allow to pass ssl-related args for zookeeper","message":"Allow to pass ssl-related args for zookeeper\n\nZookeeper does support TLS encryption and authentication for client\nconnections. There\u0027s no reason not to pass these arguments to the kazoo\nto allow encrypted connections.\n\nChange-Id: Ied29512989f477a19753afcb789e5588877fd688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a23f880da268cf49d076fa1ecd00d19944baadb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0a23f880da268cf49d076fa1ecd00d19944baadb"}]},"branch":"refs/heads/master"},"9cf04d06f961d79074decdf80736f65fd88ef5eb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-11-24 11:42:28.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/865532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/865532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/865532/2"}}},"commit":{"parents":[{"commit":"0c145b4567d3b97334ce0a8a2c289403e77b5bbf","subject":"Add grouping support in etcd to compatibility matrix.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c145b4567d3b97334ce0a8a2c289403e77b5bbf"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-11-24 11:40:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-24 11:42:28.000000000","tz":0},"subject":"Allow to pass ssl-related args for zookeeper","message":"Allow to pass ssl-related args for zookeeper\n\nZookeeper does support TLS encryption and authentication for client\nconnections. There\u0027s no reason not to pass these arguments to the kazoo\nto allow encrypted connections.\n\nI\u0027ve also sorted args in alphabetical order so it would be easier to\nread/extend in case dict will grow even more in the future.\n\nChange-Id: Ied29512989f477a19753afcb789e5588877fd688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9cf04d06f961d79074decdf80736f65fd88ef5eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9cf04d06f961d79074decdf80736f65fd88ef5eb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"91fdc4b6549b3a2179bc448cf46686b25616ee56":{"kind":"REWORK","_number":3,"created":"2022-11-24 11:53:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/865532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/865532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/865532/3"}}},"commit":{"parents":[{"commit":"0c145b4567d3b97334ce0a8a2c289403e77b5bbf","subject":"Add grouping support in etcd to compatibility matrix.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0c145b4567d3b97334ce0a8a2c289403e77b5bbf"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-11-24 11:40:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2022-11-24 11:53:40.000000000","tz":60},"subject":"Allow to pass ssl-related args for zookeeper","message":"Allow to pass ssl-related args for zookeeper\n\nZookeeper does support TLS encryption and authentication for client\nconnections. There\u0027s no reason not to pass these arguments to the kazoo\nto allow encrypted connections.\n\nChange-Id: Ied29512989f477a19753afcb789e5588877fd688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/91fdc4b6549b3a2179bc448cf46686b25616ee56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/91fdc4b6549b3a2179bc448cf46686b25616ee56"}]},"branch":"refs/heads/master"},"4b147c32d613977939ee92c688cee8ce61d38479":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-11-29 17:31:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/865532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/865532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/865532/4"}}},"commit":{"parents":[{"commit":"275a7cf2ad3de6af1989468c325dc3d51aa1926f","subject":"Loosen protobuf version that\u0027s used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/275a7cf2ad3de6af1989468c325dc3d51aa1926f"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-11-24 11:40:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-29 17:31:37.000000000","tz":0},"subject":"Allow to pass ssl-related args for zookeeper","message":"Allow to pass ssl-related args for zookeeper\n\nZookeeper does support TLS encryption and authentication for client\nconnections. There\u0027s no reason not to pass these arguments to the kazoo\nto allow encrypted connections.\n\nChange-Id: Ied29512989f477a19753afcb789e5588877fd688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4b147c32d613977939ee92c688cee8ce61d38479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4b147c32d613977939ee92c688cee8ce61d38479"}]},"branch":"refs/heads/master","description":"Rebase"},"f7bda141c6a2853e575cdd3cfb71407534380f5b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2022-11-30 17:18:30.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/865532/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/865532/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/865532/5"}}},"commit":{"parents":[{"commit":"7d50893625bd88d31c181a1c1c6f730ed64c4b3d","subject":"Loosen protobuf version that\u0027s used for docs/renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7d50893625bd88d31c181a1c1c6f730ed64c4b3d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-11-24 11:40:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-11-30 17:18:30.000000000","tz":0},"subject":"Allow to pass ssl-related args for zookeeper","message":"Allow to pass ssl-related args for zookeeper\n\nZookeeper does support TLS encryption and authentication for client\nconnections. There\u0027s no reason not to pass these arguments to the kazoo\nto allow encrypted connections.\n\nChange-Id: Ied29512989f477a19753afcb789e5588877fd688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f7bda141c6a2853e575cdd3cfb71407534380f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f7bda141c6a2853e575cdd3cfb71407534380f5b"}]},"branch":"refs/heads/master","description":"Rebase"},"f11666b66c668dd731b59369ad3a62a6d16536c6":{"kind":"REWORK","_number":6,"created":"2022-12-08 19:54:11.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/32/865532/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/32/865532/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/32/865532/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/32/865532/6"}}},"commit":{"parents":[{"commit":"7d50893625bd88d31c181a1c1c6f730ed64c4b3d","subject":"Loosen protobuf version that\u0027s used for docs/renos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7d50893625bd88d31c181a1c1c6f730ed64c4b3d"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2022-11-24 11:40:49.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2022-12-08 19:54:11.000000000","tz":0},"subject":"Allow to pass ssl-related args for zookeeper","message":"Allow to pass ssl-related args for zookeeper\n\nZookeeper does support TLS encryption and authentication for client\nconnections. There\u0027s no reason not to pass these arguments to the kazoo\nto allow encrypted connections.\n\nWe bump minimum kazoo version to 2.6.0 since change implementin SSL support\nhas been merged with [1] and was first released with 2.6.0 tag.\n\n[1] https://github.com/python-zk/kazoo/commit/35ce10669ace9d0d7e787793f0d4937d5d389f69\n\nChange-Id: Ied29512989f477a19753afcb789e5588877fd688\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f11666b66c668dd731b59369ad3a62a6d16536c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f11666b66c668dd731b59369ad3a62a6d16536c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
