)]}'
{"id":"openstack%2Ftooz~870829","triplet_id":"openstack%2Ftooz~master~I96c9822e54864cdc7cfb6e22ce069054cfef9a94","project":"openstack/tooz","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2023-01-23 19:52:45.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}},"hashtags":[],"change_id":"I96c9822e54864cdc7cfb6e22ce069054cfef9a94","subject":"tox: set allowlist_externals","status":"MERGED","created":"2023-01-17 15:32:58.000000000","updated":"2023-01-23 23:01:08.000000000","submitted":"2023-01-23 23:00:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"870829","meta_rev_id":"47155d32346c9dc55aa5ee50a309f0e9d4caac0e","_number":870829,"virtual_id_number":870829,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-23 23:00:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2023-01-23 14:12:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2023-01-18 10:58:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2023-01-23 14:12:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-17 17:09:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-18 10:58:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-01-23 14:12:59.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"ed953f89c86b38ac0b4bf3029920d419f7c4c24d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-17 15:32:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"364584b6818828d9aed5a7b7f7807e0bc8884092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 17:09:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1704037a55c14c8ab53372e64173ae89\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4645b94676194bb1aee9fb81971634d9 : SUCCESS in 1h 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/572991e5d662484db73b0ddacc91ed49 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fac72500d7c4296bc936649093503e7 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68ab28381c23474e84afb83ea66e8768 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d42c7017981e4fd9a12533a62f88eaba : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e21aeefd49049e09ab719e5847a4647 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b679cb2bb844223a8587cd6cefa3809 : SUCCESS in 5m 08s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/37d1bbc05c28430e957275f7a3f6831d : SUCCESS in 4m 32s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/e4c91cd2eaec4f19a59c4dc26066adc8 : SUCCESS in 4m 15s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2fe0134c60214b45813ccf06137c140c : SUCCESS in 4m 34s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/bb24333d9a5640bfb23080f90a5b0961 : SUCCESS in 6m 21s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/aa94a94c9dda4818b7e9d03d6f4c8026 : SUCCESS in 5m 36s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/a7529f46af1d40438d77f4ba8f18fd31 : SUCCESS in 4m 45s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/c3d65ace63af4065b9560453d65033f9 : SUCCESS in 4m 23s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/f6d3aa51ed9541949212b19544c2ff2f : SUCCESS in 4m 53s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/8bf1d70deb4f4583a77079f4335e01b7 : SUCCESS in 4m 38s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/49f6c96a3cb84554b47c83189c54e6cb : SUCCESS in 5m 17s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/66615cca9e434668a2c3bc50d2db8d88 : SUCCESS in 4m 40s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/ad070780c5cf49a39b7e15547dd2d436 : SUCCESS in 4m 55s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/f3527d9865f346ec91704ef83f9533bf : SUCCESS in 4m 24s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/aa56167ad6374a35923b82f234e0cbec : SUCCESS in 4m 35s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/cf10bd495e04491caded59984f7aa355 : SUCCESS in 5m 00s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/8cd72ecb416c4fadbe365e5851ff081f : SUCCESS in 4m 42s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8c398a7a0ad04029bb06782794c48426 : SUCCESS in 4m 40s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/60230f79909749d1a3b69a63f196e244 : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":1},{"id":"96c1c450fabc6c4670b13caa9d226d1f1c546dad","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2023-01-18 10:58:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d276d639a3fc15670c67800fc31c8a856a760339","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-01-23 14:12:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ebc4972e41e5e99e1ad143415e41ade05f108eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 14:13:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06fe00c311df961b23c99bc08fd99aea877fe42d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 15:49:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e4fa3ae186d4d86b33d8246c9ba800a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2729a03e12ab42b28be0fbf23886d173 : SUCCESS in 1h 34m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/470a808c60814a1abb2281ce3d6323d2 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5449c39b9224418a24ea42ec8fb739b : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94cee18f898f4e19b8bb8c081ac49f28 : FAILURE in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eaf3329dbac74600ae6be996e7b94b35 : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f5d843bd80847d492e32633ec7e1718 : SUCCESS in 4m 35s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/97910d35c2e14f42bee3fb7759cf69b4 : SUCCESS in 5m 30s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/2fd4311e543743fab647a829f0320899 : SUCCESS in 5m 12s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/9216a814ca6d45a3850d620b245aef2a : SUCCESS in 5m 26s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/9f7f409025fd45d8952d9e26eb1f2ab6 : SUCCESS in 6m 58s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/1db8d08e234b48bc9b99a53a3af75f2c : SUCCESS in 5m 10s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/0dc49cba71864651ba7b2f66d6e63c31 : SUCCESS in 5m 48s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/f316217d1ad8453d82fcea14e8265c6a : SUCCESS in 4m 03s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/7069b321e9c84375b5a65071d9d8d55f : SUCCESS in 5m 07s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/81536aa38bad448cb42d94dd8c253547 : SUCCESS in 4m 55s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/26443e62e8194851a8d1a663c2162ed4 : SUCCESS in 4m 46s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/078cd2173f1041d79cf70463e21041b3 : SUCCESS in 4m 41s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/459e9666743e4f77b85ec20aa376856b : SUCCESS in 4m 25s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/d45654851a9849349dda1139a9144336 : SUCCESS in 4m 44s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/b099ea07cc7e431998e41fc819d4c398 : SUCCESS in 5m 18s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4d3c3dffcd8e41c0a3c62f3b144abd3f : SUCCESS in 4m 51s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/ff93b53a87e745a19335ad00d0cdd26c : SUCCESS in 4m 11s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/d9c04b30be04482a9014e23911206364 : SUCCESS in 4m 43s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/b80fbd86c24b4ef88555f08ed08c73bf : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"557d43d18b9ef0d363b6a77168cf837068abc074","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-01-23 19:52:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0b0a4a037b5ed45a8f0efc0516cfd2f862b9aa67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 21:36:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72a6e396d4d0407f8833a50095e4041f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/707eca3ab04e415485761bbbb65ecc13 : SUCCESS in 1h 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0204f3d982d4e6bac57746350bd42c1 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb51fb43a124464483b69e3d087e4de0 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59f6bc2dc39c4fbfb65e52720b174a38 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a364362d9eaf4590b6adecba76535a0b : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b53d582d85144840b2b4f283dfadd72a : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56858849fa9f43709ce6b16b7718bc6a : SUCCESS in 3m 55s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/39e2ebbb75734992aa3c2471117a6f24 : SUCCESS in 4m 32s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/a3ba3d30758042788b2322308745c03c : SUCCESS in 5m 38s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/33c81a07c50a4609a4b291a1d31cbbc8 : SUCCESS in 4m 48s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/7f36af6e31bb485199e0ea5a09f2962c : SUCCESS in 4m 57s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/4b262f4fb7a741829e71795529190026 : SUCCESS in 4m 38s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/3bb6e19dc78441abb5928e2017b469a9 : SUCCESS in 5m 12s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/931d95f885ba4717b4d97a848e58d28c : SUCCESS in 4m 25s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/08f8aeb40cad4666baf9555bc1814134 : SUCCESS in 4m 25s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/cbd9ad302b76485e81547202212de690 : SUCCESS in 4m 51s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/752aeb0c4bda4b59adfa7cdfc2a652fb : SUCCESS in 4m 32s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/756b455043814b2ba3c43138d7f7c717 : SUCCESS in 4m 20s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/ba4ac1492658403c9dc8b8aa3bfca515 : SUCCESS in 4m 28s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/5bafa30870f74464938a05d033280d7c : SUCCESS in 4m 15s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/e994d15fcee6491a9cee118b668608e2 : SUCCESS in 4m 34s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/15513918f947439a98727321be4c6fa4 : SUCCESS in 4m 53s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/8668c23c19404596993662c6b0278d66 : SUCCESS in 4m 24s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/205df1fffee04c44a3be14b907a46d3e : SUCCESS in 4m 36s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/ffbc4bf4fd214233960280b018d0a7bc : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4ca48557f9d0dac46bc15fc2d2fe9ea1dd5d93b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 21:36:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db1d4b3abb73a8a18e2f25b786c9eb63080dca69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 23:00:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3874e02151f040b892e6200f5326136f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/708feb5fa40f4597a373aea1980f5684 : SUCCESS in 1h 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67c4746e1192480ca09863cc7314cc04 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918be34d13e84d61b97042c333586be6 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8284d89558b84fa58212b0f7ea9b2cd0 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/500b5e89927842e0bf4c2976d4451f92 : SUCCESS in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31017583ce374b91939335ed67ef1af0 : SUCCESS in 3m 31s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/8ce1fac7216349c3bfc7e7a775303aaf : SUCCESS in 5m 03s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/cde2897448fb4b2f997b31c31c1a4d82 : SUCCESS in 4m 17s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/081c121746f644828cae903f83c244a9 : SUCCESS in 4m 44s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/1eae0f5afec841e2a95ce769726bf9c6 : SUCCESS in 4m 26s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/44c659a81ca34959aa6ec89bb8a45a33 : SUCCESS in 4m 08s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/3a2c93b3f7ec440b804e3a2ff3dcccf9 : SUCCESS in 4m 52s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ce4abfbfdcb748c3beb006b1a4e1d82d : SUCCESS in 4m 19s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/f93bb28a2f8d47e1961b46eb9eaf1da5 : SUCCESS in 4m 17s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/4ad4b5bb8ae84d8d87b68879fc2694f7 : SUCCESS in 4m 14s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/594bd37541ba4716990c823ea2f477f3 : SUCCESS in 4m 05s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/0a7064467fd74d029e7c799c52a6e5e6 : SUCCESS in 4m 31s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/ea91c327622644f7bf1f5980c7122e49 : SUCCESS in 4m 21s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/170b24ab264f4fa685f80086ea6b5efe : SUCCESS in 4m 25s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/2e6f779484824ab1af9b238cd72ea5a3 : SUCCESS in 5m 00s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/93bff610b06545b8abe2e4f42c692252 : SUCCESS in 3m 51s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/86693c9e0d8b47c6a9a01285da1fec27 : SUCCESS in 4m 27s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/9b568f851ce44fdb9e8b4dff69c2b20b : SUCCESS in 4m 20s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/0871df52d4ca4ab6b403786eee0325a1 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c0ed09958172b0b59274c778a89bb266b63ffcbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 23:00:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"47155d32346c9dc55aa5ee50a309f0e9d4caac0e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-23 23:01:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03da70b1716e466c810ead579f80c8a9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fa784ab5bb04b6c83c4be6d74c26e60 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fbbfd0444c24fdca060c6f14175b2b2 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7698c582f31f78a4fab558420b18ac88e38284a1","revisions":{"7698c582f31f78a4fab558420b18ac88e38284a1":{"kind":"REWORK","_number":1,"created":"2023-01-17 15:32:58.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/29/870829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/29/870829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/870829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/870829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/29/870829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/29/870829/1"}}},"commit":{"parents":[{"commit":"f11666b66c668dd731b59369ad3a62a6d16536c6","subject":"Allow to pass ssl-related args for zookeeper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f11666b66c668dd731b59369ad3a62a6d16536c6"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-17 15:32:49.000000000","tz":0},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2023-01-17 15:32:49.000000000","tz":0},"subject":"tox: set allowlist_externals","message":"tox: set allowlist_externals\n\nChange-Id: I96c9822e54864cdc7cfb6e22ce069054cfef9a94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7698c582f31f78a4fab558420b18ac88e38284a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7698c582f31f78a4fab558420b18ac88e38284a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
