)]}'
{"id":"openstack%2Ftooz~877890","triplet_id":"openstack%2Ftooz~master~Id7e76c35c5ce7b22a8261026fe76557d509616b3","project":"openstack/tooz","branch":"master","topic":"bug/2012226","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-02-01 19:45:04.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-01 16:39:00.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-02-01 19:45:04.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-02 13:13:40.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Id7e76c35c5ce7b22a8261026fe76557d509616b3","subject":"Redis: Allow customizing ssl ca certificates path","status":"MERGED","created":"2023-03-20 05:37:59.000000000","updated":"2024-02-01 19:46:01.000000000","submitted":"2024-02-01 19:45:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877890-bug/2012226","meta_rev_id":"767817db17f6188364584bdc053685d52a61514c","_number":877890,"virtual_id_number":877890,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 19:45:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-11-27 01:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-01 15:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-01 15:09:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-20 07:50:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 09:10:27.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-02 13:12:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2023-10-02 13:12:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2023-10-02 13:12:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-02 13:13:40.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REMOVED"},{"updated":"2024-02-01 15:09:45.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"3ae945202a7bbe75d9d72dcdbe0c6c20419b449e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-20 05:37:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"528597577568dd41ab2e35c0a9e6cec2caf493de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-20 06:15:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c796972f49d00d01beff8eb3c1908e4b4cf2422b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 07:50:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/53ea2d2bde854e5c8bfa44aa3f101382\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6b55ffdea4844148f5c14cc6e866683 : SUCCESS in 1h 27m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c361d621f6274004b1cdcf3aa09562ea : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99cf2e5b8a2042288ca3b001f10c5b85 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ec11a6d7b714cd58a127330b638d232 : FAILURE in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fefc372b8294b93b59df0a2dc772c9b : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b15b56ba98f943b2b8f5555393baffa2 : SUCCESS in 4m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e40323d000f441bb6cd618637ad74d6 : SUCCESS in 3m 45s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/353d8a073d0a49fbbfa6ce845ebda047 : SUCCESS in 3m 52s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/04c1e4e3b9f74937af8be9cdb64778b5 : SUCCESS in 4m 18s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/1be332ace9694f5a9b81407b1356087e : SUCCESS in 4m 20s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/97eb29717bc54cd8bb64f5fce944fa5d : SUCCESS in 4m 47s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/484a2792e4ce4725aa2c447e05b0ffe2 : SUCCESS in 3m 45s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/5435b5d04130449eb2918fecfa20dd40 : SUCCESS in 4m 04s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/de06876587244595ac821b4cfaffa8d8 : SUCCESS in 3m 49s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/67727b27c48b4251a48382c6e6c8b59d : SUCCESS in 4m 00s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/bda09cd65ea04a3eb261ccc04f2aad4e : SUCCESS in 4m 45s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/e2d44fe37cf445aeb3ea76b982424477 : SUCCESS in 4m 26s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/354d6b2ebd7b4102b618ab61e96c1fb9 : SUCCESS in 4m 03s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/d22341a3a6e14f60b4949e215a5e2483 : SUCCESS in 4m 37s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/23047e771162481e89da8848ba026e5e : SUCCESS in 4m 03s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/6c88d14318c94a84a0aea29c177d71ef : SUCCESS in 4m 28s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/b5a2bd52638e43e49c77b2d87ce1b2ef : SUCCESS in 4m 44s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/ab91ad5cd70d422dacc42c952a3e7f45 : SUCCESS in 4m 22s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/8c91144e265a412f90917b57d0eb0da4 : SUCCESS in 4m 02s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/a4ca03dbedc441098b30e2290e488d0c : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"5dbdcdee7d6bf185c0c8e3ba376748625fa1a5f5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-20 07:52:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e665e0e9d89b8e00aeaf7caa3c5ee33b4e8b3b60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-20 09:11:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bf1e3bb39c84df197f56c832dd2c09d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f50002cb5234456ab0406cb290a8b339 : SUCCESS in 1h 18m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f50b5efeb2b54649bb2c659de9bae659 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc893ac429d4d56af13d59ba75d7bf9 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c72f5aaf65043a78ae9098c1170c4eb : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/937ab78cd4f942358be5138c6cb39c80 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/429c4824586c4c51b0af214246fb7543 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7c9804871a24c5f8100c2649162c8f1 : SUCCESS in 4m 25s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/18b8381a58d94f669b8c348183cd764f : SUCCESS in 4m 14s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/abb007e430e44bbc8cc9c92102b8f357 : SUCCESS in 4m 11s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/241a6ffc1797406bac1d7acdd2fe28e4 : SUCCESS in 4m 25s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/93981f4a106b4f81983b12d4404f09ee : SUCCESS in 4m 38s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/1d43dc06cde74bdda97710a14e3b0abd : SUCCESS in 4m 07s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/9b3fa48ecb624a228f92771de343c238 : SUCCESS in 4m 16s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/5084cc51516d40f3b4aae0a0a64ac5de : SUCCESS in 4m 27s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/ea42ffb853fd4f9ebc59ec202e357fc3 : SUCCESS in 4m 09s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/e94fa5c3a43a4e869a76d92c551a814e : SUCCESS in 4m 29s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/4304d6c6730442ccb2b67111fa764876 : SUCCESS in 4m 22s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/6ff69336241c46eeaaf9f316ca5e3d30 : SUCCESS in 4m 19s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/65e5ed8811544c549c886ca917d32026 : SUCCESS in 4m 06s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c5caae92accc4c99aef7c899161d1384 : SUCCESS in 4m 30s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/535f178921674ad58e14372fd6f286d6 : SUCCESS in 4m 37s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/6845d9a3c37b489e9bc85928990f7172 : SUCCESS in 4m 37s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/04bed1e0de524b62bd2c160ba790c56b : SUCCESS in 4m 01s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/7e106c2e99d24979a1707671df98c2d5 : SUCCESS in 4m 32s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/2d67dc9d7c3e4fa49069ba5a6a3b5039 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"22f2a31736c9ba7cdda392e24919d6fc12013292","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-03-21 09:10:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b726b2e08d5d0577bc3336d2c4371b6a69e1c06f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-02 13:13:40.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_11628\u003e.","accounts_in_message":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"_revision_number":2},{"id":"cb9770207fbea0db0c4638b04e7f2cfadded5c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 01:13:12.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c54e575c9244b4d9731a8e5dc184ef1542a3f582","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 02:55:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e901f833b1584d2eaa49b6b1ac58ae7a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a6b9ae83e2e4c4c96e6bbd703c989f2 : SUCCESS in 1h 36m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a4412314f774653b36c0f5ed27c2978 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ebe760fea5744cebb52934c734b9019 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10908691d26c44059cedd82bc120816c : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0308ad4dc6d74cd39f75f35c246f1c6d : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ca173ca2a24efe9bad6c7b19a6aff5 : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6cc21fbe514091af3cfc86bcac90ff : SUCCESS in 4m 26s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/a4ec2b6f5185420e9db63bad9faa4e11 : SUCCESS in 6m 29s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/3511fe4cabfe4fb3ab09737193a07a2f : SUCCESS in 4m 20s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/eb022d6776a84069b0bd92955565473c : SUCCESS in 4m 31s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a07229da7a3041dd841b6688d48d0e1c : SUCCESS in 5m 06s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/503f7108ac774fbea0554499adf49be9 : SUCCESS in 4m 09s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0ec1ebed04bf4ac4a81e1472e68974aa : SUCCESS in 4m 06s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/9de71152f9cd4e2f8e1696cd32f7cad3 : SUCCESS in 4m 45s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/db6f86c3399f4da39e7798ac8050aaf0 : SUCCESS in 7m 08s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/4e1844b7666f4bcaa3e6f816f8255a51 : SUCCESS in 3m 42s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/306d764a58564697876e4756bc39fa28 : SUCCESS in 4m 26s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/f3c736681d494ff3bf32b5d85d6004e4 : SUCCESS in 5m 01s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/3d30b478f9624057aad9d51d44e0d053 : SUCCESS in 4m 14s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/49e917db304d40b3aae552d897f84fef : SUCCESS in 4m 51s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/ced938c5d00c4c6b8abe0c951dba2105 : SUCCESS in 4m 19s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/52d954e64be144fb8b7e643a5666ff0d : SUCCESS in 5m 03s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/6196e7995f27480694868731743a2f10 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c1fd95ee4e3f0872bfc48740e71e69df8b14c305","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-02-01 15:09:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"71b53a7ffecffed3f9960f570a9588e3fdedee4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 15:10:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"28d88b40e5438a73f3b91e12fc15b50d04244d67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 16:31:41.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/906203c9d63349d3a5d5d01bbb6bdd52\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41f6fae3b41041f9b9c2a522e42a7744 : SUCCESS in 1h 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a15b72c521c4498b83c698d90e7d863 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6340d75e3f64cb5a0c33ada24519d3a : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7df086dcfbd42ff9007a8a3150ed983 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f2daef60d2d448e8739bbdf02f537b3 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a68a1d35207b44a6b1d74bba15e7cdb6 : SUCCESS in 4m 51s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/52ea5ea3e2454b97bd25eef443837c33 : SUCCESS in 4m 41s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/268ef5e8070744f587c16211deeebb2c : SUCCESS in 4m 46s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/406c6833d432412484ebe29efc7a8624 : SUCCESS in 6m 16s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/dd7a7d5e0a754e4099adad7de646d0be : FAILURE in 4m 39s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/98831b7b1b6645b8b998d09a26ae4b1b : SUCCESS in 4m 17s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/c597f44dd3e5418ab56a0f9b616ef67d : SUCCESS in 4m 40s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/e2d58550a4bf4c28b3727664aef13f0c : SUCCESS in 4m 17s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/6fc62c6a11ce4044a82825434080e4c3 : SUCCESS in 4m 59s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/2086aa4b041b4861a8a723e1c836fdcd : SUCCESS in 4m 15s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/f729fb5aa8ba4deaad1bf4333b7e320d : SUCCESS in 4m 17s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/62a34134e5e749fb813d897acb9c6681 : SUCCESS in 4m 30s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/464650530c134c5bb4cf864a4871c004 : SUCCESS in 4m 02s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f6b6e46f50ac4b0b99aa199f56c34acf : SUCCESS in 4m 09s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/43efd6f358944fa5a0a16c218c720951 : SUCCESS in 5m 42s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/abf3adaf89c94ed4a1f7117e833fced8 : SUCCESS in 5m 25s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/e86d31a8085d4be087b168660763770e : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":3},{"id":"3c921569d27ea64134345f8ea16cd67ee057c4b7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-01 16:39:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7f7eefba789e11585c69a0b51ffe8014caa41ac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 18:23:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4e6c82401f64428bc45e8f593892771\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/26f8f01024574321ba48a7e018fd612f : SUCCESS in 1h 41m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5b9e3d4565042518865244f475abfda : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aeeffc3732b42b3acfafa73a92da2cf : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/696afb6d6cb24f7aaa306fc98dbca235 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5d02b7d00294c69bf31d2237fd37648 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5034364141d436286d981f788331432 : SUCCESS in 4m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f7d236e773b4cb2b4876fefd94fb1ab : SUCCESS in 5m 20s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/ea33836b846e4fa5b18988f1d0974272 : SUCCESS in 5m 45s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/dcec10c6051a405eb0a3127ca848339d : SUCCESS in 4m 52s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/1eab0ecdf8a04395b3e1d9de50b4b72b : SUCCESS in 4m 15s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/9e5ae1bea43c4004b32d06899af528ed : SUCCESS in 4m 20s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/9581384f13e546c78d1de7ff3719ed70 : SUCCESS in 4m 11s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/add3d2892cbd41eb944117914f250fa9 : SUCCESS in 4m 11s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/704a913919c74362a3d2c745f6696f7a : SUCCESS in 4m 43s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/21532d1dbd514630ade61c88592df1df : SUCCESS in 4m 09s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/320f38dc0edd4a68b552e79d0cf9be87 : SUCCESS in 4m 31s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/caf124110eda4288ae977c5d9138eefe : SUCCESS in 4m 26s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/627e25ff746b42678cd7f6477f489e30 : SUCCESS in 5m 42s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/93cec688d15f4921add6109883d511f5 : SUCCESS in 5m 45s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/eb2bb62fc042449687c19f86b5cce906 : SUCCESS in 5m 31s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/36557f7240ec40bc8914fe082243faed : SUCCESS in 5m 42s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/ff7c766949fe40499a9db849d2f2cfc1 : SUCCESS in 4m 13s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/e427acbb07c949abaa2b73add11c817e : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":3},{"id":"c6f8e2fbe700288a4cfd77d0f05fd39c23fab049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 18:24:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b2d99c1007038bdf2abe9c2c6fe0b823017263b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:45:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51a4e4c129e34faaa7f5fdbc187e0f7a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8dadce2f86d4879a9528cb065a52b18 : SUCCESS in 1h 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7cb81f1a68a4a979924cf75ba4f892d : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a031d05fcc994ecc92b6a40726512745 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/551a29125ead4bf58a1be28b7a6be84a : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1a456caec7f4e9ea80e2089071bb867 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c61b9725f6646bba9ca7a19967dd512 : SUCCESS in 3m 32s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/aeff1609b4e44f659dd718c27399ecfc : SUCCESS in 5m 39s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/6b45f0ed3b2e474aa8f82d9a8e9ced0f : SUCCESS in 5m 11s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/21f642d1a1f544dba1fefa56bc90fa56 : SUCCESS in 4m 50s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/203677c7a8a84e8dba57f3954d13f543 : SUCCESS in 5m 43s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/c406f581573a43158047502f4e7f1332 : SUCCESS in 4m 31s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/117384c8dcb045a58247e7a664aa4f41 : SUCCESS in 6m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/5b673c538c9548c298395bbbad7e6376 : SUCCESS in 5m 56s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/0588e67364804e4fa45ff0a4347920c6 : SUCCESS in 4m 19s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/22baff48bf5e40af8e9ce3a5fe6083ec : SUCCESS in 4m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/26a839ac6b8a40cdb3723d5cb40018eb : SUCCESS in 4m 05s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c73cf53c2b3b43679a6cc6e2cb5fa5bc : SUCCESS in 5m 32s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/6a2f7902507a44409e06ba36eb741043 : SUCCESS in 4m 19s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/78db00cc90bf4818b68c90e5c293b6b7 : SUCCESS in 4m 26s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/1024f334f1754525a544eaf4f512752f : SUCCESS in 4m 07s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/accf2c015d0b42d88d46d90e94cf887b : SUCCESS in 4m 36s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/61ad33e6fdee4fd09e72f8704c8ffd27 : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":3},{"id":"61fe251024d87d0deab919de84c77f9ee2dfea58","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:45:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"767817db17f6188364584bdc053685d52a61514c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:46:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c227180e21254a6b80f114e9d37804fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0f1da0a8202494898115b035e3a72d4 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db03b934ab9c418eb78d00bbb3c2b116 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"38523acdc689adaebb4d302a92721abd14e7d9aa","revisions":{"f6459b2b096666ab71b7429935a875570ec373e0":{"kind":"REWORK","_number":1,"created":"2023-03-20 05:37:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/877890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/877890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/877890/1"}}},"commit":{"parents":[{"commit":"7698c582f31f78a4fab558420b18ac88e38284a1","subject":"tox: set allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7698c582f31f78a4fab558420b18ac88e38284a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 05:30:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 05:37:56.000000000","tz":540},"subject":"Redis: Allow customizing ssl ca certificates path","message":"Redis: Allow customizing ssl ca certificates path\n\nThis allows users to give ssl_ca_certs argument of redis client so that\na specific file, instead of the system global certificates, can be\nused.\n\nCloses-Bug: #2012226\nChange-Id: Id7e76c35c5ce7b22a8261026fe76557d509616b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f6459b2b096666ab71b7429935a875570ec373e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/f6459b2b096666ab71b7429935a875570ec373e0"}]},"branch":"refs/heads/master"},"c2070114ad5038a37e06609cb39894ace7a55952":{"kind":"REWORK","_number":2,"created":"2023-03-20 06:15:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/877890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/877890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/877890/2"}}},"commit":{"parents":[{"commit":"7698c582f31f78a4fab558420b18ac88e38284a1","subject":"tox: set allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7698c582f31f78a4fab558420b18ac88e38284a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 05:30:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 06:15:09.000000000","tz":540},"subject":"Redis: Allow customizing ssl ca certificates path","message":"Redis: Allow customizing ssl ca certificates path\n\nThis allows users to give ssl_ca_certs argument of redis client so that\na specific file, instead of the system global certificates, can be\nused.\n\nCloses-Bug: #2012226\nChange-Id: Id7e76c35c5ce7b22a8261026fe76557d509616b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c2070114ad5038a37e06609cb39894ace7a55952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c2070114ad5038a37e06609cb39894ace7a55952"}]},"branch":"refs/heads/master"},"38523acdc689adaebb4d302a92721abd14e7d9aa":{"kind":"NO_CHANGE","_number":3,"created":"2023-11-27 01:13:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/877890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/90/877890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/90/877890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/90/877890/3"}}},"commit":{"parents":[{"commit":"7698c582f31f78a4fab558420b18ac88e38284a1","subject":"tox: set allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7698c582f31f78a4fab558420b18ac88e38284a1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 01:13:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 01:13:03.000000000","tz":540},"subject":"Redis: Allow customizing ssl ca certificates path","message":"Redis: Allow customizing ssl ca certificates path\n\nThis allows users to give ssl_ca_certs argument of redis client so that\na specific file, instead of the system global certificates, can be\nused.\n\nCloses-Bug: #2012226\nChange-Id: Id7e76c35c5ce7b22a8261026fe76557d509616b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/38523acdc689adaebb4d302a92721abd14e7d9aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/38523acdc689adaebb4d302a92721abd14e7d9aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
