)]}'
{"id":"openstack%2Ftooz~879930","triplet_id":"openstack%2Ftooz~master~I73a909411d4896adf1fe5903afa1dec4640f8a0b","project":"openstack/tooz","branch":"master","topic":"oslo-drop-py38","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-11 11:32:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73a909411d4896adf1fe5903afa1dec4640f8a0b","subject":"Moves supported python runtimes from version 3.8 to 3.10","status":"MERGED","created":"2023-04-10 04:13:41.000000000","updated":"2023-05-03 20:03:10.000000000","submitted":"2023-04-11 11:32:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879930-oslo-drop-py38","meta_rev_id":"626c6088f29968421967ad168a5a28ae80aa5552","_number":879930,"virtual_id_number":879930,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-11 11:32:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2023-04-11 09:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2023-04-11 09:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-10 05:19:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-11 09:59:52.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"b2f6e6cf86bd02699d60a969ee818240a06ed167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-04-10 04:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a9b063d8904b92e11da92a8b946000c7cf06ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 05:19:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19bc803035a74b1782393418ba487f9c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1966efbb2408473b954a17d009ab4ed3 : SUCCESS in 1m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf0147a09e604b32917484a05423f65b : SUCCESS in 1h 00m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdafc513bdfd41a5be0179368b11c474 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3852c6c2944119a793afc1967d8443 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8445b63db8eb4a9eb1da64081bf365fe : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/210d39cb2a6d465080fefa0cd5a7cc64 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b64d998eff47a8beeeebdd892f6e36 : SUCCESS in 4m 35s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/0d17abadb65142cb9cdfa0290d797809 : SUCCESS in 4m 25s\n- tooz-tox-py310-etcd https://zuul.opendev.org/t/openstack/build/261b96aea5464a4a872dd36a983d29ba : SUCCESS in 4m 36s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/ef1ed3f376e4452183323c49a5942433 : SUCCESS in 5m 23s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/8e65789f309548b39fd42ba7d5c2da5b : SUCCESS in 4m 26s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/33c69d8adb974ae3978debc704795490 : SUCCESS in 5m 21s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/d3611e6880fb4e83935507426c2a2d30 : SUCCESS in 4m 45s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/3d4fc856054547f6b1af578d2f2b8298 : SUCCESS in 4m 37s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/68c153e80cc4417b824728f6b33535d9 : SUCCESS in 4m 11s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/1bf33c22af3547c3b44b52a01726b656 : SUCCESS in 4m 01s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/b56cc496179a4a4fa6e7ac892276c89d : SUCCESS in 4m 42s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/8455261c270042c28e5003b0295a78c8 : SUCCESS in 5m 10s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/4f4079ad0e1f45c8b88005efef9236cf : SUCCESS in 4m 53s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/148386fbde8e41c4bd5535463139bae0 : SUCCESS in 4m 06s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/357a79d0e00d454590faa1164f661738 : SUCCESS in 4m 55s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/d5f44b63a86d478e99791e12ff531022 : SUCCESS in 4m 19s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/93f9580908194babb4ec0d988e1d2769 : SUCCESS in 4m 47s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/986429b284bd4b23a287860953895abb : SUCCESS in 5m 24s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/15902fe92b37450aa55eb046ceb160a8 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"4a1e9e9b523f3b2c029291f8e828820f6489d82f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-11 09:59:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"353e0ff682983c69c770e21a1d306810a30a6ee0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 10:00:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"becbe0908212dd00be96e455e355a7d26ece6141","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:32:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16701e1823a64808844e27d5d3758a6f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/51695f1890754762bad586a353a77895 : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cba0951e9e6c4d5099c88af2e7f5e06e : SUCCESS in 1h 28m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74d6c0054014c1ca107c899e32e88a5 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f31efa768204fd88bfe74fd70f2e216 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/808d72741f75416e91bd12510cd3bfca : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68276b2785aa4117a42ebd31bc9c907d : SUCCESS in 4m 34s\n- tooz-tox-py39-etcd https://zuul.opendev.org/t/openstack/build/e34466568f374216b1e125aa1ad44b9a : SUCCESS in 4m 23s\n- tooz-tox-py310-etcd https://zuul.opendev.org/t/openstack/build/d6c56aed23014a9ab7704a024ab14c96 : SUCCESS in 4m 40s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/dff85ed4efce46bfa7476df263531006 : SUCCESS in 4m 28s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/d6f438704c3d43f6a0577b8a27bb6ae2 : SUCCESS in 4m 39s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/eaaac2c81b634b6baae5935e63773530 : SUCCESS in 4m 57s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4085529a329543fcb97e6d8c32d4d9ce : SUCCESS in 4m 04s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/02e6b27cd7ac44b4aad2b6d2efd5d8a7 : SUCCESS in 4m 42s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/483ee11fad3b4cfc827f2f011a1263dc : SUCCESS in 4m 36s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/91c4a1ed42bd45ecb89f0372332fa0e4 : SUCCESS in 4m 12s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/0e24c7ff7ca04aaab4c41e171f6481f6 : SUCCESS in 4m 44s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/a8ac13ab1a1649129fbdfcd2ffc1080b : SUCCESS in 4m 54s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/268f416fcaf54729b9a04f96a7d205a6 : SUCCESS in 3m 51s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/8e37d6c1490b4c68a263e2ec33cf849a : SUCCESS in 4m 24s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/bc5b14a70e424cdc8216dbd3cc8df3c2 : SUCCESS in 6m 35s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/c8f8c41cc7f640a1aac075d873cd75c7 : SUCCESS in 4m 14s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d60d27d83a0f43e898575df29594c1b7 : SUCCESS in 4m 32s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/2316beb3c262403e8998ea56c0e39a6a : SUCCESS in 4m 16s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/3ad3574dace146c982196760885c43c3 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8be3fc5eaccd18e771d3b776521a6d8f483ac2d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:32:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5aab95d9fc2f50d9a98db820b38522b54458e34a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 11:35:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1826fb392da460b94b64e37b4655a86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb9bca693fe4ddbabee777e1a38f9e3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"626c6088f29968421967ad168a5a28ae80aa5552","tag":"autogenerated:gerrit:revert","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-03 20:03:10.000000000","message":"Created a revert of this change as I18508891947649aa0a696372f71877c87b5c387e","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a18ae10b8d2dc164b4607fcd0728bb24516a723","revisions":{"4a18ae10b8d2dc164b4607fcd0728bb24516a723":{"kind":"REWORK","_number":1,"created":"2023-04-10 04:13:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/879930/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/30/879930/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/30/879930/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/30/879930/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/30/879930/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/30/879930/1"}}},"commit":{"parents":[{"commit":"7698c582f31f78a4fab558420b18ac88e38284a1","subject":"tox: set allowlist_externals","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7698c582f31f78a4fab558420b18ac88e38284a1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-10 04:13:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-04-10 04:13:32.000000000","tz":540},"subject":"Moves supported python runtimes from version 3.8 to 3.10","message":"Moves supported python runtimes from version 3.8 to 3.10\n\nWithin 2023.2 python version 3.9 and 3.10 are the\nsupported python runtimes [1].\n[1] https: //review.opendev.org/c/openstack/governance/+/872232\n\nChange-Id: I73a909411d4896adf1fe5903afa1dec4640f8a0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4a18ae10b8d2dc164b4607fcd0728bb24516a723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/4a18ae10b8d2dc164b4607fcd0728bb24516a723"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
