)]}'
{"id":"openstack%2Ftooz~893601","triplet_id":"openstack%2Ftooz~master~Idbe9006e843bcedcdb98d735c49b157ab5ae6b12","project":"openstack/tooz","branch":"master","topic":"fix-minus","attention_set":{},"removed_from_attention_set":{"9542":{"account":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"last_update":"2023-09-07 11:24:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbe9006e843bcedcdb98d735c49b157ab5ae6b12","subject":"Replace en-dash with ASCII minus in a docstring","status":"MERGED","created":"2023-09-04 07:45:28.000000000","updated":"2023-09-07 11:25:48.000000000","submitted":"2023-09-07 11:24:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893601-fix-minus","meta_rev_id":"85e1f377210789ccda75c0fc150bb9f4eedbb137","_number":893601,"virtual_id_number":893601,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-07 11:24:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2023-09-04 09:32:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":1,"date":"2023-09-07 09:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-04 09:32:52.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-09-04 09:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c818ff981468dde3e8148a202d76f55e94364d9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2023-09-04 07:45:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"974a16a50f7d730c508dcf2f0a75dd9a967a3f73","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-09-04 09:32:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e6270a2b1fcea83d07006237509862e791679f3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:36:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c89c500f1634a1dbe57c39cc28e4d15\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20be536b591e43ed96cbc8959b1015b7 : SUCCESS in 1h 45m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcd65bcb4a63471aa7f19359e26833af : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb97134f927a4bbea9115a2cba0b31a7 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31a5a3703d6d430b8b76963ff722779c : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e94173577eb495f8a82da4aa59e8ca5 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcb4f5b6125c48338dbe8954f2b3c7ad : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42098a3bdde544ada81040d961fd2957 : RETRY_LIMIT in 1m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d884fb67c7a744fba3287a668a8ed859 : SUCCESS in 4m 24s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/aeb996536d474c40ab43b7e7318e82d7 : SUCCESS in 4m 39s\n- tooz-tox-py310-etcd https://zuul.opendev.org/t/openstack/build/ea1f6328afae40c8b0257c0628846342 : SUCCESS in 4m 42s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/9936366a58f34cb788ce5d693163a51a : SUCCESS in 5m 02s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e14670e6f3cc46f298d15160dca3fdee : SUCCESS in 5m 27s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/d909b3c50db94743a5552aa274ce5429 : SUCCESS in 4m 43s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/179219ea017c43ed90d426687d64c3d1 : SUCCESS in 5m 00s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/21937365639841f88af0638e20671e64 : SUCCESS in 4m 34s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/cad1349367b24b9d8e7c6a8441d389fc : SUCCESS in 5m 00s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6f88b0dd7fdd41ce88aa248794b671f7 : SUCCESS in 4m 26s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/ae7613c770f041aba1f8a6bf453e60a9 : SUCCESS in 4m 47s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/0d79315812c143b5aac559d341436c73 : SUCCESS in 4m 10s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/f4a85a34145347fc95a63e148364e109 : SUCCESS in 4m 59s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/9779c133a16f4b17ba14659b7d32d7bd : SUCCESS in 4m 47s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/930e85cb2a7f49eb887ba1a18d5d4b0c : SUCCESS in 4m 27s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/d3594252ed6d40498e495744f472b61a : SUCCESS in 5m 06s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/aa4411426ef548c6af9da5f78bfca88d : SUCCESS in 4m 23s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/255a8edad1ba4905a9b9555961fbd6cc : SUCCESS in 4m 35s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/59fd74289b954fe093f5d6e75074adcc : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"7815cbf701c36f25591672d63f99a3bcf9692524","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-09-07 09:54:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e8c78e45eca54ccb4d6ac282fcc8be9c3193604","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 09:56:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f01af053a14d1c4165291a8d72440e2e6220518","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:24:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb0a8e256e92454da7e40ffdf68fcdaa\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72d8f8cd5d8a45b694431366299b715b : SUCCESS in 1h 23m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93b9e250d4004926b4a6c31abf44fab8 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/010493d9e76f4af68ba80a26dfa29571 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9326b61baa394e8f83bb64ec67ddccfd : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e26fc7da1e147cdb2afedd3a5de5c9f : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b073feca2649b2afa2167602cc3bd3 : SUCCESS in 4m 12s\n- tooz-tox-py38-etcd https://zuul.opendev.org/t/openstack/build/21a94b92b3754a2e886c0d512c6bed04 : SUCCESS in 4m 34s\n- tooz-tox-py310-etcd https://zuul.opendev.org/t/openstack/build/fe93f158aabe448abde259d52f9c3c8b : SUCCESS in 4m 16s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/a49a858d6fdf4ba6b29e2b2161fcaec0 : SUCCESS in 5m 03s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/bc365b7790304d5bbb9eda8af4744982 : SUCCESS in 5m 00s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/6ea36cc597b647d3a5f921d1e3a058d2 : SUCCESS in 4m 25s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/215cf9be52f744d8ad00a78d80694beb : SUCCESS in 4m 08s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a22adddc68ee4865a32c5f519ea6e5f1 : SUCCESS in 4m 33s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/8b8c00c29d4c467391001c5ea05832b3 : SUCCESS in 4m 54s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/53829edc28b24977ad59b3d437a0701a : SUCCESS in 4m 47s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/ba8315b0cba240bc8f2549f923ca3cec : SUCCESS in 5m 02s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/64a565a6a8714f4b9f064acf9a6cd4f8 : SUCCESS in 4m 21s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/fdf12aedd96c420894180c0ac78db5cd : SUCCESS in 5m 31s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/55f9b937a57e470aa45f110b4b0d1995 : SUCCESS in 4m 11s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/1d12bdd6b4494170aedb30013c866c3d : SUCCESS in 4m 15s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/dca8b00923c541349dabdd4516d2ac98 : SUCCESS in 4m 04s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/eccd161ff9dc4999b9b56b3d5b6f9e94 : SUCCESS in 4m 17s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/c91a79e1de634269861c2449e4d22c72 : SUCCESS in 4m 00s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/654754ecaec24869b0b6232e944d782c : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d302756358ec61a96a57cd6fd764ac4dcec6dbd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:24:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"85e1f377210789ccda75c0fc150bb9f4eedbb137","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 11:25:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef690999a8c4464b9ecd51ab26cb6da9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56ed8990aee54c75a23678cf1784308a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0feead5deb5fb28ef587b4c7683204813014ac25","revisions":{"0feead5deb5fb28ef587b4c7683204813014ac25":{"kind":"REWORK","_number":1,"created":"2023-09-04 07:45:28.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/01/893601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/01/893601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/893601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/893601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/01/893601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/01/893601/1"}}},"commit":{"parents":[{"commit":"702ea408b2d5c2b2ee27d56d4eeef1ce0263a1b9","subject":"Merge \"Change StrictRedis usage to Redis\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/702ea408b2d5c2b2ee27d56d4eeef1ce0263a1b9"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-09-04 07:40:44.000000000","tz":0},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-09-04 07:40:44.000000000","tz":0},"subject":"Replace en-dash with ASCII minus in a docstring","message":"Replace en-dash with ASCII minus in a docstring\n\ncurrently Gitea at opendev.org displays a warning about\nambiguous Unicode characters in tooz/coordination.py.\n\nWhile really minor and benign, let\u0027s fix it so it does not trigger\nand confuse future viewers.\n\nChange-Id: Idbe9006e843bcedcdb98d735c49b157ab5ae6b12\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0feead5deb5fb28ef587b4c7683204813014ac25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0feead5deb5fb28ef587b4c7683204813014ac25"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
