)]}'
{"id":"openstack%2Ftooz~897561","triplet_id":"openstack%2Ftooz~master~I1fc434bd774f940de1684a9bb7dd8ce5686828b4","project":"openstack/tooz","branch":"master","topic":"k8s-lock","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2025-04-10 08:40:13.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-04-07 12:15:17.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-04-10 06:50:44.000000000","reason":"\u003cGERRIT_ACCOUNT_16137\u003e replied on the change","reason_account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},"31245":{"account":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"last_update":"2025-04-10 08:40:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1fc434bd774f940de1684a9bb7dd8ce5686828b4","subject":"Add Kubernetes Lock support","status":"MERGED","created":"2023-10-06 14:06:05.000000000","updated":"2025-04-10 08:41:11.000000000","submitted":"2025-04-10 08:40:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"897561-k8s-lock","meta_rev_id":"556d3a060d6fb3cb07a224bad9cf9d9bdc27e7e7","_number":897561,"virtual_id_number":897561,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-10 08:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":2,"date":"2025-04-07 12:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-04-10 06:50:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-04-10 06:50:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 16:44:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-13 11:18:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"CC"},{"updated":"2023-10-13 11:18:20.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2023-10-17 05:35:38.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-02-12 16:33:19.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"CC"},{"updated":"2024-11-22 13:57:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"2978772dd9788daca00fd15eeeff8e3361245129","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:06:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b2112325977af37cd6d4d00776506efebd95922","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:06:21.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"75d7e6b680bd3fd0d8a6db063ba831ace7cf186e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:21:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d9f63bf63a02c277afb31074f3ebd0972f13ffc3","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:22:04.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d033b0d3fc461ca045ab245043979b120dba0b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:37:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"aeb44a699167b67645cff570f77702cf1500ff2f","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:37:26.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"9cdc6062d2010a580dc444c2eb2b5ee0dfced40e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:57:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b258cef17a0291f6dc05afbb78f1dbbcbf1be02a","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-06 14:57:30.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"791b226639c2c7f361c86e0a1a2731134bb676c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 16:44:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/410d169fcc6c401890c6c2576e53ac35\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/62de08dfa9d84df588a64928ef0438c8 : SUCCESS in 2m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a2b215c4a034fda95a3db4bbd9a808f : SUCCESS in 1h 41m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/613b888f85104ab4a88b546c6b414c42 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/580af45a1c444df29c2bf9ba1b3a156d : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23baa4cb1ed446a88666fac5a8181e92 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac6af523f73e4b8684e323ebf88e5014 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85ee06698d7e4176bfadb3b226ce47b8 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f10ee87a3a0d489d8789e9acf47fe2b6 : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d4b798cd824b47bf27b0d41fd0e673 : FAILURE in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b42af9ff5df749438d88d7d5b9912ce3 : SUCCESS in 4m 00s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/629d2311206c41f498abd7e8f0b3d8db : SUCCESS in 4m 50s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/bfe21c79a9ca4048b793d327923ea334 : SUCCESS in 5m 18s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/3126c9358d10467c845c26e76eb80bfd : SUCCESS in 4m 22s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4d6f62e9670a44e7997de551a324988f : SUCCESS in 4m 13s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/0087bb15904d4e68b572f98a72399456 : SUCCESS in 5m 33s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0390fac7c7734335837efd941a2efc4b : SUCCESS in 4m 57s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/4f29da7a734543ceab8a6a1d3267e7c3 : SUCCESS in 5m 35s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/bee92637581f44fc8d67fe8efcb96355 : SUCCESS in 4m 37s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/241727dad4a64ff39cdfbef44a33b5d6 : SUCCESS in 4m 27s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/2b29b3eb6a6a4d07a2c66dc6ae5563a9 : SUCCESS in 4m 22s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/048d9194b84e41c8b247a59184a410ef : SUCCESS in 4m 37s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/2e07d8d9cb7249b3958561eeb83ae338 : SUCCESS in 4m 27s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/213530893b544c6b9c69ab1fce8b50e8 : SUCCESS in 4m 49s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/b22aa8fd14334dcab5867d3f64fb309c : SUCCESS in 4m 21s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/404d508f4b6d4d889fc0a27725712242 : SUCCESS in 5m 30s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/6035779fa2324ebf89bffcb7c1201049 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":4},{"id":"7c7fd8238fd50b8d8e8da6165489a33cb7981231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-11 06:02:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"08e14657d507793f9d332241b7a6f26cd2e2d24a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 07:25:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13f326020af4491985b80d0d27129124\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/331378785eb8445398bf783b8e73863c : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f7678f390dc24469b272aec7371d51e8 : SUCCESS in 1h 21m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8118eed904b480d88fefafe9ccb0f80 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe1ca5d0873452faed5206db320c3b9 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2716d1ab0e0f4da999bc0c261fc0fbec : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8e95879e3774fbda7b152e97bd69620 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8463c47ee63943f1b8e7201e7da649f2 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a049da7a938740cf9870101576089ace : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf615d5722094ca3ba72440173da8511 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69e6efa98f73449b833b863ef134c0f3 : SUCCESS in 4m 33s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4fe651e05b854872999887d610f9aec8 : SUCCESS in 4m 57s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/cab4ece75aa54e6f90594499bae03db4 : SUCCESS in 4m 31s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/f9a5788856de4d8180757d1da012120b : SUCCESS in 4m 20s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/8dc9e75a1d6e4bf989648a9037584fe9 : SUCCESS in 4m 02s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/2d5aad9a750546c497dcb7b8b584764d : SUCCESS in 3m 50s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2850c8a0710243e48aa1c5184164ec53 : SUCCESS in 4m 19s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/74e08865a20d41b69669a9f470da9174 : SUCCESS in 4m 17s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/53f57c36d6b6436594fc9a721e2c9fdf : SUCCESS in 4m 23s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/262cfc2f9b9b4e10afdb091af1b96983 : SUCCESS in 4m 04s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/3559e630bf064cd38afd82dda3435290 : SUCCESS in 4m 18s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a7d25b0c5f724a27bc044e1a436ab291 : SUCCESS in 4m 20s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/eb8570c7f1c0441dbc25f304fd126eea : SUCCESS in 4m 06s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/c1aa29b563c5443b9ba5fdab5293432f : SUCCESS in 3m 53s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/cc1b38b0b8f840869b1a8091a5f4a40e : SUCCESS in 4m 02s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/74880be8dbe54ad4ac213264c59c0422 : SUCCESS in 4m 20s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1383f87a15114cf7b54921025162187d : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":5},{"id":"a3f652c0c294bfa5b72cd1259b19976b6b8d7d01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-11 10:06:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ff7909c4b39a9f3f4e3863e4a162e62f34e4182a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:42:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22db6ef3e7534067a837cc449ce7f4e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2f0961be3e643aebd37353bf5f3c200 : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81a5b7890050452a9d50d926543f1dce : SUCCESS in 1h 30m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7034acdcc4b5417b80c4f2154410d191 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abfd53db111d43a3b7b860b526d5bd0a : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95f7cae055574f60b73a1861b9bafd33 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d89996c24964b50acde67574c6860c3 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31c89d44dd934e618c0107e8b66e9cbf : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/014f572fac1c4387b6621d9b556c8aea : RETRY_LIMIT in 1m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f4b8ed8b614179beb2d7cab6f424b4 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ec7e8013f2f48c4824911bc19424a8b : SUCCESS in 4m 32s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/d1b88f6d31074c4e9106ccef1483d0f9 : SUCCESS in 4m 58s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/5dedf00b8d644d2eb6ba71141d47605f : SUCCESS in 5m 35s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b9ce646081cb47f5b9d51ad0cca58b26 : SUCCESS in 4m 47s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/b2f41cb027b04ca6ad188efa8fe42a98 : SUCCESS in 5m 01s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/18449484e1d94ef582d936f0e309ed7d : SUCCESS in 4m 46s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/5607d978909f46928556f120d2c4ec67 : SUCCESS in 5m 23s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/82d04d036dfb408b9fdcd9f8f977c7dc : SUCCESS in 5m 36s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/283bb48b64a6457198757964f7c0c9d4 : SUCCESS in 4m 58s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/54a21841e62d4f0cb9c14135e352728f : SUCCESS in 5m 44s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/187b75f0974a4e32a726436d3d889466 : SUCCESS in 5m 28s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/b68502b27f9f4565b990ccee941aa3da : SUCCESS in 5m 32s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/1927ca7befd84c4c949f72ab5740f80d : SUCCESS in 5m 45s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/7a51659747504d2d87df7f4157c2be76 : SUCCESS in 5m 28s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/fc391202f64f41b68ac1b190a11a4fc5 : SUCCESS in 5m 37s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/59cbad167296420d969e51600c9f9e8b : SUCCESS in 4m 43s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/e1bd9f55714448bb8199da486fedecee : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":6},{"id":"efd6000e50d421da0019d84eceb91ea339fe920e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-13 11:18:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d5b4cf4a82d3d0f4fb11e0d4b2a59ca4e8a0cae4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-13 11:18:20.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f23401c2e159b8583afc649f8c1efaed05a74888","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-17 05:35:38.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"a1073b71e84bf8d86bc130300a12473dc32048d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-17 05:41:03.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"ca487b0beb9261a4095c4a3f4b90c5b21587725a","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-17 05:43:40.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"7d376020fbfbbd5a1a8f9efdc7689fe78074c28d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 07:27:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a603e775a2a4363bc0da985d2e2c561\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e50170b37bfc4734b3e382fffaa57210 : SUCCESS in 1m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8508209762b49fc806b18243c92d86b : SUCCESS in 1h 40m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71ad6ee16db04478804ed7ddc0803cff : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892a954defe54a6aba6c179123ea33db : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a0258e14f1949adb8faf20df6ba16df : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cbab0c192794b8bafdd935a0bef49ed : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d86104d573e04b9e96e22a6f626c18bf : FAILURE in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd6eb92af251411197b1c035ec8bbf9f : RETRY_LIMIT in 1m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8379b182814b96982750d70ebea65d : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e64461a7e494f0a9f3315711527e832 : SUCCESS in 4m 45s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/6d039aa0cf0d4d31a4d33e44bbb3052e : SUCCESS in 4m 32s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/5966a0a45f89414bba6c1aebc6d554b4 : SUCCESS in 5m 38s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/da4a2173eb72468580dfc4235379193f : SUCCESS in 4m 07s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/d6e3453fb39d4acaa45350ab1e941fc4 : SUCCESS in 4m 54s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d83c7b824df3405194ec91da8d43aeb8 : SUCCESS in 4m 34s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/1582f448e5d24d7a8d9f057f32936086 : SUCCESS in 4m 46s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/0fc1c2b82bb1444c85e42be46c4857b2 : SUCCESS in 4m 38s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/edeaf02f33d64697b8e736da76ffb5cc : SUCCESS in 4m 39s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ac023d1ce5564860942bbbde06ff278b : SUCCESS in 4m 25s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/8e5dc3542a3e4371b1e7cce62c3a1c4f : SUCCESS in 4m 39s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/94fb823e82d94c4790a1a3c3ad1a4f6d : SUCCESS in 4m 44s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/6a34edb15da04e3aa47bbf745928c4e4 : SUCCESS in 4m 45s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/19338629164448ae80e4faa549fe03f6 : SUCCESS in 4m 35s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/15cf838d9ede4b87b2f065407b0f6640 : SUCCESS in 4m 48s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/fc0261e980d948c991dc24583f3127a8 : SUCCESS in 4m 39s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/a18e61fad79f4fe198e3d4d92009fa50 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":7},{"id":"e929872bb2ea67ad075d9bb1193b11185aa2de59","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-17 08:37:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1ba32235010991dffd12f2f43f5d1b154726d75d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 09:53:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecacd8fc10d74500bf7c62f350328156\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8e05b893ed844063a26ec6fabc29b132 : SUCCESS in 1m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/523981a3ad184c75bed6c2cf6a903c51 : SUCCESS in 1h 14m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca1a27d306cb4609b29c5263cb0a6efe : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4df216d24ba41bfaad9df4c41372bb9 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d89a2cb04a7e4bdd9710bb68fad5d1bc : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98adebf539e147228c9634c5999ff352 : SUCCESS in 7m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08498abdac874e19a67433bb4eed30f0 : SUCCESS in 6m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a39029564a54ab0b7ab44f9ea808852 : RETRY_LIMIT in 1m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c80b2fed214ecaa36d1084845b3905 : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd9296dd491d470dbcb457943ab2bb36 : SUCCESS in 3m 54s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/227339896b404f6f87153034076886a0 : SUCCESS in 4m 46s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/0cf2ff1aea24460a8f3d118c4210b1bb : SUCCESS in 4m 39s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c88dee1f64ff49aa868f5861c5ee04ef : SUCCESS in 4m 48s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/1a0172d8fcfa442f95e881501d413a5b : SUCCESS in 5m 55s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ecae5a09600b48438e6c9ad0394db4c9 : SUCCESS in 4m 12s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/f25bceca3b884a3cbf0307f3d39db613 : SUCCESS in 4m 39s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/3a28801e1b484623a0febed5cb202192 : SUCCESS in 4m 33s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/01523055dbb64c93ab25e2f32f10649f : SUCCESS in 6m 28s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ac49c880738f4fd594e7585a0f42be0e : SUCCESS in 4m 51s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/b94ebc5d9c484a12bc18d3593d6735b3 : SUCCESS in 4m 33s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/c05c45146b1842fcaa40143e7b66615d : SUCCESS in 4m 10s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/28956ab4dc6c49cd9c9a7b503df74eb7 : SUCCESS in 4m 35s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/f232e5227d9942d18e474505c976db1b : SUCCESS in 5m 46s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d23ab73f07344d5bb44e5d9fab28f733 : SUCCESS in 4m 40s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/0a6173e121784746ba276bf46f7c54c3 : SUCCESS in 4m 17s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/338533c167ba42f6a7ec41316d4edb69 : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":7},{"id":"726a9353e204c34847a584880826eec0a81f44a4","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-26 05:45:45.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b04e01d1fa9cd85be57f10d0761d92c8a6b90668","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-02-12 16:33:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"eccae4f4afaaf40b00ad210a26cbe62eda5a95f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-05 08:06:01.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2a374bab6148df604ecf20e167406213e4af529d","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-05 08:06:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b7d9753b547a8ec1dcd264586ebecf52a15fcb2c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-05 08:24:25.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"03df204fbc62931c63bc9481592da423572c3f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 09:41:10.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0734ef7c035c4680807858c859b6ab06\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3767d7bafe9b4f76a8534c4c051dfaa2 : SUCCESS in 1m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/911ada68a29b43f3ac4084df9e5d484e : SUCCESS in 1h 34m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49a9a9aa00334348abe243bc6e02a94f : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606c60914bf941d4ae84d238662aafc4 : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c829f0ae7d84b65aa7705404512f1ff : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55c06963327345c4b7aa2f626d6ac92f : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bd2e21c5d0d480d88be394485f2588f : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb2c2f08d17f4f43918262c7d5d066b3 : SUCCESS in 3m 23s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/87d1dd3a22404c96863223fccd2259b2 : SUCCESS in 4m 34s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2a1b4def47234c7695779a8e8da856e5 : SUCCESS in 5m 53s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/e4ff2a55d5bf44429e01815af351acf4 : SUCCESS in 4m 20s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/e5a5989069be46858d44214940846af3 : SUCCESS in 4m 12s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/d756127ee4534170a6681b8ad1fb4c15 : SUCCESS in 6m 04s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/dafa891a647545968e3b564c3db567cc : SUCCESS in 4m 19s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/215e92790fa448f6b8d5038926afdc53 : SUCCESS in 3m 53s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/6281b1b052734275964e6291cfe202f3 : SUCCESS in 4m 04s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/acaffe1d3e8a48be98e683f7f6d9a956 : SUCCESS in 4m 14s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/1857d12530794a55965873fc3f7a4e49 : SUCCESS in 3m 42s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/4ba9422058ef426f9ad527769eff7599 : SUCCESS in 4m 12s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/e807ffa7d0a34007a6336d99b51780a9 : SUCCESS in 3m 58s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/62d9e60d47954a15baff70635929a6c2 : SUCCESS in 4m 17s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/e76eaf63c20e452ab2b6eba9180c31d4 : SUCCESS in 3m 45s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/f8e491f47d2648f1acdf80ace4e76811 : SUCCESS in 4m 15s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/789281ad5d8941fbbb9a0ec33e8c118d : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":8},{"id":"67a68d95616bce51d903dcddf04d3537fc313df2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-17 17:56:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4d8d66f2f05581760f82a1134a38a3560b856227","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-08-15 10:08:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e9ab27159c68b57d1fe37781e2a66d767a165789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-11-20 02:45:57.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"42d54b4b2b1926ba2a51f339a8c3190aff343021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 02:59:40.000000000","message":"Patch Set 9: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/935735 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"34c57eb9364b3a7d42f1bd02dd38652c4370dcc5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-22 13:57:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"7914cf7c70466c7c75cb046f36a68bf7e9520d97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 15:30:41.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1b5866798254899b5603440591a3601\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8bb1348c6cbd44c0a369ac045e16a55c : SUCCESS in 1m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9eada433aee244198d66ada390bfbace : SUCCESS in 2m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7766ea098f1a4111a629c5c1887dc3a4 : SUCCESS in 1h 30m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/182b5cd17d2842568b8f9b981836135f : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f63702ab9d8400db2f6007064b465bf : FAILURE in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc54ccb2941b4024a5d3a74ca6c4ac47 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eea1bccaae554b67859d7b95099c4ddb : FAILURE in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5e874298c94b03a8bd9ffa19f38d88 : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8f87d8c34164e24abf7ed87bf3e1a11 : SUCCESS in 3m 21s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/bf131e2bdf61442e8c90b29cfd6a57c1 : SUCCESS in 4m 51s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/7583665c41a14d7785fd4d21287b15ef : FAILURE in 5m 39s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/dd3ffa2d606641c482f010ae584f812f : FAILURE in 4m 14s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/0f6fe94a4443426591702673a3e0d286 : FAILURE in 4m 54s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/3b2baae66c384a8e9790aa89d524cb06 : SUCCESS in 4m 11s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/c05cf799094548b0b8d3657f17a54840 : FAILURE in 5m 06s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/1f9defeadf384670b96fe9279cedcb5d : SUCCESS in 4m 25s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/1bfaa88c29ba4316877152b9365a2d60 : FAILURE in 3m 34s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/892cff8c24e540c9bf498adb42477ca7 : SUCCESS in 4m 05s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/b8f9fb59fc21481fbc63ead914a3cf0c : FAILURE in 4m 35s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/6d9a8762f8174cfc89dd928b95a60a7d : SUCCESS in 4m 22s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/dc864271115649338958285d7fb1a1b7 : FAILURE in 5m 07s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/6986713d46474b6ebc252035522c4188 : SUCCESS in 4m 11s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/a45379dfe9ca4383b6be46343ec03223 : FAILURE in 5m 09s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/794f1a30252649dd85fce3ec7249e266 : SUCCESS in 4m 31s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/68d662661c1e4e38a31205cc7eba3e70 : FAILURE in 5m 33s","accounts_in_message":[],"_revision_number":9},{"id":"90f933db8e5ee871899b9976becae2dcec3617ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-12-13 09:00:09.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"dccf6f10560049dbc5fb0712aa0c9a05d242d9a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 10:43:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9f16af8a1874dc5826e78fa4abcfb13\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5982f449ebf04f9e9aa994d3fca0f7ca : SUCCESS in 1m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1e92b96684324c0ab2f82d0bb4c186bd : SUCCESS in 3m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a61347fd23de466aa57b284f9be4e806 : SUCCESS in 1h 42m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da89864493bb46a0bcb4c1cecce3d54a : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774b522f79bb4439a550df56a1e6347b : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bf363640b704b16bc519d47da6b5753 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d285271082d48239c7fb22a42daed1b : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/439fc3182f6e4ecb82abc14102e60ee2 : SUCCESS in 2m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79f870a3ac6f42ea9a4dadade1161b11 : SUCCESS in 3m 10s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/eae6dcf8665d4d57aa3107b5088b4028 : SUCCESS in 5m 08s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/b93a5cc709e9452ca6cee044faae53be : SUCCESS in 5m 05s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/e45bd0da40334abeb38079b6dc2e77ae : SUCCESS in 4m 37s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/98bbaf16e6e54368abde4a06835181ad : TIMED_OUT in 41m 01s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/006da661c4d647c9b5717927bddd3fd1 : SUCCESS in 4m 36s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/3582f11a10d44db9b1e469c99f6f0211 : SUCCESS in 3m 37s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/c8341b93ff9240a599e53cae1223899e : SUCCESS in 6m 09s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/bc16515786484860a3e20ffcb0653ae1 : SUCCESS in 5m 08s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/d78c0f65a165412e988b020bd23b9be4 : SUCCESS in 4m 09s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/2240f11942a04059abb17dfd1fdf091c : SUCCESS in 5m 04s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/d441ac51f8324723856d8eb55aee9b6f : SUCCESS in 4m 27s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/cd40995baade4665982ee5b6266ff1b6 : SUCCESS in 5m 03s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/58bab3c66ac546758d00eaa33022319b : SUCCESS in 4m 20s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/2f313f42f9ae4a3aabfe7e90bbaa9922 : SUCCESS in 3m 46s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/ff4875d5c0744f05a4e2a0712801fe5d : SUCCESS in 5m 28s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/65ca5ba961a04f6481402a654643c35c : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":10},{"id":"409c31a15805d6e5cde5150cf7b0a16f046db7f0","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-12-13 15:25:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7993532fd43eddbc40abc4cdbf904322be64ab48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 17:09:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2725c76151c64749aa158d7ea2ebf8dc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cf36c46616f744efa9056719c0406c4b : SUCCESS in 1m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f0f5cc283f6a4a4187dfc65b8f63d197 : SUCCESS in 4m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82893994d030488fb3b7ccb9c25575cc : SUCCESS in 1h 43m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84c072f353a44cfd8152ae9d6d27d377 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dede585af8984403ac45fc1c39c42ec1 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f3928e367d74900aff21e08f79128e9 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d17eb887b174519957d0c337e3b0603 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7dba7a267874082a5d85f363a2dee4d : SUCCESS in 4m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29511a678d764702bc0d87217f0bd2f2 : SUCCESS in 3m 29s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/af1cba562b1c4278b087d6082bf39fc5 : SUCCESS in 6m 34s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/39438b19413d4bdaab6cd61c000b7830 : SUCCESS in 5m 44s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/442c0b6d8ce144fc9b33ee6a911aaf67 : SUCCESS in 4m 09s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/cf432c8a42894e2d999c98fe6a4cc56c : SUCCESS in 7m 11s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/9108a27e061f4e27a5c1876a7bbc13a0 : SUCCESS in 4m 16s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/f627523fda0d4b068d09671738104eab : SUCCESS in 6m 41s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/d709ad2de0394a5f99db6c3f109fbd76 : SUCCESS in 4m 51s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/412d2b987166433babac20860276411a : SUCCESS in 7m 04s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/0177dc7f47884a68b90ed002fe5387cc : SUCCESS in 5m 03s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/da0171ff8b3c4fc6baa878e84a690618 : SUCCESS in 5m 29s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/e3bbe2b853294d85a7ec137ce44a361b : SUCCESS in 5m 06s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/bdb90d06f8e74e44b164c059268d8d61 : SUCCESS in 5m 36s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/f12f1581b6d54357b91e52f64827dde6 : SUCCESS in 4m 15s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/2fef799e62f446df930f813d25ce895c : SUCCESS in 5m 53s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/0f00d15fa77940feb0e4ef759de617cb : SUCCESS in 3m 48s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/16e7ff2a00bd4be5ae711058c631b7a7 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":10},{"id":"d9e0cfa7f55c59708b61a645f5555633e360d7e1","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-12-16 09:15:47.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"52bd717e6d25456da90ff1666b4efd62959deac4","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-04-07 06:05:50.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"af82eb1fb223469b22f68e02d909c52b82370da1","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-04-07 06:09:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"679132cb6864ad94c556b076d4aa1133c7004370","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-07 12:15:17.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"826781fe8708ed50b56519e33054f1b66e08597b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-04-10 06:50:44.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"5c2bd95973caf30d30eb172851774ca7dc69efa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 06:51:25.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"52d5110207aa0c8c175638ed2e4e8345e2edb9da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 08:40:12.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e0ab518da7b490fa136fec03877d890\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/20afd14f0f4341939336b502b6087280 : SUCCESS in 2m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8de48c892a1a462c952d34d7fc1c9239 : SUCCESS in 2m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e7bbd1150d104924a451df36d04d999d : SUCCESS in 1h 44m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07b30d7170b46ab910dd4594671cc02 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4df461bab5864046b90db54b9dbe0d0b : SUCCESS in 3m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1bb23a40e4545d9b98357c42c22bde3 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d25a76474e4f74b51dc42131ca8ebd : SUCCESS in 4m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5dc51d5d06a49729ed6ad25deb437b6 : SUCCESS in 3m 00s\n- tooz-tox-py39-etcd3gw https://zuul.opendev.org/t/openstack/build/170ebd1d9935443aa8ddf3549ab4dad6 : SUCCESS in 4m 23s\n- tooz-tox-py312-etcd3gw https://zuul.opendev.org/t/openstack/build/96029b17294848beaa5fca79547d1210 : SUCCESS in 5m 19s\n- tooz-tox-py39-zookeeper https://zuul.opendev.org/t/openstack/build/f8dc90ad456343d9ab7886de9ad1e73c : SUCCESS in 4m 40s\n- tooz-tox-py312-zookeeper https://zuul.opendev.org/t/openstack/build/c450b907dae74595bf1910d5692d4f8e : SUCCESS in 5m 06s\n- tooz-tox-py39-redis https://zuul.opendev.org/t/openstack/build/fe2e2515b2594a75bef9cc361aa590d0 : SUCCESS in 5m 13s\n- tooz-tox-py312-redis https://zuul.opendev.org/t/openstack/build/a95f33adff1c4ae8a6ac807caecd1843 : SUCCESS in 5m 26s\n- tooz-tox-py39-sentinel https://zuul.opendev.org/t/openstack/build/936cd36629ea4481a660a58c46a75f65 : SUCCESS in 2m 20s\n- tooz-tox-py312-sentinel https://zuul.opendev.org/t/openstack/build/1c143e3e74714abebbd9d0a9c8175b53 : SUCCESS in 5m 44s\n- tooz-tox-py39-memcached https://zuul.opendev.org/t/openstack/build/7d15434ff16d4bfc885d86d16e7d8a43 : SUCCESS in 2m 30s\n- tooz-tox-py312-memcached https://zuul.opendev.org/t/openstack/build/90f46050fdc5431c8a0e8a3f74cb206a : SUCCESS in 5m 14s\n- tooz-tox-py39-postgresql https://zuul.opendev.org/t/openstack/build/744959f3ccbf457584831d2c325b3e74 : SUCCESS in 2m 25s\n- tooz-tox-py312-postgresql https://zuul.opendev.org/t/openstack/build/335bd83260a245ef92f1c3f65ee30a81 : SUCCESS in 5m 20s\n- tooz-tox-py39-mysql https://zuul.opendev.org/t/openstack/build/b00b3ff8280f49afa447bf5514168d16 : SUCCESS in 2m 27s\n- tooz-tox-py312-mysql https://zuul.opendev.org/t/openstack/build/eb1bda64e594458298c349abdb85838f : SUCCESS in 5m 18s\n- tooz-tox-py39-consul https://zuul.opendev.org/t/openstack/build/5ed47a2b596042d4be25a5fe8d038572 : SUCCESS in 2m 29s\n- tooz-tox-py312-consul https://zuul.opendev.org/t/openstack/build/63825be7fcd44f148b9caf9ca3f6b722 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":10},{"id":"f99021196b2e18d0680d02121a64909b6ff89642","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 08:40:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"556d3a060d6fb3cb07a224bad9cf9d9bdc27e7e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 08:41:11.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47a83202b35b4f4984c7bba47781c73d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4257d2112945cda2949ba228010462 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aac79c2765ba4dd2bab572bc465b4c73 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"7b69ca7270714ef125aeeeefe4ba39b021c8a823","revisions":{"ee7ce7364be22ee42b894b659f1c9669704d2eba":{"kind":"REWORK","_number":1,"created":"2023-10-06 14:06:05.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/1"}}},"commit":{"parents":[{"commit":"738931d026aedb846ed3970cf642a6c570a2bceb","subject":"Update the default etcd3gw endpoint to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/738931d026aedb846ed3970cf642a6c570a2bceb"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"subject":"Init Kubernetes Lock","message":"Init Kubernetes Lock\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ee7ce7364be22ee42b894b659f1c9669704d2eba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ee7ce7364be22ee42b894b659f1c9669704d2eba"}]},"branch":"refs/heads/master"},"c4f6176369fa6cd239efad35ddadd145b6bcb7b0":{"kind":"REWORK","_number":2,"created":"2023-10-06 14:21:54.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/2"}}},"commit":{"parents":[{"commit":"738931d026aedb846ed3970cf642a6c570a2bceb","subject":"Update the default etcd3gw endpoint to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/738931d026aedb846ed3970cf642a6c570a2bceb"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:21:45.000000000","tz":480},"subject":"Init Kubernetes Lock","message":"Init Kubernetes Lock\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c4f6176369fa6cd239efad35ddadd145b6bcb7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c4f6176369fa6cd239efad35ddadd145b6bcb7b0"}]},"branch":"refs/heads/master"},"c3100f991162b74cc60a815bbb9f437fc218717f":{"kind":"REWORK","_number":3,"created":"2023-10-06 14:37:16.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/3"}}},"commit":{"parents":[{"commit":"738931d026aedb846ed3970cf642a6c570a2bceb","subject":"Update the default etcd3gw endpoint to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/738931d026aedb846ed3970cf642a6c570a2bceb"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:36:57.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c3100f991162b74cc60a815bbb9f437fc218717f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c3100f991162b74cc60a815bbb9f437fc218717f"}]},"branch":"refs/heads/master"},"0621a6bc459c250735a56aa381482299587fb8c1":{"kind":"REWORK","_number":4,"created":"2023-10-06 14:57:18.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/4"}}},"commit":{"parents":[{"commit":"738931d026aedb846ed3970cf642a6c570a2bceb","subject":"Update the default etcd3gw endpoint to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/738931d026aedb846ed3970cf642a6c570a2bceb"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:57:10.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0621a6bc459c250735a56aa381482299587fb8c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0621a6bc459c250735a56aa381482299587fb8c1"}]},"branch":"refs/heads/master"},"821b1f9688bfda0392b5a28309f4e951c225b54b":{"kind":"REWORK","_number":5,"created":"2023-10-11 06:02:16.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/5"}}},"commit":{"parents":[{"commit":"738931d026aedb846ed3970cf642a6c570a2bceb","subject":"Update the default etcd3gw endpoint to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/738931d026aedb846ed3970cf642a6c570a2bceb"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-11 06:02:05.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/821b1f9688bfda0392b5a28309f4e951c225b54b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/821b1f9688bfda0392b5a28309f4e951c225b54b"}]},"branch":"refs/heads/master"},"ef396d8b1aa8bccbd7978a014b90f51ff9e5d633":{"kind":"REWORK","_number":6,"created":"2023-10-11 10:06:50.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/6"}}},"commit":{"parents":[{"commit":"738931d026aedb846ed3970cf642a6c570a2bceb","subject":"Update the default etcd3gw endpoint to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/738931d026aedb846ed3970cf642a6c570a2bceb"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-11 10:06:28.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ef396d8b1aa8bccbd7978a014b90f51ff9e5d633"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ef396d8b1aa8bccbd7978a014b90f51ff9e5d633"}]},"branch":"refs/heads/master"},"820aba12356a4e1e2be10d87ffb4dc914bf298fc":{"kind":"REWORK","_number":7,"created":"2023-10-17 05:41:03.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/7"}}},"commit":{"parents":[{"commit":"738931d026aedb846ed3970cf642a6c570a2bceb","subject":"Update the default etcd3gw endpoint to v3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/738931d026aedb846ed3970cf642a6c570a2bceb"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-17 05:40:26.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/820aba12356a4e1e2be10d87ffb4dc914bf298fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/820aba12356a4e1e2be10d87ffb4dc914bf298fc"}]},"branch":"refs/heads/master"},"7e934b90a410efaec22cf843fea43d250f4b05fb":{"kind":"REWORK","_number":8,"created":"2024-03-05 08:06:01.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/8"}}},"commit":{"parents":[{"commit":"1d0ece27a4f8b0d3aed13b603fd834eba3567183","subject":"Merge \"redis: Fix wrong type used to parse socket_keepalive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1d0ece27a4f8b0d3aed13b603fd834eba3567183"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-03-05 08:05:49.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7e934b90a410efaec22cf843fea43d250f4b05fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7e934b90a410efaec22cf843fea43d250f4b05fb"}]},"branch":"refs/heads/master"},"d18976458e8b6e21da88e0984db49e1c7b9c52c4":{"kind":"REWORK","_number":9,"created":"2024-11-20 02:45:57.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/9"}}},"commit":{"parents":[{"commit":"afb5c8bf12e7ed32f5a7775a242b8df9664b57f9","subject":"Remove Windows support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/afb5c8bf12e7ed32f5a7775a242b8df9664b57f9"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-11-20 02:45:18.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/935735\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d18976458e8b6e21da88e0984db49e1c7b9c52c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/d18976458e8b6e21da88e0984db49e1c7b9c52c4"}]},"branch":"refs/heads/master"},"7b69ca7270714ef125aeeeefe4ba39b021c8a823":{"kind":"REWORK","_number":10,"created":"2024-12-13 09:00:09.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/61/897561/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/61/897561/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/61/897561/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/61/897561/10"}}},"commit":{"parents":[{"commit":"afb5c8bf12e7ed32f5a7775a242b8df9664b57f9","subject":"Remove Windows support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/afb5c8bf12e7ed32f5a7775a242b8df9664b57f9"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-06 14:04:04.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-12-13 08:59:58.000000000","tz":480},"subject":"Add Kubernetes Lock support","message":"Add Kubernetes Lock support\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/935735\n\nChange-Id: I1fc434bd774f940de1684a9bb7dd8ce5686828b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7b69ca7270714ef125aeeeefe4ba39b021c8a823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/7b69ca7270714ef125aeeeefe4ba39b021c8a823"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
