)]}'
{"id":"openstack%2Ftooz~907652","triplet_id":"openstack%2Ftooz~master~Ibb4d26022864ece4abf91fd2d2fd665998d80bd8","project":"openstack/tooz","branch":"master","topic":"bug/2052353","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-13 05:27:01.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"28522":{"account":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"last_update":"2024-02-13 08:11:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibb4d26022864ece4abf91fd2d2fd665998d80bd8","subject":"redis: Fix parsing of sentinel fallbacks containing IPv6 address","status":"MERGED","created":"2024-02-03 16:53:31.000000000","updated":"2024-02-13 08:12:18.000000000","submitted":"2024-02-13 08:11:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"907652-bug/2052353","meta_rev_id":"eab3e93ff9a66b63c8e730933ecb932314c721d7","_number":907652,"virtual_id_number":907652,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-13 08:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-02-06 10:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-02-12 15:42:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-12 16:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2024-02-13 00:34:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-03 18:21:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 10:10:34.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-02-12 15:42:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-02-12 16:30:45.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"b68743bdb2fefd744b3d68fd47f7deda3ad7ce79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-03 16:53:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d33c42cf03e2be337d559f0c677ccf098f1dd0fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-03 16:57:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"44107c984ff9ca3e496090023c4b3073eacc3055","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-03 18:21:11.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51017ccc33ef44b4afaf53f56cbe1cdf\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05b5f0716d5541b0abf27cd61ba53b08 : SUCCESS in 1h 23m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b136b1b3ce442558f9f35c252893b44 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdd83fc7cf14d31a074aa1a85e311bc : FAILURE in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/433f680d03b24f3190a23dd2f834bc37 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e589b358c6bf4a13a302a2c4007ffcb5 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcabf15014834e90a7dbfb7be2db38fb : SUCCESS in 4m 20s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/2bf26c71325a41959121070af55af5e1 : SUCCESS in 4m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/42738b49394842d9b8ae9e8a9c1cbf74 : SUCCESS in 4m 18s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/d510d51544054c9391d74117efcaca46 : SUCCESS in 4m 05s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/acd217b8371543debd440043c1dc1463 : SUCCESS in 4m 21s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/31eebf9b20d54aa1bf32dd55a8c8be08 : SUCCESS in 4m 34s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/b4c5273838b74e42979301768a03ecc0 : SUCCESS in 4m 14s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/39fddb4f87594f3fb6108aae139d7ce6 : SUCCESS in 4m 46s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/716c3ff886234c65b4ddf54e1f23aa92 : SUCCESS in 3m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/fc11679b97bf4a5ab11f160410800d58 : SUCCESS in 4m 46s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/ef8aa011d345488e80c21d172386df8b : SUCCESS in 4m 48s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/97fe0e2eda4f413c904220981c1a0cbe : SUCCESS in 4m 00s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/9d091f2a799a4104b7ee71ac7ad7fdea : SUCCESS in 6m 34s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bccde29952944d8ca679875be7141bee : SUCCESS in 4m 41s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/fd545049fcd84edf905f120f30e04f21 : SUCCESS in 5m 03s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/56aa91670c2243aba910c8f2cbd28208 : SUCCESS in 3m 56s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/4e3f7dea6bc843cc820b291dc04e6b30 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"35d410815b6d62eb1aae3c7f1b82ed31d7bc4d81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 08:10:48.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"918ca3e42461d88b21a653ad28bbbf91e3113e98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 08:11:32.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"999808e4f7f594a319c7e120ea4cd7d5151a05a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 09:56:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb4893713aab465791b2499b677d3cb5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcc3067e0acf401191d0888764b06f90 : SUCCESS in 1h 13m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c030bb8bb55b4a02adce3e859ffb6300 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6cbb2fe50b4449b15c7f38b5140260 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b14a893845eb4f0796d68cab04fe289e : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9873212f21504c51973d47eb0c5db8a4 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33d84623a964627abc05d08aa129ce2 : SUCCESS in 4m 05s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/0347c333203441cabf5385cfa13bd5e7 : SUCCESS in 4m 40s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/d9221360eed042cd86543f79ccb22014 : SUCCESS in 4m 34s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c0e634ea976a499982864924528f3f99 : SUCCESS in 4m 27s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/910410f4a5db4e4493afedc437b6394d : SUCCESS in 4m 21s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/207dab0ae73f412e8ca6ced471637d75 : SUCCESS in 4m 38s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/97a11fa879fd42089947337d72ea1ed4 : SUCCESS in 5m 55s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/65ab32522d8c4c92a6fbbb6cd56bb648 : SUCCESS in 4m 19s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/1b3c7684ef9143c8aa7a31fa194c66e2 : SUCCESS in 4m 22s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/ab08206b59704b7f8e46ac351542cd65 : SUCCESS in 4m 02s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/15083a2e4d3f4686bb86974b4cca2e61 : SUCCESS in 5m 44s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/e5588592f7c44986ac3a7a2d9011f942 : SUCCESS in 4m 20s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/13bce09f9eae47aa9aed04bc4eaa3a0d : SUCCESS in 4m 11s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/8cc58eb9f1c44607a796b79ee0296642 : SUCCESS in 4m 18s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/330e8c43f57246298296312a8583b926 : SUCCESS in 3m 57s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/7c00156540e64d0da6ba32378d6ffee2 : SUCCESS in 5m 58s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/9a7e601d8ea54eef944197a0e664a652 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":4},{"id":"92a91f4fbf19fd61a18540066f1d164cff8dc91c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-06 10:10:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e3d588de2f04a463f6c79d306c091eab9bff1e27","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-06 10:10:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a8497539cc72bafcac65ce0874f1d6dead386c25","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-02-12 15:42:53.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"aaca4fce6e715043f325283043c1683cbc0ec18b","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-02-12 16:30:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b9a83222ddb16f6bd53167f83183468859ffc11a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-13 00:34:52.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2fb4d819b11580307f5379ce3b15b8fffa2d96ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 00:35:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a1d85985646f0ab18cba092e9efc5bd99097f368","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 02:07:40.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c79a3462045f4b809e1e748000ebefcc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/156e0d9f930b4912afe8d066246c3768 : SUCCESS in 1h 31m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038aff8a9eb747828c49b9ada1a673b8 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/965c34c4ce0541c8b08f3ff15852eeb1 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/934d0c6680014ce0b1cf11bebf255cd2 : FAILURE in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88d45da076d047dd9fb3e7aab3939305 : SUCCESS in 3m 43s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/3ac70dee785b479ebeb603f2bac31826 : SUCCESS in 4m 48s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/5e2741ad3d41421aa321b3efef569a64 : SUCCESS in 5m 47s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/8587c93144424681932d2215a1de0f12 : SUCCESS in 4m 17s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/109df0ae58204e4fbb237e2811104c29 : SUCCESS in 4m 58s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/5fe1e452b97d457eb54c5ce36a78bfd9 : SUCCESS in 4m 10s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/ed61693f66704d9399c3a7eafdf9bb41 : SUCCESS in 4m 04s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/5ddd118146674e859f4fd72991e55247 : SUCCESS in 4m 35s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/103537e1d9ac4ee1b32a1610a93c4143 : SUCCESS in 4m 12s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/aa4239f8bfca4cbcaa9d8532f0c87033 : SUCCESS in 3m 48s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/48be78c83b2c4f2f8a13e2adf3cda44a : SUCCESS in 4m 08s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a4cdccd7291a49ea96be4e67e7db8167 : SUCCESS in 3m 54s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/f1ae4c1a8ded4ecb808df45c6f43dc10 : SUCCESS in 4m 25s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/bd0243376ca74c68ae59d8e0ac21d254 : SUCCESS in 4m 08s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/9bf79cd315a74fc0b6c0a4940f426ed5 : SUCCESS in 4m 04s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/246466cefef14971861e87b52e5a0c77 : SUCCESS in 4m 42s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/794a5f5a274d49e693a0e6cf58df8025 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":4},{"id":"6a2a6851c3eccbf68564814d0d234dce4adb7dc9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-13 02:23:04.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7f16331c2c58e7dac2267bcad5f21407bc56f826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 05:21:11.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2d331e4208b4a309af182f4c82e4b26\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/edb2852a9905494880ccb59f4da4efff : FAILURE in 1h 33m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52767f2daed044bdb5eec7451baae676 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cf775316f64303839e79de84a3811d : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bfc011c8d3d041c7bbee26695e33fbaa : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1efd09758ae40c38ed85dc0093dab69 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1e328e33d24825851a25a767909af4 : SUCCESS in 4m 08s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/dea12024aeb74f7e8ad17912051b6120 : SUCCESS in 5m 17s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/205268cf2fbf4cfda29c57d019fce5d1 : SUCCESS in 4m 53s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/08e2b37f3c2c4ebc9e3fecda70c79a3d : SUCCESS in 5m 08s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/ecd0e0dd15b44850b944ab4955068873 : SUCCESS in 6m 15s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/9ab62e8ca8b8481497ab7e845e8015a4 : SUCCESS in 5m 11s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/0b9b1f6b36a542df971347503004ba48 : SUCCESS in 4m 59s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/e1d257a157d94c9fb3b72f7ccd8fcc93 : SUCCESS in 4m 59s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/cf40049a4e9640a7bf93ab9436cca813 : SUCCESS in 5m 09s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/32e282161793476cac95a2574616f629 : SUCCESS in 6m 02s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/64233718bbbc4ca4a080ad53fd037edd : SUCCESS in 5m 02s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/939fe0e558eb460086122093dafba297 : SUCCESS in 4m 35s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/23ec871c7cde4b3f91a3beb0680a7c64 : SUCCESS in 4m 33s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/a8deb02d4c4c4b9f8b864633f1123076 : SUCCESS in 4m 46s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/abad60569e2342e88bf642da3d339e39 : SUCCESS in 4m 25s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/94a76fb5232e476da80c9ba0c1126b18 : SUCCESS in 4m 27s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/8c2ef0ef67b846ebbd9877326c14f35a : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"b14b9b0d47b9857124745a7de7a93fd43e05806d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-13 05:27:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cdc1a62e935c8377ffd72f010f5cbfc2d8cb2d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 06:34:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09dc3a28375430eaf572c6c95178e09\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/581a60707a49440b82a3b1c07f74b747 : SUCCESS in 1h 05m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84ec3e5a80014709927da406961f5cad : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931b8683e60f4d8cbdb891270de5f2a2 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43b79abfdaa14ef2b7c2901695605315 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6542e82303a942f9b6cfccec9cf018cc : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a8bc46a010748508d9d7347e209eb95 : SUCCESS in 3m 46s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/7232ae8b6c2f454292dfe5d91222b651 : SUCCESS in 4m 24s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/78a2b42b99a6458ca9734f886c4a632a : SUCCESS in 5m 34s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/ef809ec78e5a4ab89b53972126ac6d2f : SUCCESS in 5m 26s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/107f1f1b72834038a02413a9feaecf12 : SUCCESS in 4m 13s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/a1777ed7de1a46e39d9b91c14ecd31b8 : SUCCESS in 5m 02s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/249292281d1041f8adb2b26a2df12141 : SUCCESS in 5m 09s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/59305af6d4ef458c82c7bbcb81292ca1 : SUCCESS in 6m 26s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/47fcc9a8658e49cca4acffe87443f9f8 : SUCCESS in 3m 54s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/4fd074795720421a9d19cb6a060216aa : SUCCESS in 4m 14s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/996c0acbee4e4ff4aa66533ae4afc01d : SUCCESS in 5m 05s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/ce7746679c974170a4d80f8158df50ef : SUCCESS in 4m 02s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/8f6a9782083c43408c16e9aa3a6a07ee : SUCCESS in 3m 58s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/8ac03a8936f148dd9b9fbf2b3a57e473 : SUCCESS in 5m 24s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/737ecbe9585340c8b900ddd2695f221a : SUCCESS in 3m 55s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/941329856b7b422f864c6720479330e9 : SUCCESS in 5m 01s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/3b99fd9fefb3408689fa7e926c6c70c1 : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":4},{"id":"b478cc0b266dde3c1176efc2fe80ba15394a01ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 06:35:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"31bf98617426eef0cd0c8fd2f985393a57f3eb89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:11:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704f32b378194365b1aa905a0d8094c6\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a917d9a22c704aff8c4991a6476d57b8 : SUCCESS in 1h 30m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb55fd278f24c2dab95aa6d5dc69f9b : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a921b257f91640579be70427c1f6c634 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f709bdbb2a84344a15de28a83624099 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0146e50b36b4982bb2e9865e2f07d96 : SUCCESS in 4m 53s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/de5da5813ce84e0da820a49b2f38a50b : SUCCESS in 5m 49s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/c45e0875025d441bbba55937b8c73899 : SUCCESS in 5m 59s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/35dccfa2ccdb47a88fe4deb11b8cb6cd : SUCCESS in 3m 42s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/a382edb761d5487ebacf62c7d50e35bf : SUCCESS in 5m 20s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/85e434eee1f2478ba28c1ed7fbd186b1 : SUCCESS in 5m 42s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/c62beccef9a1437abdd7ef7e7a871f90 : SUCCESS in 5m 09s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/851742b7df984568be34cce00a7a8541 : SUCCESS in 5m 25s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/9f2c585e9c5a48188fc21140f3f641a2 : SUCCESS in 4m 08s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/e7e3de5958fd40298b3fb8551fb4a749 : SUCCESS in 4m 03s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/9f9833357d304dbfaa502a62ba6ee453 : SUCCESS in 4m 13s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/b405c505aef940a08aa17629a149a861 : SUCCESS in 3m 57s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/5c355902a519427289ac36ebd58a58e1 : SUCCESS in 4m 01s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/acb4b0ef4a4d44559e15ebb0d90d1f14 : SUCCESS in 4m 14s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/caa9119598c04adc909e76944632bfce : SUCCESS in 3m 55s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/2243719e7e14470b857a4db532bdadbb : SUCCESS in 4m 00s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/3d8be1208cee47f792ccd09aed3baeab : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":4},{"id":"26eb488714af20ef8cc7e611163377ff860bb5c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:11:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"eab3e93ff9a66b63c8e730933ecb932314c721d7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 08:12:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a972649a1f914c68ad635d39f580105c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57eb86f07a484cdeba6791e0ab02f7f8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2c0d37c2e58456ccaa5660f82f708166e21b2502","revisions":{"cdf03c3c1c6cb4381041ff48d31372df9c3b5f3e":{"kind":"REWORK","_number":1,"created":"2024-02-03 16:53:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/907652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/907652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/907652/1"}}},"commit":{"parents":[{"commit":"8943f944085fa07774be018dec16175320180187","subject":"Move driver tests to drivers directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/8943f944085fa07774be018dec16175320180187"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 16:51:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 16:51:16.000000000","tz":540},"subject":"redis: Fix parsing of sentinel fallbacks containing IPv6 address","message":"redis: Fix parsing of sentinel fallbacks containing IPv6 address\n\nThe sentinel fallbacks option may include IPv6 addresses like\n[\u0027[::1]:26379\u0027], and bare split can\u0027t handle this case properly.\n\nThis ensures the parse logic distinguishes the separator of host and\nport and the characters in IPv6 addresses.\n\nChange-Id: Ibb4d26022864ece4abf91fd2d2fd665998d80bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/cdf03c3c1c6cb4381041ff48d31372df9c3b5f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/cdf03c3c1c6cb4381041ff48d31372df9c3b5f3e"}]},"branch":"refs/heads/master"},"044a3565207d55cdc7afc95c20fe42697732b8ae":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-02-03 16:57:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/907652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/907652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/907652/2"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 16:51:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 16:57:11.000000000","tz":540},"subject":"redis: Fix parsing of sentinel fallbacks containing IPv6 address","message":"redis: Fix parsing of sentinel fallbacks containing IPv6 address\n\nThe sentinel fallbacks option may include IPv6 addresses like\n[\u0027[::1]:26379\u0027], and bare split can\u0027t handle this case properly.\n\nThis ensures the parse logic distinguishes the separator of host and\nport and the characters in IPv6 addresses.\n\nCloses-Bug: #2052353\nChange-Id: Ibb4d26022864ece4abf91fd2d2fd665998d80bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/044a3565207d55cdc7afc95c20fe42697732b8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/044a3565207d55cdc7afc95c20fe42697732b8ae"}]},"branch":"refs/heads/master"},"a070cfda89ab1680838652b0b5c908b1939e4cb7":{"kind":"REWORK","_number":3,"created":"2024-02-04 08:10:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/907652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/907652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/907652/3"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 16:51:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 08:10:48.000000000","tz":0},"subject":"redis: Fix parsing of sentinel fallbacks containing IPv6 address","message":"redis: Fix parsing of sentinel fallbacks containing IPv6 address\n\nThe sentinel fallbacks option may include IPv6 addresses like\n[\u0027[::1]:26379\u0027], and bare split can\u0027t handle this case properly.\n\nThis ensures the parse logic distinguishes the separator of host and\nport and the characters in IPv6 addresses.\n\nCloses-Bug: #2052353\nChange-Id: Ibb4d26022864ece4abf91fd2d2fd665998d80bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a070cfda89ab1680838652b0b5c908b1939e4cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a070cfda89ab1680838652b0b5c908b1939e4cb7"}]},"branch":"refs/heads/master"},"2c0d37c2e58456ccaa5660f82f708166e21b2502":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-04 08:11:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/907652/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/52/907652/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/52/907652/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/52/907652/4"}}},"commit":{"parents":[{"commit":"a267b2811b94690f152511edadc5db86178849ce","subject":"redis: Add username","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a267b2811b94690f152511edadc5db86178849ce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-03 16:51:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 08:11:32.000000000","tz":0},"subject":"redis: Fix parsing of sentinel fallbacks containing IPv6 address","message":"redis: Fix parsing of sentinel fallbacks containing IPv6 address\n\nThe sentinel fallbacks option may include IPv6 addresses like\n[\u0027[::1]:26379\u0027], and bare split can\u0027t handle this case properly.\n\nThis ensures the parse logic distinguishes the separator of host and\nport and the characters in IPv6 addresses.\n\nCloses-Bug: #2052353\nChange-Id: Ibb4d26022864ece4abf91fd2d2fd665998d80bd8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2c0d37c2e58456ccaa5660f82f708166e21b2502"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/2c0d37c2e58456ccaa5660f82f708166e21b2502"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
