)]}'
{"id":"openstack%2Ftooz~907656","triplet_id":"openstack%2Ftooz~master~I9b53c3a314dcffbe49535799ba6d7a1fdddb8d84","project":"openstack/tooz","branch":"master","topic":"bug/2052371","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-12 17:17:39.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I9b53c3a314dcffbe49535799ba6d7a1fdddb8d84","subject":"redis: Add username","status":"MERGED","created":"2024-02-04 06:11:06.000000000","updated":"2024-02-12 20:19:28.000000000","submitted":"2024-02-12 20:18:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907656-bug/2052371","meta_rev_id":"a9eb8e0aeec6d7572d7a24a7741c715ef0035c63","_number":907656,"virtual_id_number":907656,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-12 20:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2024-02-06 10:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-02-12 15:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-02-12 15:38:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-04 09:14:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-06 10:09:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-02-12 15:38:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"13beba0c5b7b416d249c6adbb1e2dc337be90fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 06:11:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e64e2cc5245933fa57cae268c656f2b30b8d47c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 07:30:30.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b097a31c5c9b2391b5a4594cd66202dc4d4b34eb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 07:30:35.000000000","message":"Topic set to bug/2052371","accounts_in_message":[],"_revision_number":2},{"id":"5084c5617baad04d73a892775460e4c099222fa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 07:32:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b7b9fb2b174dc5cc8dc1d8eddb71007baa80d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 08:11:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"6a6708f055bd9a3eea02045188f8d7ecd7ab6a55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 08:11:23.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"00a9e5f6b394496eb48fedcbdfedc881080318c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 09:14:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab700ce59a3c44f597314159bd9f0ad9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e990aa9185b9428b992e9042b8489388 : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5f6df7e323b2437a9ac215a365684a0d : SUCCESS in 1h 02m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34e64ee7807545c59799820104ba5f51 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60d156f26824987807b4a5e2533d178 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33986753d7d34d23bab6703a3cdd41f9 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/071dcd03ebf94993ac6575d7ba3d9afe : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb61f47bc83e4dce87e50743c87c930c : SUCCESS in 4m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df7a8e9932ae4413877f17c7a575c7a5 : SUCCESS in 3m 52s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/c50727da32e1429e86e7bdb5002f2e81 : SUCCESS in 4m 46s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/fbf6f0d5b09e4ddeb71755bc35945de5 : SUCCESS in 4m 24s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/aed2d3196f37424d968f1a9169f297ea : SUCCESS in 4m 12s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/2d20533934c144c6af5e3124d387117a : SUCCESS in 4m 06s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/cb4f7a65b937405e8dfb20be8c843332 : SUCCESS in 4m 28s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6b6f7f174a2448998df252dc0405f456 : SUCCESS in 4m 07s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/050cac41d3074838bbcd7cbdad645fc4 : SUCCESS in 4m 19s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/740c06e3886742adb88be918df3cc82b : SUCCESS in 4m 20s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/5af62bfdc2084559a9303723c5c3ce68 : SUCCESS in 5m 00s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/6a148fff06174726956c24eb97198ad1 : SUCCESS in 5m 10s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/bdf64eadfc544a049c77926ceaa78ed4 : SUCCESS in 4m 25s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/73435027549e4abdb8b9fe264fbb1995 : SUCCESS in 4m 26s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/671e820c4ffd4724bc025f3f59434721 : SUCCESS in 5m 04s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/81eb3042d96d45be98fbf33efaacf1e0 : SUCCESS in 4m 03s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/42cdd3ac708540d68863baf72243a54a : SUCCESS in 5m 04s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/782378d5b3d24813a33d1f6bbd675462 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":5},{"id":"5044d9843575a59251c129839edf7ff6e3df0fef","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-02-06 10:09:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"120a3580dd1a3c7f325548c3e1dfccbc3c0ec136","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-02-12 15:38:46.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"34bd60660c15af1c6124894b9993e3fff8150bf6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 15:39:16.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c81eb1191892f140a2e9eeb0b3a2eaf3d5421fa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 17:15:56.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c19a4287bd33460a82d79e06b777062d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/36b3f31daa164a22b0f8cbf12643b648 : SUCCESS in 2m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a38c8989a3614a51a696b938a3e2498c : SUCCESS in 1h 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88953457e5c445eabc01f1e3556ded40 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c0b76b96f60463787eb0e49ef10886a : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ba1c4d2d6d74338a0fcad8813764be7 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74bcf525eccb4e4a8348386be0b7a379 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9db3d4fc19114d8faefd0eef6a4c6445 : SUCCESS in 4m 58s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4e0607c0393b48b7bc64a68229e10e15 : SUCCESS in 5m 20s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/2da79a24228247c79d151018526f1eba : SUCCESS in 5m 13s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/7a50eaf90d5a49d68989c6e5e35179ac : SUCCESS in 4m 35s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/54fb53f796eb448693f64f640639b5d7 : FAILURE in 4m 12s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/ebc388ad1d514223a0ee270114b471f4 : SUCCESS in 4m 30s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6af19afa00bb40139e2d4a0c2d5de523 : SUCCESS in 4m 48s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6af9fb6750f844ba91a6ccacb35336fb : SUCCESS in 4m 28s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/c6657021109a4b269ebc4fcd2f125099 : SUCCESS in 4m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/699d7f10e3864c3d84cf44859838fe1e : SUCCESS in 4m 24s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/986a29616f4c40468730127f0e057409 : SUCCESS in 4m 45s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/00baacbf819b4e57b96ae90fef9b9fbf : SUCCESS in 4m 40s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/91cfcc9c72e2426dbf1271f5f84e4750 : SUCCESS in 4m 22s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/1c171d2480694c0c899c2fb315668fe3 : SUCCESS in 4m 50s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/acb6c3bf54284185aca5015a509bf4bc : SUCCESS in 4m 53s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/c67de4f96c6e49b6b1fd1110a07e3ad2 : SUCCESS in 4m 46s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/9929e26be9d34672bbe9e57030e9292b : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":5},{"id":"ec4aae1d20571383ec15a825c504bb9d7779b698","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-12 17:17:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"12321448a774c595bd178b8b081e4633281a0508","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 18:34:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b119d894ffd46bcbe39c40bc58cfdec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8fac0eba80fb41b3be38eb62274bc722 : SUCCESS in 1m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4271aba096e645eaa45099aea8351081 : SUCCESS in 1h 14m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05e072151df0497fa0561bc60a47d9e1 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b2b3425f5b24b8987afb98f83980393 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d297bc8540444d32952cf54f40f8e0a5 : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e893d356f99c4ea2ba56188d669b12c5 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9df0818812cc422a8678b7178934605c : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7be66bf1518b4ed9862fdec5ddecaddb : SUCCESS in 3m 20s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/a24555196a834a6ea0f0e97fed6397a6 : SUCCESS in 5m 40s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/e70ccc0ed24146c188b7278f95497921 : SUCCESS in 5m 01s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/9112b462dba14809af6ff3abc8b57b86 : SUCCESS in 5m 42s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/838ea8648cd843de9ddc18305b15024e : SUCCESS in 4m 09s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/8e20694ca3e24f46be29da1005f57411 : SUCCESS in 6m 06s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/43361a6c17184ef29515d5b7582f96a4 : SUCCESS in 4m 12s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/81d23655a5964ec9a1831906f042af26 : SUCCESS in 3m 53s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/2b46d7b8384d498182357c2b0148779a : SUCCESS in 5m 27s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/6724012350f343ebbfdacb7cb5e2182f : SUCCESS in 4m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/85eaf993187b4e799f6da8b92d4561c4 : SUCCESS in 5m 49s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/efc5deffaba043d99d3dcb98a59b2408 : SUCCESS in 4m 34s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/d55475014b3d405e87c63761d533c432 : SUCCESS in 5m 24s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/d72de746389a4034964c767ee311e742 : SUCCESS in 4m 12s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4b9426c5268b46aea398d566ca644183 : SUCCESS in 4m 38s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/315b1c3d266d4e0bb2e4a8c04045b2f9 : SUCCESS in 5m 32s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/dcc0451c4bfc4aed82c8613ebce826d4 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":5},{"id":"2ad71d972bd24b39d1c00a663f7be475af0efdaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 18:34:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"42b373505067508b5a77712faf9acd5d8ca4bc93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 20:18:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4153876e06f145ab8eabfc6ff9b29d0e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b6db2978ea94a9eb71ad4d6be113733 : SUCCESS in 1m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/415888763a8f4f788e6f9d536c3ace1e : SUCCESS in 1h 33m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9379fc67da6945099e3933fe6fe6fcec : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/692a5c3f675c426299f72c50d5f2b7dc : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6778c997eef4895836bf4359f9286a4 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b5434bacba348889e22c55b923e18e7 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/436e99b0cbee4f68866ee796cb5520b0 : SUCCESS in 3m 26s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/4c6dfdb7e2e9457c9c36dfc9fdc2b94b : SUCCESS in 5m 39s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/41b5bcbd1bfb43bfa26e4ea192bdb399 : SUCCESS in 4m 42s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/98b2032c578b47f3a6623e069634b5fc : SUCCESS in 4m 36s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4de97457f3e140a8b918b2b723cc138f : SUCCESS in 4m 17s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/6b2458c09ee44ee0a5765e8ce375f31c : SUCCESS in 5m 40s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/416f67932149483bb370f12056b1a203 : SUCCESS in 5m 46s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/52861f939e9647ada8449021175c3424 : SUCCESS in 4m 52s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/011ed60c4f16466885788928a9ba2210 : SUCCESS in 4m 05s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/2e2abcfd2de647c588e0b53544f19ac0 : SUCCESS in 5m 30s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/b87827a98e004e0f946fe830f67b54e4 : SUCCESS in 5m 14s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/a3c259d832f547b9bfa7828b850b59fc : SUCCESS in 5m 38s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/02fbfe081d4740d18bcb24cf2bf1ca43 : SUCCESS in 4m 35s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/8f562b20e7eb478f8ddee024b691eff3 : SUCCESS in 5m 28s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/4ecb29af878d42818b1cac548593be8d : SUCCESS in 5m 37s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/54801dd6e19d47ce8d68fbf533543ad7 : SUCCESS in 5m 00s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/f454e402e86544169f3e87a3e6ad8c8c : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":5},{"id":"d0b25c756304c7f9d54a7ffd3e67827554cbd1a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 20:18:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"a9eb8e0aeec6d7572d7a24a7741c715ef0035c63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 20:19:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a22ac4f8d1cd49b4838631654b7c68ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/856ab18c82d74d1cb79f26d46e99ccd6 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a03c56df88ee40d4a376d5cc9dbfcf66 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a267b2811b94690f152511edadc5db86178849ce","revisions":{"886323cc0580a73371b136ce4c9c05536100596e":{"kind":"REWORK","_number":1,"created":"2024-02-04 06:11:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/907656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/907656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/907656/1"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:10:56.000000000","tz":540},"subject":"redis: Add username","message":"redis: Add username\n\nRedis introduced ACL feature in 4.0.0, and this feature is supported by\nredis-py since 3.4.0[1]. When ACL is enabled, authentication requires\nusername in addition to password.\n\n[1] https://github.com/redis/redis-py/commit/8df8cd54d135380ad8b3b8807a67a3e6915b0b49\n\nChange-Id: I9b53c3a314dcffbe49535799ba6d7a1fdddb8d84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/886323cc0580a73371b136ce4c9c05536100596e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/886323cc0580a73371b136ce4c9c05536100596e"}]},"branch":"refs/heads/master"},"da977d071025d67a0a176c22f5ee2f703ff728c8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-02-04 07:30:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/907656/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/907656/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/907656/2"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 07:30:30.000000000","tz":0},"subject":"redis: Add username","message":"redis: Add username\n\nRedis introduced ACL feature in 4.0.0, and this feature is supported by\nredis-py since 3.4.0[1]. When ACL is enabled, authentication requires\nusername in addition to password.\n\n[1] https://github.com/redis/redis-py/commit/8df8cd54d135380ad8b3b8807a67a3e6915b0b49\n\nCloses-Bug: #2052371\nChange-Id: I9b53c3a314dcffbe49535799ba6d7a1fdddb8d84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/da977d071025d67a0a176c22f5ee2f703ff728c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/da977d071025d67a0a176c22f5ee2f703ff728c8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9a3324a4260641c70654378e68b8fb8cc9025887":{"kind":"REWORK","_number":3,"created":"2024-02-04 07:32:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/907656/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/907656/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/907656/3"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 07:32:35.000000000","tz":540},"subject":"redis: Add username","message":"redis: Add username\n\nRedis introduced ACL feature in 4.0.0, and this feature is supported by\nredis-py since 3.4.0[1]. When ACL is enabled, authentication requires\nusername in addition to password.\n\n[1] https://github.com/redis/redis-py/commit/8df8cd54d135380ad8b3b8807a67a3e6915b0b49\n\nCloses-Bug: #2052371\nChange-Id: I9b53c3a314dcffbe49535799ba6d7a1fdddb8d84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9a3324a4260641c70654378e68b8fb8cc9025887"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/9a3324a4260641c70654378e68b8fb8cc9025887"}]},"branch":"refs/heads/master"},"eff11c4c331259e364ec1ba1ad1a48fb8ffa5ea2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-02-04 08:11:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/907656/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/907656/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/907656/4"}}},"commit":{"parents":[{"commit":"a070cfda89ab1680838652b0b5c908b1939e4cb7","subject":"redis: Fix parsing of sentinel fallbacks containing IPv6 address","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a070cfda89ab1680838652b0b5c908b1939e4cb7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 08:11:04.000000000","tz":0},"subject":"redis: Add username","message":"redis: Add username\n\nRedis introduced ACL feature in 4.0.0, and this feature is supported by\nredis-py since 3.4.0[1]. When ACL is enabled, authentication requires\nusername in addition to password.\n\n[1] https://github.com/redis/redis-py/commit/8df8cd54d135380ad8b3b8807a67a3e6915b0b49\n\nCloses-Bug: #2052371\nChange-Id: I9b53c3a314dcffbe49535799ba6d7a1fdddb8d84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/eff11c4c331259e364ec1ba1ad1a48fb8ffa5ea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/eff11c4c331259e364ec1ba1ad1a48fb8ffa5ea2"}]},"branch":"refs/heads/master","description":"Rebase"},"a267b2811b94690f152511edadc5db86178849ce":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-04 08:11:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/56/907656/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/56/907656/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/56/907656/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/56/907656/5"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 06:10:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 08:11:23.000000000","tz":0},"subject":"redis: Add username","message":"redis: Add username\n\nRedis introduced ACL feature in 4.0.0, and this feature is supported by\nredis-py since 3.4.0[1]. When ACL is enabled, authentication requires\nusername in addition to password.\n\n[1] https://github.com/redis/redis-py/commit/8df8cd54d135380ad8b3b8807a67a3e6915b0b49\n\nCloses-Bug: #2052371\nChange-Id: I9b53c3a314dcffbe49535799ba6d7a1fdddb8d84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a267b2811b94690f152511edadc5db86178849ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a267b2811b94690f152511edadc5db86178849ce"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
