)]}'
{"id":"openstack%2Ftooz~907663","triplet_id":"openstack%2Ftooz~master~I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96","project":"openstack/tooz","branch":"master","topic":"MEMCACHED_DRIVER","attention_set":{},"removed_from_attention_set":{"36759":{"account":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"last_update":"2024-08-22 08:18:37.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-06 03:25:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96","subject":"Add clustered memcached servers support for tooz.","status":"ABANDONED","created":"2024-02-04 09:52:22.000000000","updated":"2024-08-22 08:18:37.000000000","total_comment_count":10,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"cffab07986ae0e1a59a7115e2f63f69349f0c428","_number":907663,"virtual_id_number":907663,"owner":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-02-06 04:09:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2024-02-06 03:25:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-04 12:41:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 02:30:00.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"47237d0b439f395fff013d67590c9ac7faae04c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 09:52:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f105e38e10f2fb3149c2cb288744d2bc2514468","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 09:54:19.000000000","message":"Topic MEMCACHED_DRIVER removed","accounts_in_message":[],"_revision_number":1},{"id":"55246b1dee8265d7735e8547d5551eebc794e18d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 10:28:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8afb9b1a6775c3c82f335c5d0df864dfa58ce694","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 10:28:44.000000000","message":"Topic MEMCACHED_DRIVER removed","accounts_in_message":[],"_revision_number":2},{"id":"51c52761715b51a3bdf5fa6d3ad45dc04f470a7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 10:34:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a4853d7bc6148e7b2bd3587f15fce4896e22a09e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 10:39:05.000000000","message":"Topic MEMCACHED_DRIVER removed","accounts_in_message":[],"_revision_number":3},{"id":"de091ac9bd10f384f25ea1883562dd81d891e4e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 10:47:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e83e188d8e0bdb994d94c04608335b3154b26ace","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 10:48:03.000000000","message":"Topic MEMCACHED_DRIVER removed","accounts_in_message":[],"_revision_number":4},{"id":"cc753568d38b319548f5b88779600f95d3217636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-04 11:05:37.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent tree, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e118df5783fa2977d03979ec2bb79dcd4ecb2119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 12:41:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/daa09053f76a4c3399894e86300eaa18\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24a2d0c415854a53a14741e92368197a : SUCCESS in 1h 29m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f18bb31da6cb4a049f6b14984f670598 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac459e036284383a129df96588493b8 : FAILURE in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc5f8a21ce1a495e94915a225d4142c7 : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d9ac6ef28104cbd91fe3586d372bf1f : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569deaf3617f4ea1bd7559e315b5e8e8 : SUCCESS in 4m 14s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/1934bd9e16cb4ebbb0c2c79be990450e : SUCCESS in 4m 42s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/948f91bb80d642d89cb7e53fc0827eac : SUCCESS in 4m 39s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/903667e7fd494336b6848d9640dca210 : SUCCESS in 4m 24s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/c18078ef939c4e7cb134e957a576dbd0 : SUCCESS in 4m 12s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/77f06cb81825461686ba345a7006b33a : SUCCESS in 4m 07s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/cd546b147904435c805a81b1dfac245e : SUCCESS in 4m 04s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/7472cc32a57548ae92f9922e756b9947 : SUCCESS in 4m 26s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/189bae2448514f6ea90cf9f5006dc104 : SUCCESS in 4m 00s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/721d06ce7e41409a97d4498a1b914214 : SUCCESS in 4m 17s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/538188d9430749a0a4021f4953253215 : SUCCESS in 4m 07s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/9a46ebf258054232b4b05f8c13fd192e : SUCCESS in 4m 08s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/1694a810a7e646ad8798329617eacd70 : SUCCESS in 4m 08s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/b0e5424667084bb48c85ed57c5fef2bc : SUCCESS in 4m 18s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/47449119a5c24907b59590419806dbd1 : SUCCESS in 4m 01s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/a11ce2c9aacf4a7aabc9f962e5790bd5 : SUCCESS in 4m 08s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/bb7997a983944af68e1479763d25da2e : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"528fa40cb9887637eac064ce103c2081178b709c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-05 01:27:15.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"1b3668916a37acd97c96efbf07f9b7586949179c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-05 01:29:34.000000000","message":"Topic MEMCACHED_DRIVER removed","accounts_in_message":[],"_revision_number":6},{"id":"c85118401121759698d9d7ff4cc83bce88c65bd2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 02:30:00.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8f16ede03a4aae2e580bee03e21f6937d7d290ac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 02:34:24.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7cb6cd8a111775c32a6d37d7ee6f3e504ce8a0dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 02:41:11.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49382f248aeb4776a9e7e4b7f7fa3709\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef6626a0b9de4f409ff30b33c6067dc3 : SUCCESS in 1h 12m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a09c8dddb5d4b31a1355ded1a216ce1 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abbef45e70634723b9eb1ec5396a84f3 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c9a88a0dde54657995e833b3dfc7db0 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54a9098f125145f2a92401bba1438a16 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b890e8ff9a349bd8849f1d41ab4d309 : SUCCESS in 3m 53s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/56bebd274a634b77bd78f3256fec3df1 : SUCCESS in 4m 27s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/1f542fc497584b088c98fdfc0cd2e5f7 : SUCCESS in 5m 10s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/c957ac9c7047442f957869cbc50087ff : SUCCESS in 4m 06s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/44cd6cf9374e4276b53f9105deb49e0a : SUCCESS in 4m 42s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/3c0edbd27eb24ccb9537a44d7643b02a : SUCCESS in 4m 04s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/27e5ecddc4034a1f8cb94f7af002218e : SUCCESS in 3m 56s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/41cf275142fd41e184f80db0e4b09f94 : SUCCESS in 5m 27s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/93190fd90a24462c87980ba977a9734a : SUCCESS in 4m 44s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/abf7ee2f99c24ec3abe97fd5d7adc187 : SUCCESS in 4m 40s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/796fafc55b254e29803c8b7a982c5973 : SUCCESS in 4m 11s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/317502ffdcbc4b95a1a85ea95590fcfa : SUCCESS in 4m 36s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/cd0570d02ce54ac8948bf95fd76fec25 : SUCCESS in 4m 34s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0fa8a5c5f51942288b36406fa8e0e4d1 : SUCCESS in 4m 39s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/00667c698a814e35ab17189bc043aeae : SUCCESS in 7m 57s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/90ce3498d67945b48d2f7c9fa7e284dd : SUCCESS in 7m 34s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/942f3deaf9b84441a5efae2c2979d1a7 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":6},{"id":"47c4e44a7a2638cb3d1234d5fe865b8cbfa609d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-06 02:11:56.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"a80eeb98f88f08f163f3bf5b7683febcf0c1a3b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-06 02:16:43.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"079f499b37ec3c0a3d00cac929af13b47498e309","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-06 02:23:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"842f85469444335f672b282c31f62846411f5820","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-06 02:25:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1807455317fa88b487e475163691e12b393a4ce9","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-02-06 02:26:28.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f589779405c9f0d7f57a3f4be7eb0d6b6cb56783","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 03:25:55.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"50c349cfbf2d4cd789cff1351dcdf4fa603e821d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 03:26:31.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5fe2a6e5f09d2a916da58350353f9531b2859444","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-06 03:30:44.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"621638758494cca972f8769094ef7467557c679c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 04:09:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1296893b2cc40808fb561220463acf5\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49937ef56a234d37a666831550984629 : SUCCESS in 1h 37m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb8cf8d46a0544bb9ff043e33f08f4bc : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e8b24fd3cfc45ceafd4b029f60153ce : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d99ddabbd6140d481b89ebc042aaf0f : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e5191b2e44204aa8b589673a8e6004b2 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6338b75a495f44b783f6fa078b5c1c71 : SUCCESS in 4m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46aa1c82e3bc4d60abc6ee406a09b336 : SUCCESS in 4m 26s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/362d528e5a1f4096881b16edf5e68fe6 : SUCCESS in 4m 30s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/5ce15a2c835a4af3a3396b0be6e061d2 : SUCCESS in 4m 44s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/1ad548f69d464dab83cda833862421e3 : SUCCESS in 4m 26s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/85524959da0c44ea9198ee0c94ed5abe : SUCCESS in 5m 12s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/0c8710b63c2d4b559c61b2043829ef67 : SUCCESS in 5m 28s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/2944b0e18dd7400c9bf48b53c51050f7 : SUCCESS in 4m 19s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/89da68f8cc954c29869051fd6e06347e : SUCCESS in 4m 33s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/aec1de11125d48f29080e4615fc7eddc : SUCCESS in 5m 55s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/1060f9fcb01d4054892de7eba23f3db3 : SUCCESS in 4m 26s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/09475be8278c4dc8ba0277d9d866c23d : SUCCESS in 4m 10s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/185d0827cc8b484385ed8a0b3b423888 : SUCCESS in 4m 19s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/69fa45964fae4dc49200ee483311b8b7 : SUCCESS in 4m 19s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/007001a906b3445a831be0e146ff2e48 : SUCCESS in 4m 11s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/39dbc365d51d472f853c0d607398c8c8 : SUCCESS in 4m 36s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/be2a0d9cf58644d3822847fe62d84262 : SUCCESS in 4m 38s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/1a2f9ce60958452886992c3e548cdebc : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":8},{"id":"cffab07986ae0e1a59a7115e2f63f69349f0c428","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"date":"2024-08-22 08:18:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"fcac07f7544259c665ebdea7652e23065811dc5b","revisions":{"1695f1faa754cd77ef45b54abc6d6405b19c015c":{"kind":"REWORK","_number":1,"created":"2024-02-04 09:52:22.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/1"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1695f1faa754cd77ef45b54abc6d6405b19c015c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1695f1faa754cd77ef45b54abc6d6405b19c015c"}]},"branch":"refs/heads/master"},"eb61409bdd1eaa7724e31fe63658a06d26caeb43":{"kind":"REWORK","_number":2,"created":"2024-02-04 10:28:05.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/2"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 10:27:42.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/eb61409bdd1eaa7724e31fe63658a06d26caeb43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/eb61409bdd1eaa7724e31fe63658a06d26caeb43"}]},"branch":"refs/heads/master"},"81f3e3e11d6f5b16ad28ce96db15f5d2eaa33e49":{"kind":"REWORK","_number":3,"created":"2024-02-04 10:34:54.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/3"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 10:34:39.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/81f3e3e11d6f5b16ad28ce96db15f5d2eaa33e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/81f3e3e11d6f5b16ad28ce96db15f5d2eaa33e49"}]},"branch":"refs/heads/master"},"570933d2ea346dca7828e6010ba11a2cb661ccb1":{"kind":"REWORK","_number":4,"created":"2024-02-04 10:47:13.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/4"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 10:46:56.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/570933d2ea346dca7828e6010ba11a2cb661ccb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/570933d2ea346dca7828e6010ba11a2cb661ccb1"}]},"branch":"refs/heads/master"},"0dac7d9d4717ac179201606fd3d9a42febc5a2b1":{"kind":"NO_CHANGE","_number":5,"created":"2024-02-04 11:05:37.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/5"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 11:05:22.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0dac7d9d4717ac179201606fd3d9a42febc5a2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/0dac7d9d4717ac179201606fd3d9a42febc5a2b1"}]},"branch":"refs/heads/master"},"ba3a7531467418387b26f6bc0794b59728b02172":{"kind":"REWORK","_number":6,"created":"2024-02-05 01:27:15.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/6"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-05 01:26:01.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba3a7531467418387b26f6bc0794b59728b02172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/ba3a7531467418387b26f6bc0794b59728b02172"}]},"branch":"refs/heads/master"},"c17a41d629571f0d4c367d245eacb002f7c17a7b":{"kind":"REWORK","_number":7,"created":"2024-02-06 02:11:56.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/7"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-06 02:11:22.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c17a41d629571f0d4c367d245eacb002f7c17a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/c17a41d629571f0d4c367d245eacb002f7c17a7b"}]},"branch":"refs/heads/master"},"fcac07f7544259c665ebdea7652e23065811dc5b":{"kind":"REWORK","_number":8,"created":"2024-02-06 02:16:43.000000000","uploader":{"_account_id":36759,"name":"Xu Qi","email":"xuqi_yewu@cmss.chinamobile.com","username":"XuQi"},"ref":"refs/changes/63/907663/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/63/907663/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/63/907663/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/63/907663/8"}}},"commit":{"parents":[{"commit":"22361cb48d112040499e7e3548b5fb71af016b95","subject":"Merge \"tox: Bump python runtime versions in the default envlist\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/22361cb48d112040499e7e3548b5fb71af016b95"}]}],"author":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-04 09:51:57.000000000","tz":0},"committer":{"name":"XuQi","email":"1530865185@qq.com","date":"2024-02-06 02:16:20.000000000","tz":0},"subject":"Add clustered memcached servers support for tooz.","message":"Add clustered memcached servers support for tooz.\n\nChange-Id: I7ac6cdfecb6ad1d5fa2a78cfa6b236c45e3d4a96\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fcac07f7544259c665ebdea7652e23065811dc5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/fcac07f7544259c665ebdea7652e23065811dc5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
