)]}'
{"id":"openstack%2Ftooz~911648","triplet_id":"openstack%2Ftooz~master~Iaa5f09e8dcfff9e65853f6f89976cf4c025bc864","project":"openstack/tooz","branch":"master","topic":"bug/2056429","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-07 16:32:05.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-07 16:32:05.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-03-07 16:32:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaa5f09e8dcfff9e65853f6f89976cf4c025bc864","subject":"Fix broken redis sentinel support","status":"MERGED","created":"2024-03-07 08:29:26.000000000","updated":"2024-03-07 16:33:00.000000000","submitted":"2024-03-07 16:32:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911648-bug/2056429","meta_rev_id":"a30776cb38035c3734b6bdb94303f472035f2cac","_number":911648,"virtual_id_number":911648,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-07 16:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":2,"date":"2024-03-07 13:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-03-07 14:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-03-07 14:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 08:46:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-03-07 09:04:53.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2024-03-07 10:52:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 12:23:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-03-07 13:50:58.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2024-03-07 14:28:55.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"}],"messages":[{"id":"7c4e3ddf63c8483b7d9ce9d924995e27119c356c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 08:29:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34e8083f2d4309e9f9124a78f0cd2d213d677d44","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-07 08:46:14.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ce14097f6fb998e23f8632a70ab322b742f100c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 08:52:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"610227126bdff1c39846cc665ca78e40fc1a5347","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 08:53:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"785fbc9aba77ee2804cdcc2624166cae191b6d99","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-03-07 09:04:53.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4224cf1403ff230f52e0e16a4f5be81b5158c740","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 09:24:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9347f7b91b80098a6f8e9dc7570f4b2c50c5c94","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 09:32:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8391952b2694233b0ecac5ffca065f0124b6279e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 10:52:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d154b6d9cc943ad947f4c28d5ef6d5f\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aad7b85e4bf0493fae4d5ecf69b52e5d : SUCCESS in 1h 53m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5e66c90b287449e9a16c93dc5ed19f8 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc0d09020bc473a9fbfc303e00b3dc7 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/923d9b9ee35b4ec98a5af6d67c84cd4e : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f840cb00849f451c8d1a7ce20d336a79 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc021ad1413420592f3e6be6fb4f313 : SUCCESS in 3m 37s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/59342fc219b34e34985f22d0bc2e7892 : SUCCESS in 4m 33s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/b6dccee1ac0b40a691652dec19b58f03 : SUCCESS in 4m 53s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/236eb7c8828f4bb093b1f40282acf4d6 : SUCCESS in 4m 27s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/4d8fabb86e1e47c49aea46e58221e6a0 : SUCCESS in 4m 25s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/313c514f7582443984b958cff94cd20a : SUCCESS in 4m 09s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/79323d2629c34adeb3ef4a60918315d6 : SUCCESS in 4m 13s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/6ef041e0922947bca7719b65081928ae : SUCCESS in 4m 05s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/6bbedf46c8804237a95fd300396f406b : SUCCESS in 4m 20s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/40ff70ac36c84d61a6d19a5cde3ef6a2 : SUCCESS in 4m 41s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/c690bd8524824e768456cd3f316e344a : SUCCESS in 4m 06s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/88be478e9b314c3ba595c3305e919b35 : SUCCESS in 4m 11s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/899cf78100994b89b9c9283d568396ea : SUCCESS in 4m 01s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/0177df39e6d943959404493d0c4fc403 : SUCCESS in 4m 38s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/330b9ae5d1144d9eab55fb95619831d7 : SUCCESS in 4m 03s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/88ceeb8802bd426fae0785ff36ff8816 : SUCCESS in 3m 51s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/16547f95c0014c7f9f14dd952e68f9fc : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3b533f18569ffe66a0f1b95adb0c5df1b5ff021b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 10:53:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b5ff857db90ec6c1dcb4e28ddc5edc25e11acb6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2024-03-07 13:50:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b71809b2c95fcdf86bc742b1ff4c6cfdb0666e8e","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-03-07 14:28:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"94491b94c56a701b364ac8578cdb6a5740051566","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2024-03-07 14:31:23.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed5c0cc738fa93f904937d0433fe8db1ad86564e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:33:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28f07222c0aa05c23edadfe95db78f16d2e98366","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:32:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fcba086c3d14c998c4a46795aaa97ee\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eb85c38b35014e008d708fe79d932610 : SUCCESS in 1h 37m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cf604a305ff460488d27fe9978f4d70 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edf19e3588a8492082f4f4717491106e : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b383314e534845e5bb78f0125a685ef3 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658ddcc10a8b417686ae14e4bfda0420 : SUCCESS in 5m 19s\n- tooz-tox-py38-etcd3gw https://zuul.opendev.org/t/openstack/build/8a16f98338934d35919b6e950cd7381e : SUCCESS in 7m 50s\n- tooz-tox-py310-etcd3gw https://zuul.opendev.org/t/openstack/build/47479a47cde845a385f45bed2ed57e0c : SUCCESS in 6m 19s\n- tooz-tox-py38-zookeeper https://zuul.opendev.org/t/openstack/build/b7eaa2aa985c4c8e9db6c25bd065b437 : SUCCESS in 5m 44s\n- tooz-tox-py310-zookeeper https://zuul.opendev.org/t/openstack/build/cce86244a0c14b448fc5d1a6280bbc10 : SUCCESS in 7m 23s\n- tooz-tox-py38-redis https://zuul.opendev.org/t/openstack/build/eb59e0870c8a43b7b864d567d0794ee0 : SUCCESS in 7m 42s\n- tooz-tox-py310-redis https://zuul.opendev.org/t/openstack/build/6e62baa1dc5a4597b4c5e9c53ad62739 : SUCCESS in 5m 37s\n- tooz-tox-py38-sentinel https://zuul.opendev.org/t/openstack/build/4048251f993742b4ac3dd25133fe9ce5 : SUCCESS in 5m 29s\n- tooz-tox-py310-sentinel https://zuul.opendev.org/t/openstack/build/a4a376b8ecb24a16a6b8b256c7847ea1 : SUCCESS in 4m 23s\n- tooz-tox-py38-memcached https://zuul.opendev.org/t/openstack/build/f991e7debdcf497fb8972e9bdad30ef0 : SUCCESS in 7m 24s\n- tooz-tox-py310-memcached https://zuul.opendev.org/t/openstack/build/d4c61fe7efea4b2d96180cf5a7dbd454 : SUCCESS in 4m 34s\n- tooz-tox-py38-postgresql https://zuul.opendev.org/t/openstack/build/33b121962c15450d972814245ddb5cd4 : SUCCESS in 6m 03s\n- tooz-tox-py310-postgresql https://zuul.opendev.org/t/openstack/build/c90fb58fbb8d4d178ac07d0a31030556 : SUCCESS in 7m 03s\n- tooz-tox-py38-mysql https://zuul.opendev.org/t/openstack/build/4948a0ebabb54971b83c014e3bb29b97 : SUCCESS in 4m 21s\n- tooz-tox-py310-mysql https://zuul.opendev.org/t/openstack/build/d0d19be4af464d6d911e3df00033877f : SUCCESS in 6m 19s\n- tooz-tox-py38-consul https://zuul.opendev.org/t/openstack/build/7ca41b37cf034b3f83aabcabcae582cc : SUCCESS in 4m 59s\n- tooz-tox-py310-consul https://zuul.opendev.org/t/openstack/build/5155d03b2aee4a8d8722eb8786207f05 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":2},{"id":"833e5cb2084017b5409cb233d49f768087ad9114","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:32:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a30776cb38035c3734b6bdb94303f472035f2cac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:33:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8201986c188641ee88caad6d96cb5a1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b7fc62dea92400c9a91c1e4837008fa : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a20ca740cf5ad29c74026937d75314a8823d3361","revisions":{"94d299c77736dafb9eadb80f080aec3d974d3edf":{"kind":"REWORK","_number":1,"created":"2024-03-07 08:29:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/911648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/48/911648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/48/911648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/48/911648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/48/911648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/48/911648/1"}}},"commit":{"parents":[{"commit":"1d0ece27a4f8b0d3aed13b603fd834eba3567183","subject":"Merge \"redis: Fix wrong type used to parse socket_keepalive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1d0ece27a4f8b0d3aed13b603fd834eba3567183"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 08:25:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 08:29:08.000000000","tz":540},"subject":"Fix broken redis sentinel support","message":"Fix broken redis sentinel support\n\nThis fixes the regression caused by [1] and ensures host and port are\nremoved from kwargs, to avoid conflict with underlying usages.\n\nThis also makes the current implementation consistent with taskflow.\n\n[1] 4954e284b9616f5e0c2cea77d94bbe18e0b8fd39\n\nCloses-Bug: #2056429\nChange-Id: Iaa5f09e8dcfff9e65853f6f89976cf4c025bc864\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/94d299c77736dafb9eadb80f080aec3d974d3edf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/94d299c77736dafb9eadb80f080aec3d974d3edf"}]},"branch":"refs/heads/master"},"a20ca740cf5ad29c74026937d75314a8823d3361":{"kind":"REWORK","_number":2,"created":"2024-03-07 08:52:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/911648/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tooz","ref":"refs/changes/48/911648/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tooz refs/changes/48/911648/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tooz refs/changes/48/911648/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tooz refs/changes/48/911648/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tooz refs/changes/48/911648/2"}}},"commit":{"parents":[{"commit":"1d0ece27a4f8b0d3aed13b603fd834eba3567183","subject":"Merge \"redis: Fix wrong type used to parse socket_keepalive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/1d0ece27a4f8b0d3aed13b603fd834eba3567183"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 08:25:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-03-07 08:52:57.000000000","tz":0},"subject":"Fix broken redis sentinel support","message":"Fix broken redis sentinel support\n\nThis fixes the regression caused by [1] and ensures host and port are\nremoved from kwargs, to avoid conflict with underlying usages.\n\nThis also makes the current implementation consistent with taskflow.\n\n[1] 4954e284b9616f5e0c2cea77d94bbe18e0b8fd39\n\nCloses-Bug: #2056429\nChange-Id: Iaa5f09e8dcfff9e65853f6f89976cf4c025bc864\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a20ca740cf5ad29c74026937d75314a8823d3361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tooz/commit/a20ca740cf5ad29c74026937d75314a8823d3361"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
